)]}'
{"id":"openstack%2Foctavia~706396","triplet_id":"openstack%2Foctavia~stable%2Fstein~I611d9cd5a519abc6a330ec2c46a2f5a1fdb89313","project":"openstack/octavia","branch":"stable/stein","hashtags":[],"change_id":"I611d9cd5a519abc6a330ec2c46a2f5a1fdb89313","subject":"Add a periodic image build job","status":"ABANDONED","created":"2020-02-06 21:56:20.000000000","updated":"2020-06-11 16:21:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0775133748ec5ed38e687458cb1c7d88cd5bf60f","_number":706396,"virtual_id_number":706396,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-11 09:32:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":-1,"date":"2020-06-11 09:46:16.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-11 09:32:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-11 09:46:16.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"77ecb20e0abc9afa5a43a9f4e1b88418249e275b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-06 21:56:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf3f52e30ae9e5afe89a54570f556325eadc09c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 23:23:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c80861c3e36345628638b64cc47d1a30 : SUCCESS in 10m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be9f38bc0e75435984a9263992cc1350 : SUCCESS in 11m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3be1a0cccb5146c6b061314a56c06ab8 : FAILURE in 6m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/34840650613a4902824c455fb365bc21 : SUCCESS in 7m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4dd0edb2d9e240938a4c789675c816b5 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df6eb58de5194f43875500970847f4f6 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19318119d05648e688efbf1ef4ad2ad6 : SUCCESS in 9m 26s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/c35ad69f4775470a8bed4cf1d4986302 : SUCCESS in 1h 18m 18s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/db4d48629eec455c90a637f75c6eba01 : SUCCESS in 1h 18m 42s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/08cf5baf419743a19649403a71a1fb06 : FAILURE in 35m 55s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/e4fdb465c2c14f5ab6f64cd0851eb65f : FAILURE in 34m 30s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c13ce5c8f83f4bd1a5cbe07d124b2727 : SUCCESS in 9m 17s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/54800e68e33844b0a8125d3a3897f2c4 : SUCCESS in 9m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9ab34415e7e945bcb97a96dd44e9e0e4 : SUCCESS in 10m 06s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/da95a87c1fe1477b88b05898fcefdc61 : FAILURE in 35m 20s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/bd730c607469439fb107e4f8f31efc59 : FAILURE in 40m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a11ac4d4fb6f4d4fafc26993bbaac45f : SUCCESS in 48m 00s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/9715fdfcb08549699156fe32b8d39714 : SUCCESS in 45m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/705d7b82fb68405e98cea68c012ee6cf : FAILURE in 34m 29s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/fc7d557ad6a04f33bb94e345f9f727a8 : FAILURE in 32m 22s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/950b45fdf63743968fda370d61f6806f : FAILURE in 23m 10s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/20cf4cac985049c6ad5f13fb3efb834a : FAILURE in 25m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2c821963e1814f518422027ec640aa1a : FAILURE in 33m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/328e5dac55bd4762bbab416fc3aa0a48 : FAILURE in 36m 05s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/7d01260fc7fd436b8611c948251b5be1 : FAILURE in 36m 44s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ea05dece1cd842c79d984422f7bfc841 : FAILURE in 39m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bc4ba846a10548ff8071206c1f401ad3 : FAILURE in 36m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"313edcdc7d429b388f67aa556f83580d6abab86b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-05-01 09:44:00.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"28a385ba84a2918b127e7ede3dd0240eee269872","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 11:32:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc4cc85e516247c3a7994c5661b739d1 : SUCCESS in 8m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8282439914094272bd75867bdfdba520 : SUCCESS in 9m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2b3b7f056364f768b62e45285294bac : SUCCESS in 5m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/51883ead9cc0443db1b53a4124573280 : SUCCESS in 4m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/126ebeb63fce41509e6773be33d05115 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3259fd604b64ccf8e5c63e0b2e4549d : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b589281ecfb3431fad23a5b224d17ab6 : SUCCESS in 6m 53s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/02d6ad5f412843e7a7217a1362f0f62d : SUCCESS in 1h 16m 06s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/63720e7331a047b9906cdf28de420891 : SUCCESS in 1h 17m 08s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/913a08085dcc44f59dc13e7f6c40b21b : SUCCESS in 1h 27m 06s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/4fa2bd9d9a394592a8e22a31179790a2 : SUCCESS in 1h 22m 34s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6c4f6c3fb7e54dde8a23b5282788436a : SUCCESS in 6m 57s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/4fc1f38c090c47fba11d14aa9d076f80 : SUCCESS in 6m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0df1950cd66f42dd8cd9e00540be9f7e : SUCCESS in 6m 22s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4041a90207f245e696be89c9b93d4e58 : SUCCESS in 1h 02m 42s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/53e6a9e8b3b543aa9ae7f458bec45fc9 : SUCCESS in 1h 01m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b808536be8164087a4cda3a9c4d940fd : SUCCESS in 42m 20s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/79af2be0788c4da9a7ed40c150823049 : SUCCESS in 41m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/31e60207db1e4f768405b5ffcddbed7c : SUCCESS in 1h 37m 17s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/914aec8a775d4eec94310eca14650339 : SUCCESS in 1h 39m 45s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3e5cfa47ad00428bbdd9710f93d48d23 : FAILURE in 20m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e65e2c7c23234a74973e210bc46f7eb1 : FAILURE in 32m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cd9aa7ef7f3a4c7bbb02c7beb7a1a622 : SUCCESS in 1h 42m 23s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/a363178c44c74ec69d6ad3fc82ed9b9a : SUCCESS in 1h 43m 35s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cfec9c82bd494bb18a58a249b504bedc : FAILURE in 39m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f91173657dda4cbf907b15b6c7e52f8a : FAILURE in 56m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a85179d8304d90a93f61490b8f9992565a13c887","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-11 06:28:05.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"04a7f63cc8fc17597fa906d3da0cd71b015ffda2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 09:32:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5dd74f171c54ae3b8417ee80be4d29a : SUCCESS in 9m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/995f67eacb4141fb90fd18b78b26e291 : SUCCESS in 10m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e65d16bbde1459f8758e9d89948436d : SUCCESS in 8m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5d8efcc7d3424537ab2441561f8f4ea5 : SUCCESS in 8m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/dcfe97868036438e97a150073137e08a : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc6ff38f15034f0c9e80500459edbc18 : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/581fb5f0a0164de5abb72c62ef8dd3c3 : SUCCESS in 8m 42s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/c5c2c2ffd742495b94bf64beb70891b9 : SUCCESS in 1h 17m 31s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/ec18e0fcb89749b78a17f14f358116fc : SUCCESS in 1h 17m 47s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9b279568e7784870ae578739b5fbe392 : TIMED_OUT in 2h 11m 23s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/200790072f7745de9d574e91fb386435 : SUCCESS in 1h 23m 18s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/aa0a1bb2661b4229b2fcda21cb63012d : SUCCESS in 9m 42s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/4c6deac4555a47a8940cf4dc90c25069 : SUCCESS in 8m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bae34c9d7d4d4e41adac20bd82bd6371 : SUCCESS in 12m 19s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/49179409838c4920badc17dd88bb374d : SUCCESS in 1h 14m 36s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/af4ddd301bc24c9b9e02da529700e8ce : SUCCESS in 1h 00m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2fc8d7433be9419d92a448d49b369fb8 : SUCCESS in 47m 11s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/de5405f00a1948bfa5c8086cafebb465 : SUCCESS in 1h 04m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c23620b870db4ee9ad0364c22e4634d3 : SUCCESS in 2h 08m 04s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2b6039f96430499aaf0415fe0b162e5e : SUCCESS in 1h 33m 03s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/56abb18de97943f6a1db34e4e7538adb : FAILURE in 14m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b8b54d2f85b40d9b45f3066af58c6a1 : FAILURE in 38m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/52938b871214428cbaf2d8942161a9d8 : SUCCESS in 1h 56m 35s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/36d8c3a955e4444e815ea6176a12d0b6 : SUCCESS in 1h 46m 26s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fe9804c2ee7b4e41a9b6e9e587d33318 : SUCCESS in 1h 13m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/aa7499462e8d4530936b95049e3e57bc : FAILURE in 59m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9915859f562cacc9343ae2c885d4a4b033e5bfdd","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-11 09:46:16.000000000","message":"Patch Set 3: Code-Review-1\n\nWith DIB now pinned in our stein jobs, do we still need this?","accounts_in_message":[],"_revision_number":3},{"id":"0775133748ec5ed38e687458cb1c7d88cd5bf60f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-11 16:21:32.000000000","message":"Abandoned\n\nI agree, now that DIB is pinned this is less important.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0a85a7b1177097110b6d357a33407fb1d878c217","revisions":{"f9d151af921443f2514cd9958abb0eeb4f300736":{"kind":"REWORK","_number":1,"created":"2020-02-06 21:56:20.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/96/706396/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/706396/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/706396/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/706396/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/706396/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/706396/1"}}},"commit":{"parents":[{"commit":"c563d24681b814be05e2aa519a777b0247adb995","subject":"Merge \"Add listener and pool protocol validation\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c563d24681b814be05e2aa519a777b0247adb995"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-06 21:50:56.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-06 21:56:19.000000000","tz":-480},"subject":"Add a periodic image build job","message":"Add a periodic image build job\n\nThis patch adds a periodic job to build the amphora image using the\nreleased version of diskimage-builder (our other jobs use Git master).\nThis job will not publish the produced image, it is only to test the\nbuild.\n\nChange-Id: I611d9cd5a519abc6a330ec2c46a2f5a1fdb89313\n(cherry picked from commit 1f9915037ad68356231691607b461c30b28df3b7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f9d151af921443f2514cd9958abb0eeb4f300736"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f9d151af921443f2514cd9958abb0eeb4f300736"}]},"branch":"refs/heads/stable/stein"},"2d63b2fc3bae7aec88b159b0b7037f2f99159100":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-01 09:44:00.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/706396/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/706396/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/706396/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/706396/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/706396/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/706396/2"}}},"commit":{"parents":[{"commit":"922c4987fb4db019212c7db322a65c3998f0c1a9","subject":"Remove xenial based jobs from stein gates.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/922c4987fb4db019212c7db322a65c3998f0c1a9"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-06 21:50:56.000000000","tz":-480},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-05-01 09:44:00.000000000","tz":0},"subject":"Add a periodic image build job","message":"Add a periodic image build job\n\nThis patch adds a periodic job to build the amphora image using the\nreleased version of diskimage-builder (our other jobs use Git master).\nThis job will not publish the produced image, it is only to test the\nbuild.\n\nChange-Id: I611d9cd5a519abc6a330ec2c46a2f5a1fdb89313\n(cherry picked from commit 1f9915037ad68356231691607b461c30b28df3b7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2d63b2fc3bae7aec88b159b0b7037f2f99159100"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2d63b2fc3bae7aec88b159b0b7037f2f99159100"}]},"branch":"refs/heads/stable/stein"},"0a85a7b1177097110b6d357a33407fb1d878c217":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-06-11 06:28:05.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/96/706396/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/706396/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/706396/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/706396/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/706396/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/706396/3"}}},"commit":{"parents":[{"commit":"aefbef124586d624d8978342942585b6da2049af","subject":"Do not install diskimage-builder from Git","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aefbef124586d624d8978342942585b6da2049af"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-06 21:50:56.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-11 06:28:05.000000000","tz":0},"subject":"Add a periodic image build job","message":"Add a periodic image build job\n\nThis patch adds a periodic job to build the amphora image using the\nreleased version of diskimage-builder (our other jobs use Git master).\nThis job will not publish the produced image, it is only to test the\nbuild.\n\nChange-Id: I611d9cd5a519abc6a330ec2c46a2f5a1fdb89313\n(cherry picked from commit 1f9915037ad68356231691607b461c30b28df3b7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a85a7b1177097110b6d357a33407fb1d878c217"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a85a7b1177097110b6d357a33407fb1d878c217"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
