)]}'
{"id":"openstack%2Foctavia~706397","triplet_id":"openstack%2Foctavia~stable%2Frocky~I611d9cd5a519abc6a330ec2c46a2f5a1fdb89313","project":"openstack/octavia","branch":"stable/rocky","hashtags":[],"change_id":"I611d9cd5a519abc6a330ec2c46a2f5a1fdb89313","subject":"Add a periodic image build job","status":"ABANDONED","created":"2020-02-06 21:59:15.000000000","updated":"2020-02-26 15:55:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a7a5fd96de61a41abd7875e927ac65bec790b680","_number":706397,"virtual_id_number":706397,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-02-06 22:38:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-06 22:38:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c650fdb61f52e97d36713ad68c06b78dc5a60266","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-06 21:59:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec2eb2576d158b3ec21adbfacf54d388d7ea1643","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 22:38:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5096f09f60e34548b4ad2fa52632c8bb : SUCCESS in 9m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/532ad7ae06f846a188deb734d669f793 : SUCCESS in 8m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/72956159e78e4598a7ebb5439209c530 : SUCCESS in 6m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d5fce78e514d430ea00146234bd7ec06 : SUCCESS in 7m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71a839ec1d894c82bba46be2e8af5c69 : SUCCESS in 8m 00s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/d7487fcb8f834b50acc20a0257b727d8 : FAILURE in 19m 51s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/b65454c5988e423cb00d5e244bf6758c : FAILURE in 20m 58s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/60714910631b482a95ec24f2882ef244 : FAILURE in 20m 01s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/2f9e7a253a414cdab6e5857046aab359 : FAILURE in 20m 22s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e961e7b608ab4fa6aa5c1e7e660d3269 : SUCCESS in 8m 13s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/b9da1a8f49964435b3240295ac6f158b : SUCCESS in 7m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6e82f9593b7e426caf383614cbacb889 : SUCCESS in 9m 25s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0e13d1de3ff84c6b83b12e4c8e640a1b : FAILURE in 19m 12s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/3fcd22a38f6e46b0a47a0cdeeabd463d : FAILURE in 19m 41s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/32d89e1f2ead40ecbcc77875c237b49f : FAILURE in 21m 58s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/cca597863f0542dcad723cb43bb1af5f : FAILURE in 24m 57s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/3942e006250542d99e9331fb19a10cfd : FAILURE in 19m 25s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7e0f880c3445486086b9ff09e282f71a : FAILURE in 17m 31s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/496858044e844d3ca3a06039f1918ed4 : FAILURE in 20m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/47b8e355ef12417daaf5650b7f5f0c11 : FAILURE in 18m 20s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/3ae328fd2f1743e8a79e44c692a83638 : FAILURE in 19m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6a8f5148658342c48e072e685bb5f631 : FAILURE in 19m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/710053517ac9410cb0b57491875c3c20 : SUCCESS in 11m 05s","accounts_in_message":[],"_revision_number":1},{"id":"a7a5fd96de61a41abd7875e927ac65bec790b680","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-26 15:55:02.000000000","message":"Abandoned\n\nAbandon, rocky is going to EM","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"27f3a6a1416f2925e7d86da4cc4d5f725e59f387","revisions":{"27f3a6a1416f2925e7d86da4cc4d5f725e59f387":{"kind":"REWORK","_number":1,"created":"2020-02-06 21:59:15.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/97/706397/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/97/706397/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/706397/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/706397/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/706397/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/97/706397/1"}}},"commit":{"parents":[{"commit":"34545524b6cc194c788172aaf3e565513f2e4d74","subject":"Add listener and pool protocol validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34545524b6cc194c788172aaf3e565513f2e4d74"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-06 21:50:56.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-06 21:59:14.000000000","tz":-480},"subject":"Add a periodic image build job","message":"Add a periodic image build job\n\nThis patch adds a periodic job to build the amphora image using the\nreleased version of diskimage-builder (our other jobs use Git master).\nThis job will not publish the produced image, it is only to test the\nbuild.\n\nChange-Id: I611d9cd5a519abc6a330ec2c46a2f5a1fdb89313\n(cherry picked from commit 1f9915037ad68356231691607b461c30b28df3b7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27f3a6a1416f2925e7d86da4cc4d5f725e59f387"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27f3a6a1416f2925e7d86da4cc4d5f725e59f387"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
