)]}'
{"id":"openstack%2Foctavia~707091","triplet_id":"openstack%2Foctavia~stable%2Fqueens~I094e9125fd00b229ac852f24da41aec0b27dc9a2","project":"openstack/octavia","branch":"stable/queens","hashtags":[],"change_id":"I094e9125fd00b229ac852f24da41aec0b27dc9a2","subject":"Test","status":"ABANDONED","created":"2020-02-11 10:16:00.000000000","updated":"2020-02-11 11:48:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"06e6d225da79b7df8c11f30c98af330ec62258e5","_number":707091,"virtual_id_number":707091,"owner":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-02-11 11:40:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-11 11:40:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"954a45c4400f4701aa332cf11923f35530d8da2a","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-02-11 10:16:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"516ec4206390e07dc16b167e8bbacded117d63df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 11:40:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d210eaf268a842d6ab908eb1f87be551 : SUCCESS in 10m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/89721cc1e4b34cf5bb397e77c68a85d9 : SUCCESS in 6m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2ac68cf6a4984692b390487c5dfd7d91 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5be2a50f80314e60ad10739d8f937654 : SUCCESS in 9m 18s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/6fdfc06c008344568e3d09b87847fc2c : FAILURE in 11m 41s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/04910fd18eb44919aa891912472038ea : SUCCESS in 1h 17m 13s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9cf8577b5b4c49e2a39d35233854318e : FAILURE in 13m 01s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/c5d4a391f2a94a90aea57169ef52ce74 : FAILURE in 33m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6a16dab41f24b70accdfe3d7b83210e : SUCCESS in 9m 20s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/138ce301358f40b899c2f5d3025a8083 : SUCCESS in 10m 26s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/e4b62fb2752449329b7f0adf6df39f2e : SUCCESS in 7m 04s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9f949f470a3244d582bf4ccacb4bf70c : FAILURE in 14m 44s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/582b4d81a7374d0ab0748682ec48b122 : FAILURE in 35m 04s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/d0dd488957a24f389c8325196ff5fb07 : FAILURE in 16m 26s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/7880ea05233046bdabfd5b997621fe65 : FAILURE in 37m 05s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/a3f63e9e457a4d65908de71c5e35a1af : FAILURE in 13m 04s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/54e6e630bc5342539a6037315afa2d12 : SUCCESS in 44m 03s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/ceb23547b5a44190afaaedebd249e145 : SUCCESS in 43m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d95caf911c03464d9a6d6a36613575f7 : FAILURE in 30m 12s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/46401278da85445b891ab9c94cb7c459 : FAILURE in 31m 29s","accounts_in_message":[],"_revision_number":1},{"id":"06e6d225da79b7df8c11f30c98af330ec62258e5","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-02-11 11:48:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"56417464c1efb1c402ccd54e14f6b22097b35026","revisions":{"56417464c1efb1c402ccd54e14f6b22097b35026":{"kind":"REWORK","_number":1,"created":"2020-02-11 10:16:00.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/91/707091/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/707091/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/707091/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/707091/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/707091/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/707091/1"}}},"commit":{"parents":[{"commit":"e3ddec776da4654517a237287c05298a9280e52f","subject":"Add listener and pool protocol validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3ddec776da4654517a237287c05298a9280e52f"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2020-01-30 10:14:55.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-02-11 10:15:20.000000000","tz":240},"subject":"Test","message":"Test\n\nChange-Id: I094e9125fd00b229ac852f24da41aec0b27dc9a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/56417464c1efb1c402ccd54e14f6b22097b35026"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/56417464c1efb1c402ccd54e14f6b22097b35026"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
