)]}'
{"id":"openstack%2Foctavia~708232","triplet_id":"openstack%2Foctavia~master~I6794144b2ba66f3bbd2bc948d95d1019286c9889","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I6794144b2ba66f3bbd2bc948d95d1019286c9889","subject":"Fix a ValueError with mock 4.0.1","status":"ABANDONED","created":"2020-02-17 19:00:35.000000000","updated":"2020-02-17 21:47:57.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"38b4e37894c5d3bd0b822dec5cc1aabbdeccd5fc","_number":708232,"virtual_id_number":708232,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"date":"2020-02-17 20:18:55.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":0,"date":"2020-02-17 20:21:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-02-17 20:19:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-02-17 20:21:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2020-02-17 20:21:35.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"date":"2020-02-17 20:46:41.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":-2,"date":"2020-02-17 20:21:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-17 20:18:55.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-02-17 20:19:01.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-02-17 20:21:35.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-02-17 20:21:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7ee3b341bad40ac6388be3f9c67a79ddf2c0117e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-17 19:00:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6427918402c461d3d99eadfcb80edf605e3479c0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-17 19:02:02.000000000","message":"Patch Set 1: Workflow-1\n\nNot sure this is really the right fix. Asking for comments/reviews","accounts_in_message":[],"_revision_number":1},{"id":"a9669a12f0181b61093161c3c8965a3df8915729","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 19:20:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32917fcc479346b0bbd694022f6469a6 : SUCCESS in 10m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fefe8960898f421a89a4805e574306dc : SUCCESS in 9m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9790a5b010d48da8aaee41c0ecf5d1c : SUCCESS in 9m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dd99180477a4e34966dd5b4fe1842d6 : SUCCESS in 6m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/be0074734b5f446aaf1411ad393f715e : SUCCESS in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87dc817c882a43a99fb02521f6c75d0b : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0de4201cb0945baa2966a4010085949 : SUCCESS in 14m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/96e45f2676ee4887b827aa0277c548a3 : SUCCESS in 6m 56s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/28f4ab464665437f80c639a23f55d17c : SUCCESS in 8m 42s","accounts_in_message":[],"_revision_number":1},{"id":"02848cc1bb62832649bc686a3e8675768935fe6b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-17 20:16:29.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"7583fd9aee08ef878d97fb0a780f782ca5f14abd","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-02-17 20:18:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c35d3bb4102879f06470016e795c8cc8dbbefa44","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-02-17 20:19:01.000000000","message":"Patch Set 1: Code-Review+2\n\nYeah who cares what random BS we have to do with Mock, so long as it works. This is more explicit testing still than we do almost anywhere else.","accounts_in_message":[],"_revision_number":1},{"id":"fb91707e2bb63ca26e726ff257d5ac32990fd2df","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-17 20:21:35.000000000","message":"Patch Set 1: Backport-Candidate-2 Code-Review+2 Workflow+1\n\nNo need to backport. Stable branches have mock upper constrained to 3.0.5","accounts_in_message":[],"_revision_number":1},{"id":"be7c3e72cb6e8947b97451145f15e20158f9478b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 20:21:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1d18ec0dedc1a2dfe0b5312dec3f4107a700d341","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-17 20:24:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"99d8861bb7d261b29487fc0c95d169d2e27386b3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-17 20:46:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"38b4e37894c5d3bd0b822dec5cc1aabbdeccd5fc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-17 21:47:57.000000000","message":"Abandoned\n\nWe are blacklisting mock 4.0.1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dcb383c6af0a612efd3327c2fcb3da7e0b390206","revisions":{"dcb383c6af0a612efd3327c2fcb3da7e0b390206":{"kind":"REWORK","_number":1,"created":"2020-02-17 19:00:35.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/32/708232/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/708232/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/708232/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/708232/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/708232/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/708232/1"}}},"commit":{"parents":[{"commit":"bb0efa2a5d4e30636b328d27c76f204a0f40407f","subject":"Add docs warning for PING health monitors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bb0efa2a5d4e30636b328d27c76f204a0f40407f"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-17 18:58:04.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-17 19:00:33.000000000","tz":-480},"subject":"Fix a ValueError with mock 4.0.1","message":"Fix a ValueError with mock 4.0.1\n\nMock version 4.0.1 raises a ValueError with the current mock \"spec\"\nfor the secrets.Secret[1]. This patch resolves this issue by running\n\"dir(secrets.Secret)\" for mock. This is what mock should be doing\nautomatically, but this object appears to be confusing it.\n\n[1] https://zuul.opendev.org/t/openstack/build/ \\\n    c5fd6cadc9bf4b3b9ea81372ef047246/log/job-output.txt#4605\n\nChange-Id: I6794144b2ba66f3bbd2bc948d95d1019286c9889\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dcb383c6af0a612efd3327c2fcb3da7e0b390206"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dcb383c6af0a612efd3327c2fcb3da7e0b390206"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
