)]}'
{"id":"openstack%2Foctavia~708265","triplet_id":"openstack%2Foctavia~master~Ie3f11f240e57ce3b272a2c0c79d89172e9ddc363","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ie3f11f240e57ce3b272a2c0c79d89172e9ddc363","subject":"Do not prematurely remove build dependencies","status":"NEW","created":"2020-02-17 22:02:45.000000000","updated":"2020-02-17 23:53:16.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cc50d8bc045f9b86ab64c6672c3fe3aca21911f4","_number":708265,"virtual_id_number":708265,"owner":{"_account_id":28397,"name":"Curt Moore","email":"curt.moore@gmail.com","username":"jcmoore"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-02-17 23:53:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":28397,"name":"Curt Moore","email":"curt.moore@gmail.com","username":"jcmoore"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-02-17 22:10:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2020-02-17 22:03:29.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28397,"name":"Curt Moore","email":"curt.moore@gmail.com","username":"jcmoore"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":28397,"name":"Curt Moore","email":"curt.moore@gmail.com","username":"jcmoore"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":28397,"name":"Curt Moore","email":"curt.moore@gmail.com","username":"jcmoore"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2020-02-17 22:03:29.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28397,"name":"Curt Moore","email":"curt.moore@gmail.com","username":"jcmoore"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28397,"name":"Curt Moore","email":"curt.moore@gmail.com","username":"jcmoore"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-17 22:10:25.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-02-17 23:53:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"877ebaa8aa680eeef0743749b6d4f9e7b7096c18","author":{"_account_id":28397,"name":"Curt Moore","email":"curt.moore@gmail.com","username":"jcmoore"},"date":"2020-02-17 22:02:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"595ce22f01d2e58e13de405e09b609b848447c46","author":{"_account_id":28397,"name":"Curt Moore","email":"curt.moore@gmail.com","username":"jcmoore"},"date":"2020-02-17 22:03:29.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bb884f322404010514d13cd149a54eb9e8f61d0c","author":{"_account_id":28397,"name":"Curt Moore","email":"curt.moore@gmail.com","username":"jcmoore"},"date":"2020-02-17 22:05:07.000000000","message":"Patch Set 1:\n\nThis was initially observed in stable/train so it would be nice to have this backported there as well.","accounts_in_message":[],"_revision_number":1},{"id":"0443785e5871ea57684d22fe8567ba51cfbd6582","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-17 22:09:54.000000000","message":"Patch Set 1: Code-Review-2\n\nI think this is a bug in the dkms element in diskimage-builder.\n\nPer the diskimage-builder documentation: https://docs.openstack.org/diskimage-builder/latest/developer/developing_elements.html\n\n\"Elements in the upstream diskimage-builder elements should not create executables which run before 10- or after 90- in any of the phases if possible. This is to give downstream elements the ability to easily make executables which run after our upstream ones.\"\n\nSo, really dkms should run prior to the 9x- phases.","accounts_in_message":[],"_revision_number":1},{"id":"9edd638673caad440935b9522fe823faaa449792","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-17 22:10:25.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"88107df0ad84443095d3f87540d143e1d9ae62d2","author":{"_account_id":28397,"name":"Curt Moore","email":"curt.moore@gmail.com","username":"jcmoore"},"date":"2020-02-17 22:40:37.000000000","message":"Patch Set 1:\n\n\u003e I think this is a bug in the dkms element in diskimage-builder.\n \u003e \n \u003e Per the diskimage-builder documentation: https://docs.openstack.org/diskimage-builder/latest/developer/developing_elements.html\n \u003e \n \u003e \"Elements in the upstream diskimage-builder elements should not\n \u003e create executables which run before 10- or after 90- in any of the\n \u003e phases if possible. This is to give downstream elements the ability\n \u003e to easily make executables which run after our upstream ones.\"\n \u003e \n \u003e So, really dkms should run prior to the 9x- phases.\n\nThanks for the feedback.\n\nWith that in mind, there are currently numerous post-install upstream executables which fall into those ranges:\n\n $ find elements -type f -wholename \"*/post-install.d/9*\" -o -wholename \"*/post-install.d/0*\" | sort\n elements/dkms/post-install.d/97-dkms\n elements/dracut-ramdisk/post-install.d/01-ensure-drivers\n elements/dracut-ramdisk/post-install.d/99-build-dracut-ramdisk\n elements/grub2/post-install.d/01-delete-grubenv\n elements/hpdsa/post-install.d/95-hpdsa\n elements/hpdsa/post-install.d/98-hpdsa-post-dkms\n elements/iso/post-install.d/01-copy-bootloaders\n elements/openssh-server/post-install.d/99-harden-sshd-config\n elements/package-installs/post-install.d/00-package-installs\n elements/package-installs/post-install.d/95-package-uninstalls\n elements/pypi/post-install.d/00-unconfigure-pypi-mirror\n elements/ramdisk/post-install.d/99-build-ramdisk\n elements/ramdisk-base/post-install.d/01-ensure-binaries\n elements/rpm-distro/post-install.d/06-network-config-nonzeroconf\n elements/ubuntu-signed/post-install.d/90-get-signed-kernel\n elements/yum/post-install.d/99-reset-yum-conf\n elements/zypper/post-install.d/98-zypper-clean-cache\n elements/zypper/post-install.d/99-zypper-no-keep-packages\n elements/zypper-minimal/post-install.d/01-locale-cleanup\n\nShould we change the remove-build-deps script in Octavia or work to change the executable indices in upstream diskimage-builder?\n\nMy fear is that changing the indexes upstream may have have larger fallout.","accounts_in_message":[],"_revision_number":1},{"id":"cc50d8bc045f9b86ab64c6672c3fe3aca21911f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 23:53:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/904b55b10b96464dbb92fda6ba23e5dc : SUCCESS in 12m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06f0e7a318aa4434acbaefcca04de6ae : SUCCESS in 10m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77091e5edba44805b8085df09934b246 : SUCCESS in 8m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b98f30351f044640b93187c28b875679 : SUCCESS in 5m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e46020c3ef0b43e2b6f30bf5aa5a4898 : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9180191e72864d4ba559c6c5a134398d : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7092adffa45e4e2695315e96b2b9de37 : SUCCESS in 14m 19s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0174f99411464518ab39534e9bff74ae : SUCCESS in 7m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8c58a3244182490ba518d58a76fe312b : SUCCESS in 9m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6f4beffe9efd4f2d8cec4582a11eb797 : SUCCESS in 8m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8cecb14d0118452b91f9d059fbc6ebfc : FAILURE in 45m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6f47ef8568df4f9cbbf670866c6cfbb1 : FAILURE in 1h 34m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ee46c8be7d0a479b83803b4cdd5c6e29 : SUCCESS in 55m 18s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1da85ef8326b45bda0bc6aa89ae19edb : SUCCESS in 55m 16s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fb33fa5de389416b8294ea8f124f4221 : SUCCESS in 1h 01m 06s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/51727f1fd63c4b7fbe25231e74d83c58 : SUCCESS in 1h 41m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ca12710393e44219ad4030180f938eaa : SUCCESS in 1h 41m 56s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/45643bcea8544f17905c430ca4e8e279 : FAILURE in 58m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/003d79780f6e4e2b88698e4f919d7cb9 : SUCCESS in 1h 33m 23s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ec426da2d98386b1074da8594cff6dd1137d1794","revisions":{"ec426da2d98386b1074da8594cff6dd1137d1794":{"kind":"REWORK","_number":1,"created":"2020-02-17 22:02:45.000000000","uploader":{"_account_id":28397,"name":"Curt Moore","email":"curt.moore@gmail.com","username":"jcmoore"},"ref":"refs/changes/65/708265/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/708265/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/708265/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/708265/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/708265/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/708265/1"}}},"commit":{"parents":[{"commit":"bb0efa2a5d4e30636b328d27c76f204a0f40407f","subject":"Add docs warning for PING health monitors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bb0efa2a5d4e30636b328d27c76f204a0f40407f"}]}],"author":{"name":"Curt Moore","email":"curt.moore@gmail.com","date":"2020-02-17 21:54:07.000000000","tz":-360},"committer":{"name":"Curt Moore","email":"curt.moore@gmail.com","date":"2020-02-17 21:59:37.000000000","tz":-360},"subject":"Do not prematurely remove build dependencies","message":"Do not prematurely remove build dependencies\n\nThe dkms element has a post-install.d entry with an index of 97.\nWhen the build dependencies outlined in this script are removed,\ndkms is removed as a dependency, causing the dkms step to fail.\n\nChange-Id: Ie3f11f240e57ce3b272a2c0c79d89172e9ddc363\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ec426da2d98386b1074da8594cff6dd1137d1794"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ec426da2d98386b1074da8594cff6dd1137d1794"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28397,"name":"Curt Moore","email":"curt.moore@gmail.com","username":"jcmoore"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28397,"name":"Curt Moore","email":"curt.moore@gmail.com","username":"jcmoore"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
