)]}'
{"id":"openstack%2Foctavia~708269","triplet_id":"openstack%2Foctavia~master~I72520a2ca010c2c27315d9dff839a4f9d7540b6b","project":"openstack/octavia","branch":"master","topic":"native-mock","hashtags":[],"change_id":"I72520a2ca010c2c27315d9dff839a4f9d7540b6b","subject":"Remove the dependency on the \"mock\" package","status":"MERGED","created":"2020-02-17 22:28:38.000000000","updated":"2020-02-18 19:41:07.000000000","submitted":"2020-02-18 19:38:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708269-1582054706139-2176cf01","meta_rev_id":"dab24db0defd88e5b15261b32fb34d4a8d0501c9","_number":708269,"virtual_id_number":708269,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"date":"2020-02-18 16:01:57.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-18 19:38:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-02-17 23:01:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-02-18 14:29:29.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2020-02-18 14:29:29.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-17 22:31:05.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-02-17 23:01:40.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-02-18 14:29:29.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2020-02-18 16:01:57.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-02-18 19:38:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aa5b6b684016779b5e190969b27ce347696406ad","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-17 22:28:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b7bc5a068e6aa8ec4b6d8d3607cc4eb3af7223b0","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-02-17 22:48:22.000000000","message":"Patch Set 1: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1789c81d2384e853418ebcda080ecb83cc5bf0f9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-17 22:55:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dc377ac6a00930fb27b7a8a6fa1cecc533c31a06","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-02-17 23:01:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"25014a0e0e04bfced457c16a43f2271f57562150","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 00:46:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5aab1e31dc624cae864a43b83fd5e2b7 : SUCCESS in 7m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2b6d55647f2742bfbce8cd926254c680 : SUCCESS in 3m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc5d49f518924bf580ce5d85f3ce1a14 : SUCCESS in 14m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ecef6eb48014bafa271ea160f84adeb : SUCCESS in 11m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/102c8dd7548840a8909a5bf51e040d4f : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbefd265433c400da2c07d6585211a34 : SUCCESS in 5m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf6bb4ee18bb4503a6eec8a53c2ff590 : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e33f4b5af30342c6b932bc120e0b147c : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0342f004dc794aa6a7dfe4f0c2bb6d5a : SUCCESS in 13m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea123a42ce62440bb0496a98bcb04368 : SUCCESS in 11m 46s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bad98b4af47d46368118279e92b1c854 : SUCCESS in 11m 29s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bfc1b48df1fb41aaa282dcade57c3391 : SUCCESS in 12m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1b77f3b7f7fa4d0f8d653e685eb86f07 : SUCCESS in 8m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/04885cfe78404854969a949d5d37e2d8 : SUCCESS in 44m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ca995d7ef0e546bc9d76c2df9126a1c5 : SUCCESS in 1h 44m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/691bbf9975eb493c86b6571c5840c33d : SUCCESS in 1h 06m 39s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/cbdbde63ea8c4f168bf2908c4e4b8d37 : SUCCESS in 58m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/456d7b834f93417ea6c19b2e52fa4d49 : SUCCESS in 1h 08m 20s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9e92c4b941444926937137f41d378252 : SUCCESS in 1h 43m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6a24a77edca54c199b02e47e9b0776fa : FAILURE in 1h 25m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/588d4579b040477ab4ae068aca7c9479 : SUCCESS in 1h 40m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d7bc8b5aca524cbbbd1368498c9ab3d93b70887f","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2020-02-18 14:29:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fc15cbf1f62383ae01fa5e667c6475ac54f67eaf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 14:29:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"53ea86d76dca0d26dfb6db39b6d08c848cd7fdd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 15:02:03.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d11b420427b54dc08a3316afe44f6531 : SUCCESS in 8m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/968eb85b57fd42ce8446e8fc637c5a20 : SUCCESS in 5m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ca0474f1ddc4e14b39ff0407c409488 : SUCCESS in 12m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c49d52e74dd64c529d7f88b660e8467f : SUCCESS in 9m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ebf8d5e591945a39dcfac28afc5946b : SUCCESS in 7m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2039c9702e55481e9990f8bca5c6f045 : SUCCESS in 8m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/071e49ec0441494894d5d0b354d70a63 : SUCCESS in 14m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24c034ee12ac4505bc14761a5784d506 : SUCCESS in 8m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/db46f4a1bb7a412aad8d0582db873166 : SUCCESS in 10m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/098d8795bcc44955b2c47a6d45683a57 : CANCELED in 28m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/28ae301eeb8f4e9c9346e29ce07d0062 : CANCELED in 29m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3ae15695f5254968b38a4511b93dfa83 : CANCELED in 27m 08s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8bee6d46103b4d73b4cadc357abf2c26 : FAILURE in 29m 14s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f494cfc25d9f4f87901f45f0c8dded44 : CANCELED in 25m 55s","accounts_in_message":[],"_revision_number":2},{"id":"714652347de6df35e7d71b4922e14f843cd6b07b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-18 16:01:57.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d6bef063c8d6d7516fee5b9e90ad89665fd7eef9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 18:08:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8f882921fe184ee4a9c2ebe34e148e43 : SUCCESS in 11m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7fe55905df314ef79881220e06b3ba99 : SUCCESS in 4m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/272ecd20c9a648b9899547e019e73ba8 : SUCCESS in 10m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e27b43f1ae6c43e0b2af41880b6c783b : SUCCESS in 10m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42c1d6f8b7b341d29c26e246bd6a69f7 : SUCCESS in 10m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/535cd7bb7e414b9c8d5ce69fa54dd72b : SUCCESS in 7m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6433072ee08c437e88dc00606c0258c4 : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae85d911cc9147309aa6292cea99d63e : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62fa838299dd46d1a385cbecdfe46a3e : SUCCESS in 13m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c89f2a60dd68413889725bd72d2e6629 : SUCCESS in 9m 27s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/86c733c520f54e85af0577abc1d616a5 : SUCCESS in 6m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7f7025f46fc0476fb08b71fff5049afa : SUCCESS in 9m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5096d97509ac4e51a80542af6be82d9a : SUCCESS in 8m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cf9f45ddfcb0489aae929006ec9c234a : SUCCESS in 50m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c5d22a8d0a154c35b7a156742fa81ac4 : SUCCESS in 1h 44m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c5cc44f6315a4a088b1ccbcca079407e : SUCCESS in 1h 13m 16s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7baf5086d0bf4025af410e5dc50ab9d2 : SUCCESS in 1h 09m 27s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b90695609cfb4977aaaaa38cf6b2171a : SUCCESS in 1h 04m 12s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dc9829d9e2054e9382be5ad8b772df05 : SUCCESS in 1h 37m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3bc53b5d9f5f4c48a9684ec37ff32765 : SUCCESS in 1h 59m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1af9e463bfec4b31956c902be4a8bed7 : SUCCESS in 1h 39m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fb438d97044ca9e5d359694f95aad7fbdff104ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 18:10:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1281ce1214144d3fda5ee6fa21b67c51ca8073bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 19:38:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d47872e60ea346f5b2d2b7cca7c218ef : SUCCESS in 8m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8a9749cf5e2d409bbe162d441d5a9cc3 : SUCCESS in 6m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d9d98fac63145a680b62099883a3b3e : SUCCESS in 10m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe8cd8d5d8ab4a6ea33c5c8279dcaa95 : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c22be92cc23408ebb433f8dbc297204 : SUCCESS in 7m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/34a890d8fc654fb5b910ae5776841d02 : SUCCESS in 10m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a42f99adbe7448f79d4f1bacc5c67b15 : SUCCESS in 14m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a5cfc5007e44fc5bc0ceaf21b107d01 : SUCCESS in 10m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e72c9cf2882e4241ab16ad69f8b0f92f : SUCCESS in 10m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2646393edf1d4fbea7941ad4c552da1d : SUCCESS in 45m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d8eb51d472904ce48680fab3b5aaa3fe : SUCCESS in 1h 24m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9c4bbba7d3d04ffbb9d1e099e5df28da : SUCCESS in 1h 11m 37s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7f7349b2bea8422db0de256d0f968dc8 : SUCCESS in 53m 01s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bcec235f2f70423890336b4e3da79cca : SUCCESS in 1h 15m 15s","accounts_in_message":[],"_revision_number":2},{"id":"fca6874f838f4cb5ca7bd7bd93ea4e7a25e4002d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 19:38:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"fab25c1938811ebf5287e005bed225857cc4461e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 19:41:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/3fd182828f8c4115b69384f25b17c46d : SUCCESS in 1m 07s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71e88935d381431fbea2951094fda854 : SUCCESS in 2m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0e4ceaacb1b4ef98c149657d922df4c : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"49036845c118aec11740f18680b779f99113bcfc","revisions":{"b41f4e5b05d2a825d9ec872ff4c027429cbdb78e":{"kind":"REWORK","_number":1,"created":"2020-02-17 22:28:38.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/69/708269/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/708269/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/708269/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/708269/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/708269/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/708269/1"}}},"commit":{"parents":[{"commit":"bb0efa2a5d4e30636b328d27c76f204a0f40407f","subject":"Add docs warning for PING health monitors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bb0efa2a5d4e30636b328d27c76f204a0f40407f"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-17 22:23:29.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-17 22:28:30.000000000","tz":-480},"subject":"Remove the dependency on the \"mock\" package","message":"Remove the dependency on the \"mock\" package\n\nNow that we are python3 only, we should move to using the built\nin version of mock that supports all of our testing needs and\nremove the dependency on the \"mock\" package.\n\nThis patch moves all references to \"import mock\" to\n\"from unittest import mock\". It also cleans up some new line\ninconsistency.\n\nChange-Id: I72520a2ca010c2c27315d9dff839a4f9d7540b6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b41f4e5b05d2a825d9ec872ff4c027429cbdb78e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b41f4e5b05d2a825d9ec872ff4c027429cbdb78e"}]},"branch":"refs/heads/master"},"49036845c118aec11740f18680b779f99113bcfc":{"kind":"REWORK","_number":2,"created":"2020-02-17 22:55:25.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/69/708269/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/708269/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/708269/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/708269/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/708269/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/708269/2"}}},"commit":{"parents":[{"commit":"bb0efa2a5d4e30636b328d27c76f204a0f40407f","subject":"Add docs warning for PING health monitors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bb0efa2a5d4e30636b328d27c76f204a0f40407f"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-17 22:23:29.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-17 22:55:19.000000000","tz":-480},"subject":"Remove the dependency on the \"mock\" package","message":"Remove the dependency on the \"mock\" package\n\nNow that we are python3 only, we should move to using the built\nin version of mock that supports all of our testing needs and\nremove the dependency on the \"mock\" package.\n\nThis patch moves all references to \"import mock\" to\n\"from unittest import mock\". It also cleans up some new line\ninconsistency.\n\nChange-Id: I72520a2ca010c2c27315d9dff839a4f9d7540b6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/49036845c118aec11740f18680b779f99113bcfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/49036845c118aec11740f18680b779f99113bcfc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
