)]}'
{"id":"openstack%2Foctavia~709160","triplet_id":"openstack%2Foctavia~master~I4d20274d30eff8b02508bde09c5d76c27d64467a","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I4d20274d30eff8b02508bde09c5d76c27d64467a","subject":"Amp create should be linear with anti-affinity","status":"ABANDONED","created":"2020-02-21 15:42:21.000000000","updated":"2020-04-08 23:31:28.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2f70ca4bbd2dba1c0e0ca4c78b226b16a0b2ed7a","_number":709160,"virtual_id_number":709160,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-02-21 17:32:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-2,"date":"2020-02-21 16:07:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-21 16:07:43.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-02-21 17:32:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5ae4a3ed00a00f0a55f77af8fb7aa0494487c820","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-02-21 15:42:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3fd26277648518866e0e215275239b3f07dde8c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-21 16:07:43.000000000","message":"Patch Set 1: Code-Review-2\n\nThis is clearly a Nova bug and not something we should change. We are properly calling the nova APIs. If they are not honoring those calls it is a bug in their code.\nI have opened: https://bugs.launchpad.net/nova/+bug/1863190","accounts_in_message":[],"_revision_number":1},{"id":"511fb87c97de36591331694f34f0cdb92e2ae28e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 17:32:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/352b603de21f4b82822112db9bb96731 : SUCCESS in 12m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6098861066314f2e999635261f1a07eb : SUCCESS in 11m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1651bc26b6f843c5a041595041082f3f : SUCCESS in 10m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5db72c249c214b2bacd9e13944a20cc6 : SUCCESS in 6m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d3002468c43d410f938df39b583c7884 : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/661d63f9185442a2941c49a6b393e64f : FAILURE in 8m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf32ed8c406d4bc4a06334180617f7ac : SUCCESS in 15m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/289482d5c1e1440d839e621b2244b7b2 : SUCCESS in 8m 23s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9926d43a02004f14ba748680f795b202 : SUCCESS in 9m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c0d107ce3b8a48fc8f5077e7aee38125 : SUCCESS in 8m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cc794a17e2344d64ba91b4f1e3a97d6e : SUCCESS in 51m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bf8ec974d81b4a7d973a7f95c239af7a : SUCCESS in 1h 45m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/df3fc95480994057935e1cc1fe0fafac : SUCCESS in 1h 05m 41s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f358c13a05ba4311bec7d0ace81fa26c : SUCCESS in 56m 19s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9a87e1d9984c42c1a922cedec5b2a4cf : SUCCESS in 1h 06m 54s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/38b5f91fcf3f48d0b12432009e8fb243 : FAILURE in 1h 32m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b745ef1518d2490a92f685da02d5354b : SUCCESS in 1h 46m 12s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2be7881006e148fda3905f869f9a66c1 : FAILURE in 1h 01m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/86f8d64af00a4f5db0187b746b6d997a : SUCCESS in 1h 29m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2f70ca4bbd2dba1c0e0ca4c78b226b16a0b2ed7a","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-04-08 23:31:28.000000000","message":"Abandoned\n\nYeah this is gross, but with pack mode enabled in nova, soft-anti-affinity will never work on creates (just an FYI).","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"825f61ef6b0bfccb67974c47f2c057d215367238","revisions":{"825f61ef6b0bfccb67974c47f2c057d215367238":{"kind":"REWORK","_number":1,"created":"2020-02-21 15:42:21.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/60/709160/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/60/709160/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/709160/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/709160/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/709160/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/60/709160/1"}}},"commit":{"parents":[{"commit":"49036845c118aec11740f18680b779f99113bcfc","subject":"Remove the dependency on the \"mock\" package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/49036845c118aec11740f18680b779f99113bcfc"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-02-21 15:17:35.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-02-21 15:23:27.000000000","tz":-480},"subject":"Amp create should be linear with anti-affinity","message":"Amp create should be linear with anti-affinity\n\nIf we use an unordered flow for ACTIVE_STANDBY amp creates and\nanti-affinity is enabled, nova will not be able to properly schedule,\nand probably will require a reschedule (which is not always enabled) due\nto the two amps having a friendly scheduling race.\nUsing ordered flow is slightly slower, but is the only way to guarantee\nthe scheduler will be able to pick correctly for the second amp.\n\nChange-Id: I4d20274d30eff8b02508bde09c5d76c27d64467a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/825f61ef6b0bfccb67974c47f2c057d215367238"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/825f61ef6b0bfccb67974c47f2c057d215367238"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
