)]}'
{"id":"openstack%2Foctavia~709531","triplet_id":"openstack%2Foctavia~stable%2Ftrain~Ifb6c34426e0927534d332a8bbf2c66aac6c002c5","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"Ifb6c34426e0927534d332a8bbf2c66aac6c002c5","subject":"Fix uncaught DB exception when trying to get a spare amphora","status":"MERGED","created":"2020-02-24 13:52:02.000000000","updated":"2020-02-24 18:53:26.000000000","submitted":"2020-02-24 18:49:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709531-1582570161781-eaa21e4c","meta_rev_id":"706a287ca9060d4c3b23a0a329dae13281d6ced8","_number":709531,"virtual_id_number":709531,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-24 18:49:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":2,"date":"2020-02-24 16:14:26.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":2,"date":"2020-02-24 14:04:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":1,"date":"2020-02-24 16:14:26.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-24 14:04:16.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-02-24 16:14:26.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"},{"updated":"2020-02-24 18:49:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"73779b128a637c756f29c55c2312f4d623ac7850","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-02-24 13:52:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1507d37b93e4fbd7267e59e6f04699c6f1ccb229","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-24 14:04:16.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"48d2c372b07a5ecdc689b24c592965c9d3fcb8ec","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2020-02-24 16:14:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2297ebe07e21ae8b3aa59d7828a74b42c0cb8c58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 16:59:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c162fe95747457b92128dc5695a8b60 : SUCCESS in 11m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5dfb33b14aac45658e43588c7e5fa23f : SUCCESS in 13m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dd7193fb934451cb73a5bedac5a90a5 : SUCCESS in 7m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d1df993284b54d4480eb3cf914375a58 : SUCCESS in 11m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41e44640dac3417da82f9bad1d8b2fdc : SUCCESS in 6m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/52adc77081c8442dbb2c6a8a85896df8 : SUCCESS in 8m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ba3d80a725b4963aa89959cac55aeee : SUCCESS in 16m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f0f9a51ce336489ca07e39dcd11b1d3f : SUCCESS in 8m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2c792f42abe4486ba0630cdf3927de48 : SUCCESS in 10m 34s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3de62c9910274fa2a9ee34d0866859c8 : SUCCESS in 9m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/780cb71067cc47fc851a922b5f695a77 : SUCCESS in 8m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/31301f127e4a41b0874f326f650e2d0a : SUCCESS in 52m 48s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c8f392ebe06f4ab4be15bdf6e164351e : SUCCESS in 46m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a2a1f0a8a04e478db2d0099b25b6b4fd : SUCCESS in 1h 41m 21s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a3364ca9f7bf469f84deb0ba23812694 : SUCCESS in 1h 51m 12s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/6f2656e9d4874250b90ffe8502b9c9e8 : FAILURE in 19m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d61b58564d8f4215a9d4ffd544a77a8b : SUCCESS in 2h 08m 20s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/75d7b117a8d744aebc06c426f2333d59 : TIMED_OUT in 2h 34m 37s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b496ccb6ca1945489d48f6185582d3fd : SUCCESS in 1h 21m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b7801ba2cd8f4e42a2ecf63a84cb3cd3 : SUCCESS in 59m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3de982385e094b199b362cf9eb2ee56d : FAILURE in 1h 51m 46s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2985ece3058b41948761e549e2ed66f3 : SUCCESS in 1h 10m 05s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/914e2a0767ad4cc29e7126ee76c9a8fc : FAILURE in 1h 22m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/119f776b032f41bf97869b956f06a603 : FAILURE in 28m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f34c999897016197c55a55b75ab19b6c27108ec2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 16:59:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ecc6c29c272ea026316eeea4d95c012ec20fddc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 18:49:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f64071dbb154e4fb5b09c6f526ab4b5 : SUCCESS in 12m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bea545c62a64d63a6650ce7a53b44f9 : SUCCESS in 7m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5fbfa59f848c4d2f98e35d5335b8a281 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f5d4607cc6d40259f0b8e8f616f4125 : SUCCESS in 5m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c2305fe2c7834c73a84326b21944414c : SUCCESS in 8m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6b163ec7a44436a9cabbe6377881b34 : SUCCESS in 12m 14s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/875cca612edf4c00bea707ddaf42f3fe : SUCCESS in 8m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f000352657134b19a4acdb4d9a0694b5 : SUCCESS in 7m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dd8431d312274250a75637cab709451a : SUCCESS in 53m 07s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/601b0c0f20cd4b2aa3e36e8992f32fc1 : SUCCESS in 45m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ced728c9eb6549cb8a55889a598186a8 : SUCCESS in 1h 35m 17s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/51cbcd2ff4ed4c68b9736e4b14ca08bb : SUCCESS in 1h 43m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/68922d9fb6394da382e9dbecf60e46d4 : SUCCESS in 1h 19m 00s","accounts_in_message":[],"_revision_number":1},{"id":"fc42ee71f9fc533f1c0f8ffd6e545dfae674fddf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 18:49:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"706a287ca9060d4c3b23a0a329dae13281d6ced8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 18:53:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/673c17497ad0416a9cc3b4db5294e8bf : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0c1cb417e904daa75deedaabc72a20d07b138eb9","revisions":{"0c1cb417e904daa75deedaabc72a20d07b138eb9":{"kind":"REWORK","_number":1,"created":"2020-02-24 13:52:02.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/31/709531/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/31/709531/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/31/709531/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/31/709531/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/31/709531/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/31/709531/1"}}},"commit":{"parents":[{"commit":"a741eb1fbb86cfebd3da0082fbf3b1f173a9cf17","subject":"Merge \"Fix house keeping graceful shutdown\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a741eb1fbb86cfebd3da0082fbf3b1f173a9cf17"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-02-21 09:48:43.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-02-24 13:51:15.000000000","tz":60},"subject":"Fix uncaught DB exception when trying to get a spare amphora","message":"Fix uncaught DB exception when trying to get a spare amphora\n\nThe MapLoadbalancerToAmphora task performs some database queries but\ndoesn\u0027t catch any exception from it.\n\nIn case something goes wrong with the database or the query, this commit\ncatchs any exception from the DB call, logs it and then returns None to\nswitch to the next decider branch (boot a new amphora).\n\nStory 2007320\nTask 38831\n\nConflicts:\n\toctavia/controller/worker/v1/tasks/database_tasks.py\n\nChange-Id: Ifb6c34426e0927534d332a8bbf2c66aac6c002c5\n(cherry picked from commit 3bbd32a2a5588ab1502efdb6f9556c646d6e8f19)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c1cb417e904daa75deedaabc72a20d07b138eb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c1cb417e904daa75deedaabc72a20d07b138eb9"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
