)]}'
{"id":"openstack%2Foctavia~709567","triplet_id":"openstack%2Foctavia~stable%2Fstein~Ifb6c34426e0927534d332a8bbf2c66aac6c002c5","project":"openstack/octavia","branch":"stable/stein","hashtags":[],"change_id":"Ifb6c34426e0927534d332a8bbf2c66aac6c002c5","subject":"Fix uncaught DB exception when trying to get a spare amphora","status":"MERGED","created":"2020-02-24 16:18:58.000000000","updated":"2020-02-26 04:50:41.000000000","submitted":"2020-02-26 04:49:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709567-1582692556730-2a34db55","meta_rev_id":"b734c61ee9de5389037e6def9a330e037b872e49","_number":709567,"virtual_id_number":709567,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-26 04:49:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":2,"date":"2020-02-26 03:09:56.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":2,"date":"2020-02-24 19:44:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":1,"date":"2020-02-26 03:09:56.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-24 19:44:14.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-02-26 03:09:56.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2020-02-26 04:49:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fd365db42f29c99f5c11c3f7de4db83afa7876e9","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-02-24 16:18:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12f070bba4300504fa86e82a963e6879e2c1f5e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 18:11:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d323cd96f35140aa826dcc838aa74817 : SUCCESS in 12m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/238daed0862945a8809179bbe1ce995c : SUCCESS in 12m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1f1e6e2cf6341bfbc6c7bfa0e95ac15 : SUCCESS in 8m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8c83d995901b4169bb6bce5b7633ba51 : SUCCESS in 7m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/55b40aed62b047828acc46b7e7279dbc : SUCCESS in 9m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4145e8cca4a347de86793c174f605659 : SUCCESS in 8m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b76034847f74fab85d35a5594e0e53d : SUCCESS in 9m 24s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/86c37d6177a540a1a66189da6e42e132 : SUCCESS in 1h 16m 08s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/3b569694b03842858dfbfbe8681b950c : SUCCESS in 1h 22m 28s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8db74598e4654a9cb8b37e93216c2468 : FAILURE in 1h 19m 14s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/2436e1521bf643879394c5049fc0081a : FAILURE in 1h 17m 40s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f07192d3c26c4c0f858a933a27dda590 : SUCCESS in 10m 29s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/5f323acd9a7a4fc4ad7574669c24f1d9 : SUCCESS in 10m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/93a502ad4ea54508bde7a71b78b71d55 : SUCCESS in 8m 20s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0d968c3184924cbf9e42d430596c34a4 : SUCCESS in 1h 06m 47s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/6b9293a79bf644d29df1b9ab0062a327 : SUCCESS in 1h 06m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/22e0a96b5d304fd684599ea55b8d6a41 : SUCCESS in 45m 46s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/6e7ac8909182408387e1afeb881da361 : SUCCESS in 51m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/96463777ba9e47c9a2aff0f407b869b9 : SUCCESS in 1h 40m 27s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4bc642c0b847456cb70ee03aaffd755e : SUCCESS in 1h 29m 14s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/04c5df0ad4614d7487d6756825b2a5b6 : SUCCESS in 1h 44m 39s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/689286725e954d378a19fd019a0fc068 : FAILURE in 25m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e00cce12f0a3443b94532455a0acea94 : FAILURE in 35m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7663ddf60e074c1f9b7d22fc8664bd5c : SUCCESS in 1h 41m 58s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/e9778970f90d446397b0557ff244118a : SUCCESS in 1h 33m 34s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e934636de6cd4631aad53e922f138cd4 : SUCCESS in 1h 29m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/51e6d20c3f684a29b2691a50410b4034 : FAILURE in 59m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5f259c6ee2f32224be5c884ce2b0c6c3a77d3b19","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-24 19:44:14.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"b8cbc77e97a8932020b7b64f91ec50126070b408","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2020-02-26 03:09:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"981110f56a6305bfc229cfc47f8d4616defb7586","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 03:10:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"22272af81d520b9f16647a9a924e7634127a3e5e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 04:49:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2254958bcb5494ad9e56f593fb5d0ed484505f26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 04:49:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41552ca39e634255a7440317c881802b : SUCCESS in 10m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/085da1d0de2e4df4a6b59ef42860e2b6 : SUCCESS in 6m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8d12ecddbd664d62b74903b01449374f : SUCCESS in 5m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/86a998e82c3340fa803f6ec407e4da1f : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/856cf1f0b9b14d4a9f6638c09e2319a3 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5914c28c0fb4cd7842db4e393f3d0af : SUCCESS in 7m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/885ebf1f9bd6453cb1b3ad11249c29bb : SUCCESS in 7m 32s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/6034833e43564dd3bd453a0f37ffc372 : SUCCESS in 7m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4df8d94245604ce390fa99d4a7aa460e : SUCCESS in 7m 47s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e878d4aa641d4989906f37760082d5ed : SUCCESS in 58m 16s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/3a3d58539c3849d3b7b08a88c5d3eeff : SUCCESS in 1h 06m 20s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/567cc8d6c8a3482694353afc58582670 : SUCCESS in 1h 15m 00s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/3caaa6dba6da4e108fa879e8d95d0797 : SUCCESS in 1h 16m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ff1bbb7614fe491f82220b8de578e63e : SUCCESS in 48m 19s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/ba9d358487ad422c9eda2c2dc016f0d3 : SUCCESS in 46m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c338048f35b54538bf6c8abb1030447f : SUCCESS in 1h 26m 52s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ecf996dbcb874950a553df1ed0ce9760 : SUCCESS in 1h 28m 15s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/7db467aced81439eb9498e77177f53b4 : SUCCESS in 1h 34m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6c3f36eb44324117a664c363b287ae8d : SUCCESS in 1h 13m 38s","accounts_in_message":[],"_revision_number":1},{"id":"b734c61ee9de5389037e6def9a330e037b872e49","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 04:50:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3b04d1d9f7a4ed0b041ef7658270ea3 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"372e047a840334ea61c23e21f24fb3c9bfbed902","revisions":{"372e047a840334ea61c23e21f24fb3c9bfbed902":{"kind":"REWORK","_number":1,"created":"2020-02-24 16:18:58.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/67/709567/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/67/709567/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/709567/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/709567/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/709567/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/67/709567/1"}}},"commit":{"parents":[{"commit":"bc7d78582f95107ac9919c1a9d01030ecec4b600","subject":"Merge \"Fix house keeping graceful shutdown\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc7d78582f95107ac9919c1a9d01030ecec4b600"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-02-21 09:48:43.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-02-24 13:53:24.000000000","tz":60},"subject":"Fix uncaught DB exception when trying to get a spare amphora","message":"Fix uncaught DB exception when trying to get a spare amphora\n\nThe MapLoadbalancerToAmphora task performs some database queries but\ndoesn\u0027t catch any exception from it.\n\nIn case something goes wrong with the database or the query, this commit\ncatchs any exception from the DB call, logs it and then returns None to\nswitch to the next decider branch (boot a new amphora).\n\nStory 2007320\nTask 38831\n\nConflicts:\n\toctavia/controller/worker/v1/tasks/database_tasks.py\n\nChange-Id: Ifb6c34426e0927534d332a8bbf2c66aac6c002c5\n(cherry picked from commit 3bbd32a2a5588ab1502efdb6f9556c646d6e8f19)\n(cherry picked from commit 0c1cb417e904daa75deedaabc72a20d07b138eb9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/372e047a840334ea61c23e21f24fb3c9bfbed902"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/372e047a840334ea61c23e21f24fb3c9bfbed902"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
