)]}'
{"id":"openstack%2Foctavia~709569","triplet_id":"openstack%2Foctavia~stable%2Fqueens~Ifb6c34426e0927534d332a8bbf2c66aac6c002c5","project":"openstack/octavia","branch":"stable/queens","hashtags":[],"change_id":"Ifb6c34426e0927534d332a8bbf2c66aac6c002c5","subject":"Fix uncaught DB exception when trying to get a spare amphora","status":"MERGED","created":"2020-02-24 16:20:43.000000000","updated":"2020-03-08 13:31:00.000000000","submitted":"2020-03-08 13:29:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709569-1583674169130-cb471cbf","meta_rev_id":"8532af9b5b301955e0160045e8ea4c8dc88fd0a7","_number":709569,"virtual_id_number":709569,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2020-03-02 10:04:49.000000000","_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"date":"2020-03-02 20:54:13.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-08 13:29:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-02-28 04:57:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-02-28 16:20:07.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2020-02-28 16:20:07.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-28 04:57:25.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-02-28 16:20:07.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-03-02 20:54:13.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-03-08 13:29:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a54a9c3e5b6c36466570f2edad00c9757171083c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-02-24 16:20:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5bc128da5ef166d3c0327536b20c87a89e7293dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 18:10:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37fb6a81c2204603b43b15cfa0f7be6c : FAILURE in 12m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7e794ff25136446b9b7eedd9d16e78ac : SUCCESS in 6m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3e5ad8f771294fdba3b694a5172958d4 : SUCCESS in 9m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0342b1c3457a41128c821de4c8ca06b6 : SUCCESS in 8m 09s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/827b6807476e4131a1e65b6682a5f35d : SUCCESS in 1h 17m 44s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/7d1a4eda5d7b4021a565eb3ba0284ffb : SUCCESS in 1h 23m 13s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7320c1aa05ea4c28a47eb50f3ae16462 : SUCCESS in 1h 17m 25s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/b77e147ca4dc4446afd19324241a0f7e : SUCCESS in 1h 29m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5a42e8b4e1046ae9dcf54918b51e0ba : SUCCESS in 12m 42s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/51af3005fbeb473c920cc139459bea54 : SUCCESS in 11m 21s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/c9088c8475ba4f22bf2606417dc51455 : SUCCESS in 8m 51s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/96059937acfd47a6b2f5257ce0c62967 : SUCCESS in 1h 01m 16s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/8e283c9c68734d538ae2442cafdb8095 : SUCCESS in 1h 00m 13s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/1db7d8bb28c641f497d9e0eb1bd512e5 : SUCCESS in 1h 09m 30s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/95240d8bc07b4e04b8b372c96af9a56a : SUCCESS in 1h 28m 06s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/0d4b8822dcd94f4ba0f769a1a7db88ac : SUCCESS in 1h 18m 39s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c4246cd716144ae69ee388db6850137d : SUCCESS in 46m 30s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/3933fcb13236419e80cd7bda781db05a : SUCCESS in 45m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/18e7cd33308b4ef498e12b4b593f3b17 : SUCCESS in 1h 36m 18s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/8fe90a159ef84a72bb0ee26910e14091 : POST_FAILURE in 1h 37m 08s","accounts_in_message":[],"_revision_number":1},{"id":"0fc83e2a4f9e8dcef416cd6d5859ed4e39431e80","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-02-25 08:17:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2c8f3c71e78272fcd9fac5e65c5c5244f03f4dc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 10:00:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a859d3cbc4b2443396635a6020ecf8f1 : SUCCESS in 10m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d3366c895a40407f8586500c6cd3c46f : SUCCESS in 6m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6f9da045a16649ed8a8b18332c29ac61 : SUCCESS in 8m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97e09fbd88284f6fa7b56c52790ff227 : SUCCESS in 7m 06s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/e532437b62fa4d849e4c9de800143801 : SUCCESS in 1h 15m 01s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/64f077bf3b754afeb9cacb827932fe4c : SUCCESS in 1h 23m 55s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e601568f12bc49d6ab907793539f4c1a : SUCCESS in 1h 35m 33s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/37ab60e306154764a9ae5bda9861a8b6 : FAILURE in 1h 18m 19s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92d28e948dda4a9b81247424f5cbf51c : SUCCESS in 10m 22s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f44c5b1bb4cd44288836633a7d5f97f7 : SUCCESS in 7m 52s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/4d9210b1c2f84864aec999a5b209fcc5 : SUCCESS in 7m 36s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/71f3b39ab432450d9c3619c441be8cc1 : SUCCESS in 56m 01s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/ca7f2b2cda4f4eb68ec5d538fdca5e33 : SUCCESS in 1h 06m 11s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/6f5d0977cf21432c93d882e765babdf2 : SUCCESS in 1h 14m 50s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/9eda1d46cab14f86a3057919c73fd467 : SUCCESS in 1h 19m 16s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/001d66acee924f008b6869f0bebdc404 : FAILURE in 59m 55s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1eb26f5394344d99817098243845d43f : SUCCESS in 48m 48s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/3f0aa27e3a29484697cfccc80f90f137 : SUCCESS in 47m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ac62da7dcbef49aebb8d2b734e3848c0 : FAILURE in 1h 36m 45s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/c50a36ee7bfc4892bfe6cc50d63e52e7 : FAILURE in 1h 28m 18s","accounts_in_message":[],"_revision_number":2},{"id":"e0280b6d4e0b657bdc13f01bb1fe1fbcc0e9811a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-02-25 10:43:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4f99b8f7d81531890f74679350a44e191c19d653","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 12:21:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/713f877b70814d37bf599aff7b64a5c6 : SUCCESS in 12m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/70fb21dc36ba43e1aae81c13e03189ad : SUCCESS in 6m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1dcf77bc7e964ff5bc6c9b7eade670f7 : SUCCESS in 8m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7111bc2fc2eb4bde823f123b57c39415 : SUCCESS in 9m 33s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/5333a4e57c284a4c93dd7d7ea1716e79 : SUCCESS in 1h 21m 00s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/6d77580a71294ec4b2777998dd8a0dd1 : SUCCESS in 1h 18m 16s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d3c2a1c7657d4aa0ac6e23c81aae84bf : SUCCESS in 1h 15m 19s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/ba9ff077791e4798a1a6dc7e4d1909f6 : SUCCESS in 1h 18m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d610a447c4334e01933a64c42d9ff3fb : SUCCESS in 12m 11s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d9f430be721845d0888c588cd9530406 : SUCCESS in 8m 56s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/ba253b9b6a3c42dfb064a3d1d06eba68 : SUCCESS in 9m 18s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b5855750dd9d4c09b2d939d75a423226 : SUCCESS in 1h 03m 24s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/5e7abe2e6052425d9b5fe02e6c7b11f7 : SUCCESS in 57m 45s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/e84ffc21f1dc42d783ca64693c098fdd : SUCCESS in 1h 10m 21s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/46a2c0fc9d124092bdb31aae5503f81b : SUCCESS in 1h 09m 42s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/0f82a9e468774296a301400a4de2a80c : SUCCESS in 1h 07m 22s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bd69c038b3374f36a8e869a104854947 : SUCCESS in 46m 27s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/4a0e596c2b644f3785abd78311ba9d16 : SUCCESS in 45m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b9906940cfea44c7891c55cfab8a7920 : SUCCESS in 1h 31m 16s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/21de6ce2a24d45888c4a3fbc3714bbe5 : SUCCESS in 1h 19m 37s","accounts_in_message":[],"_revision_number":2},{"id":"25cc07d6d9b9965bc1e1b4e2347c42f7da65b417","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-02-28 04:57:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a49ec4038e22ed0aebbd367f09e766df0bbda2f9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-28 16:20:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"97a8df2aa086092909529f9612226b251593f39f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 16:20:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"996d992387e52e30d33cde338027347b17fb77c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 19:32:47.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/001c38bc227b44d087069797a4a1f1c6 : SUCCESS in 11m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d11008e1a3ba42b7ad1a8783a899f011 : FAILURE in 7m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3656d5b08c44404cbed580080d1ff218 : SUCCESS in 8m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac23b64ecc094bdebe177003bf06385e : SUCCESS in 8m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09e920adba2743858d6c01bd0726948c : SUCCESS in 10m 07s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/77755b179a7b42a197f2a6a21c16d1d3 : SUCCESS in 7m 25s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/44574d596ad744c69597f1078140d5e5 : SUCCESS in 9m 03s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/c3d176e33a26400a8c639f5ae556b4c0 : SUCCESS in 1h 15m 53s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/249cf6a7182f4c138c96b4fe4b0005ea : SUCCESS in 1h 23m 05s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5371f89014514b4ba9986971a91b03a6 : FAILURE in 43m 12s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/f51604cd80954744bb4477c8ca670634 : SUCCESS in 58m 34s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/9cacc3a27d81465fa83e0dfb782f736c : FAILURE in 48m 50s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/9e92583eb5934263927b63b3ab9b1aed : SUCCESS in 1h 07m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fcc75246967f4c4891d34eb73a341e27 : SUCCESS in 54m 34s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/e6c01657a1704f4ea5ebae7988eac631 : SUCCESS in 45m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/28e454a24ec04f9bb35b4d9cb74efb4e : SUCCESS in 1h 21m 01s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/29b35c7118dd40f3b5588b14c4338d61 : FAILURE in 1h 58m 27s","accounts_in_message":[],"_revision_number":2},{"id":"9d9f5fa55811c2bbe6f5a8fd9c87de0795668413","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-03-02 07:47:03.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f440297444a56e786724380fa9d2a1d72b9d64cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 09:26:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2e971846e3148c89f7c1adc82a15de2 : FAILURE in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e3c17b8d60b8448bae8f0ccdef80ec57 : FAILURE in 3m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cc5f45c8443e496ca7141d3300736bb3 : FAILURE in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a3649a6090245a79ade8852e3d33a20 : FAILURE in 4m 23s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/a388f04bb512491ca909b4b5fff72851 : SUCCESS in 1h 10m 22s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/c036bb4bb2214bb9b91da6bd004076a7 : SUCCESS in 1h 11m 52s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/658658802f144d93abfa0a2feac15a98 : FAILURE in 49m 17s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/60f4b596a5cd4eaf82efc69dd18ee9fd : FAILURE in 26m 06s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/606ee28de33a421781707ad0411317da : FAILURE in 4m 36s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9de4e83b0b0f44418ea9dd7a3c3752f3 : FAILURE in 3m 53s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/326fc40cdfc04d4d904cbb73f234d552 : FAILURE in 3m 19s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/582d4891022d469cb3642dbe554b4200 : SUCCESS in 53m 04s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/4b9181dc9820418cb2d25a30994f953f : SUCCESS in 1h 08m 32s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/ed962d10fda2496b842d9cc52ce1f275 : SUCCESS in 1h 15m 52s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/f0aaef66cc1f4cca9d8311444b9a093f : SUCCESS in 1h 20m 44s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/bfac435fd2b642918a26477c2b9c1a3c : FAILURE in 39m 48s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1523fe8ec70e4ed9b35624eba68dcb24 : SUCCESS in 56m 13s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/38eae965b0a542c08e97002cff0ff719 : SUCCESS in 57m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/21fb5e8667814cd5bf78bf90e0180278 : SUCCESS in 1h 33m 13s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/ad0d943251734c3e86e17034004bed6b : SUCCESS in 1h 36m 27s","accounts_in_message":[],"_revision_number":2},{"id":"49269bcbbac0a819e2c01c6a8578733186502918","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-03-02 10:04:49.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4da3e092105bd0d22762c9496fbcd0618fac0844","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 11:53:44.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b97583b9749041128a1741ee225e3c56 : FAILURE in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1bfaf954838342b9a9b284c2347a2959 : FAILURE in 4m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/adfc92fff09b43138ce0d5ac254bbb55 : FAILURE in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/659d770d6ffe4300bf949989489c7a25 : FAILURE in 4m 24s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/7c4a2f2062a14e8fbcacb33236d1f5cc : SUCCESS in 1h 16m 03s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/eaf0510c6659442292f0b48a5bfc480f : SUCCESS in 1h 18m 14s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e142ea3fd1fe4fb2b3e87d5c6e8447fd : SUCCESS in 1h 25m 59s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/5e82084d790a4c2a876114254ba77bdf : SUCCESS in 1h 13m 18s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1cb4c6c1d84048349d194741897251b9 : FAILURE in 5m 50s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/71a96dc7f7fe4e87a689bec702f84bd5 : FAILURE in 4m 21s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/9efc669c16f04750bc69da1adbea430f : FAILURE in 4m 04s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/99037bf1e1b344d9a45d9ac84f78d8dd : SUCCESS in 57m 09s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/7ae42f9bd18141cb98f1821edbe8a375 : SUCCESS in 56m 10s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/63a1129471c3489f89df57cd16de955f : SUCCESS in 1h 08m 25s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/3c8dd6e874eb46cbbdc1cd43499f09f6 : SUCCESS in 1h 15m 14s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/0d4fe83752294ca09a985194f2a70bf4 : FAILURE in 30m 40s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1549ef9dcd7745368d48b98219700ca6 : SUCCESS in 45m 46s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/aee0fdb7034b4fd79c61b58d83a9b6c3 : SUCCESS in 1h 03m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/714785fc975c4b5dbee8c0a31d855422 : SUCCESS in 1h 32m 09s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/b0ad36474ad74908a67bb5b46ccd6b47 : SUCCESS in 1h 31m 36s","accounts_in_message":[],"_revision_number":2},{"id":"76d5520126b372b1e7d67d0e9ee7b48109338d85","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-03-02 20:54:13.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1ec2239d46919879678b2510fed21f662f88f3d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 22:42:16.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d97178e55e1b4627abd81462eda5cade : SUCCESS in 11m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bac4242974ad4edab1f276e3c1338ddf : FAILURE in 10m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3aff084fb2fe47568a862e966595119a : SUCCESS in 8m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc94fd2247c0488b8336f6e32105abee : SUCCESS in 10m 45s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/6de89a46effd43c3bde58c956d109b50 : SUCCESS in 1h 21m 14s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/1525397056514e38a249396b8ff8db9b : SUCCESS in 1h 20m 34s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cd81e7b438894ad7b83de5c4f8e6c05c : SUCCESS in 1h 19m 18s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/96f9dd7fe99d4cd5830a2f1f9a25eb04 : SUCCESS in 1h 25m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f979b5c092064309b7c93554a161c669 : SUCCESS in 10m 57s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1249c28fb6704514b2606a6a7c845f2e : SUCCESS in 8m 12s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/6c0a4c66020a410faf5228844cb1a4b5 : SUCCESS in 9m 01s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4a25bc2817af4bd0bfed4ea1f4331286 : SUCCESS in 1h 00m 56s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/541adaaa8adc4029bcb7d231d28bdf9e : SUCCESS in 1h 03m 27s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/ab61aa6d53e74fd4872e9771d6bb4052 : SUCCESS in 1h 06m 06s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/581d017bb63c4c7c9d3d21ad8b45fa7a : SUCCESS in 1h 21m 59s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/31fe979ae7e042d3aa493ab9a4e150c6 : FAILURE in 31m 58s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5529eae987044c8b8416e8c6f9c25930 : SUCCESS in 48m 12s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/d65aed87617d4d3295b1f18bb7a32a12 : SUCCESS in 47m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/45bc8e61597a4a54aac7169dfd759c4e : SUCCESS in 1h 29m 10s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/be61de2e4ceb486eb3fa1fdef2e1188a : SUCCESS in 1h 38m 46s","accounts_in_message":[],"_revision_number":2},{"id":"f53e2bb2b9b762a7929b0f4ece38226726124bca","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-08 10:25:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ecfa7660204339aeb410cc817e2fd09b4e52de2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-08 11:53:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39829f90ce044b63a3e33448aa78eca5 : SUCCESS in 7m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6f529a455efb443c963ad8ad532b6a38 : SUCCESS in 7m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c5260119ff00401fbe88f7a4dbeff523 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a78e12930e54ee78f6a85498164b971 : SUCCESS in 6m 07s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/1d96caa0d74345a084ea7d2876d6000f : SUCCESS in 1h 16m 22s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/b8f82f774ce74942aafefbb4094d03ed : SUCCESS in 1h 17m 28s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ce0819d8a17546f3a203ae7c964ef7b5 : SUCCESS in 1h 16m 15s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/c7c50507c9674101940a68170d43f956 : SUCCESS in 1h 09m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb079ce807bb4aa3a02f46853adf990e : SUCCESS in 9m 09s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b38599f94aef4427b6e3549f56ab04d0 : SUCCESS in 7m 03s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/94cb1f33697b44a8956a942e5598229d : SUCCESS in 6m 07s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f89336729a7140e58a0d853f582d636f : SUCCESS in 54m 58s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/706921a3bd6d49769cda4c4cda27277f : SUCCESS in 55m 07s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/8855838a4ac447b88cc0af648dd9714b : SUCCESS in 57m 31s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/a3e9b9a030364f08b44453eb4ebb4b7c : SUCCESS in 1h 08m 09s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/05001a72e86a4f219619c0f11f84d67b : FAILURE in 27m 18s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6dd775ad934f498fb503c304e6878b26 : SUCCESS in 42m 39s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/2842253428cd410993f47a7bd0aed3c5 : SUCCESS in 38m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0714e81097ea4a01a54e4c220218e7c6 : SUCCESS in 1h 10m 58s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/8a245101b10a444db717e06751203314 : SUCCESS in 1h 23m 32s","accounts_in_message":[],"_revision_number":2},{"id":"06cab494dd5af27492dfc067eb5b2bfe230a9405","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-08 11:53:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6a0f72ebb8492711a06edf705f36a9ae08ac73d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-08 13:29:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faf4fc3df5b6434b916580c00fa7c5b3 : SUCCESS in 11m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e1bb01af04004d20b83313eb2209255d : SUCCESS in 5m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/73b10fd52a0748da8d805954a5fd39a0 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/673293102ea94549892803fa770104d4 : SUCCESS in 5m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/834da454274044c88582112626031f49 : SUCCESS in 8m 25s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9042432c134c4f929815314584622aab : SUCCESS in 10m 00s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/b2f541e511d94e34aa37a07fa0f32ab4 : SUCCESS in 8m 34s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/971d9cb5e4004fe983fc8861a2e9727b : SUCCESS in 1h 18m 42s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/7a246090988b4bbbac2922f1c0bade00 : SUCCESS in 1h 23m 59s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ddb7d28b84d043f893bf13495a45261c : SUCCESS in 53m 41s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/26d8666ce670420e92e0db5ba47d318e : SUCCESS in 59m 43s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/b0554d2e532a444998cda2b1dc413887 : SUCCESS in 1h 05m 03s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/0633103cfd4349b68fb79821d951fe20 : SUCCESS in 1h 14m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fa7bc6bd9e5e4cae8ca293efbe498d41 : SUCCESS in 43m 40s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/26541a031bc24ed68348f4722bf74156 : SUCCESS in 46m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/de73fbb9eaa14e0c9dcd8742f388f7fc : SUCCESS in 1h 23m 29s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/01d8f790500f4d0d9b33172d8a6f2481 : SUCCESS in 1h 21m 03s","accounts_in_message":[],"_revision_number":2},{"id":"07a70e5466ce7f0bc0361295e6ad91fe2e637458","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-08 13:29:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"8532af9b5b301955e0160045e8ea4c8dc88fd0a7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-08 13:31:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ded180dfe4504afbaf85186ad0ef9ee0 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5528268e2f360c2eb921d665313a326387c2e1b7","revisions":{"5a7ea4660bc25ccd3e91dea7841053253f18b054":{"kind":"REWORK","_number":1,"created":"2020-02-24 16:20:43.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/69/709569/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/709569/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/709569/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/709569/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/709569/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/709569/1"}}},"commit":{"parents":[{"commit":"7d8472776c5397d1a164e6b8bb151101d51f87fb","subject":"Merge \"Fix house keeping graceful shutdown\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d8472776c5397d1a164e6b8bb151101d51f87fb"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-02-21 09:48:43.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-02-24 16:20:36.000000000","tz":60},"subject":"Fix uncaught DB exception when trying to get a spare amphora","message":"Fix uncaught DB exception when trying to get a spare amphora\n\nThe MapLoadbalancerToAmphora task performs some database queries but\ndoesn\u0027t catch any exception from it.\n\nIn case something goes wrong with the database or the query, this commit\ncatchs any exception from the DB call, logs it and then returns None to\nswitch to the next decider branch (boot a new amphora).\n\nStory 2007320\nTask 38831\n\nChange-Id: Ifb6c34426e0927534d332a8bbf2c66aac6c002c5\n(cherry picked from commit 3bbd32a2a5588ab1502efdb6f9556c646d6e8f19)\n(cherry picked from commit 0c1cb417e904daa75deedaabc72a20d07b138eb9)\n(cherry picked from commit 372e047a840334ea61c23e21f24fb3c9bfbed902)\n(cherry picked from commit e7376be7afec98192da969dc9382ad860b1cbe01)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5a7ea4660bc25ccd3e91dea7841053253f18b054"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5a7ea4660bc25ccd3e91dea7841053253f18b054"}]},"branch":"refs/heads/stable/queens"},"5528268e2f360c2eb921d665313a326387c2e1b7":{"kind":"REWORK","_number":2,"created":"2020-02-25 08:17:12.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/69/709569/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/709569/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/709569/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/709569/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/709569/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/709569/2"}}},"commit":{"parents":[{"commit":"7d8472776c5397d1a164e6b8bb151101d51f87fb","subject":"Merge \"Fix house keeping graceful shutdown\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d8472776c5397d1a164e6b8bb151101d51f87fb"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-02-21 09:48:43.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-02-25 08:16:02.000000000","tz":60},"subject":"Fix uncaught DB exception when trying to get a spare amphora","message":"Fix uncaught DB exception when trying to get a spare amphora\n\nThe MapLoadbalancerToAmphora task performs some database queries but\ndoesn\u0027t catch any exception from it.\n\nIn case something goes wrong with the database or the query, this commit\ncatchs any exception from the DB call, logs it and then returns None to\nswitch to the next decider branch (boot a new amphora).\n\nNote: In Queens, string format method in the LOG.error call was changed\nto please pep8/pylint.\n\nStory 2007320\nTask 38831\n\nChange-Id: Ifb6c34426e0927534d332a8bbf2c66aac6c002c5\n(cherry picked from commit 3bbd32a2a5588ab1502efdb6f9556c646d6e8f19)\n(cherry picked from commit 0c1cb417e904daa75deedaabc72a20d07b138eb9)\n(cherry picked from commit 372e047a840334ea61c23e21f24fb3c9bfbed902)\n(cherry picked from commit e7376be7afec98192da969dc9382ad860b1cbe01)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5528268e2f360c2eb921d665313a326387c2e1b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5528268e2f360c2eb921d665313a326387c2e1b7"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
