)]}'
{"id":"openstack%2Foctavia~709696","triplet_id":"openstack%2Foctavia~master~I5d3c4d5280916916e95120cfba6fc076a1650cf4","project":"openstack/octavia","branch":"master","topic":"jobboard","hashtags":[],"change_id":"I5d3c4d5280916916e95120cfba6fc076a1650cf4","subject":"[Amphorav2] Fix noop driver case","status":"MERGED","created":"2020-02-25 11:36:45.000000000","updated":"2020-03-17 17:39:06.000000000","submitted":"2020-03-17 17:36:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709696-1584466572163-9c168db8","meta_rev_id":"6beb95b18da03a5362a6ae7bb9b383f518b8faee","_number":709696,"virtual_id_number":709696,"owner":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-17 17:36:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-03-17 15:21:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-03-16 10:54:06.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2020-03-17 15:21:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":-2,"date":"2020-03-16 10:54:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-16 10:54:06.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-03-17 15:21:25.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-03-17 17:36:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5008be140eeed1db49618783a374c6cc7dc27c93","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-02-25 11:36:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4cbec4f951dfa2c119f5d90fd51ae58e3e369b3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 13:38:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/541074478f524c468cb7e8db3de19d35 : SUCCESS in 10m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d302ec616934de4bb930d508e9354ef : SUCCESS in 12m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dd6119f30504ef1bef94f052ae6620c : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/795f49d635ea4bcebbf915c5a59e282f : SUCCESS in 5m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/99df9fc82d2341ce89d20d2870764826 : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de06d9fcd8b34980995f419d0a68964f : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a728eb8160ac479989a5fd0f368e052f : SUCCESS in 17m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5dd0ac0ffd474281a476b898e0430006 : SUCCESS in 8m 53s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9338a8f99b9a4a3c9e46865c7dc7aefd : SUCCESS in 8m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eaa522556d2c461e8c20457abb419aa7 : SUCCESS in 7m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8b8995f59fc942d3839434e5d41fc31c : FAILURE in 32m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bbad7f27461f4dee9621b585c3a1b73c : SUCCESS in 1h 32m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6645541545d045f4a0be422f0b3422e6 : SUCCESS in 55m 47s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b99b5891d92145ebac65168a552b1367 : SUCCESS in 57m 19s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a3c22d424d054040a7dbdc623de0dc54 : SUCCESS in 1h 01m 41s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/775eb091bc884ee69fb02b81c5704317 : SUCCESS in 1h 42m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/763e91c242d7477b841823aeac54d5f7 : SUCCESS in 1h 53m 18s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fae0df5d80c6482d890d37eacaf43083 : FAILURE in 1h 09m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0cb0159e83c9487793963b0dae329bc9 : SUCCESS in 1h 45m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cf2d71025b8083ac5fcac7826a7c9414f8eef5eb","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-02-26 08:15:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f5eab9b9045758c0588fe1fc5133a3a8fea4f87a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 10:45:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d463711e0b8743fb885d21c129299a2c : SUCCESS in 10m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16cfba6db99e43988794b12a5cf2aa0d : SUCCESS in 13m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a093e9c550bf4941ba6b5a6534a15f69 : SUCCESS in 8m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ee8bf6c374946ce9b8760d4218705fd : SUCCESS in 7m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/513c102af66d47258f91f497df2e6284 : SUCCESS in 9m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/958cfabef07d4325bf594f1bec4db0cc : FAILURE in 7m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/614bbef1fb35416d8348ee080359670a : SUCCESS in 15m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e01c75d2528c48c099363721ee4ebadd : SUCCESS in 6m 24s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5702fb4b72cf493da75f79dfa52482d6 : SUCCESS in 10m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/01f8222b2fa34232ab50611953817981 : SUCCESS in 8m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8094f1bcdf7441f58b0e8786aafeaf2e : FAILURE in 43m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e1004b6eca4f45289855cf0135213d4c : SUCCESS in 1h 55m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f7a946bbae6b4e66ae9cef83e7c2efb4 : SUCCESS in 1h 00m 58s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7ffe16811a0c4601a0321555aff82a4a : SUCCESS in 58m 25s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/11fb357cc1404dc4af7076ed384adca3 : SUCCESS in 1h 00m 35s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/79f1b96faaaf4020b4b7574b091d672a : SUCCESS in 2h 04m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0285284794d54c35a185b02ebb91755a : SUCCESS in 2h 25m 09s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/20096feff05d4e539d94f55fbad4372c : FAILURE in 1h 04m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/be2cb787c57d44f1904e9fd23fc541e8 : SUCCESS in 1h 55m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b39a9be3315325df6864af140bf4afae327aba91","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-02-27 08:02:36.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7ed9870d0d9adac97014a74ae45549ff295e2122","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-02-27 08:02:56.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"3dba6ece6739b7cedc41d60e51c3cba38673816f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 14:41:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5be781b60c7b417a913ba9f5a0f8c717 : SUCCESS in 10m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f3af77710704cffa79099202c3f9f32 : SUCCESS in 10m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cbdb6128d894d35a13fdcca52d7db16 : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e278c08245ac4c20b1631844b1311a77 : SUCCESS in 5m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/798f9cfa3265400b99ba16765c1565c8 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dedc981327c54418a07917cb444568a1 : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f6b380de8224dd388c52e5ceda65b85 : SUCCESS in 14m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/49ecc6e801bb4994ae2d2b96caed0965 : SUCCESS in 5m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/045ffd3885034495981ca9d322604df3 : SUCCESS in 9m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a43f008ad8004a26b3db7d11297eef9e : SUCCESS in 7m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/85a9a69220e848d3bd68f46b22faf57c : SUCCESS in 47m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/104479188511415abbdb0027ea8d3172 : RETRY_LIMIT in 1h 30m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a55b73db690f42028c31b17e3f0a1316 : SUCCESS in 1h 15m 37s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/28f1068194124e1aac17beec117a33f6 : SUCCESS in 56m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/30615a38bc4541bc8d21fbb422d96d75 : SUCCESS in 1h 05m 09s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1822c45d0a924d30938a0970e6e57033 : RETRY_LIMIT in 19m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a56e910582b147f5974587d54d23eab2 : SUCCESS in 1h 29m 05s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a595dccd83cd445f90ed8e4c20c0385f : FAILURE in 1h 17m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b7dff39bba6048ea98a1bd94a495a1cd : RETRY_LIMIT in 1h 30m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"348e53b4476247015c414241a722f4a237e3d096","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-03-02 10:24:39.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"cb55047d08b30f54f868722b57bc4137508af233","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 12:44:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74565f8ea6194e628567be42edf6350e : SUCCESS in 11m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c2deb7fbc0ef4acaa9b553d09b0d8093 : SUCCESS in 11m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/193969e03d8044d5b1f25b633a996d9c : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/412b60510aca46a1bb1ce6ed802be547 : SUCCESS in 6m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/45b16be1e1da41b0b5972778fd581cb1 : SUCCESS in 8m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c43c8003f72b436eb7c5b15b40391ec1 : FAILURE in 8m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/469fcefe1d824d1f8eddd444663ea66c : SUCCESS in 16m 50s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8a90091830cb495fbd09622e2b0216c0 : SUCCESS in 8m 58s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/19e9bc54a2464c79b5dc1b020ceacd0e : SUCCESS in 10m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f5bd9ec64da044aea49e46e9ecd4b340 : SUCCESS in 9m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e904d84691f5414eb064caa1aa04b321 : SUCCESS in 48m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8e9174591f4645fea37fcd9a5ad9be56 : SUCCESS in 1h 46m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/98376abf46f14ea1b0a9337b2d2d4790 : SUCCESS in 1h 02m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/455642e72d4a4fbda29d6e1460477698 : SUCCESS in 55m 42s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d78978c8188541f4873900eebd631354 : SUCCESS in 1h 09m 24s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dff9707f6d91482e8e7ea28bbdeb80a1 : SUCCESS in 1h 55m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5520d7c4c1a744a8a513725f4a535bd0 : SUCCESS in 2h 12m 25s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/70f836282b9d4ddbbde054edab241cc5 : FAILURE in 1h 05m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/288dc24c6a1d4c6e8502c36ea6f0c7d8 : SUCCESS in 1h 59m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"92e8e0ba34c6cf3f091fd6c04f631c5f53e9d05e","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-03-04 08:22:38.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"91ea0385d6df13a438aeb4dc57dfb60c039908cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 10:31:22.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/470a34be7e1245b496d1aa930cb879fd : SUCCESS in 10m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6ee469595ef4180a80214c9a2666bb0 : SUCCESS in 9m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cfbb6a2d13e4b889d65e2476088b797 : SUCCESS in 7m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9396eed2a2d048aa86baf7363b0d5da2 : SUCCESS in 5m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bbeb3a8cad6d4bb88561b7123bec619c : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/034cb9ff50124386857540a6dc40866d : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78fe5fbbcdde4f16851b57bfced8f04d : SUCCESS in 12m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fa7c74956e724e0f9b0af1e89f131ef3 : SUCCESS in 4m 58s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d0bcce0497a94f0b87b35e4df5b5c193 : SUCCESS in 8m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f5155ef7cd0f49d48944bf2b822fcfb4 : SUCCESS in 8m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8f8959243b1d4f61a57a402135e5aa01 : SUCCESS in 47m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2cc23ef8c7644647a2c466e0b2d5bd47 : SUCCESS in 1h 41m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/557f61a11e024c45bc1cde2d1e406639 : SUCCESS in 1h 03m 33s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/960db9ebc76f42859c96f7eecd8e0e95 : SUCCESS in 56m 02s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7293f57512364b40845198c39f3745f5 : SUCCESS in 1h 05m 25s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a42d6c92385b4515ae408187eb2e5a1a : SUCCESS in 1h 52m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cd821b6862c54af78dd37cec166cc44f : FAILURE in 1h 31m 38s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/38298861fec24a008522869ffd7657b2 : FAILURE in 1h 04m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a1d90f691b164fa3b7d50dcfa4f2bbf2 : SUCCESS in 2h 07m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"715c46b6ed59ab221fcd5c9530edd14787a930b2","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-03-04 16:53:41.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)\n\nWhoops, looks like this was missed when this change was made a long time ago. T_T","accounts_in_message":[],"_revision_number":5},{"id":"0a2ef2c9cb02a47fe1330a6f9fcb5e4a68d32785","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-03-05 07:19:07.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"278bd820e338dc5ad21c69d6b141224908b93e43","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-16 10:34:37.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)\n\nJust a nit in the docstring, otherwise LGTM.","accounts_in_message":[],"_revision_number":5},{"id":"529c62bac589e89478707903ad5ed8a520fb62b8","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-03-16 10:51:24.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"f8cf2d812cb8d7db767247103464228563ac0104","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-03-16 10:51:31.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"23ab69efa112d16c6ee217932c4ba2b0501046af","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-16 10:54:06.000000000","message":"Patch Set 6: Backport-Candidate-2 Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":6},{"id":"e6d359f03ab0b3bb00baf58916aa901ce99e7cce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 13:28:34.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/283fdf002e21467abfcde471776ff239 : SUCCESS in 15m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/52e296e5cb174b438a80bc757f7db50f : SUCCESS in 12m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c508744981348d4bf5937a3546cde92 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd208d0abec34176b18338ec134aac93 : SUCCESS in 5m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dc217e21702b44f888914e46caf21775 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85128f85e21945b480f59142ad190989 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9f46d035e7244a3bc5735e42d4b79b3 : SUCCESS in 12m 04s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9c0477a60b5c4c509552fe6bab71abfd : SUCCESS in 5m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8d9a1021abe345cb9183be278cda11fb : SUCCESS in 8m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9d276bb5e02a44f0bb249f583ab155f5 : SUCCESS in 7m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9e8bca566dea44e7812d22dea3a9eb19 : SUCCESS in 44m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5aab64fb859043488e5b445bc7cff0a2 : SUCCESS in 2h 13m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0a23255cd61c440fb87bbf4b4d11142b : SUCCESS in 1h 03m 18s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/bad4e98bc2f149d987eaa030a7ec3bad : SUCCESS in 54m 22s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9f6ef6c35bda450686d15156bba3ef8e : SUCCESS in 1h 11m 39s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7032b89ffaea4346a2718177540cf6b8 : SUCCESS in 2h 30m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/77870a66971a4884885231359a77d1d8 : FAILURE in 2h 18m 49s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/549e7cf64a514dd2aa5f78f4df192e99 : FAILURE in 1h 06m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2581d1f9a20a47858ccd76464335f756 : SUCCESS in 1h 45m 10s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5c30caaddf14a760accf3deb2c74d9dd5cb6beb5","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-03-17 15:21:25.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"89a9890a86f559aca8a3d93e2b2c77a69385c380","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 15:21:37.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"0328d0bc6ef6370a95b991a831dbf0b2084caf0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 17:36:11.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e9a4edd0df248a7b8d3c4e935f0e5b4 : SUCCESS in 11m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/472a0cd4f160452c92585c4c582d3755 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4129f04ca35e4174b301777b7ca01081 : SUCCESS in 7m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/70b6a16b916246a88fcb8f5177c5e323 : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a371721fa454cd08abb2b698b69f2b2 : SUCCESS in 12m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4b82a7a49afb4c1d8218cb84dfb82a1c : SUCCESS in 8m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/244d76a3875c4838a6990153d620a52a : SUCCESS in 56m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/55e9d8d50607410aa8a26f74ba7540ed : SUCCESS in 2h 07m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/10b3ed9d3cd84bd0b4c8758b39b75e7c : SUCCESS in 58m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ed1a5f8dba7d4b9c86d4b6b3bee6fe03 : SUCCESS in 53m 00s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ef80b6068132436fb236ece4770b8b75 : SUCCESS in 1h 08m 00s","accounts_in_message":[],"_revision_number":6},{"id":"c232b3ae0154b30ba505a41d5782cfd945320482","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 17:36:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"846db37d866ff984414e044fd865f82461c2d41c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 17:39:06.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b46a01b9040149a1a9b0a52569f17a37 : SUCCESS in 2m 25s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"6bca6bef1b6a17571ae90c98df174e477d89db6b","revisions":{"f5871e35638fc6e092db7a6eea31bc253c8feef5":{"kind":"REWORK","_number":1,"created":"2020-02-25 11:36:45.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/96/709696/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/709696/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/709696/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/709696/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/709696/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/709696/1"}}},"commit":{"parents":[{"commit":"e0ae8a3cac837beda479e23f6015ee6584093e9b","subject":"Merge \"Fix uncaught DB exception when trying to get a spare amphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e0ae8a3cac837beda479e23f6015ee6584093e9b"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2020-02-25 11:33:27.000000000","tz":240},"committer":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2020-02-25 11:33:27.000000000","tz":240},"subject":"[Amphorav2] Fix noop driver case","message":"[Amphorav2] Fix noop driver case\n\nFix amphora creation with amphora_noop_driver\nand network_noop_driver.\n\nChange-Id: I5d3c4d5280916916e95120cfba6fc076a1650cf4\nStory: 2005072\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f5871e35638fc6e092db7a6eea31bc253c8feef5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f5871e35638fc6e092db7a6eea31bc253c8feef5"}]},"branch":"refs/heads/master"},"88babb7f3c15eba496e1c73afcfffc2f043efa05":{"kind":"REWORK","_number":2,"created":"2020-02-27 08:02:36.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/96/709696/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/709696/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/709696/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/709696/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/709696/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/709696/2"}}},"commit":{"parents":[{"commit":"e0ae8a3cac837beda479e23f6015ee6584093e9b","subject":"Merge \"Fix uncaught DB exception when trying to get a spare amphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e0ae8a3cac837beda479e23f6015ee6584093e9b"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2020-02-25 11:33:27.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-02-27 08:02:19.000000000","tz":0},"subject":"[Amphorav2] Fix noop driver case","message":"[Amphorav2] Fix noop driver case\n\nFix amphora creation with amphora_noop_driver\nand network_noop_driver.\n\nChange-Id: I5d3c4d5280916916e95120cfba6fc076a1650cf4\nStory: 2005072\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/88babb7f3c15eba496e1c73afcfffc2f043efa05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/88babb7f3c15eba496e1c73afcfffc2f043efa05"}]},"branch":"refs/heads/master"},"9930f458d1b45ce90c13276ba0dc6c3233e02855":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-02-27 08:02:56.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/96/709696/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/709696/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/709696/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/709696/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/709696/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/709696/3"}}},"commit":{"parents":[{"commit":"2d30dced39700fc0aa0d75d820f902a8b515a671","subject":"Jobboard based controller","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2d30dced39700fc0aa0d75d820f902a8b515a671"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2020-02-25 11:33:27.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-02-27 08:02:56.000000000","tz":0},"subject":"[Amphorav2] Fix noop driver case","message":"[Amphorav2] Fix noop driver case\n\nFix amphora creation with amphora_noop_driver\nand network_noop_driver.\n\nChange-Id: I5d3c4d5280916916e95120cfba6fc076a1650cf4\nStory: 2005072\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9930f458d1b45ce90c13276ba0dc6c3233e02855"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9930f458d1b45ce90c13276ba0dc6c3233e02855"}]},"branch":"refs/heads/master"},"93f0799250c5a7a2de2f142459bc989aec569668":{"kind":"REWORK","_number":4,"created":"2020-03-02 10:24:39.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/96/709696/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/709696/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/709696/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/709696/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/709696/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/709696/4"}}},"commit":{"parents":[{"commit":"2d30dced39700fc0aa0d75d820f902a8b515a671","subject":"Jobboard based controller","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2d30dced39700fc0aa0d75d820f902a8b515a671"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2020-02-25 11:33:27.000000000","tz":240},"committer":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2020-03-02 10:11:55.000000000","tz":240},"subject":"[Amphorav2] Fix noop driver case","message":"[Amphorav2] Fix noop driver case\n\nFix amphora creation with amphora_noop_driver\nand network_noop_driver.\n\nChange-Id: I5d3c4d5280916916e95120cfba6fc076a1650cf4\nStory: 2005072\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/93f0799250c5a7a2de2f142459bc989aec569668"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/93f0799250c5a7a2de2f142459bc989aec569668"}]},"branch":"refs/heads/master"},"be1eef4928848691bab3f26eb10de9476c668440":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-03-04 08:22:38.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/96/709696/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/709696/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/709696/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/709696/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/709696/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/709696/5"}}},"commit":{"parents":[{"commit":"2c76209003d5ca09a302881a40a35fd14762db55","subject":"Select the right lb_network_ip interface using AZ","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c76209003d5ca09a302881a40a35fd14762db55"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2020-02-25 11:33:27.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-03-04 08:22:38.000000000","tz":0},"subject":"[Amphorav2] Fix noop driver case","message":"[Amphorav2] Fix noop driver case\n\nFix amphora creation with amphora_noop_driver\nand network_noop_driver.\n\nChange-Id: I5d3c4d5280916916e95120cfba6fc076a1650cf4\nStory: 2005072\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/be1eef4928848691bab3f26eb10de9476c668440"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/be1eef4928848691bab3f26eb10de9476c668440"}]},"branch":"refs/heads/master"},"6bca6bef1b6a17571ae90c98df174e477d89db6b":{"kind":"REWORK","_number":6,"created":"2020-03-16 10:51:31.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/96/709696/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/709696/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/709696/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/709696/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/709696/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/709696/6"}}},"commit":{"parents":[{"commit":"2c76209003d5ca09a302881a40a35fd14762db55","subject":"Select the right lb_network_ip interface using AZ","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c76209003d5ca09a302881a40a35fd14762db55"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2020-02-25 11:33:27.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-03-16 10:50:28.000000000","tz":0},"subject":"[Amphorav2] Fix noop driver case","message":"[Amphorav2] Fix noop driver case\n\nFix amphora creation with amphora_noop_driver\nand network_noop_driver.\n\nChange-Id: I5d3c4d5280916916e95120cfba6fc076a1650cf4\nStory: 2005072\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6bca6bef1b6a17571ae90c98df174e477d89db6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6bca6bef1b6a17571ae90c98df174e477d89db6b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
