)]}'
{"id":"openstack%2Foctavia~710161","triplet_id":"openstack%2Foctavia~stable%2Frocky~I2fd51664336dca51f134b3fccd3e8c936b809839","project":"openstack/octavia","branch":"stable/rocky","hashtags":[],"change_id":"I2fd51664336dca51f134b3fccd3e8c936b809839","subject":"HTTPS HMs need the same validation path as HTTP","status":"ABANDONED","created":"2020-02-27 00:31:34.000000000","updated":"2020-07-24 11:41:55.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3a7e36f464c962063e6d412867d0ba73db6ecd9d","_number":710161,"virtual_id_number":710161,"owner":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"date":"2020-05-04 06:58:36.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"date":"2020-05-01 16:26:51.000000000","_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-02-27 03:24:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},{"_account_id":2846,"name":"meera-belur","email":"meera.belur@suse.com","username":"meera-belur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-15 17:00:29.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2846,"name":"meera-belur","email":"meera.belur@suse.com","username":"meera-belur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},{"_account_id":2846,"name":"meera-belur","email":"meera.belur@suse.com","username":"meera-belur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":0,"date":"2020-02-27 00:39:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2020-02-27 00:38:17.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-03-04 12:12:58.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"date":"2020-02-28 04:53:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2020-02-27 14:03:50.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"value":1,"date":"2020-04-15 19:49:03.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},{"value":0,"date":"2020-02-27 00:32:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":2846,"name":"meera-belur","email":"meera.belur@suse.com","username":"meera-belur"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},{"_account_id":2846,"name":"meera-belur","email":"meera.belur@suse.com","username":"meera-belur"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-27 00:32:14.000000000","updated_by":{"_account_id":2846,"name":"meera-belur","email":"meera.belur@suse.com","username":"meera-belur"},"reviewer":{"_account_id":2846,"name":"meera-belur","email":"meera.belur@suse.com","username":"meera-belur"},"state":"REVIEWER"},{"updated":"2020-02-27 00:38:17.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-27 00:39:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-27 03:24:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-27 14:03:50.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-02-28 04:53:59.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-03-04 12:12:58.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2020-04-15 17:00:29.000000000","updated_by":{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},"reviewer":{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},"state":"REVIEWER"},{"updated":"2020-04-15 19:49:03.000000000","updated_by":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"reviewer":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"state":"REVIEWER"},{"updated":"2020-05-04 06:58:36.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"8ccc081c1f2ecea1a7914b985d0df1ee06259026","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-02-27 00:31:34.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6bb27af2a1a2931f2c51a89b110666ce8bb5e58f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 03:24:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2dfacb5c68a348adbece747d10d65fcc : SUCCESS in 9m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1329fef283bc4e1b9ce9d7b144465b04 : SUCCESS in 8m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/77d364dd09894f2db90ab44f4efa59f4 : SUCCESS in 7m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/57ea3effe7f041ba8b9038e01d398b21 : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f08ec05ae2e4538b10d3bf2917613b8 : SUCCESS in 7m 27s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/69186a6684494584af254154f4f3a661 : SUCCESS in 1h 14m 01s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/757f723f602e47b4a1cceef6fd9b1d99 : SUCCESS in 1h 16m 25s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dadef53e09e34002b32d29b466cb0e31 : FAILURE in 1h 16m 27s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/a8b51cc371b44919a59dfb3aa388d9ab : SUCCESS in 1h 20m 18s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/44c46960c00b4fb990c015d31d703394 : SUCCESS in 7m 12s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/ed02da94b58b4e5ba39091457917868e : SUCCESS in 8m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b12c3b1820cb48cebcbd850093b79405 : SUCCESS in 7m 05s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e8fc9de15af543fd8c337d1f309bcc13 : FAILURE in 47m 01s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/932f2bcac7d9474cb84cae262d9b76ff : SUCCESS in 1h 02m 02s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/2e20106770b74fa5bf6ea1bcdfcb3b5e : FAILURE in 1h 29m 30s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/50304e4749094f3c8470d97ff45e4b2b : FAILURE in 1h 46m 54s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/d99a456f2105455fbc89cf48c277e255 : FAILURE in 28m 44s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7b0d3457d6ec4385a085aaa4cdb714d9 : SUCCESS in 49m 36s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/3aa5cef8bd964ea6afaea3d691c715aa : POST_FAILURE in 53m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/49a925b21614432eb23f54139d9e4417 : SUCCESS in 1h 30m 18s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/cc059ec4182846b6acaf6eb7ca0e111f : SUCCESS in 1h 43m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/564208d84be740128d0b8257765d5655 : SUCCESS in 1h 15m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2889bee84fab4b908f9ee6c5c4d366e3 : SUCCESS in 11m 40s","accounts_in_message":[],"_revision_number":1},{"id":"88ad457e8e5cdcad592e4962514e442aa5dbaf77","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-02-27 14:03:50.000000000","message":"Patch Set 1: Backport-Candidate+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"83c2fb58fa380f080e2598b4a528f777d08ccdd5","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-02-27 15:22:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a813bb9a4e8b01fa14bb9f0371335fb79d444653","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 18:21:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d77b222be1a42f3afb3625f50343829 : SUCCESS in 11m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22ae707943d3411d9139adf89414eb4d : SUCCESS in 9m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fd3969fb7d0742e18045dc038336f59d : FAILURE in 9m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8da0a58ff04b472d9b6343be68bee6b1 : SUCCESS in 8m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b948c2c1a9fd43089f277d4275e603e4 : SUCCESS in 8m 25s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/d026c64b07e044c1bad0051238a9e6ee : RETRY_LIMIT in 27m 26s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/e43ce2f13dc2472683f7bb2ccca8d91a : RETRY_LIMIT in 17m 58s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bb727641550a4ecab17e77c7fa856238 : RETRY_LIMIT in 18m 38s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/47b5b77710c14183bb67a0ea8c6e1a9e : RETRY_LIMIT in 44m 05s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/97020b01d3564f0c97aca5153c14c613 : SUCCESS in 9m 44s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/99ad9b9f2e5146baa9fdf0697d051b0d : SUCCESS in 7m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ffb9c93d050b4e44aff7c96fe58f257e : SUCCESS in 7m 53s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d3d8e98320f649b6be37428731442fa0 : RETRY_LIMIT in 22m 47s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/a4a357bfd005498a8b1cb21629ca3a58 : RETRY_LIMIT in 41m 40s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/1c56456215984dcca963f6202bcf92f5 : RETRY_LIMIT in 30m 39s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/f63cabba6a594007a7c20387945db030 : RETRY_LIMIT in 27m 35s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/4997285866844641a49af6024a14866d : RETRY_LIMIT in 25m 44s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0d563a033f96423986ef2db3030c1653 : RETRY_LIMIT in 27m 27s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/d84b9d77e43f4b4bb4d492dd34c0167f : RETRY_LIMIT in 24m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/86b8a255bacd4bd6adf9a9db34b08a51 : RETRY_LIMIT in 3m 38s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/b69f58fbf44e4469b9d86097369f6b99 : RETRY_LIMIT in 26m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7a450587cf734d85a5bc9221e1e146d6 : RETRY_LIMIT in 26m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/395ee27b77324cfcaa5170dfce2f7ed4 : SUCCESS in 10m 15s","accounts_in_message":[],"_revision_number":1},{"id":"287925894191a0316c24e4003a61fd1f70e6947e","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-02-27 18:25:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"60c5ed6063db33bed313634146af39354240f1c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 21:59:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28ace87774b74bba8b0be1b89f1cf7bd : SUCCESS in 10m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/355e91ce9a2a4b7d81f3f6f5c92213b5 : SUCCESS in 10m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/21e00e943fec49068ba074a254df1abe : FAILURE in 7m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1127d13e1f3c40859b1437d47e0c1928 : SUCCESS in 10m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10e4d0ea232b410cb2e85f5cf4f7bd38 : SUCCESS in 8m 16s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/9db999c65da94ee8b08743f3e63f65e2 : SUCCESS in 1h 35m 35s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/2cbfc48fdc01451180be93c81c1ae461 : SUCCESS in 1h 40m 14s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5ef4cbc76fb04398896b6863a4ffca19 : SUCCESS in 1h 44m 16s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/fd583a8c74d64dd0a172c48612c96290 : SUCCESS in 1h 03m 39s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/610f1c6b13604a2faffbbf4cc7081636 : SUCCESS in 8m 24s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/34f7d1b253fe4cb98c054d39f65cd1cb : SUCCESS in 8m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eb27e297c07d4929ba6c58330b2970cf : SUCCESS in 8m 05s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cdaacc38e9824bcebff1bf0669d0daf8 : FAILURE in 43m 06s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/9451ae39fc164a1c889b3f3b84568a06 : SUCCESS in 1h 03m 46s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/fb0975fa9eca425e884304322eb2c354 : FAILURE in 1h 30m 11s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/d1a22919cf9f49f6978ae98acc77d214 : FAILURE in 2h 05m 58s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/b3ea79b49104498aa663d334a60444e7 : FAILURE in 1h 03m 52s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0a35ed425c7041388b2b20db3fb07a0e : SUCCESS in 1h 16m 34s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/242bbca66362410fbdf471413b26a961 : SUCCESS in 1h 16m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/394db03123f940198dceff9c1d9dcf86 : SUCCESS in 2h 08m 50s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/f588090b4374466ba67f7d0e09b59b34 : SUCCESS in 2h 12m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/47bbf0ca5a784598a3225b05d44930d7 : SUCCESS in 1h 52m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e73e49d51804ce2961dc2d590749e2d : SUCCESS in 13m 59s","accounts_in_message":[],"_revision_number":1},{"id":"97a18c1cbe380325238af661e71a451acf75b1e4","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-02-27 23:26:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d42e8b44ffec5b0e0e493628b73c9a2d4c37e763","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 01:35:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3901ff079c7443f89823968cb66ffa2 : SUCCESS in 10m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82399c0a741e4b65b4eb5e19f411d85f : SUCCESS in 9m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8e6f8fa2dca74183a424afaf59992cc6 : FAILURE in 7m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/922bff0058294591810f747e3f3df70c : SUCCESS in 7m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/338fab82805640b6a37a06affed258d4 : SUCCESS in 7m 56s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/7fea3461b02d42b8b7afcbe36887f94e : SUCCESS in 1h 19m 40s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/4cdacfd4bd7047df92c09d214adc46fd : SUCCESS in 1h 19m 42s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fd2586cda5e24740aebacd7b1694501a : SUCCESS in 1h 29m 05s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/53ccf02c39d24c9f8832ab593c502a4a : SUCCESS in 1h 24m 17s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6a0db0c600e149709792ec26aeac526d : SUCCESS in 10m 12s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/e011e972f0e44a938a4fe09e162b6212 : SUCCESS in 7m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8e2e3a2180e54d98a222c610041be093 : SUCCESS in 6m 46s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a38861171fd84798a2444db99cab181f : SUCCESS in 1h 03m 17s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/ee7d6ed31d1649328778251e2cb81612 : SUCCESS in 1h 09m 54s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/ef06601af8f84fb8b0e4b86a2c868427 : FAILURE in 1h 21m 20s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/548ce95db05d43a0b94e2e46f4df6d70 : FAILURE in 1h 22m 23s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/49c18eacbfa346138472744374db24e6 : FAILURE in 31m 10s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/86700928733248bdae9316bf60ef24ac : SUCCESS in 55m 32s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/1bdf91f63449432fb527e32a26bc94f5 : SUCCESS in 49m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bf6f48a849ce4b58bc26d9b198e1baa7 : SUCCESS in 1h 58m 57s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/3a27592873984cf498668b771dc5a5ac : SUCCESS in 1h 25m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f33a48de940e4d8092a12c6e3c13cbd0 : SUCCESS in 1h 18m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7dbc773d7d4848d0a31557c161009845 : SUCCESS in 10m 09s","accounts_in_message":[],"_revision_number":1},{"id":"979b4bcc2e59667e126794f890086e2a2bf0e3e0","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-02-28 01:53:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e9820fc8dc83a29ea73dc3b5f43c702fe25763e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 03:54:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ce2470f64904855830969d12a0cd3dc : SUCCESS in 10m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0954e7b0642447c48f92646b455bb4fc : SUCCESS in 9m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/badc43da3f374d799fc11bf5178df0e8 : FAILURE in 7m 52s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/28cefeaa0b55447e9c30c2e0560b6d79 : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b94ba0c7ab9142429b17d2c6f452b75c : SUCCESS in 8m 44s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/a8f1b1c717c843fe959e93454757ddfe : SUCCESS in 1h 16m 45s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/ebd1b559a5684bc8a05e673994e102b1 : SUCCESS in 1h 15m 23s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/171bf7d12c814ee3865c10f84c29901a : SUCCESS in 59m 17s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/0d8562fcf4f14402b2f36e430517793c : SUCCESS in 1h 27m 43s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6e6be422accb45848a092d0d49faaf21 : SUCCESS in 11m 37s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/8756977652254751891ae27e7a9425ab : SUCCESS in 11m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ec6423776b684a469b83e9328143f928 : SUCCESS in 10m 35s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c06344fdc6a54e659e86069ee26db6cb : SUCCESS in 1h 02m 57s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/68e437f9499e4291a4829ca18c3af453 : SUCCESS in 1h 01m 10s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/eb9b0c120b9d43f4bf8283293c8a2d6f : FAILURE in 1h 18m 40s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/9b7edd408497442084fb8515295884e6 : FAILURE in 1h 34m 14s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/db388a708ddb4016bebc996a804e2891 : FAILURE in 33m 03s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b5307f4a19f741558a1ad68d8074142a : SUCCESS in 49m 13s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/bfd0e141db7e4d27a0fb2b0e93070198 : SUCCESS in 47m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e6dbf13e346e429a9fe50ff06ec0b493 : SUCCESS in 1h 34m 22s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/1e18f2ab0cfc4623bbb93611d6c31997 : SUCCESS in 1h 39m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6ae8f2b10b9a4efbbad52faf49b157ad : SUCCESS in 1h 08m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/160ea69da79848649ad6030ba3c40a34 : SUCCESS in 10m 45s","accounts_in_message":[],"_revision_number":1},{"id":"640a9248f3806e111dabce76221e31a5d767230e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-02-28 04:53:59.000000000","message":"Patch Set 1:\n\nI think the python27 issue might be the newer versions of virtualenv breaking steveador driver loading? I haven\u0027t looked too closely here, but we\u0027ve noticed similar issues elsewhere. Either way, I don\u0027t think it\u0027s related to this patch.","accounts_in_message":[],"_revision_number":1},{"id":"32564e6e2b5ee99743ee5bc9fe87f748d6f023f0","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-02-28 17:27:38.000000000","message":"Patch Set 1:\n\n\u003e I think the python27 issue might be the newer versions of\n \u003e virtualenv breaking steveador driver loading? I haven\u0027t looked too\n \u003e closely here, but we\u0027ve noticed similar issues elsewhere. Either\n \u003e way, I don\u0027t think it\u0027s related to this patch.\n\nYeah, it would be nice to get this one landed so we can have HTTPS healthmonitor in stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"9e0b12fa60c730cfd31c1cf954678b38e78106f2","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-02-28 17:51:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ec74aaf096ae185fce619036e7a00c656d5b593b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 20:16:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8fccf2ea3bcf4fcb9d5a3f0a81d98ab5 : SUCCESS in 8m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6aadb9831ec04aaba852f2fdd680e4fd : SUCCESS in 10m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/606622b345a44f439f7d6f653ba5fcc7 : FAILURE in 7m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/aa9aadb005f247d7a38a770232de2824 : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90489d891fd14c7a9db82680b42967fc : SUCCESS in 10m 45s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/1fb4e8f681d3400782ae4de8a8bf331f : SUCCESS in 1h 39m 22s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/c775e5ff93cf427bb144e49394859612 : SUCCESS in 1h 28m 50s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/513ca60260964bf5ac029e8f82a08f72 : SUCCESS in 1h 40m 28s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/2a376843fba4483bbef2498e170de860 : SUCCESS in 1h 59m 17s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/67dd93017991437c8c0c66e0bcbfd7fe : SUCCESS in 9m 15s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/b4c0a59c47d84129a2c22395f7215ce1 : SUCCESS in 8m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3aa241d13a1a44158c04292e2919f8c0 : SUCCESS in 6m 51s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a65382e9a84e4a8788da9907e48d3832 : SUCCESS in 1h 05m 19s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/c4f45c42b95e4e5e82d7b1553d23f6a5 : SUCCESS in 58m 54s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/80fe841d6cea4d0d84369843a37b88c3 : FAILURE in 1h 18m 33s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/1876dc4d059741d89ced6836ef2d408c : FAILURE in 1h 39m 05s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/60bc546ac4a54cf79317f2c408ff71cd : FAILURE in 43m 21s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/73527e2f457b4684aae252954ce7e47d : SUCCESS in 44m 57s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/6f2677dc03f94275befa423c69acf6e6 : SUCCESS in 48m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/37e320633d444e69ad9a49b827d5165e : SUCCESS in 1h 55m 37s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/800ce9042b7f4210a05c2aa83deb1989 : SUCCESS in 1h 39m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f34eac8ef528460ba648a344d4aa1bf1 : SUCCESS in 1h 09m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f22d0f0221d4f03b2405c7b8e660fb8 : SUCCESS in 9m 09s","accounts_in_message":[],"_revision_number":1},{"id":"ac0a1dbf88bfa1a35ddf2fa66127828c35d73991","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-03-02 16:17:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"36fe5606831271687488af777c1a40cd9fc59171","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 17:45:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76fc28909d3e4f51a443607d172f91be : FAILURE in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e08d9a9e8ba4fd884b54eb827247f91 : FAILURE in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0cff2679a4a0411696d4c8e2fb68ef41 : FAILURE in 4m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bbedec46c9b54c4c8171dbc67eb7fd1c : FAILURE in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b5bf75e2d594618bf4747f784545ce8 : FAILURE in 5m 38s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/aad414cdb56c4b1d954a0b397d36e758 : SUCCESS in 1h 16m 20s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/6cb28bce41884660928cae6a667ac873 : SUCCESS in 1h 22m 32s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ed848b3381d842839f97d4ef7d048c2e : FAILURE in 33m 23s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/d850e5ba234e4a5aa7f6c4d325cdf8f5 : FAILURE in 37m 14s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/22d6b918c40b456d95357d33798a65dd : FAILURE in 4m 57s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/49b48de7730f41e387493aef2f5cd30c : FAILURE in 6m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a19cf04e36a7452aa3d73145049d0854 : SUCCESS in 9m 28s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/acd7aad35d654370bff3d93cf9b798b0 : FAILURE in 37m 03s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/ca5df4bb03e942b6a397ef2f82218d72 : FAILURE in 39m 18s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/0702814724df4970959f256201b653d4 : FAILURE in 40m 41s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/d5b07f851ecc496fa24d11cbb22244fb : FAILURE in 44m 49s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/c169f8d44a4a4492b1ffde703b4ca472 : FAILURE in 32m 54s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b8303e117374421bb769e60cc0ddbe58 : SUCCESS in 47m 03s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/1079534ebc4a4f8aabca503ed762c3dc : SUCCESS in 44m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d197cf72b2eb4427a5ba18a1bd7b111c : FAILURE in 34m 52s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/ea459c1322004c13be86bcdcb1149e1e : FAILURE in 36m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7a39b43a58124844979925679673a6ec : SUCCESS in 1h 15m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6700152c9af4e12b69012ed6df8cbe6 : FAILURE in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"b3795052b4d3b9836ee5fc9647ae532ca684ff06","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-03-03 02:02:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4e5787b78f479b638af4437916f1d59237945c7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 03:56:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/05771f5aa7ba47978fcc3d0443e910e7 : SUCCESS in 8m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2349f0b7c11e448ab48e18e4906b2094 : SUCCESS in 10m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6664c9634b0c4f1986dd52620a1ed440 : FAILURE in 7m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/688fb243d4fc448b9192e60c76dfb9ac : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ee7ab9a4f774475a9a5113c2f10eb05 : SUCCESS in 8m 20s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/6114796c58a34c95b32b6602a3864388 : SUCCESS in 1h 12m 05s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/3017441d80634315a1765bcc0e7e4a39 : SUCCESS in 1h 16m 01s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/38b90dff927d43c497e4920aee86a005 : SUCCESS in 1h 10m 37s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/56cf5f656e9d4006a16baf0e4a3e3ce6 : FAILURE in 43m 19s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c3dd9dbb5fd44fa4bb9fd0e05958d8ce : SUCCESS in 6m 42s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/733e2669e5814ba0af076bbadd5621ef : SUCCESS in 10m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d7c912bbb12144f386dc4de3f835f556 : SUCCESS in 6m 07s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/47cae299c99447f994ea240ba895a7d4 : SUCCESS in 56m 51s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/78370033a49b4976add23cda39041932 : SUCCESS in 1h 15m 41s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/19d2e98deba9400b98aa4c01bc55f32d : FAILURE in 1h 35m 17s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/d0bf9fedb0a344abb501990da7bfb15f : FAILURE in 1h 25m 22s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/d3be3528fa004b0ab9b2a4db834e6ee8 : FAILURE in 29m 16s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/74d4092ee51e4790871ecff0a86d2760 : SUCCESS in 50m 44s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/4db7630b5eb041eaa38d03f42ae0ab89 : SUCCESS in 50m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7203c50a31b1420a800a90f794dad83d : SUCCESS in 1h 48m 42s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/2a9745438d81456a9e4183310bf74c45 : SUCCESS in 1h 35m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fbabd788c55b41089219ea4224f4d7be : SUCCESS in 1h 24m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/232fe27f9d27446c808fd91a161cd656 : SUCCESS in 12m 49s","accounts_in_message":[],"_revision_number":1},{"id":"bdbdd1f9b256c0e3e207bc362eafe354d9dec149","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-03-04 12:10:17.000000000","message":"Patch Set 1: Backport-Candidate-1\n\nWhere is the train and stein backports?","accounts_in_message":[],"_revision_number":1},{"id":"6c0fe5212abe7bca295a484480d876d6e9ba4231","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-03-04 12:12:58.000000000","message":"Patch Set 1: -Backport-Candidate","accounts_in_message":[],"_revision_number":1},{"id":"e74427f85712a28fe1bc6b526965f9a21393fc6f","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-03-04 16:15:58.000000000","message":"Patch Set 1:\n\n\u003e Where is the train and stein backports?\n\nThe bug was fixed in Stein but was never backport to Rocky. Without it, we can\u0027t create HTTPS HMs.","accounts_in_message":[],"_revision_number":1},{"id":"282971a4d012d673c59aca88d620708a47ac76d8","author":{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},"date":"2020-04-15 17:00:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8ae4ba5f417a39529834207ba74965b396106648","author":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"date":"2020-04-15 19:49:03.000000000","message":"Patch Set 1: Backport-Candidate+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"c5afd971d90826ed8c5ff62d725eac1bb30bc728","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-05-01 16:26:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"acc3ad54a27b733f35554cd5796ab86cf3ceaf83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 18:01:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08454a6915b24fb2b14f889fd2c46380 : SUCCESS in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc28b3461068489a8fe2809cc6782097 : SUCCESS in 8m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2ce42c532e664735bf64918acf431171 : SUCCESS in 6m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e6aa85fb670644919b69d59b9e1ebed4 : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb7754b28e0445e7a642578216b3cf77 : SUCCESS in 7m 37s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/a29bc08bf44a49d8ad1c32ffd76d7e23 : SUCCESS in 1h 14m 58s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/0e6824cf34744de9befdf1512319adb3 : SUCCESS in 1h 16m 54s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/256eaceb51b042dcaf554d33a843f610 : FAILURE in 1h 18m 49s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/141b746a670942d9a64c8d38b902f718 : SUCCESS in 1h 23m 04s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b4652a2659df4d8b982d079414298782 : SUCCESS in 7m 36s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/edc9bd27b6ec4afaaf49220b9b0d912f : SUCCESS in 6m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3d69810a572a4ce69db97e30fb09d516 : SUCCESS in 6m 38s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/546d669211714124bb0011d8e8ec082d : SUCCESS in 55m 03s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/8586f468f23c403f970348c21d9cb7d9 : SUCCESS in 1h 13m 19s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/5d3fde9282bf466eb6e72c632cab123c : FAILURE in 1h 20m 57s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/b68651bb259f4e518eb83de590e148cf : FAILURE in 1h 27m 07s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/e4a43b7e4b1e46219135b8aee4d98db0 : FAILURE in 27m 46s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5a1a5757407741c9bb3a984045779c68 : POST_FAILURE in 28m 33s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/5479292042c84a00992c7f371de930a0 : POST_FAILURE in 33m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/595b0667ac95436aaa42b9ad57679a02 : POST_FAILURE in 38m 37s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/47c6ff0200844e4c9598cdebe0b78da2 : POST_FAILURE in 41m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/01221ab79f0f493c82b6afae99d35235 : FAILURE in 1h 01m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f28d40aa6454931b8b685d5c39fd970 : SUCCESS in 9m 34s","accounts_in_message":[],"_revision_number":1},{"id":"cbe4a31830f85f781a60cc5deaffd909c70e6a5a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-05-04 06:58:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6e1cb225d3732fadbd55dd983a9e28e5f5326efb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 11:36:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ca420f88cbc4ceaa9196ba31623b464 : SUCCESS in 12m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe029d83a0dc4f3b847775deac7a2f2f : SUCCESS in 6m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7bb7790f7e7546659bfe4c850bb8a15b : SUCCESS in 6m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b65b59d5cb0b4ebb9510a64d8f4f6b73 : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6018d13749214c3fbd9684ec42df84c1 : SUCCESS in 7m 50s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/0edcf2378b254b339cbc7fe7001a70b9 : SUCCESS in 1h 21m 52s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/ffdab744e669421bb158ded47e025232 : SUCCESS in 1h 20m 50s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/36b379756f0b4d2fba107c0d686383f8 : SUCCESS in 1h 24m 02s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/4b5ed6cec3a6451aab1d2ce1feb420c5 : SUCCESS in 1h 28m 06s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f556e34dcd064295a907814ab132e1d3 : SUCCESS in 8m 38s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/ed8498baad21427c8cec19b8961bc884 : SUCCESS in 7m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/926077f2a76347298e342d30ec460939 : SUCCESS in 5m 38s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8da8fee9c1854ecb8d7bff2555e2f3c6 : SUCCESS in 1h 00m 11s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/e8dc324fe3094a8187c32d8642e60266 : SUCCESS in 59m 24s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/450a2d9a2cfa4b82891084c66aa622b5 : FAILURE in 1h 31m 08s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/d69642a0e8ad48c78001e1a7a6c251f4 : FAILURE in 1h 42m 32s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/14e0a372e8d94488b6d6e020f1de3a42 : FAILURE in 29m 15s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/989425cdecf144ea84721716d8e16ead : FAILURE in 32m 03s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/a2564257fe5b447c87102a2c216ede1e : FAILURE in 32m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d568047df6ab4fddac6407462e4754da : FAILURE in 40m 51s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/4843f6e7017842f7aa4fb5d2e76780b3 : FAILURE in 45m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/475e1a56941f4b6b8a096d0329b80efb : FAILURE in 1h 02m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82f988b6fa664da7b6f8132b0dd54de5 : SUCCESS in 9m 23s","accounts_in_message":[],"_revision_number":1},{"id":"3a7e36f464c962063e6d412867d0ba73db6ecd9d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-24 11:41:55.000000000","message":"Abandoned\n\nRocky Octavia reached End of Life (EOL) state. This branch no longer accepts changes.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d273100ed72dff6f36dff52745a180a426e3788d","revisions":{"d273100ed72dff6f36dff52745a180a426e3788d":{"kind":"REWORK","_number":1,"created":"2020-02-27 00:31:34.000000000","uploader":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"ref":"refs/changes/61/710161/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/61/710161/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/710161/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/710161/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/710161/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/61/710161/1"}}},"commit":{"parents":[{"commit":"e7376be7afec98192da969dc9382ad860b1cbe01","subject":"Fix uncaught DB exception when trying to get a spare amphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e7376be7afec98192da969dc9382ad860b1cbe01"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-09-24 22:14:17.000000000","tz":-420},"committer":{"name":"guang-yee","email":"gyee@suse.com","date":"2020-02-27 00:31:34.000000000","tz":0},"subject":"HTTPS HMs need the same validation path as HTTP","message":"HTTPS HMs need the same validation path as HTTP\n\nChange-Id: I2fd51664336dca51f134b3fccd3e8c936b809839\n(cherry picked from commit 7dd6f8b846015e6fbe75af09ee3a5108741b1626)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d273100ed72dff6f36dff52745a180a426e3788d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d273100ed72dff6f36dff52745a180a426e3788d"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
