)]}'
{"id":"openstack%2Foctavia~711154","triplet_id":"openstack%2Foctavia~master~I7231392431008a23bebe254c9af41adc24e36ded","project":"openstack/octavia","branch":"master","topic":"tls-options","hashtags":[],"change_id":"I7231392431008a23bebe254c9af41adc24e36ded","subject":"WIP Add parameters for configuration of TLS protocols","status":"ABANDONED","created":"2020-03-04 03:14:40.000000000","updated":"2020-03-29 01:18:49.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"67e75d06a3a57e3d62a786d247aecf63c4b0cc2e","_number":711154,"virtual_id_number":711154,"owner":{"_account_id":31563,"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","username":"luketollefson"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31563,"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","username":"luketollefson"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-03-04 05:31:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31563,"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","username":"luketollefson"},{"value":-1,"date":"2020-03-04 23:12:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":31563,"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","username":"luketollefson"},"all":[{"value":-1,"date":"2020-03-04 23:09:02.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":31563,"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","username":"luketollefson"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31563,"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","username":"luketollefson"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31563,"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","username":"luketollefson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-04 05:31:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-04 23:12:45.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"ec496684df987853f7ccec660a9170e5a4b5beac","author":{"_account_id":31563,"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","username":"luketollefson"},"date":"2020-03-04 03:14:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e98993d3d3c35da9bab0ea1878d7caf67d882924","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 05:31:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/430dd85969414507bc4cdb0fedb99849 : SUCCESS in 8m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec61e5e44fd44314ae22bf5cebc06f65 : SUCCESS in 8m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c38c00c08d1e4886ad153c0335587a6b : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59123143d3464e6381d1531159e1b50b : SUCCESS in 5m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7903aaef422b42199dd79c14286cb3f0 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3829a0d936440f2af0b268f026e3493 : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adfd79cb4ea3473385fe6b4d965dc38c : SUCCESS in 11m 57s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/eee011115f354e03977bdcbac5899b24 : SUCCESS in 5m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/624df868fce0434d97fb88aa29d6810e : SUCCESS in 7m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e27a0c6c35724228b5b11af8ccfdcb4a : SUCCESS in 5m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/957532bb6d7a4bb2ad8735194f91bb36 : SUCCESS in 41m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/41124d080fbe4693a3a3b325eb4e4834 : SUCCESS in 1h 48m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/564daf0e93e24f4caa1e39740850dc9c : SUCCESS in 1h 05m 12s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ee64a75e2a104492b9bb457554ed3ce1 : SUCCESS in 55m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0541a5fd66a04042b4fedaed67700f33 : SUCCESS in 1h 06m 59s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e59fe866b6af45f3abd152154c937245 : SUCCESS in 1h 51m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ccefae03357843a6989a04d6f3ce1e09 : SUCCESS in 2h 11m 17s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4ce240ac152f43008dcc24b94f230ff3 : FAILURE in 1h 02m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b53e8c77745a40d29c89f7a53b3ffe0d : SUCCESS in 1h 59m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9facf9e32d293bc0dc847374d099396640aeaa54","author":{"_account_id":31563,"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","username":"luketollefson"},"date":"2020-03-04 23:09:02.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"b0141bcb63a14044e1760427fd12418b1199b63b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-03-04 23:12:45.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nLooks good so far, only needs the Story and Task tags on the commit message.","accounts_in_message":[],"_revision_number":1},{"id":"9bdc746dff22a1418195c356353a96e278db4aaf","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-03-04 23:15:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4a1ef0d5338b8919823a8595669238a49b5de92d","author":{"_account_id":31563,"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","username":"luketollefson"},"date":"2020-03-29 01:18:49.000000000","message":"Abandoned\n\nThe majority of this commit belongs in octavia-lib","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1a2e8f61fc8bf0784e1edfb60485239cbc12e53e","revisions":{"1a2e8f61fc8bf0784e1edfb60485239cbc12e53e":{"kind":"REWORK","_number":1,"created":"2020-03-04 03:14:40.000000000","uploader":{"_account_id":31563,"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","username":"luketollefson"},"ref":"refs/changes/54/711154/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/54/711154/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/711154/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/711154/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/711154/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/54/711154/1"}}},"commit":{"parents":[{"commit":"2c76209003d5ca09a302881a40a35fd14762db55","subject":"Select the right lb_network_ip interface using AZ","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c76209003d5ca09a302881a40a35fd14762db55"}]}],"author":{"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","date":"2020-03-04 02:58:52.000000000","tz":-360},"committer":{"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","date":"2020-03-04 03:13:27.000000000","tz":-360},"subject":"WIP Add parameters for configuration of TLS protocols","message":"WIP Add parameters for configuration of TLS protocols\n\nThis is quite incomplete. When finished the user will have the\nability to specify the TLS protocols used by listeners though\noctavia.conf.\n\n37169\n2006733\n\nChange-Id: I7231392431008a23bebe254c9af41adc24e36ded\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a2e8f61fc8bf0784e1edfb60485239cbc12e53e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a2e8f61fc8bf0784e1edfb60485239cbc12e53e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
