)]}'
{"id":"openstack%2Foctavia~711275","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I161a522abad4a2aa521ea46cb1065c5b05a2cd2e","project":"openstack/octavia","branch":"stable/train","topic":"tox-cleanup-stable/train","hashtags":[],"change_id":"I161a522abad4a2aa521ea46cb1065c5b05a2cd2e","subject":"Fix batch member update error on empty change list","status":"ABANDONED","created":"2020-03-04 16:49:16.000000000","updated":"2022-04-14 22:53:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7d8bd8697ed84706842b9c7de81b3039b2f879d1","_number":711275,"virtual_id_number":711275,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-03-05 15:58:53.000000000","_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-03-05 18:26:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":-2,"date":"2020-04-15 17:01:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2020-04-15 07:17:20.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-05 18:26:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-15 07:17:20.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-04-15 17:01:31.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"9fef40c0f54578341665ea29a7e2b47e566cce83","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-03-04 16:49:16.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"80d3f1049b8765ad75acd4334c9a26ffdecb0240","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 19:19:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/622205aec8794d03b666e788e07e1d06 : FAILURE in 9m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c8184c4ad2349e8bc5f76506b397b8d : SUCCESS in 10m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a5325c85d324983a8a708217b6f6fe1 : SUCCESS in 8m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cb04e68744e8491cb2fe05baab54ac01 : SUCCESS in 7m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9686211d4eea4d4a90cb49b7ae2949aa : SUCCESS in 7m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a8313d0224564a0d891017c8199fe348 : SUCCESS in 7m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59a45aa3bbfe4fa5b339a654d1526d14 : SUCCESS in 12m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3e42e893657f40c1b7fb635942b8ce84 : SUCCESS in 7m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/24f361c264d140e194b462ea874d3077 : SUCCESS in 9m 51s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c5d40af82a2f4263ae2ef376682396d6 : SUCCESS in 9m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4abe57c41adf4840a7cddf62cf758720 : SUCCESS in 9m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/934b0dc46ba1442a9c5cf9f9e07f08df : SUCCESS in 49m 39s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b1639b7c2bbb4ac598f7d954b9b09319 : SUCCESS in 48m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4616f1863b9d4a869d06390c86e79560 : SUCCESS in 2h 08m 50s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/9f9d863283884c7cb21e08fb8dfbc6cf : SUCCESS in 2h 03m 21s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f470ca309d6743debca5d923ce36ee42 : FAILURE in 24m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/28a9e0e112324a9c85aebb15596cd489 : SUCCESS in 1h 22m 24s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/60dd4b0ce22043f2b56ace125e35e119 : FAILURE in 1h 52m 29s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cfe47ac8644c49babead1a8afd6805e0 : SUCCESS in 1h 21m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/19f7f8e76ebc4f448b9881ca1827f4e0 : SUCCESS in 1h 01m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/435c5d5acf0649dc9a7286d602f84c00 : SUCCESS in 2h 14m 48s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2e256b50a8db480ba9f6f4b05a0f3291 : SUCCESS in 1h 27m 48s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6a33f5480e1848aaa7bab32122f070c8 : SUCCESS in 1h 12m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c88175dd5797463291ef1dd3b51092fe : FAILURE in 24m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2385c5946101ccb06fa5288d98c88d50a04b2f13","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-03-05 15:58:53.000000000","message":"Patch Set 1:\n\nrecheck mirror problem","accounts_in_message":[],"_revision_number":1},{"id":"a9a893fe66b50f20f2fd9e854b1cd05cb172e6dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 18:26:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/990268e1f0e94f14ab382e7eb453c36c : SUCCESS in 8m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23f7031c08a641e9aacdda9d9e0fc3e4 : SUCCESS in 9m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/857c9597897c404abf622fd4b067c452 : SUCCESS in 6m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/66cc3a44a06846989f2e152fe8c0b765 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7003f698fc104a539a9f123f0db8c492 : SUCCESS in 8m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a3028d306ac248c9beecbc4c303ed40c : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7684e0a1e714b89a5586d54440ecce3 : SUCCESS in 11m 41s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2d7cb7370e3b4fcaaf393a647fe5ee37 : SUCCESS in 6m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0b0ff86c9c954d27b5ef1e45be5fafb8 : SUCCESS in 8m 16s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/bd8e97a197d1431ab21eaff9488c176a : SUCCESS in 7m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ff45297285d34f6e8cfcfe0c4027578e : SUCCESS in 7m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/064f481ad26b4450bcae771218bc0726 : SUCCESS in 55m 01s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c4d2a92fd1f94d0baf6db2857aa05705 : SUCCESS in 47m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b5e261ed2b5f43e0b027ab1ffa4f860f : SUCCESS in 2h 12m 44s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/622a84dec2684df2ba0b203a1df9631e : SUCCESS in 1h 50m 55s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d15755181b5f4469bfcf376045f3a0ca : FAILURE in 19m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d1f9e29717df407caa3fd71f9d6d6c5e : SUCCESS in 2h 08m 38s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/a95d5cfadb5848b3a02b6d3e63474035 : SUCCESS in 2h 05m 07s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3c75e04ed5224dcbb75f0f0d72234e74 : SUCCESS in 1h 35m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2269c3fc26d446698f9de2e568121f09 : SUCCESS in 1h 01m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/063d41eaf55142cd8e958577d9464cad : SUCCESS in 2h 19m 30s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f0237992288c4f84b9ea62391ec51407 : SUCCESS in 1h 17m 37s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/dbead2f8021b45cfa05e54ed9e58fdae : SUCCESS in 1h 17m 41s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/e8c3562f6698428aa0467324355f3d37 : FAILURE in 27m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1719ee9f2429b9c0642d2737d319b6888b93dea3","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-03-10 00:42:43.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"c1c43ad154d90134a0b43df6bebbef3c06d19658","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-10 07:12:13.000000000","message":"Patch Set 1: Code-Review-1\n\nThe depends-on patch needs to be backported to stable/train first. Octavia-lib is a branched repository.","accounts_in_message":[],"_revision_number":1},{"id":"ee21301f6318e81d65ec1a1a0f048d56ba06b0c4","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-10 07:16:25.000000000","message":"Patch Set 1:\n\nI am not sure, though, how to release stable/train octavia-lib in a way that it could be consumed by stable/train octavia and would respect upper-constraints (octavia-lib\u003d\u003d\u003d1.4.0).\n\nhttps://github.com/openstack/requirements/blob/stable/train/upper-constraints.txt#L141","accounts_in_message":[],"_revision_number":1},{"id":"128aa31b21064ff85ffdc386f6a2c0b75199e8f8","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-03-10 08:40:48.000000000","message":"Patch Set 1:\n\n\u003e I am not sure, though, how to release stable/train octavia-lib in a\n \u003e way that it could be consumed by stable/train octavia and would\n \u003e respect upper-constraints (octavia-lib\u003d\u003d\u003d1.4.0).\n \u003e \n \u003e https://github.com/openstack/requirements/blob/stable/train/upper-constraints.txt#L141\n\nMaybe bump the octavia-lib version in https://github.com/openstack/requirements stable/train first? If octavia-lib is backward compatible.","accounts_in_message":[],"_revision_number":1},{"id":"61603b520744324ff99b0d52e1816d0941720cbf","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-03-10 21:07:19.000000000","message":"Patch Set 1:\n\nIf octavia-lib goes with branches, maybe make a new release for branch stable/train, then the https://github.com/openstack/requirements/blob/stable/train/upper-constraints.txt could be updated automatically, that\u0027s what I remember.","accounts_in_message":[],"_revision_number":1},{"id":"13a0d56e4f8861927eba9a1783a6e51e41561b69","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-04-15 07:17:20.000000000","message":"Patch Set 1: -Backport-Candidate\n\nHi any update on this one?","accounts_in_message":[],"_revision_number":1},{"id":"7af96196932e2210c035e07f0a2633d9a08d806d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-15 07:28:51.000000000","message":"Patch Set 1:\n\nI added an item to today\u0027s meeting agenda -- https://wiki.openstack.org/wiki/Octavia/Weekly_Meeting_Agenda#Meeting_2020-04-15","accounts_in_message":[],"_revision_number":1},{"id":"fd19c38d1b8e1f815b88545ca95bfbcaf6037add","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-15 17:01:31.000000000","message":"Patch Set 1: Code-Review-2\n\nThis patch cannot be backported. It depends on a non-backportable octavia-lib patch. For more info, please see meeting notes: http://eavesdrop.openstack.org/meetings/octavia/2020/octavia.2020-04-15-16.01.log.html#l-70","accounts_in_message":[],"_revision_number":1},{"id":"df176165aba709cc6f0f24b044a0832ba5a87431","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-24 11:52:00.000000000","message":"Patch Set 1:\n\nAre we comfortable to abandon this patch? If not, please work with us to find a possible way to proceed with this change.","accounts_in_message":[],"_revision_number":1},{"id":"7d8bd8697ed84706842b9c7de81b3039b2f879d1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-14 22:53:49.000000000","message":"Abandoned\n\nAbandoning as this does not meet the OpenStack stable branch policy.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"495749dbd5f85e9fd30b0697095e671829440728","revisions":{"495749dbd5f85e9fd30b0697095e671829440728":{"kind":"REWORK","_number":1,"created":"2020-03-04 16:49:16.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/75/711275/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/75/711275/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/711275/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/711275/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/711275/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/75/711275/1"}}},"commit":{"parents":[{"commit":"0c1cb417e904daa75deedaabc72a20d07b138eb9","subject":"Fix uncaught DB exception when trying to get a spare amphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c1cb417e904daa75deedaabc72a20d07b138eb9"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-14 19:43:32.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-03-04 16:49:16.000000000","tz":0},"subject":"Fix batch member update error on empty change list","message":"Fix batch member update error on empty change list\n\nIf the list of changes was empty, the worker would fail to fetch the\npool because it was retrieved implicitly from one of the changed\nmembers. Pass it explicitly instead, and also short-circuit on NOOPs.\n\nStory: 2006719\nTask: 37090\n\nDepends-On: https://review.opendev.org/#/c/688546/\nChange-Id: I161a522abad4a2aa521ea46cb1065c5b05a2cd2e\n(cherry picked from commit 01dfb8ffaf6f22e48915e05f70553870c39ea96a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/495749dbd5f85e9fd30b0697095e671829440728"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/495749dbd5f85e9fd30b0697095e671829440728"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
