)]}'
{"id":"openstack%2Foctavia~711863","triplet_id":"openstack%2Foctavia~master~I578cb9e4827aab3078e7cac1e72f037476127300","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I578cb9e4827aab3078e7cac1e72f037476127300","subject":"WIP: Adding The abllity to set a cipher blacklist","status":"ABANDONED","created":"2020-03-08 23:33:18.000000000","updated":"2020-05-01 20:58:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3a500c0743fd5436df08bf7fe1075f7d8ed2cfb0","_number":711863,"virtual_id_number":711863,"owner":{"_account_id":31565,"name":"Noah Mickus","email":"noah.mickus@gmail.com","username":"nmickus"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-03-09 16:50:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31565,"name":"Noah Mickus","email":"noah.mickus@gmail.com","username":"nmickus"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31565,"name":"Noah Mickus","email":"noah.mickus@gmail.com","username":"nmickus"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":31565,"name":"Noah Mickus","email":"noah.mickus@gmail.com","username":"nmickus"},"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-03-09 14:17:31.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":31565,"name":"Noah Mickus","email":"noah.mickus@gmail.com","username":"nmickus"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31565,"name":"Noah Mickus","email":"noah.mickus@gmail.com","username":"nmickus"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31565,"name":"Noah Mickus","email":"noah.mickus@gmail.com","username":"nmickus"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-09 16:50:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"099c122a16eb5db223f1357cc4332a3bc2bc8827","author":{"_account_id":31565,"name":"Noah Mickus","email":"noah.mickus@gmail.com","username":"nmickus"},"date":"2020-03-08 23:33:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce7fb9df7e014de20461a7685e7d5cdb9db39087","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 01:32:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f90826bd6ea847c39fda817faf74226a : SUCCESS in 7m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b5394f8e9af4cf6b0fc86c99b30e7a6 : SUCCESS in 8m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/559358a22fe44b319cca62293931882f : FAILURE in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40997901c90046d5a48cc7eecc622965 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/84c5d916a15e45d68da98abb66c21c84 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c0b37024c0e4a54871e22925350ebda : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b6b3e4985a34e54ab3d8579e263fa7b : SUCCESS in 9m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/80ca6cc6f69747079b50e64317bbadf3 : SUCCESS in 4m 23s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/189183a273234aee9486f34208000cb2 : SUCCESS in 6m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/01058318bffc4803b8c64cb68c783b46 : SUCCESS in 6m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/748d6579d55b47409005d35259a1ae6e : SUCCESS in 40m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4a10b84912ce4335bc7edcb575ff70dc : SUCCESS in 1h 39m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4cfc0b4be9bd4253970de682a043d032 : SUCCESS in 55m 13s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/41db500c926d455a93374f86562e2ad1 : SUCCESS in 51m 56s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9d028965e6844bbc8d7b25a1e829209c : SUCCESS in 1h 01m 51s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/abb02229cfad41b092be92282c259968 : FAILURE in 1h 45m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3ad2963cfd434ed8b5f40e0496e02f46 : SUCCESS in 1h 15m 13s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dd9c2550835848bca37e33ef64b27013 : FAILURE in 1h 07m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0b990426111f47fd8ca0831055f29295 : SUCCESS in 1h 58m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e0507f8ac116cedc9e9ffcd02749bf83c8ac2d0","author":{"_account_id":31565,"name":"Noah Mickus","email":"noah.mickus@gmail.com","username":"nmickus"},"date":"2020-03-09 14:00:20.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"aa3c234e0a87a3bf79bd49b74d93c96b4ce8dbd3","author":{"_account_id":31565,"name":"Noah Mickus","email":"noah.mickus@gmail.com","username":"nmickus"},"date":"2020-03-09 14:16:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"30be84c28ceb3827246a33f3519a5d6800e7d68e","author":{"_account_id":31565,"name":"Noah Mickus","email":"noah.mickus@gmail.com","username":"nmickus"},"date":"2020-03-09 14:17:31.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"b267756c5edc91f5b4972607eb94bd2c1fc45bed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 16:50:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8512168e15b425eaf012c71cc53a76d : SUCCESS in 9m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41186bf9f98842f59d9b336ab786378e : SUCCESS in 10m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e34d1113825f4e5eb6543db8dd18cc28 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6cf937812fbf44579af0802dfa036ab8 : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e22d882ed6924b20be03d8ef1cc12015 : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/776b5c0218f648c9864122548d970026 : FAILURE in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/593241cc7b7f4ef6b25a0c7342af1b49 : SUCCESS in 13m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/011f6fa04c584ca0b5d57b4d112e14dc : SUCCESS in 8m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5410d767f681437eb4ae2303ccde2880 : SUCCESS in 11m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bcb6a6fbc4464fc0a8a0c2128028014f : SUCCESS in 9m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f2da7aae4942448a97cc88ccd0e8c8b3 : SUCCESS in 44m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b3d7f02d25d847338560dd33f6e3fc18 : FAILURE in 2h 32m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d5aec869d1a348b8803ffc0e09344fd8 : SUCCESS in 57m 56s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0e1c631739664f19ab57492efdca45bb : SUCCESS in 1h 06m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c513bc46b43b40f9a63c4bc5ce4081f1 : SUCCESS in 1h 04m 15s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3b9f836cc6e046debb9fdeb7330cb183 : SUCCESS in 2h 12m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/11f9601388ef4b218782c456b6f01398 : SUCCESS in 1h 24m 05s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9d5ea146d136497bb9159cd4953d7ce4 : FAILURE in 1h 02m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bc355a632e83426b98ab4df38afd02d5 : POST_FAILURE in 31m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3a500c0743fd5436df08bf7fe1075f7d8ed2cfb0","author":{"_account_id":31565,"name":"Noah Mickus","email":"noah.mickus@gmail.com","username":"nmickus"},"date":"2020-05-01 20:58:33.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"62d5ff6e6c4c0323c53c3a1499a04d71336339c5","revisions":{"676dcb48e00fcdef0b94f8efb99e4beba663b6a7":{"kind":"REWORK","_number":1,"created":"2020-03-08 23:33:18.000000000","uploader":{"_account_id":31565,"name":"Noah Mickus","email":"noah.mickus@gmail.com","username":"nmickus"},"ref":"refs/changes/63/711863/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/711863/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/711863/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/711863/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/711863/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/711863/1"}}},"commit":{"parents":[{"commit":"af9edc30d7f2f93fec2e0fd2a430e31f6846541e","subject":"Merge \"Complete dropping py27 support goal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/af9edc30d7f2f93fec2e0fd2a430e31f6846541e"}]}],"author":{"name":"Noah Mickus","email":"noah.mickus@gmail.com","date":"2020-03-08 23:33:05.000000000","tz":0},"committer":{"name":"Noah Mickus","email":"noah.mickus@gmail.com","date":"2020-03-08 23:33:05.000000000","tz":0},"subject":"WIP: Adding The abllity to set a cipher blacklist","message":"WIP: Adding The abllity to set a cipher blacklist\n\nChange-Id: I578cb9e4827aab3078e7cac1e72f037476127300\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/676dcb48e00fcdef0b94f8efb99e4beba663b6a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/676dcb48e00fcdef0b94f8efb99e4beba663b6a7"}]},"branch":"refs/heads/master"},"62d5ff6e6c4c0323c53c3a1499a04d71336339c5":{"kind":"REWORK","_number":2,"created":"2020-03-09 14:16:40.000000000","uploader":{"_account_id":31565,"name":"Noah Mickus","email":"noah.mickus@gmail.com","username":"nmickus"},"ref":"refs/changes/63/711863/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/711863/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/711863/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/711863/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/711863/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/711863/2"}}},"commit":{"parents":[{"commit":"af9edc30d7f2f93fec2e0fd2a430e31f6846541e","subject":"Merge \"Complete dropping py27 support goal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/af9edc30d7f2f93fec2e0fd2a430e31f6846541e"}]}],"author":{"name":"Noah Mickus","email":"noah.mickus@gmail.com","date":"2020-03-08 23:33:05.000000000","tz":0},"committer":{"name":"Noah Mickus","email":"noah.mickus@gmail.com","date":"2020-03-09 14:09:56.000000000","tz":0},"subject":"WIP: Adding The abllity to set a cipher blacklist","message":"WIP: Adding The abllity to set a cipher blacklist\n\nThis patch so far includes the changes to config.py\nas a start to adding a \"blacklist\" of disallowed ciphers\n\nChange-Id: I578cb9e4827aab3078e7cac1e72f037476127300\nStory: 2006627\nTask: 37168\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/62d5ff6e6c4c0323c53c3a1499a04d71336339c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/62d5ff6e6c4c0323c53c3a1499a04d71336339c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
