)]}'
{"id":"openstack%2Foctavia~712091","triplet_id":"openstack%2Foctavia~master~I90ed1a6c8105cef206941143963360cc46b91408","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I90ed1a6c8105cef206941143963360cc46b91408","subject":"Check","status":"ABANDONED","created":"2020-03-10 13:52:53.000000000","updated":"2020-03-12 10:15:07.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f7b808d0ad8d851b95cc8639dc09fab86459a4e7","_number":712091,"virtual_id_number":712091,"owner":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-03-11 14:43:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-11 14:43:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"194a84b3989c86d04886d24ace04d9a9b2c1a962","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-03-10 13:52:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b2557e3b2a6e9eaff96bddf03d31b3888ded347","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 16:20:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eef9874119d142f4b16e5173b07f58a0 : FAILURE in 6m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96c6a16e17444ee7b092dee5ee1f1659 : FAILURE in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1ecb973c84b465a9570fadfe099a7e5 : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4229004b8d6a4c3cb8bad798086554d1 : FAILURE in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9e036af3ca834ff1a21d139844330077 : FAILURE in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f747e602f0634113b008ff9df6743191 : FAILURE in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccb90d0a4df14ddab9b63f0e7d3d8b7a : SUCCESS in 12m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/06fc6c5cc7324c948563d3093ade9773 : FAILURE in 5m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/533bb2bb62dd40fca2055ed4f1becb26 : SUCCESS in 8m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fa800024217b4473b60f2bdc54801aa1 : SUCCESS in 7m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cbdbf0d046c642739fb7c1708db08533 : SUCCESS in 44m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/944bfcb9d5ed4eab994cb83df5ce8f61 : SUCCESS in 1h 46m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7b8968f6cf92482c96cef0223f73627b : SUCCESS in 1h 04m 31s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/cea253ceefc94ca0b235e0f6c54b48c1 : SUCCESS in 56m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0e088c6e3102438b8fbb3adc3737f8a5 : SUCCESS in 1h 04m 18s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/771240f5d9874c009ce2ce875cb07128 : SUCCESS in 2h 20m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/01adb3a9092b4715869782d64bf51b7a : SUCCESS in 2h 08m 30s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/50f8ad0873b441a9ad701d6a4fc07eed : FAILURE in 1h 02m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/075140dc50304f7d80e17a8d6a542c32 : SUCCESS in 1h 56m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"be64c76dae62cbe403130715c3e6cc1de556d7d2","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-03-11 10:06:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"64b4b5e20ba101d30f8551f7ff0117609ea37a37","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-03-11 12:01:14.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"cb490d622d31ba18e9b65441e30d9b18e2424a1d","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-03-11 12:03:35.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b552406adefaf26d361629fc2914baaecde59719","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 14:43:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bed3706045034c47b3e6ff71186ff43e : SUCCESS in 9m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a360acb89ff24ecda8fbc50f24aedee3 : SUCCESS in 10m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0187026e87264df797d80c6619873a59 : FAILURE in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d64bd514a6e4e7b8a8e7b5d3b0ef07d : SUCCESS in 6m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe19953cafe84896b65a0a942358ead5 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a64b3c6e37fe47fc9bf5d359c6354471 : FAILURE in 6m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cce6b4953014089a3eeafb1a6467bd8 : SUCCESS in 14m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/de3972fa7d1d4e75b22d0d2ec4560e84 : SUCCESS in 7m 44s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/48605ba877554387be6515702a827a3f : SUCCESS in 10m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/90263e08cf7e4d148d301435a7a4f1f2 : SUCCESS in 8m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/02f0032319e844de9c6aa60ce19c1a71 : SUCCESS in 1h 09m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fb0abd50dee6434c95ef6919bda2ef66 : SUCCESS in 1h 53m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3a17ad345aa44e3080d93e684d3e2ea0 : SUCCESS in 59m 38s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d8bb2e01b0da41588efdf8b160255913 : SUCCESS in 1h 04m 55s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7f7d15db007d49e0a9f404c1fe2a93aa : SUCCESS in 1h 15m 51s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7fc6aa5eaf5145d2a3a923ee66f27303 : SUCCESS in 2h 13m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bc8c3f04ccee4e569f673a0f5388fb4e : TIMED_OUT in 2h 35m 36s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4bc06f7934834adbbf654ec118f0825b : FAILURE in 1h 05m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/08c716b0c56c4cf9ba5f1d37fb66287a : SUCCESS in 1h 49m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f7b808d0ad8d851b95cc8639dc09fab86459a4e7","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-03-12 10:15:07.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"318c1dabf898e9531ad39b9bed6dfa5f7fcd8ad0","revisions":{"3de48a981aecdc861818ac7a45cce14aa8b83740":{"kind":"REWORK","_number":1,"created":"2020-03-10 13:52:53.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/91/712091/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/712091/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/712091/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/712091/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/712091/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/712091/1"}}},"commit":{"parents":[{"commit":"13d2765a2fd03092177cfc217f25ddad301a4e8f","subject":"Add option to set default ssl ciphers in haproxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/13d2765a2fd03092177cfc217f25ddad301a4e8f"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2020-03-10 13:52:35.000000000","tz":240},"committer":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2020-03-10 13:52:35.000000000","tz":240},"subject":"Check","message":"Check\n\nChange-Id: I90ed1a6c8105cef206941143963360cc46b91408\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3de48a981aecdc861818ac7a45cce14aa8b83740"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3de48a981aecdc861818ac7a45cce14aa8b83740"}]},"branch":"refs/heads/master"},"318c1dabf898e9531ad39b9bed6dfa5f7fcd8ad0":{"kind":"REWORK","_number":2,"created":"2020-03-11 12:03:35.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/91/712091/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/712091/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/712091/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/712091/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/712091/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/712091/2"}}},"commit":{"parents":[{"commit":"13d2765a2fd03092177cfc217f25ddad301a4e8f","subject":"Add option to set default ssl ciphers in haproxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/13d2765a2fd03092177cfc217f25ddad301a4e8f"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2020-03-10 13:52:35.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-03-11 12:03:08.000000000","tz":0},"subject":"Check","message":"Check\n\nChange-Id: I90ed1a6c8105cef206941143963360cc46b91408\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/318c1dabf898e9531ad39b9bed6dfa5f7fcd8ad0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/318c1dabf898e9531ad39b9bed6dfa5f7fcd8ad0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
