)]}'
{"id":"openstack%2Foctavia~712790","triplet_id":"openstack%2Foctavia~master~If5b6411a0b7c75441a406234c2792ea68d35d0fe","project":"openstack/octavia","branch":"master","topic":"story/2007421","hashtags":[],"change_id":"If5b6411a0b7c75441a406234c2792ea68d35d0fe","subject":"Fix listener update with SNI certificates","status":"MERGED","created":"2020-03-12 20:33:52.000000000","updated":"2020-07-16 19:59:44.000000000","submitted":"2020-07-16 19:56:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712790-1594929395955-f352bdee","meta_rev_id":"cfdd6d278ff65bd53ff228186493143cd167dd6d","_number":712790,"virtual_id_number":712790,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-16 19:56:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},{"value":2,"date":"2020-07-16 12:07:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-08 06:13:37.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2020-07-16 18:09:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2020-07-16 18:09:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-18 06:43:52.000000000","updated_by":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"reviewer":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"state":"REVIEWER"},{"updated":"2020-06-04 09:25:36.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-07-16 12:07:26.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-07-16 18:09:01.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-07-16 19:56:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2155a1fd39b75fae1dc64bf9fa00ece4c52eb6bb","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-12 20:33:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d325776fcb91d0e2a4ca15b07ed23a71b5b94a59","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-12 20:40:58.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"a79297f6ea30c4518c67c0dea632d245395f7904","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 22:48:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1946c92e25664164913c320a2e4759f8 : FAILURE in 8m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e5b8415f3dc465cbfb92aefdb6dfc3e : FAILURE in 10m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6a01a5a5ed04a4f9263f1cf2eb48893 : FAILURE in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75272403bbf04c68bf2cd69b7f2e5210 : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5d4e2d9bc56348bcb2ba55fad2530364 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/844872aea5cf42a5a388c51f9be3d0cf : FAILURE in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be6509eae8824d47b9e1d1ed4a65b5af : SUCCESS in 12m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f8c62d86f0b547d98f46ac49a7c7ee16 : SUCCESS in 6m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b9e99dea979c43b89c55bb41ee532f49 : FAILURE in 7m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ed660b18480c4e48b03a107acbe611dd : FAILURE in 7m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cfde05238ad0441fa986581be45f0bd5 : SUCCESS in 51m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/02392a0c3d8f432ca0618838a08034c6 : SUCCESS in 2h 02m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/65ef19cc4f9044da83257456f2e91303 : SUCCESS in 1h 06m 33s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/696cbe54efc84e71876920fe74a96b4f : SUCCESS in 1h 13m 44s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/de732e2e2d2e469c9f7b32f2845cce98 : SUCCESS in 59m 15s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a855c2c47bd24ae29e61338fd3f97e13 : SUCCESS in 1h 57m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ac2b7ab3383447c3ad992c744143df17 : SUCCESS in 1h 20m 04s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1412e4fe726d4fa2a0ba2fc405b817ee : FAILURE in 1h 17m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7e9b4e6ad79e455f8447f574f0cea3e1 : SUCCESS in 1h 45m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"15da0297196a31c13bfc06e8f9e434e72c396178","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-13 15:27:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a64196c55581a5f579d9ddf37e51aa779e77325d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 17:41:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78cbe31feb604fac98dcca531805e97e : FAILURE in 4m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5ee58dc2cdb41ebb414c2b3c0d38dfa : FAILURE in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa6d8af8c18449cabee150d347fdc383 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7e79441e1a34d6bbcb83421aebb9a12 : FAILURE in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b2f6b8a14cd74102900a94d608e29719 : FAILURE in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/640d736446514708b566595f4bfa704e : FAILURE in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e09e4aa3869e4b13b1f737d96c316c33 : SUCCESS in 12m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1f3da45451194f45854c79c5eedd1bed : FAILURE in 6m 53s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/87bd5009c08d44ab8f63cf360eec2a90 : FAILURE in 7m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6695a023954c4046a2327bd5d3035387 : FAILURE in 9m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b5afdbaa540d4f13a294591f395c19b5 : SUCCESS in 42m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/69e5d89e87ba480cbc8a985f5485e972 : SUCCESS in 1h 51m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/97598e6c24c34be883274d01b1c991f0 : SUCCESS in 1h 03m 46s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/47e51aaf389247e2b85c93a535b6710c : SUCCESS in 53m 39s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/971a178dc12249ac91e4095b325068ce : SUCCESS in 1h 04m 21s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ede0a4289b2d4f9daf62dc9dd1f35a5e : SUCCESS in 2h 07m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/edb0ddaaee064491b1ac09dcebd84daa : SUCCESS in 1h 20m 33s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/518f7b0969d243d7bdf7fd559a38eb9d : FAILURE in 1h 08m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7923181e9bab46e899beb8498bf3f2c7 : SUCCESS in 2h 02m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8354c08b2c761f4ac6b29c36917fc6401ed79175","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-13 23:10:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7f42d211a25b6f73d8177cd624822389570e5d89","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-13 23:11:43.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9724c056a15879365d33cb009a42ab3c3be3e1ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-14 01:18:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1a33e4b785241df8928ae9ea61e0f84 : SUCCESS in 7m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04875a454ea44215ba145cc56cd0e0cd : SUCCESS in 13m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3298370109b345b18c86e94f96337604 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83dedef0bd7e4bd9a399e7df78c43913 : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/86bb1a4f9e7c4b75864e2c7b39a53feb : SUCCESS in 7m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f347ae87f0d74fd59bd267d112156e7e : FAILURE in 8m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7258d6de0664c9db9faf064b94e47bd : SUCCESS in 12m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c11db3b493c745a79f726ef7e655b9cb : SUCCESS in 6m 27s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1c18825c0e424c45a04dc3a4e425cd98 : SUCCESS in 5m 17s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/00dc877ce90d49c0bba382ecc7f3fecc : SUCCESS in 9m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4bf945a2e04342fb8b36b63baf9c39f5 : SUCCESS in 9m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/acb0ea5e25d34bee89dc110f15de3c84 : SUCCESS in 38m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/72268d7a34ed4ec0aba2bcb0d7879403 : SUCCESS in 1h 57m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4593e270b545429fa567a1d988b8213e : SUCCESS in 1h 02m 34s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3859faa976c84ac88852e47fe4dc5be4 : SUCCESS in 50m 05s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f157a8497c1349d28777cb130dee22f0 : SUCCESS in 1h 09m 14s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a7257b4a8f0f40dfaaff907b956bfe14 : SUCCESS in 1h 48m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/31576c0aa5694dde826026bb83e6c899 : SUCCESS in 1h 19m 04s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0663776c7d4246c287f894116715d02d : FAILURE in 58m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/87fa532c2c58459c83e3e223ea2ec1d9 : SUCCESS in 2h 05m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1b0a081f752bf9e81caeedc469b1e732f043d0a7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-31 13:28:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"643f9e11ec25f1fbe09b883ee0706f6e9681073f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 15:32:55.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e2806055833422b98f425e8d251b092 : SUCCESS in 11m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0b405325d6843ffbffc7ca6fde44ef4 : SUCCESS in 11m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecf07f87ab374ab9bcdb0aea8349d230 : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4498149800f84be48bc61f27a253bb44 : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/26a54cb80db2483d81a8f37e3efeb9a7 : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/868775d2c576497ca7728aa7f19e6d06 : FAILURE in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e78b361235b42b09f27bee21da63414 : SUCCESS in 11m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09a22dabc0d346558e2d33ac70ba2778 : SUCCESS in 6m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7622c872fc9745cfa71f1a61e97322b1 : SUCCESS in 7m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2ca2b62af32e4506930119709433412d : FAILURE in 9m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c32d5cfc301b48dc80d85c249597c3ed : SUCCESS in 7m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/add1c58f99c34d72998472813b9d790a : SUCCESS in 48m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fffb0f2769fa49b2aecb5873a8917518 : SUCCESS in 2h 00m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2a2b5dac15694521a8c908f4850d3811 : SUCCESS in 1h 08m 35s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/45789ff08dba41969e8b7c7cc3e2a6da : SUCCESS in 53m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dab1e969823445bab0b3f7634fd9b378 : SUCCESS in 1h 17m 42s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a23571da15a940e8b7a51cf8f52a0d0c : SUCCESS in 1h 55m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8245509a1f6b45d6add7705e80dd23c2 : FAILURE in 1h 26m 50s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4aa66d5f94614a5f83d508577c7e87ca : FAILURE in 1h 06m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3aebed0fdf4748d49ebbdd72c84a0ac5 : SUCCESS in 1h 50m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7883e8c94b597a33a41b8931aee77cb3e8df62bc","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-02 11:22:11.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"12f8b1cb4d932477e969a85127695bff5076cb3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 13:48:54.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6057431478534fdbbde56e5ac0d1957c : SUCCESS in 7m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c1e85776e534b7ebe5dd9ed10a289e5 : SUCCESS in 10m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c8c0cfa8fc64dda83d6e5d761590dba : SUCCESS in 8m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0538a75d610c423c87cde29e7e4a98de : SUCCESS in 6m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/77d07f33dd9a4cb5ae8083994d875f8c : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/218851f17c144d878a77c0c558ede197 : FAILURE in 6m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba1e59acb9f04b4f83f8a1502270882f : SUCCESS in 12m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a408b0af1004d9eafc72522f4fdd599 : SUCCESS in 7m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4acec9f703a14bd9ac7e7ea2581f569f : SUCCESS in 7m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a700a50e5cc049e48e123aff45ab548a : FAILURE in 10m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cea6ae39a1514ddeacac2c8527adc780 : SUCCESS in 7m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9acffe6b3cf94d3990d0ae5ccb650c37 : SUCCESS in 52m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d6b468b423984090a703d14df4adae5a : SUCCESS in 2h 10m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a3ee535107cb4941a09fd1c1b37bd174 : SUCCESS in 59m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/98d9fec83e7d461595cc5155512f1da7 : SUCCESS in 55m 32s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4060011d4cbe4976a2a2a2581edbca82 : SUCCESS in 1h 06m 51s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7347c0687ef64ed3bb6489f4cfaf2912 : SUCCESS in 1h 56m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/82c7935c67814272876427aa3d4ebed4 : SUCCESS in 2h 16m 34s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d33f5b7a9d824124925b25544f49b7d4 : FAILURE in 1h 22m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6262634ee7be4a94a715d39aaebf69bd : SUCCESS in 2h 11m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5d03ea4eebc1eccd0e6c276a0e90ef3eb2221856","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-04 13:52:37.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"012e2ffe565c265ca7b5c9cf499c660d3dfe0f30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 16:14:54.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b79dee663c104ac9ab47948401443f49 : FAILURE in 8m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49a388e82063406fa7806ef0f458a472 : SUCCESS in 9m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1e402156a124a66b42e9068bcd1e3dc : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eef13a7c8fdf4973b873f9585eb5ddd3 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/095d622eeb4847058a97a6e38d7ecae0 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd2a362e1ed34947b20f8fd25a1d51e2 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/678e898ab12a405ab11b6b202276a3ac : SUCCESS in 11m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7fe493d93914e459cd3d2d9587ccd09 : SUCCESS in 5m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8d454b398f4c4565a5589279deea0796 : SUCCESS in 5m 24s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/79c085f0d14d44ea9f57725d1a211e37 : FAILURE in 8m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d3052fca818f4ece948e367b6f304f6b : FAILURE in 7m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ef37af347bf6454287400c5e47c59137 : SUCCESS in 43m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/07ff8db5e1914349bf4a252cc00f9be8 : SUCCESS in 1h 54m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b065b68a043c4256b34eed58381cfd5e : SUCCESS in 1h 00m 37s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7f9bb8395ced43e780137571f0cba134 : SUCCESS in 50m 27s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7591e2e05b334c249ac9e8749de89333 : SUCCESS in 1h 01m 41s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d19b5f44b1e14a75a0752b2242aa5938 : SUCCESS in 1h 44m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/77de0c7d5aef45cea4630c2b4e593cf7 : SUCCESS in 1h 56m 21s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/28f711afc7854ea69354f48a92a63152 : FAILURE in 1h 03m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7bd340b47b7f4d98a6b6f67fad92e1a6 : SUCCESS in 2h 11m 01s (non-voting)\n\nWarning:\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/oslo_context/context.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/oslo_db/sqlalchemy/enginefacade.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/debtcollector/renames.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/unittest2/compatibility.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/taskflow/atom.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/oslo_context/context.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/taskflow/atom.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/oslo_db/sqlalchemy/enginefacade.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/debtcollector/renames.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/unittest2/compatibility.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/oslo_db/sqlalchemy/enginefacade.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/oslo_context/context.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/taskflow/atom.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/taskflow/atom.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/dns/hash.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/debtcollector/renames.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/pep8/lib/python3.6/site-packages/bashate/bashate.py in the commit\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/pep8/lib/python3.6/site-packages/bashate/bashate.py in the commit\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/pep8/lib/python3.6/site-packages/bashate/bashate.py in the commit\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/pep8/lib/python3.6/site-packages/bashate/bashate.py in the commit\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/pep8/lib/python3.6/site-packages/bashate/bashate.py in the commit\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/pep8/lib/python3.6/site-packages/bashate/bashate.py in the commit\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/pep8/lib/python3.6/site-packages/bashate/bashate.py in the commit\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/pep8/lib/python3.6/site-packages/bashate/bashate.py in the commit\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/pep8/lib/python3.6/site-packages/bashate/bashate.py in the commit\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/pep8/lib/python3.6/site-packages/bashate/bashate.py in the commit\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/pep8/lib/python3.6/site-packages/bashate/bashate.py in the commit\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/pep8/lib/python3.6/site-packages/astroid/interpreter/_import/spec.py in the commit\u0027\n  Job octavia-tox-py37-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job octavia-tox-py37-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/unittest2/compatibility.py in the commit\u0027\n  Job octavia-tox-py37-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/oslo_db/sqlalchemy/enginefacade.py in the commit\u0027\n  Job octavia-tox-py37-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job octavia-tox-py37-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job octavia-tox-py37-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/eventlet/patcher.py in the commit\u0027\n  Job octavia-tox-py37-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job octavia-tox-py37-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job octavia-tox-py37-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/taskflow/atom.py in the commit\u0027\n  Job octavia-tox-py37-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/oslo_context/context.py in the commit\u0027\n  Job octavia-tox-py37-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/debtcollector/renames.py in the commit\u0027\n  Job octavia-tox-py37-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/taskflow/atom.py in the commit\u0027\n  Job octavia-tox-py37-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job octavia-tox-py37-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job openstack-tox-functional-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional-py36/lib/python3.6/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job openstack-tox-functional-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional-py36/lib/python3.6/site-packages/octavia_lib/api/drivers/data_models.py in the commit\u0027\n  Job openstack-tox-functional-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional-py36/lib/python3.6/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job openstack-tox-functional-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional-py36/lib/python3.6/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job openstack-tox-functional-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional-py36/lib/python3.6/site-packages/oslo_context/context.py in the commit\u0027\n  Job openstack-tox-functional-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional-py36/lib/python3.6/site-packages/oslo_db/sqlalchemy/test_base.py in the commit\u0027\n  Job openstack-tox-functional-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional-py36/lib/python3.6/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job openstack-tox-functional-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional-py36/lib/python3.6/site-packages/wsme/api.py in the commit\u0027\n  Job openstack-tox-functional-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional-py36/lib/python3.6/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job openstack-tox-functional-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional-py36/lib/python3.6/site-packages/oslo_db/sqlalchemy/enginefacade.py in the commit\u0027\n  Job openstack-tox-functional-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional-py36/lib/python3.6/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job openstack-tox-functional-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional-py36/lib/python3.6/site-packages/wsmeext/pecan.py in the commit\u0027\n  Job openstack-tox-functional-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional-py36/lib/python3.6/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-functional-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional-py36/lib/python3.6/site-packages/oslo_db/sqlalchemy/exc_filters.py in the commit\u0027\n  Job openstack-tox-functional-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional-py36/lib/python3.6/site-packages/sqlalchemy/orm/unitofwork.py in the commit\u0027\n  Job openstack-tox-functional-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional-py36/lib/python3.6/site-packages/webob/acceptparse.py in the commit\u0027\n  Job octavia-tox-functional-py37-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional-py37/lib/python3.7/site-packages/oslo_context/context.py in the commit\u0027\n  Job octavia-tox-functional-py37-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional-py37/lib/python3.7/site-packages/oslo_db/sqlalchemy/test_base.py in the commit\u0027\n  Job octavia-tox-functional-py37-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional-py37/lib/python3.7/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job octavia-tox-functional-py37-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional-py37/lib/python3.7/site-packages/webob/acceptparse.py in the commit\u0027\n  Job octavia-tox-functional-py37-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional-py37/lib/python3.7/site-packages/webob/acceptparse.py in the commit\u0027\n  Job octavia-tox-functional-py37-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional-py37/lib/python3.7/site-packages/sqlalchemy/orm/unitofwork.py in the commit\u0027\n  Job octavia-tox-functional-py37-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional-py37/lib/python3.7/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job octavia-tox-functional-py37-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional-py37/lib/python3.7/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job octavia-tox-functional-py37-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional-py37/lib/python3.7/site-packages/wsme/api.py in the commit\u0027\n  Job octavia-tox-functional-py37-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional-py37/lib/python3.7/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job octavia-tox-functional-py37-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional-py37/lib/python3.7/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job octavia-tox-functional-py37-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional-py37/lib/python3.7/site-packages/oslo_db/sqlalchemy/enginefacade.py in the commit\u0027\n  Job octavia-tox-functional-py37-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional-py37/lib/python3.7/site-packages/octavia_lib/api/drivers/data_models.py in the commit\u0027\n  Job octavia-tox-functional-py37-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional-py37/lib/python3.7/site-packages/wsmeext/pecan.py in the commit\u0027\n  Job octavia-tox-functional-py37-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional-py37/lib/python3.7/site-packages/oslo_db/sqlalchemy/exc_filters.py in the commit\u0027\n  Job octavia-tox-functional-py37-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional-py37/lib/python3.7/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/oslo_db/sqlalchemy/exc_filters.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/debtcollector/renames.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/sqlalchemy/orm/unitofwork.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/wsmeext/pecan.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/oslo_db/sqlalchemy/enginefacade.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/oslo_db/sqlalchemy/test_base.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/octavia_lib/api/drivers/data_models.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/oslo_context/context.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/wsme/api.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/debtcollector/renames.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/cliff/app.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/sqlalchemy/orm/unitofwork.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/oslo_policy/_checks.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/babel/localtime/_unix.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/wsmeext/pecan.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/eventlet/support/greendns.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/oslo_messaging/_drivers/pool.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/octavia_lib/api/drivers/data_models.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/wsme/api.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/oslo_db/sqlalchemy/test_base.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/flask/config.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/oslo_context/context.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/unittest2/suite.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/setuptools/dist.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/babel/plural.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/oslo_db/sqlalchemy/enginefacade.py in the commit\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/docs/lib/python3.6/site-packages/sadisplay/describe.py in the commit\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/docs/lib/python3.6/site-packages/sadisplay/describe.py in the commit\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/docs/lib/python3.6/site-packages/wsme/api.py in the commit\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/docs/lib/python3.6/site-packages/sadisplay/describe.py in the commit\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/pdf-docs/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/pdf-docs/lib/python3.6/site-packages/wsme/api.py in the commit\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/docs/lib/python3.6/site-packages/wsmeext/pecan.py in the commit\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/pdf-docs/lib/python3.6/site-packages/sadisplay/describe.py in the commit\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/docs/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/pdf-docs/lib/python3.6/site-packages/sadisplay/describe.py in the commit\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/pdf-docs/lib/python3.6/site-packages/wsmeext/pecan.py in the commit\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/pdf-docs/lib/python3.6/site-packages/sadisplay/describe.py in the commit\u0027\n  Comments left for invalid file .tox/pep8/lib/python3.6/site-packages/astroid/interpreter/_import/spec.py\n  Comments left for invalid file .tox/functional-py36/lib/python3.6/site-packages/octavia_lib/api/drivers/data_models.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/unittest2/compatibility.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/oslo_db/sqlalchemy/enginefacade.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_load_balancer.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_driver_lib.py\n  Comments left for invalid file .tox/functional-py37/lib/python3.7/site-packages/oslo_db/sqlalchemy/test_base.py\n  Comments left for invalid file .tox/cover/lib/python3.6/site-packages/oslo_db/sqlalchemy/test_base.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/unittest2/suite.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/cover/lib/python3.6/site-packages/oslo_db/sqlalchemy/enginefacade.py\n  Comments left for invalid file .tox/cover/lib/python3.6/site-packages/debtcollector/renames.py\n  Comments left for invalid file .tox/functional-py37/lib/python3.7/site-packages/rfc3986/_mixin.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/dns/hash.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/oslo_context/context.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/babel/plural.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/oslo_db/sqlalchemy/enginefacade.py\n  Comments left for invalid file .tox/functional-py37/lib/python3.7/site-packages/octavia_lib/api/drivers/data_models.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/debtcollector/renames.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/pdf-docs/lib/python3.6/site-packages/wsmeext/pecan.py\n  Comments left for invalid file .tox/functional-py36/lib/python3.6/site-packages/oslo_db/sqlalchemy/test_base.py\n  Comments left for invalid file .tox/functional-py36/lib/python3.6/site-packages/oslo_db/sqlalchemy/exc_filters.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/oslo_db/sqlalchemy/enginefacade.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/rfc3986/_mixin.py\n  Comments left for invalid file .tox/pep8/lib/python3.6/site-packages/bashate/bashate.py\n  Comments left for invalid file .tox/functional-py36/lib/python3.6/site-packages/oslo_db/sqlalchemy/enginefacade.py\n  Comments left for invalid file .tox/functional-py36/lib/python3.6/site-packages/oslo_context/context.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/oslo_db/sqlalchemy/enginefacade.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/babel/localtime/_unix.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file .tox/cover/lib/python3.6/site-packages/sqlalchemy/orm/unitofwork.py\n  Comments left for invalid file .tox/cover/lib/python3.6/site-packages/rfc3986/_mixin.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/debtcollector/renames.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/debtcollector/renames.py\n  Comments left for invalid file .tox/cover/lib/python3.6/site-packages/wsme/api.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/oslo_context/context.py\n  Comments left for invalid file .tox/functional-py37/lib/python3.7/site-packages/webob/acceptparse.py\n  Comments left for invalid file octavia/amphorae/backends/utils/keepalivedlvs_query.py\n  Comments left for invalid file .tox/pdf-docs/lib/python3.6/site-packages/sadisplay/describe.py\n  Comments left for invalid file .tox/cover/lib/python3.6/site-packages/webob/acceptparse.py\n  Comments left for invalid file .tox/cover/lib/python3.6/site-packages/oslo_db/sqlalchemy/exc_filters.py\n  Comments left for invalid file .tox/functional-py36/lib/python3.6/site-packages/sqlalchemy/orm/unitofwork.py\n  Comments left for invalid file .tox/docs/lib/python3.6/site-packages/eventlet/patcher.py\n  Comments left for invalid file octavia/common/context.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/oslo_context/context.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/webob/acceptparse.py\n  Comments left for invalid file .tox/cover/lib/python3.6/site-packages/oslo_context/context.py\n  Comments left for invalid file .tox/functional-py37/lib/python3.7/site-packages/sqlalchemy/orm/unitofwork.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/sqlalchemy/orm/unitofwork.py\n  Comments left for invalid file .tox/cover/lib/python3.6/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/wsme/api.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_base.py\n  Comments left for invalid file .tox/cover/lib/python3.6/site-packages/octavia_lib/api/drivers/data_models.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/taskflow/atom.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/taskflow/atom.py\n  Comments left for invalid file .tox/functional-py37/lib/python3.7/site-packages/oslo_context/context.py\n  Comments left for invalid file .tox/docs/lib/python3.6/site-packages/wsmeext/pecan.py\n  Comments left for invalid file .tox/pdf-docs/lib/python3.6/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/unittest2/compatibility.py\n  Comments left for invalid file .tox/cover/lib/python3.6/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file .tox/docs/lib/python3.6/site-packages/wsme/api.py\n  Comments left for invalid file .tox/docs/lib/python3.6/site-packages/sadisplay/describe.py\n  Comments left for invalid file .tox/functional-py36/lib/python3.6/site-packages/rfc3986/_mixin.py\n  Comments left for invalid file .tox/functional-py37/lib/python3.7/site-packages/oslo_db/sqlalchemy/exc_filters.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/flask/config.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/wsmeext/pecan.py\n  Comments left for invalid file .tox/cover/lib/python3.6/site-packages/wsmeext/pecan.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/cliff/app.py\n  Comments left for invalid file .tox/functional-py37/lib/python3.7/site-packages/oslo_db/sqlalchemy/enginefacade.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/debtcollector/renames.py\n  Comments left for invalid file .tox/functional-py37/lib/python3.7/site-packages/wsme/api.py\n  Comments left for invalid file .tox/pdf-docs/lib/python3.6/site-packages/wsme/api.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/oslo_messaging/_drivers/pool.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/oslo_db/sqlalchemy/test_base.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/eventlet/support/greendns.py\n  Comments left for invalid file .tox/functional-py37/lib/python3.7/site-packages/wsmeext/pecan.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/rfc3986/_mixin.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/rfc3986/_mixin.py\n  Comments left for invalid file .tox/functional-py36/lib/python3.6/site-packages/wsme/api.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/oslo_context/context.py\n  Comments left for invalid file .tox/functional-py36/lib/python3.6/site-packages/wsmeext/pecan.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_noop_driver.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/oslo_policy/_checks.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/eventlet/patcher.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_provider.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/setuptools/dist.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/octavia_lib/api/drivers/data_models.py\n  Comments left for invalid file .tox/functional-py36/lib/python3.6/site-packages/webob/acceptparse.py","accounts_in_message":[],"_revision_number":4},{"id":"0f27c04cf310549d26a38bf45d1dcac275a5b7ce","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-07 16:57:58.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"b6928f3e5ae953fadeb193ce6fd57bb1a7985f53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 19:16:58.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/edeaa0485b4e4395a97777b6475160e9 : SUCCESS in 14m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f95f0708748d49d18172ce87f139a485 : SUCCESS in 12m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d07b7cf355b542748b1f640a66f1a812 : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6db3cc661d8549fbbd7a15cf8f3f1031 : SUCCESS in 7m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/11d0eea8ddc846d1b28e2ee05f864b5c : SUCCESS in 8m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51155a603f16486a9f9f64d1f3feabb3 : FAILURE in 8m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31a6868f1eee4c2e945320e32f3f59cf : SUCCESS in 14m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/655d1d7b4fc84f1db664d1bc6252b826 : SUCCESS in 8m 16s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fe4eed00f7d0482bb4e6dbba50416a35 : SUCCESS in 7m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e329d85bc70142238d386e2ac78eea64 : SUCCESS in 10m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/93db2f5f998148a69e1711b6b02cb37d : SUCCESS in 9m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/eb32095a1d364aeb8f103f037b59c96f : SUCCESS in 51m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0f3c5e64ef7e4b7f8c9529d9bb4ecc85 : SUCCESS in 1h 57m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0b543cb1f8e44004bf639552e1c250b8 : SUCCESS in 58m 53s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9f9d9fca00654153a35c75301f945a75 : SUCCESS in 50m 12s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9b5dcebafafb433096b31a7bfedb114d : SUCCESS in 1h 06m 26s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e14e750a9fe741eca69c04b17d906b92 : SUCCESS in 1h 53m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/13a65bbde6b94eb8ab0d2cbb1ef4bf5d : SUCCESS in 2h 00m 12s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8f79e7f9290d48fd9891ddaba99c0d69 : FAILURE in 1h 07m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c37e2a88bd77454188aa95f1f804ea06 : SUCCESS in 2h 06m 30s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/tests/functional/api/v2/test_provider.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_base.py\n  Comments left for invalid file octavia/common/context.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_load_balancer.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_noop_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_driver_lib.py\n  Comments left for invalid file octavia/amphorae/backends/utils/keepalivedlvs_query.py","accounts_in_message":[],"_revision_number":5},{"id":"60b10b18301bd843714ccee6548413da2060d696","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-06-04 09:25:36.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"ff6bb32b2fd8e1487dc1a32d3620f06d9cb892c6","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-08 06:13:37.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"d8b8b9bbc1bf1a2c445f1375b90e8c3df180a9dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 08:52:57.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b3253fc7a7a477b9950e5d01eabefea : SUCCESS in 11m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b451da8c3278418ab42c75ff5e2d831f : SUCCESS in 13m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1270e396102d47f7a4a2067d2058a951 : SUCCESS in 8m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85cc8f6b69f244a7a1b5c844dce6db98 : SUCCESS in 8m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4232960b746f46509cc1c0e5dc3d7c32 : SUCCESS in 7m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e14131035d0940f1bd08349820c0d518 : SUCCESS in 12m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3c8c2360a0f43a5bd7df3473efaf7ff : SUCCESS in 7m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6cb445e2d350483ab6ecce3dfb276fbf : SUCCESS in 10m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/936fc858a61d4584ae786042431e2fb3 : SUCCESS in 10m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8b9a21be09b446de9c747869d929d7bd : SUCCESS in 8m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e2c3c60ab91f4cb7a04d94b9dbbb31bb : SUCCESS in 45m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b55d1925fa1247abb4abdba30ed70acc : SUCCESS in 1h 57m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a09cbca3401445879ed93bdc81a9e651 : SUCCESS in 56m 48s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6b1a8e487ae5483eb074cddf5ed0c878 : SUCCESS in 59m 13s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/221a109158cb4722802e09c78cf26870 : SUCCESS in 1h 07m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7c0000162ebc4eac9188014de5f660d0 : SUCCESS in 1h 47m 33s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4fb030ff37ce4ce2b91b06abf943c663 : SUCCESS in 1h 57m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f62accbaecb94b8eb5bbc2bfd2248cdf : SUCCESS in 1h 26m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/252f5b3adc7640e38fb413b5e3032ba6 : SUCCESS in 1h 01m 01s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"71ce53f00d0be516eff8f5fb3dd4bdbf940acf1b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-16 12:07:26.000000000","message":"Patch Set 6: Code-Review+2\n\n(3 comments)\n\nLooks fine to me. I want to be surprised we had no tests that actually checked SNI stuff, but I really am not surprised at all.","accounts_in_message":[],"_revision_number":6},{"id":"0b029f473a771d14b5ca294ff576992584a023e9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-16 18:09:01.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(2 comments)\n\nI\u0027m ok with this.","accounts_in_message":[],"_revision_number":6},{"id":"4163133ad30a00d3347032671ddd07eb5106d06e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 18:09:14.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"88e90dad87e9ea4be4c9c80dcaf6b6069f0ff81c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 19:56:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"bd55021bda7ac736b1d58287345e173a49a10b52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 19:56:35.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb14d9119c85446c8651f58e513e7085 : SUCCESS in 13m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bc2820e65eb47f0b10c6ea46afb0d15 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89f0a904c0ca4e4a823a1187bb42bb42 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f15c663868ba478197fad3bda124ea31 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/464ea5f2c0484074bb37a64ad8b09116 : SUCCESS in 12m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/346846ef72904a18ba665ce4c0929ae0 : SUCCESS in 5m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/809852a4070a4c97b9a0ee66da3d5fe4 : SUCCESS in 7m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b272cc8ca255433fbfbd410d19469b77 : SUCCESS in 45m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9b79a91f5f4d407db4fb3b31fe52ebda : SUCCESS in 1h 13m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3a9cfeec1cfa4a12aac60dd0ba88833d : SUCCESS in 41m 42s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c9bb4cc285e94921b35849a3211f5c5e : SUCCESS in 37m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1bb9485345014098a26a1094b896beae : SUCCESS in 40m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fee9854be62547b883199f80a2e7b695 : SUCCESS in 1h 41m 28s","accounts_in_message":[],"_revision_number":6},{"id":"05dafa40fe65f7a4c0860384e283b08aa8987f1b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 19:59:44.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/255eea26afa5406ca43ac38dc4ebbfaf : SUCCESS in 2m 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29c3e7dfa49b41af8137ea24536187a1 : SUCCESS in 2m 18s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"9e070e6e6df6468a0b8f792cb089de0d9fc180b5","revisions":{"8b231d1fb92856a4f4b4d3800cea961acba67276":{"kind":"REWORK","_number":1,"created":"2020-03-12 20:33:52.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/90/712790/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/712790/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/712790/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/712790/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/712790/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/712790/1"}}},"commit":{"parents":[{"commit":"801b99ed0069f438646c106389bb9b84f37fb619","subject":"Update the available provider drivers list.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/801b99ed0069f438646c106389bb9b84f37fb619"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-03-12 20:30:31.000000000","tz":0},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-03-12 20:33:06.000000000","tz":0},"subject":"WIP: Fix SNI certificates on listener update","message":"WIP: Fix SNI certificates on listener update\n\nTask: 39042\nStory: 2007421\n\nChange-Id: If5b6411a0b7c75441a406234c2792ea68d35d0fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b231d1fb92856a4f4b4d3800cea961acba67276"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b231d1fb92856a4f4b4d3800cea961acba67276"}]},"branch":"refs/heads/master"},"5e8a7d6303469606505bd0b5e8f420c1fde9e865":{"kind":"REWORK","_number":2,"created":"2020-03-13 15:27:10.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/90/712790/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/712790/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/712790/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/712790/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/712790/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/712790/2"}}},"commit":{"parents":[{"commit":"6a535c979bdef181ee0acc6a3e7e507564cdf9fb","subject":"Merge \"Support hacking 2.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6a535c979bdef181ee0acc6a3e7e507564cdf9fb"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-03-12 20:30:31.000000000","tz":0},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-03-13 15:24:03.000000000","tz":0},"subject":"Fix listener update with SNI certificates","message":"Fix listener update with SNI certificates\n\nSNI certificates were not being set in the database on listener update. A\nlistener GET would not show the certificates in the sni_container_refs\nattribute. Also, the API was allowing set of SNI certs on non TERMINATED_HTTPS\nlisteners.\n\nTask: 39042\nStory: 2007421\nStory: 2007430\n\nChange-Id: If5b6411a0b7c75441a406234c2792ea68d35d0fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5e8a7d6303469606505bd0b5e8f420c1fde9e865"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5e8a7d6303469606505bd0b5e8f420c1fde9e865"}]},"branch":"refs/heads/master"},"cc213716740296665e0ef541a975f210c51940e0":{"kind":"REWORK","_number":3,"created":"2020-03-13 23:10:56.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/90/712790/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/712790/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/712790/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/712790/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/712790/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/712790/3"}}},"commit":{"parents":[{"commit":"6a535c979bdef181ee0acc6a3e7e507564cdf9fb","subject":"Merge \"Support hacking 2.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6a535c979bdef181ee0acc6a3e7e507564cdf9fb"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-03-12 20:30:31.000000000","tz":0},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-03-13 22:45:02.000000000","tz":0},"subject":"Fix listener update with SNI certificates","message":"Fix listener update with SNI certificates\n\nSNI certificates were not being set in the database on listener update.\nA listener GET would not show the certificates in the sni_container_refs\nattribute. Also, the API was allowing set of SNI certs on non\nTERMINATED_HTTPS listeners.\n\nTask: 39042\nStory: 2007421\nStory: 2007430\n\nChange-Id: If5b6411a0b7c75441a406234c2792ea68d35d0fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cc213716740296665e0ef541a975f210c51940e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cc213716740296665e0ef541a975f210c51940e0"}]},"branch":"refs/heads/master"},"e005336ce5ac11dd5f81da6b740a7fcc7c9c13ab":{"kind":"REWORK","_number":4,"created":"2020-03-31 13:28:58.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/90/712790/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/712790/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/712790/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/712790/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/712790/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/712790/4"}}},"commit":{"parents":[{"commit":"cd4a52f50d5965d285d7de3f528dbbcd17932777","subject":"Merge \"Ussuri contributor docs community goal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cd4a52f50d5965d285d7de3f528dbbcd17932777"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-03-12 20:30:31.000000000","tz":0},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-03-31 13:28:41.000000000","tz":120},"subject":"Fix listener update with SNI certificates","message":"Fix listener update with SNI certificates\n\nSNI certificates were not being set in the database on listener update.\nA listener GET would not show the certificates in the sni_container_refs\nattribute. Also, the API was allowing set of SNI certs on non\nTERMINATED_HTTPS listeners.\n\nTask: 39042\nStory: 2007421\nStory: 2007430\n\nChange-Id: If5b6411a0b7c75441a406234c2792ea68d35d0fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e005336ce5ac11dd5f81da6b740a7fcc7c9c13ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e005336ce5ac11dd5f81da6b740a7fcc7c9c13ab"}]},"branch":"refs/heads/master"},"231b2982f9d331352172ac66d90d4567af9698fe":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-04-07 16:57:58.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/90/712790/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/712790/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/712790/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/712790/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/712790/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/712790/5"}}},"commit":{"parents":[{"commit":"8c6d64baa7c35f6c29ede2706f178dedb540bf37","subject":"Merge \"Add ability to set TLS cipher list for listeners\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c6d64baa7c35f6c29ede2706f178dedb540bf37"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-03-12 20:30:31.000000000","tz":0},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-04-07 16:57:58.000000000","tz":0},"subject":"Fix listener update with SNI certificates","message":"Fix listener update with SNI certificates\n\nSNI certificates were not being set in the database on listener update.\nA listener GET would not show the certificates in the sni_container_refs\nattribute. Also, the API was allowing set of SNI certs on non\nTERMINATED_HTTPS listeners.\n\nTask: 39042\nStory: 2007421\nStory: 2007430\n\nChange-Id: If5b6411a0b7c75441a406234c2792ea68d35d0fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/231b2982f9d331352172ac66d90d4567af9698fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/231b2982f9d331352172ac66d90d4567af9698fe"}]},"branch":"refs/heads/master"},"9e070e6e6df6468a0b8f792cb089de0d9fc180b5":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-06-08 06:13:37.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/90/712790/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/712790/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/712790/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/712790/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/712790/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/712790/6"}}},"commit":{"parents":[{"commit":"409b89f1417632433f7a5e55549343d5bcde96c3","subject":"Merge \"Update the feature matrix for new features\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/409b89f1417632433f7a5e55549343d5bcde96c3"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-03-12 20:30:31.000000000","tz":0},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-08 06:13:37.000000000","tz":0},"subject":"Fix listener update with SNI certificates","message":"Fix listener update with SNI certificates\n\nSNI certificates were not being set in the database on listener update.\nA listener GET would not show the certificates in the sni_container_refs\nattribute. Also, the API was allowing set of SNI certs on non\nTERMINATED_HTTPS listeners.\n\nTask: 39042\nStory: 2007421\nStory: 2007430\n\nChange-Id: If5b6411a0b7c75441a406234c2792ea68d35d0fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9e070e6e6df6468a0b8f792cb089de0d9fc180b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9e070e6e6df6468a0b8f792cb089de0d9fc180b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
