)]}'
{"id":"openstack%2Foctavia~714043","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I48315f8f293811e1d99584ea36da05c4211cf275","project":"openstack/octavia","branch":"stable/train","topic":"695589","hashtags":[],"change_id":"I48315f8f293811e1d99584ea36da05c4211cf275","subject":"Fix load balancer update with provider filtered params","status":"MERGED","created":"2020-03-20 08:09:42.000000000","updated":"2020-04-24 01:37:59.000000000","submitted":"2020-04-24 01:36:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714043-1587692189454-c713193e","meta_rev_id":"59b107ed430b95240e4272a348a36b3afd6812d8","_number":714043,"virtual_id_number":714043,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-24 01:36:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-03-20 19:53:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-04-23 23:31:22.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2020-04-23 23:31:22.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-20 19:53:17.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-04-23 23:31:22.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"},{"updated":"2020-04-24 01:36:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"115c141e1b5ec0b48244fd7674814ad9ba10b811","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-20 08:09:42.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"fcb8af5a5c6b8bb26fee3dc800d20dbdc6db3227","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-20 08:10:40.000000000","message":"Topic changed from 695589-stable/train to 695589","accounts_in_message":[],"_revision_number":1},{"id":"66083ef0e75e8d5e56647ef291c22991a9acc5de","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-20 09:20:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"288284ae1773a0d5f2864ae5ea8e32819d8ba5ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 11:58:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6f6e9bacb544ca780b778f5d7dc20f9 : SUCCESS in 8m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d86293ebbaf43c5972c1cfd04993bac : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3169ff89302454d9e34c7ddbb1606de : SUCCESS in 5m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1262f67564f64e7596792b112a97150f : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/002f373016a847d9aa537c3d2589f8ca : SUCCESS in 5m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/06c0ac21336d4117ae9b8c8805347829 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/941744275e244e068f3c9be5145b2302 : SUCCESS in 11m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afca1e68c4fb4782bd975dd17c36256e : SUCCESS in 6m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b41778826f8044f296c4f79b7d3b466f : SUCCESS in 6m 36s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ee9333a293484a2993048a36e174442a : SUCCESS in 7m 57s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c3759356724b4aa496c3a3eaf889a0dd : SUCCESS in 7m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1107d51fc7d94612b25320f680f22ee5 : SUCCESS in 6m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/902434bbdf5a4565aefd18cbe47e9876 : SUCCESS in 42m 47s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/da674dbe5ae54393872c1a2ae5cb1db6 : SUCCESS in 43m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d2dd4d29f4d0442f80700bdbdc43746c : SUCCESS in 1h 59m 16s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a367051da140493ea3bf3cb4f2acd5a6 : SUCCESS in 1h 53m 58s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/29a6962fb9c446f6ab07706da3b474fb : POST_FAILURE in 2h 32m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d7d9bf16c22849fbada12a18284ba143 : SUCCESS in 2h 15m 16s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/487787252e4f41e083bd32a9142f7e04 : SUCCESS in 1h 54m 05s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/41c8356e63e049dfaeab01039677abbc : SUCCESS in 1h 21m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/361f0f2511994cbf8c1c2b1145b13fae : SUCCESS in 1h 02m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b60659df333241fb9391e59a88766307 : SUCCESS in 1h 53m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5a8fb44929454726b33984273df773d1 : SUCCESS in 1h 14m 51s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1c5242afe3634dc5a314c6cc4e839a25 : SUCCESS in 1h 07m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f145b280a82f45d4ab00890d9bb94c82 : POST_FAILURE in 1h 35m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8137bbfbf2d4299e9c2ff731b0875de8f544a82c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-03-20 19:53:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a1c72d88c7a2fcaf9f42eb3774b83f9a13ad3527","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2020-04-23 23:31:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a5632b40b3e686588f29c5fd72a728194e61daf5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 23:33:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d33e9e259b44c76cd68d4a1715f25e7829f10980","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 01:36:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82153fbb5c894962a06e07c56a09f559 : SUCCESS in 9m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afe523332aa248d19c34574916f596de : SUCCESS in 6m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/47673e1f64404ab4a81d1efbea532541 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/108d1db5bf75430294e608d838667bd2 : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/864de2875d604b60a6a78ecbd39c0a7d : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71dca76ace624e9c8b522a441e5dc124 : SUCCESS in 10m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93f15999dc4e4e458bc94af19847788b : SUCCESS in 5m 21s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/549601ebfe084f97b44214d88c528728 : SUCCESS in 7m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b38ffea535134e4cbb5eae93aa72e71a : SUCCESS in 6m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/48e2573e7127456bb21089e1ac21986f : SUCCESS in 48m 09s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/55d4e30bf397402ea0acf1240dc1785c : SUCCESS in 44m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ea82db9b0c4c4a47adbbb66de7af3eed : SUCCESS in 1h 48m 52s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/9a6a482cb881455ba367ddaf81a38a0a : SUCCESS in 1h 57m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cc7fc130f37440158f6062b379417112 : SUCCESS in 1h 14m 48s","accounts_in_message":[],"_revision_number":2},{"id":"de040aa909161a114b1352f1f2e254a66861b173","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 01:36:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"59b107ed430b95240e4272a348a36b3afd6812d8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 01:37:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72647f19fd674caabd026bb6e472d3f7 : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0304d4e984b740f680e2336afd9fcada : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"535dc7c69885d661fed2a59ac7d4b8b3ce89e064","revisions":{"a491285ce4413aff868a6a17151e3c22227e9c2b":{"kind":"REWORK","_number":1,"created":"2020-03-20 08:09:42.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/43/714043/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/43/714043/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/714043/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/714043/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/714043/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/43/714043/1"}}},"commit":{"parents":[{"commit":"0c1cb417e904daa75deedaabc72a20d07b138eb9","subject":"Fix uncaught DB exception when trying to get a spare amphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c1cb417e904daa75deedaabc72a20d07b138eb9"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-11-21 21:31:26.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-03-20 08:09:42.000000000","tz":0},"subject":"Fix load balancer update with provider filtered params","message":"Fix load balancer update with provider filtered params\n\nWhen only setting tags to an existing load balancer, the amphora driver\nwould try to apply QoS policy on VRRP ports even if no policy is\ndefined. This raised an NetworkException that led load balancers to go\ninto ERROR.\n\nTask: 37589\nStory: 2006922\n\nChange-Id: I48315f8f293811e1d99584ea36da05c4211cf275\n(cherry picked from commit 2cfcd71a8659583d91cda3e488c60ded078c9677)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a491285ce4413aff868a6a17151e3c22227e9c2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a491285ce4413aff868a6a17151e3c22227e9c2b"}]},"branch":"refs/heads/stable/train"},"535dc7c69885d661fed2a59ac7d4b8b3ce89e064":{"kind":"REWORK","_number":2,"created":"2020-03-20 09:20:34.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/43/714043/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/43/714043/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/714043/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/714043/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/714043/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/43/714043/2"}}},"commit":{"parents":[{"commit":"0c1cb417e904daa75deedaabc72a20d07b138eb9","subject":"Fix uncaught DB exception when trying to get a spare amphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c1cb417e904daa75deedaabc72a20d07b138eb9"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-11-21 21:31:26.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-03-20 09:18:28.000000000","tz":60},"subject":"Fix load balancer update with provider filtered params","message":"Fix load balancer update with provider filtered params\n\nWhen only setting tags to an existing load balancer, the amphora driver\nwould try to apply QoS policy on VRRP ports even if no policy is\ndefined. This raised an NetworkException that led load balancers to go\ninto ERROR.\n\nTask: 37589\nStory: 2006922\n\nChange-Id: I48315f8f293811e1d99584ea36da05c4211cf275\n(cherry picked from commit 2cfcd71a8659583d91cda3e488c60ded078c9677)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/535dc7c69885d661fed2a59ac7d4b8b3ce89e064"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/535dc7c69885d661fed2a59ac7d4b8b3ce89e064"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
