)]}'
{"id":"openstack%2Foctavia~714240","triplet_id":"openstack%2Foctavia~master~Ia9fec4e72c00f7086489b245d9dc50ed9c27f12a","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ia9fec4e72c00f7086489b245d9dc50ed9c27f12a","subject":"Fix padding logic for UDP health daemon","status":"MERGED","created":"2020-03-20 23:57:16.000000000","updated":"2020-04-27 11:51:36.000000000","submitted":"2020-03-23 03:16:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714240-1584933416491-5004bf0a","meta_rev_id":"0ec2ae2cadba779d603902b58f07b3ea0105023a","_number":714240,"virtual_id_number":714240,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-23 03:16:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":2,"date":"2020-03-21 03:16:28.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":2,"date":"2020-03-23 01:24:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":1,"date":"2020-03-23 01:24:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":2,"date":"2020-03-23 01:24:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-21 03:16:28.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"},{"updated":"2020-03-23 01:24:03.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-03-23 03:16:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eb10a38587b847ba526c6488f6b86a287172d626","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-03-20 23:57:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e628b159029c2376ba61c2e9db50cdd82e4400a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 02:01:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae952563837545a68ec044e7d2b1177b : SUCCESS in 9m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/533230ba07ad4f5f86e3da8aeeb9d915 : SUCCESS in 9m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6ea377bed5a49dcab2473b1d164c601 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4d0694677a74171ba6da47072b86a62 : SUCCESS in 6m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4dbf645652ed4fb58a6a4a83fe4b7158 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a841770b1bf542c6aa7d6dea2030bd25 : FAILURE in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93685b04d8d24ba49a586311f23332de : SUCCESS in 14m 13s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f8cddb7ad749490aa516d02c851cf709 : SUCCESS in 5m 43s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c0aaaee393a8401b8fbdce52bbf1b8d4 : SUCCESS in 8m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9f6396b0a9e746af956b9185c53d6238 : SUCCESS in 10m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7bc682f6977641a18b6699811ad4a5d4 : SUCCESS in 44m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a2016bcc62724d8684d769997a64a466 : SUCCESS in 1h 50m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/21b468b7529b454e916210afbba9a10d : SUCCESS in 59m 21s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e344cffc369c4dbfa9b572225b202022 : SUCCESS in 55m 14s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8cb199c5025248a092796a86102fad60 : SUCCESS in 1h 04m 47s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8c8a374d1ff14a03a2ff2e7d2eba0ea9 : SUCCESS in 1h 56m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4f17e15b819346ac847ab6847a8d54e3 : SUCCESS in 1h 56m 58s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c0d4c7f3ba904615966e72a2e3f7fbb3 : FAILURE in 1h 11m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/43a7bf1c02d5444ca5eaf9771bac4866 : SUCCESS in 1h 42m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d5a4848852541513c88b2b21e975ac584b36dc8b","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2020-03-21 03:16:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"aa3c04c4dc05a62c6e1c81664bc238eda418ca5c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-03-23 01:24:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"0bcbbd32755bcb43678894c5e690c6f8fc32de27","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-03-23 01:24:08.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"323a96b7dfde141789d2fec01b8b19e7a7ee4f6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 01:24:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9c7f6e7e9c06cec94126f610b94f94e153fd265b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 03:16:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f93a9289be95a0d043c16fb743b6a94b34721d9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 03:16:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/feecccbf2cd34d56a669863bb0c5a3b3 : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbf06e20bcf64e97ac58edcaf5dcfed0 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0631d71644f40daaa0540d6356535fa : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d42b4bfe73e4bb2b1b90479cdc00441 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/509f5cb9152e4389ab04f39466f4d0f9 : SUCCESS in 11m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2337a73eb3ff4a3bb767a40b512d27f2 : SUCCESS in 6m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8ef05d6c0cff4b3080da14efbaa737b6 : SUCCESS in 46m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a7f774dbe3904467bd3c8511ab58bcce : SUCCESS in 1h 49m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/15219162d5b84563b7728a7d339e88aa : SUCCESS in 56m 06s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d6be72f2dc044ab0b948028d8426eb13 : SUCCESS in 55m 07s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/38b45c27630c467295af990968f40bd7 : SUCCESS in 1h 04m 13s","accounts_in_message":[],"_revision_number":1},{"id":"aeef1e7505579ceaf6970eb168657a3b6ba973ec","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 03:18:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05ca59d20d3249cab7f5c8388d8464c1 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1},{"id":"0ec2ae2cadba779d603902b58f07b3ea0105023a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-27 11:51:36.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 49d58bb81e097ef87f5ae525620695242ee9f312","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d27ee3f0ee98934e6ab354704edc47f61fffdcda","revisions":{"d27ee3f0ee98934e6ab354704edc47f61fffdcda":{"kind":"REWORK","_number":1,"created":"2020-03-20 23:57:16.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/40/714240/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/40/714240/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/714240/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/714240/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/714240/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/40/714240/1"}}},"commit":{"parents":[{"commit":"d2b432b07e1cccfc38deac9be681411921416071","subject":"Merge \"Fix load balancer update with provider filtered params\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d2b432b07e1cccfc38deac9be681411921416071"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-03-20 23:51:44.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-03-20 23:51:44.000000000","tz":-420},"subject":"Fix padding logic for UDP health daemon","message":"Fix padding logic for UDP health daemon\n\nShould have done \"pad to 8 characters\" on the hex conversion, but it was\ninstead hardcoded to pad a single `0`, which is right in a lot of cases\nbut not all.\n\nFor example:\n\u003e\u003e\u003e ip1 \u003d ipaddress.ip_address(\u002798.136.140.23\u0027)\n\u003e\u003e\u003e ip2 \u003d ipaddress.ip_address(\u002710.1.1.1\u0027)\n\u003e\u003e\u003e \"%X\" % ip1._ip\n\u002762888C17\u0027\n\u003e\u003e\u003e \"%X\" % ip2._ip\n\u0027A010101\u0027\n\nChange-Id: Ia9fec4e72c00f7086489b245d9dc50ed9c27f12a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d27ee3f0ee98934e6ab354704edc47f61fffdcda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d27ee3f0ee98934e6ab354704edc47f61fffdcda"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
