)]}'
{"id":"openstack%2Foctavia~714854","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I2f680a00be695c3828166e6803882765af79b79e","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"I2f680a00be695c3828166e6803882765af79b79e","subject":"Pick stale amphora randomly","status":"MERGED","created":"2020-03-25 08:06:55.000000000","updated":"2020-03-31 13:21:00.000000000","submitted":"2020-03-31 13:18:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714854-1585660711371-65716cc9","meta_rev_id":"1b8b2d58819e4ec7887c85498731ffbb2f37dbb6","_number":714854,"virtual_id_number":714854,"owner":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-03-25 11:10:57.000000000","_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-31 13:18:31.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-03-31 11:18:43.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-03-27 16:38:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2020-03-31 03:03:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2020-03-31 11:18:43.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-27 16:38:53.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-03-31 03:03:35.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-03-31 11:18:43.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-03-31 13:18:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"255212bf614fcf542629bd7126e9ee515211fefb","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-03-25 08:06:55.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"bdd4c86c0323c5a02ac26086faef807b2519b003","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 11:08:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76befb1a13d04ea7b42edce96548c72d : SUCCESS in 9m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5f341aebbfd4d41874afa2bcff03b27 : SUCCESS in 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1abfa06e3a9f4368b188f0adcc94bb94 : SUCCESS in 6m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/27e8f2c7cf7a4929944200cb65bc864b : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0157aaccb8d248c19d8092080b9711b3 : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aca363577ba24539ba6f63d25f0104a7 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a717d3f74cc422997a0bed9fa7c54e8 : SUCCESS in 11m 46s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/10cd3a9dd0204e8987d9206976bc8a3a : SUCCESS in 6m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b4dac464706a4092ab410bb844051534 : SUCCESS in 9m 54s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/91e88b26ccd14b9eaf3fe2c3c9dedd15 : SUCCESS in 7m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/452d43e46f9a455ba79a31ed351397f5 : SUCCESS in 12m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/073fd1d0ac7041ba8f50a1a849b4c409 : SUCCESS in 49m 03s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/757be60f5874436190725fcf0cdecaed : SUCCESS in 50m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aeae16bd5ef94bf1bbeed52f636527fe : FAILURE in 2h 36m 36s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/76e301800ba64961a8156d77ff68f12e : FAILURE in 2h 19m 31s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f767d074bcb84be59297c1d42bb01d7b : POST_FAILURE in 2h 25m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ff6fb09a72fe4389a13aea7b78e941e9 : SUCCESS in 2h 00m 44s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/44e66d37edc845a28ff6315d9be86f1f : FAILURE in 1h 38m 56s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c30a261579c34bf4b0db7b799870a412 : SUCCESS in 1h 14m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5e9a415fe48e4c98a83ed39322ea394b : SUCCESS in 1h 12m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f89b77e8f5a64424a1cf46e703c5db9f : SUCCESS in 1h 46m 41s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2664e1430207459b9e4d374c1bf8dc2a : SUCCESS in 1h 04m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c5bf994382c44d40ae9b1c8337d4e7e0 : FAILURE in 1h 30m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/19a6b59eb5e84c0598f676a87a775ec4 : POST_FAILURE in 1h 16m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1ee8735b48bd18a654ae3faddaf6412f89ceb2e0","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-03-25 11:10:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b11e57003868cc97332cc548c91a53864bb568fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 13:41:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8967054b6cf346048a3025777ff2f00f : SUCCESS in 11m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/243a8ebac0f14dd1aff131f71ac467f5 : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94b9b85e7856432588df7827fd5bb9e8 : SUCCESS in 6m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0b80a63d7fd54b3e9bf337f06a3cdfd5 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f65907eda28d4de4a3d305f7f1cb0628 : SUCCESS in 5m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0471ab824502489d80235f6cab884779 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc8f367c872849c88d45dd04e6b03c34 : SUCCESS in 11m 28s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4d1a91c241ea425f9dd97fbdfb4cb9bd : SUCCESS in 6m 53s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d3bce7c51a8349a3aafe570626cb2288 : SUCCESS in 7m 49s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/27123ccca5314ca3a96c8bb5d9830a0c : SUCCESS in 7m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f43a932cdf4f445aa9cc67095be99e9a : SUCCESS in 6m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3878ce25ebd14c95851fb23929f8cef4 : SUCCESS in 44m 28s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/9fdbfaafd35d4d7a8fc27e4eb33194d0 : SUCCESS in 46m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/65b294edf58d42cdb0fd07c6017a6c33 : SUCCESS in 1h 44m 52s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/94b6983343dc41fcbc2b934a3f4086bb : SUCCESS in 1h 55m 27s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/b8284c0134db481183a03fc273701a12 : POST_FAILURE in 2h 25m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7025f5d353144d33aa29256289202189 : SUCCESS in 2h 07m 09s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/6d268a957a5a4730b0cba9e75780bc9f : SUCCESS in 2h 16m 05s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5eeaea88d7584d0db08cb1a8f9abbe57 : SUCCESS in 1h 23m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1fbde3ca166e4d938b1cdfff7e0c92ba : SUCCESS in 1h 00m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9add65ceea5348349c4c8c408f057998 : SUCCESS in 1h 54m 51s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/93444d9344fd4f3e9adaf91adfe4ca90 : SUCCESS in 1h 11m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/69e590cb66e44304a95a64c11ed04f95 : SUCCESS in 1h 05m 05s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/ed0ec301480f4abdb4385a6807268540 : POST_FAILURE in 1h 17m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6435aa554b404e91e7ed161e3ea301ad5c1b5809","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-03-27 16:38:53.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"c10e825c6e69c30799aed7e362ad96342d785246","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-03-31 03:03:35.000000000","message":"Patch Set 1: Code-Review+1\n\nthanks, this patch also benefits us.","accounts_in_message":[],"_revision_number":1},{"id":"d494e014f019ea8433f11b70e6d27b7d1dde32c1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-31 11:18:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"727cb1fdd1e907907cb6b60ee2bce1434845a177","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 11:18:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"749bb9a31875354a7ff2d279d0feeb014b48f827","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 13:18:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e2549e4b85e696825c16b7f88e507a2449701eee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 13:18:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8dee37492e924918a54007fd3078cb7b : SUCCESS in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4617e2ecf43048d89eb90be355c2c3c3 : SUCCESS in 6m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/878455f2d33f4d2ea385204de44fd5b8 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/802b8e11d48442efa3697307a957d548 : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d53e6326725f4f8ab9848d417c582a01 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1da150d93a5441468dfa5db74a007e97 : SUCCESS in 12m 36s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e5e448497fe54f3899f45f04abeca998 : SUCCESS in 6m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/43ed79a6464a461e9614889509df11c9 : SUCCESS in 6m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4badb05c544a40fd8035ff52c434ecd5 : SUCCESS in 43m 34s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b90c747be596478090313673395770eb : SUCCESS in 43m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a063ae41e32e4446a0b189bae69d443f : SUCCESS in 1h 43m 32s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/68af616984a24219b9b34bb3a2344e16 : SUCCESS in 1h 58m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b68bcc3c7103421e8e2fa7db9f954264 : SUCCESS in 1h 16m 09s","accounts_in_message":[],"_revision_number":1},{"id":"1b8b2d58819e4ec7887c85498731ffbb2f37dbb6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 13:21:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9c7bd5ba0d94d3ab76f8fbce9ca2403 : SUCCESS in 2m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5bdecaafa7de47aeef96441a4edba414d7609aa1","revisions":{"5bdecaafa7de47aeef96441a4edba414d7609aa1":{"kind":"REWORK","_number":1,"created":"2020-03-25 08:06:55.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/54/714854/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/54/714854/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/714854/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/714854/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/714854/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/54/714854/1"}}},"commit":{"parents":[{"commit":"cb55a8cf78605c3d0b9c483b922f0156a6b3286f","subject":"Remove the barbican \"Grant access\" from cookbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cb55a8cf78605c3d0b9c483b922f0156a6b3286f"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2020-03-06 11:20:50.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-03-25 08:06:55.000000000","tz":0},"subject":"Pick stale amphora randomly","message":"Pick stale amphora randomly\n\nAvoid getting the same amphora all the time if its loadbalancer\nstuck in PENDING state and could not be failovered.\n\nStory: 2007340\nTask: 38877\n\nChange-Id: I2f680a00be695c3828166e6803882765af79b79e\n(cherry picked from commit 0f7d03feca4e3a4796c9c1a5a55d8da2adc44ebd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5bdecaafa7de47aeef96441a4edba414d7609aa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5bdecaafa7de47aeef96441a4edba414d7609aa1"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
