)]}'
{"id":"openstack%2Foctavia~714855","triplet_id":"openstack%2Foctavia~stable%2Fstein~I2f680a00be695c3828166e6803882765af79b79e","project":"openstack/octavia","branch":"stable/stein","hashtags":[],"change_id":"I2f680a00be695c3828166e6803882765af79b79e","subject":"Pick stale amphora randomly","status":"MERGED","created":"2020-03-25 08:08:34.000000000","updated":"2020-06-11 10:58:07.000000000","submitted":"2020-06-11 10:56:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714855-1591872963954-03911b87","meta_rev_id":"96c148ef11d38d3f37c072a88afcf806da651b19","_number":714855,"virtual_id_number":714855,"owner":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-11 10:56:03.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-06-02 06:44:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-06-02 06:44:07.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-06-11 06:26:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-31 11:18:53.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-04-10 11:04:25.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-06-11 06:26:21.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-06-11 10:56:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eec5e467b5c93e993f494f2fb19bdbd7fbafcb8b","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-03-25 08:08:34.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c37542adbc6a95d4c23ee571ec309ea6b69404aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 10:25:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5bc03b4da7574d5d8dd03d35a2c9a7b8 : SUCCESS in 8m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/693ac8072bf94984bc889bc50118f1d7 : SUCCESS in 9m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d405f51d0664973bacbd82793dcc0a6 : SUCCESS in 8m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9056fdf4ccf64347bc8e1e67482c42ca : SUCCESS in 4m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ae93352b53d44a72aefd118639a88b18 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe358d34bb854916a67eac67cb86ec0d : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf0477a93e6041a59617b9eb94cf2f1a : SUCCESS in 7m 47s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/cf7d5a98e6df436185d08c0b9b129939 : SUCCESS in 1h 14m 28s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/8658e0a042264a1880a1090a5e646a1c : SUCCESS in 1h 14m 14s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aedf5cfd9cbb499fa9603edaf5c7e0a2 : FAILURE in 1h 15m 19s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/fa0a7ba173784dff82eee504184af912 : FAILURE in 1h 17m 36s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f28e1adcc85443ac9a12721614518d9f : SUCCESS in 7m 37s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/6bd3e4e50e5b44ed8a5890ee0b0d3b21 : SUCCESS in 8m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a28ce5bbd37f4707a3cce32f004bcb0a : SUCCESS in 6m 55s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1de5596453b7415cbff60c405a827f60 : SUCCESS in 1h 00m 53s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/e15fd88d7cb047cdaf513607bd71c8d1 : SUCCESS in 1h 03m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f6fd09d939ba4e668f575a03e891d341 : SUCCESS in 57m 46s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/43f5b27c4c914427b683a3811ce4aa0f : SUCCESS in 44m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/805edb7ca4f543e09de87e51baf3f8c3 : SUCCESS in 1h 52m 10s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/42dbcc871c294b758f02ec9ceb999997 : SUCCESS in 2h 15m 37s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/99d5f64998f742c48e997aa4ede60e5d : FAILURE in 1h 41m 18s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/0a7a72b2fe6244e0919b6637805de440 : FAILURE in 21m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20a8aef665d5463a920262e85e9c3650 : FAILURE in 36m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f5e56bf83bb54f8f858ecd00ad8f81e2 : SUCCESS in 1h 43m 23s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/b1810354d8de4c28aa21ef2bdeadd086 : FAILURE in 1h 56m 55s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c29dd4303864480e9c3f9f460bb574a0 : SUCCESS in 1h 10m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1c866368d730432a8dd9481ee492f059 : FAILURE in 58m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"97ded1a19b4676807ec778de84baecfa02f852a8","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-03-25 11:10:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"daaf33ec21c96cece31c7bf2a31d6e39d8626408","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 13:18:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23646ae2956f4c01bbee4bbfd82aeb4e : SUCCESS in 12m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/600d632923d140e1955196513c8efd4a : SUCCESS in 8m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eeef26e8fb814e57878c0e1e2e12954a : SUCCESS in 5m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9e94b9765b5a4bd393e9266fea5cf561 : SUCCESS in 6m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/64a6998b1c764d8c98c08ea658cb6f29 : SUCCESS in 7m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/762cf1a5ff834f89a0ef6526221c7693 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b81c5c03e69430f81adb473d8a0eb5d : SUCCESS in 8m 03s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/293a96161eab4163a026d62ee99d2cd8 : SUCCESS in 1h 19m 21s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/57bcf1ba4a0c4b318f785f33454cec60 : SUCCESS in 1h 25m 56s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9f8676d9180344119256bac7bdc8857d : SUCCESS in 1h 20m 37s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/d3983e734d924e24954905befc773cf6 : SUCCESS in 1h 29m 18s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8878935d60e5435582377b84de66d5e5 : SUCCESS in 7m 20s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/651d8bb2286c441bad850de5bc303c3d : SUCCESS in 8m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/147eb29855e842458e1090d99930aeb9 : SUCCESS in 9m 06s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dc0f13164a0b437381da927a916d84e2 : SUCCESS in 1h 03m 39s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/02b0d0add38142d8a4d90f6b857543a0 : SUCCESS in 1h 06m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e155de5e99c3476da88656232b1faaf2 : SUCCESS in 41m 46s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d150150b375341e6a12962553dddd9e2 : SUCCESS in 51m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3edcf8727e87474ca404571bd249fd4c : SUCCESS in 1h 47m 37s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/995d66ebb0b24e19a0b2a77c3b755cc4 : SUCCESS in 1h 55m 07s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/a73dc88cfd604a02b63671b66d4c108c : FAILURE in 1h 53m 39s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/4ada57d0dec345e8a0010610019cc7bb : FAILURE in 24m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81b44736f8bd477492cfe8e592fe3916 : FAILURE in 39m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/86564d3f656a4b7c8fbc530638567135 : FAILURE in 1h 45m 21s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/36aafeb270c847f08cb1428af9caa54e : SUCCESS in 2h 01m 47s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/121e32a4ac55459b810758005b0c081a : SUCCESS in 1h 21m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d645f8c96d524f71b84ed833c6fc9258 : FAILURE in 58m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"920913993fef499b54a0f842b7ced4193ec79eb6","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-03-26 06:57:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0c9ef76d771eb4c997e30e59607945e862d889bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 09:01:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5a77ec2ba744675bdf714532ec7f158 : SUCCESS in 9m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d30d3c023e9f4e7db4bf18a7c03d2b95 : SUCCESS in 9m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8db296c15d44c8ebbc0b820ed315125 : SUCCESS in 6m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7ee888471bae41fbad8896ce72f93620 : FAILURE in 3m 24s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c403d99f44f84094a3cdc4ab1b98ba63 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c83ea98786a44278dd21ea71a96291b : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b565665bc1ef431b86f7b6b94134350d : SUCCESS in 6m 35s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/caf84e3a58094a3f98b3b1348eeba03f : FAILURE in 21m 59s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/6981e0d17c464503b4d801229985c61a : SUCCESS in 1h 33m 25s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0236e835f2f942a2bc2803633576147d : FAILURE in 17m 16s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/efcdbcc803c945168019b782098a73ef : FAILURE in 1h 18m 33s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/da6d07c5bf6d4d3ab5e5e6d2714cbfce : SUCCESS in 9m 42s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/9b45dd596f9e4037b2f50e510a40248e : SUCCESS in 7m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1b62b58baedc435f926d90129e5862a3 : SUCCESS in 6m 42s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d44c5dd4466243a39d659ba43e538c61 : SUCCESS in 1h 16m 34s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/0e863f996ab2432eb2c1ca29cbd9f1a8 : SUCCESS in 57m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ae26f2cae27e447b8b9440d771835c06 : FAILURE in 14m 42s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/09977cd939364d078d916563b0b2e3e1 : FAILURE in 15m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3ac68e3c123c41a5b7f1a3f9ac8a92cf : FAILURE in 9m 43s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/05733a5d8e13488c88b773957db6dcb2 : SUCCESS in 1h 59m 19s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/5b1b2d7bc3b84a44b0dc285e82030445 : FAILURE in 2h 00m 43s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/67687226bac141f797fd669a7c8c1ccb : FAILURE in 12m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4f8da13df1b648c293040ab7f590393a : FAILURE in 14m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/79c72629aab14d45b532b328303ccfca : FAILURE in 17m 46s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/ea4bfafeaaaa4b7b91c9e091be395557 : FAILURE in 1h 47m 00s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/186687303acf4a7388869bd3945908f6 : FAILURE in 17m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a9bf7ca7a6234ea2995aef885a12aafb : FAILURE in 15m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"863fbce967eca852b8129648915a86c0e8d8427d","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-03-30 06:39:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a9661b84a1186a897209b168be8d4b9650eb68eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 09:11:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65ac6d42827f4525ab5f48c7b552c359 : SUCCESS in 9m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e071c2664cbd446488e63d2613e342d3 : SUCCESS in 11m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/258f31428117451c9c26b6431c03af76 : SUCCESS in 5m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9058bb9d87d44f958cbd02fb6a501b1c : SUCCESS in 5m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1e8465a5b41b4e02b264ea41680b1fed : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1bf1dfcc2c34998a9309882f010a4dd : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54babf0ffe274437aec0c3a81227bde1 : SUCCESS in 7m 43s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/02641bc20bea47918d67b588801271e8 : SUCCESS in 1h 22m 35s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/5b82618995ff477ba081b766febf5355 : SUCCESS in 1h 19m 43s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f84cc225ce644382aa2385df6b0156ce : SUCCESS in 1h 24m 34s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/fd444418c3064346ba29fea5f541cf01 : SUCCESS in 1h 27m 09s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fc0f44d72bc04f2f8fecb7234ba243a3 : SUCCESS in 7m 02s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/f2ff1b3b85c444bc8da3cdc3c55ea091 : SUCCESS in 7m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a46a0724dae34474a3727acd41bc0afe : SUCCESS in 7m 08s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0f3e7fe697e248928f7c48a3261928c0 : SUCCESS in 1h 05m 36s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/04c82b3529c2425a84e5e1c43b8e5393 : SUCCESS in 1h 07m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/370a0bafbfef40baabba3465177f0bfa : SUCCESS in 46m 53s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/60d58de72e2b4b5cb896fb34088dfc02 : SUCCESS in 48m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ca373091fef1429baf0450cecc024444 : SUCCESS in 1h 57m 14s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5d7a5ff3b2a8418089752d5572069297 : SUCCESS in 2h 28m 12s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/c771c390e2f9403593d7c40342f402ac : SUCCESS in 2h 12m 30s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/27d297656a1d4819966c568fb04d18be : FAILURE in 23m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/310cbc8c55c947efa0f6b4d8cd748056 : FAILURE in 35m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c1ae80fb20aa4a6da146e063db504b55 : SUCCESS in 1h 58m 54s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/ba1d13fbc2d845569af04adae584a909 : SUCCESS in 2h 12m 33s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8c311de158de440184e435300958acf5 : SUCCESS in 1h 24m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/91784c409bab4ffa995d514c911810df : FAILURE in 58m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"13128c39cbcd1f93d11c9e612505ac8064e210b9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-31 11:18:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d19053c1ab572910c7e88dfe79eece6761fcd216","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-04-10 11:04:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9a93418a1cfb1813f25f7570fe060c27c37cbb3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 11:04:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6171b411fab5e5739891be80ff240bbe1afbb42a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 13:16:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/655a7c990e0144c4b7754c750016f363 : SUCCESS in 8m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6a470996e884f10b1a4e3a13a023f33 : SUCCESS in 5m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/82428ee7c8f04a9792006719d339d5f3 : SUCCESS in 4m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/07c440e78c624e9cb0ea2acf0299a99d : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b5a5ea223f74d698786b1f377eaa5ba : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/705b6a091cc44e49a212532790364f18 : SUCCESS in 6m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/dece3810b66e4a63a972589618c036d3 : SUCCESS in 7m 09s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/9171be23f849480d98278bf794196772 : SUCCESS in 7m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7f31056ceac04c63ab7b662638eafcd1 : SUCCESS in 6m 49s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/07f0869055794b9e89ca3f6705fef19b : SUCCESS in 57m 23s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/72067e164eb54e75a3ef2be276fa79fa : SUCCESS in 1h 03m 14s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/4968b1184402489b981910b25d0d68ee : SUCCESS in 1h 15m 51s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/ef0266f43f6c4875abb1ee81e7396a12 : SUCCESS in 1h 17m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a976bff312724844ab2c35de4c8fddf4 : SUCCESS in 46m 43s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/996a9d28fbe04c3b8a309989932b89c5 : SUCCESS in 40m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a857b08a58984f428c45bef189e95b54 : SUCCESS in 1h 42m 11s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2af25595e50645fbbdd87b81f797d0d9 : SUCCESS in 2h 10m 18s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/b791eed69ab146f0a6e00bd3571cef92 : FAILURE in 17m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/03dd1f0fa95c4df0b4a62ba509a78e4c : SUCCESS in 1h 22m 27s","accounts_in_message":[],"_revision_number":1},{"id":"658ba00649c87bd98d93f9736c449b8aa18a1fc7","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-04-10 13:40:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"88ba3869a69dc9e68b1c246e00d84cf8da4bfea9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 15:46:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/616d52e121d845ae977dbead0adb2eb5 : SUCCESS in 8m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef0b3bf892c14af29da4c8015887b2c8 : SUCCESS in 8m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0665aaba5f7e4970b28a614a9489d1bf : SUCCESS in 6m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ae3b6ac4764c4ce7be6318fdb1d283f2 : SUCCESS in 5m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/05b41de8bedf45c68d843b5d8ca1cb19 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aacbaac769b84ba7938abeac1b39d46d : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59de4381187e4dbf873735a437d65a5a : SUCCESS in 7m 32s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/d0828439856547f7816a4bc925c2880c : SUCCESS in 1h 18m 01s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/2caf6276c7364791a308f0ab319ee474 : SUCCESS in 1h 16m 30s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/88339918fe9546399230ce1215d2caaa : SUCCESS in 1h 19m 56s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/338c40e05b574bfb893fea5f09595d05 : SUCCESS in 1h 21m 25s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e4d0d05e73464d94ad30ac4b675c21cf : SUCCESS in 7m 11s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/b2df9b321e184b9d80f02247608a13c3 : SUCCESS in 8m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/503a6013f41b4e1899ae478e31a7f949 : SUCCESS in 6m 08s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f3b655091f7f416a8610c5b10701bc28 : SUCCESS in 59m 36s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/6e7e95708da441288a79e69aac0e5532 : SUCCESS in 1h 04m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/72b73db8d5114314a34db1cc8af15f45 : SUCCESS in 43m 36s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/03e9968a05174628a9a742536a8b0518 : SUCCESS in 41m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b740ef2906834d9c8d47d9826b639597 : SUCCESS in 1h 46m 06s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/92ae33deb8da41ec9ea16d52b0c33cb6 : SUCCESS in 1h 47m 50s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/5bfd525a6d9549c387b68344ec3bb3c5 : FAILURE in 18m 32s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/7f9beb6106194a73a5926d07d27f518f : FAILURE in 22m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e11c0ea8a58e4b9b981f30a3a094946e : FAILURE in 32m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/46262a2805ce4d5c935b1df877f0ab88 : SUCCESS in 1h 57m 35s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/1dae540cb3de4a519a66d053c90b1bd5 : SUCCESS in 2h 00m 32s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6b1804b64c904b26b8f66e4e12996261 : SUCCESS in 1h 16m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/082eaf585e734a309dd41d2d196c239e : FAILURE in 1h 00m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"caee19fda9d38ae69f5c7c3624d506560a1d0d37","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-04-13 07:24:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fa371c2ef84f15ed0f52d00e4c51ac53579a9fd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 09:48:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6050f06a7734560a33ce9888c49d299 : SUCCESS in 8m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d497536b32147399af120275a20caa6 : SUCCESS in 12m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ead84055c74f4ee8b1e95f566793caa3 : SUCCESS in 6m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7d83810660d945b1aea564a053517cf8 : SUCCESS in 4m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cd5390b04aa3417084086840802f176e : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55cd413227ff47b087946b1bdd3fd52c : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a02e15db0edb4575addd9c0ae842a6d9 : SUCCESS in 8m 03s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/2618c18e80f1483d9e318edeb5f2c2cf : SUCCESS in 1h 38m 19s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/103a067f700b460c9e1c71e9bcfb0449 : SUCCESS in 1h 39m 16s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4d86e13164474828b25ef477b0bb8561 : FAILURE in 1h 54m 01s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/61aceefc7d2b4d27bd5829d41dae3e24 : SUCCESS in 1h 25m 56s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d9e7162fc13b4ad789c1c9316a4f64ae : SUCCESS in 7m 44s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/f94c54f8ff204990a84e806f5079cc61 : SUCCESS in 8m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/455b7e5ab17b4c418455cdb331f16555 : SUCCESS in 6m 29s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6eb537ffa52342b6b9fd223dd194541f : SUCCESS in 1h 27m 41s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/78117a70435b4bafb7f689eb8eaaccbd : SUCCESS in 55m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7db6e041c5d84180aea0782b88ef34bb : SUCCESS in 1h 06m 06s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/0c9a5a2df9704acaaa54e8c4c6c934ee : SUCCESS in 1h 04m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/404d0a13879f445894afa0bfe73a4cb2 : SUCCESS in 2h 02m 59s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8d58bf059cb946f8abe1c12cbf1aaa8d : SUCCESS in 2h 02m 49s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/73f40f98a66a4f14b087e4357b07c414 : FAILURE in 29m 58s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/0efab6c19f114678a02f6190e34c7b80 : FAILURE in 29m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f2def371fb2642459b75266b69393dfd : FAILURE in 35m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/60c697cb48c94e8d9d838f21f9e542e8 : SUCCESS in 1h 17m 58s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/3b82bd2caae14283bf8b08a31ebdb403 : SUCCESS in 2h 03m 02s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d5a5e8ae687a4e68832226087acd96a5 : SUCCESS in 1h 47m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/01e35b80898d43a985e14484b51cc2eb : FAILURE in 1h 01m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f080bcedab92167fde17584a0b0ce15f9f62fba6","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-04-14 09:57:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5c1fa4c34be2b76554c8df27f9eed6fcadfed364","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 12:17:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43065373bc5b44eebc00202f57c6ae3a : SUCCESS in 10m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e62345fab10244aea73145c0a3628400 : SUCCESS in 8m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f6857d5f68c4bb0a0b72e5f51a35c70 : SUCCESS in 5m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/43e4a170b0a44d219203704db70fcd0c : SUCCESS in 4m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/37fa73e185d142499c20891343120b6e : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b23e7245a0254440958d0a4e15af5bbb : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec3870e3b9784650846b0bbab5bb90c7 : SUCCESS in 6m 39s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/037940f0a56640448a7c886ecc59e6e9 : SUCCESS in 1h 17m 18s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/4022c910c36740e39b0d41c5401da602 : SUCCESS in 1h 20m 29s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c3d9d9ae0f2e4efeb28a8ad269a6d5c3 : SUCCESS in 1h 26m 57s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/aca042f0a4c542d795c82c7dcda3992e : SUCCESS in 1h 27m 07s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/599ea3f9bb76492c9dbed1c0de4114f5 : SUCCESS in 6m 13s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/ebcdd14a92da44f7ad35a11e2189b795 : SUCCESS in 7m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f5bb368afba949f0a51fbfb012635170 : SUCCESS in 6m 44s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/310f5a87c0774078b62af2e1b63da430 : SUCCESS in 1h 03m 38s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/87046dfad6e34a6c94916457371def02 : SUCCESS in 1h 05m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/10695d561e894835856bbdd864a233ff : SUCCESS in 44m 52s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/f247e830c5ac411e9f8a0016fc144627 : SUCCESS in 46m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1f8e83966c4b4fbbaed5fe40493830a3 : SUCCESS in 1h 41m 24s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/dc9e6cb4bd35479fb50ff14879e2aa8e : SUCCESS in 1h 46m 22s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/257637edbc0147c78b9dd802b764bd4c : FAILURE in 22m 12s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/30977022500047019078e90430cf1e52 : FAILURE in 25m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8ec20a1932ce41c7892d86cc43b22a7e : FAILURE in 34m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7cb2b01269c44476b1bdf09eff450293 : SUCCESS in 1h 50m 45s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/8deb4f8b70df4fd8b8137d51d9efcefc : SUCCESS in 1h 50m 52s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0bf01845954f4cef9097929ceada247e : SUCCESS in 1h 20m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fe1e922744e34e828b3ed69119b4416a : FAILURE in 1h 00m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"80a6a8751d2dab3ca9c73eff106c6e033132fdc0","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-04-15 20:08:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ff0132e4db81f9d60d0a7c91c9d2d7c78d2d7b41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 22:19:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8f0f1e6ab2b4db39c055024d88c991d : SUCCESS in 11m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c17611ddaba641b1a3c60ea57d580607 : SUCCESS in 12m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1ba9ba8a06344e388c86f6754f70926 : SUCCESS in 7m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/af0e0f192f9a4ba6b2c8c6fe17985646 : SUCCESS in 6m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/93cb8c73f8884b1d9013d0db52b043b3 : SUCCESS in 7m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/263b36b9982949748ae09bbe93b04dba : SUCCESS in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a79c2084242f49e2a6cc21794976c777 : SUCCESS in 8m 39s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/00a9f285e26a45bb8a6e965cfe3a2db0 : SUCCESS in 1h 17m 11s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/c9255f054fbd4a9c86fbcb0a57c9a9f0 : SUCCESS in 1h 20m 11s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/630e334aff854ab6ab22935cf3253023 : SUCCESS in 1h 26m 14s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/b8199ae849bd4861bf9f61686721797b : SUCCESS in 1h 19m 17s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/585a590b795a4016aee5264c72be5572 : SUCCESS in 9m 57s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/8df13f293ced411ab48960579d70d621 : SUCCESS in 9m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/da968a021c4b4a4c85e5f862a5bcfec3 : SUCCESS in 8m 57s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/47d7a679b18543ffbad059e2d0fd6189 : SUCCESS in 57m 34s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/3f8b4058c860406eaf744d49ee9f1b57 : SUCCESS in 58m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5498ec28c4464c879e6f01c20c15dc6b : SUCCESS in 43m 54s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/48c01c9438534bd69828f7aeee65f79a : SUCCESS in 44m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a02df3a1d30c43f195934a6d1675a14a : SUCCESS in 1h 46m 12s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/991c7b23cc75469ab63036ec86bda370 : SUCCESS in 1h 46m 53s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/076730451505488a98b1bd1dbdc66263 : FAILURE in 21m 07s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a2211f3f2ce74384b62bfb3b7d1c88bb : FAILURE in 22m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c84bd64a358e47ed897fc3823462d95d : FAILURE in 35m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/abc187a132f24f68b9ebb9c36e95702c : SUCCESS in 1h 44m 51s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/54026c2ac87746a0b6dc07dd4a74bfb1 : FAILURE in 2h 06m 54s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2d15c468349145f488ce52217ba7e5c4 : SUCCESS in 1h 11m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/64fc3680ba154755bda1b7d2330ffa6d : FAILURE in 57m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ac6f1368506d547a609e2ef3cbc87e4ba0b91277","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-04-16 06:29:09.000000000","message":"Patch Set 1:\n\n\u003e Obtaining file:///opt/stack/tempest\n\u003e tempest requires Python \u0027\u003e\u003d3.6\u0027 but the running Python is 3.5.2\n\n\n:(","accounts_in_message":[],"_revision_number":1},{"id":"5cc3e279fdcd14682b5e79bf37948a16be849881","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-16 08:02:06.000000000","message":"Patch Set 1:\n\nDevstack should be pulling tempest 20.0.0 as per constraints set in stable/stein upper-constraints.txt: https://github.com/openstack/requirements/blob/stable/stein/upper-constraints.txt#L392\n\nTempest 20.0.0 supports Python 3.5: https://github.com/openstack/tempest/blob/20.0.0/setup.cfg#L19\n\nThis patch https://review.opendev.org/#/c/718131/ should have fixed the problem but apparently there are also other users reporting they\u0027re still seeing it (same as us here): https://bugs.launchpad.net/neutron/+bug/1871327","accounts_in_message":[],"_revision_number":1},{"id":"5ff4e95eb4d448a9a8dc2a982b9cf48ae6e6770a","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-04-16 08:19:13.000000000","message":"Patch Set 1:\n\nThanks for checking this!\nI added comment to bug you\u0027ve mentioned.","accounts_in_message":[],"_revision_number":1},{"id":"84bde3b0718720232ac0a14aec010e06b65badb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 09:58:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b32fd2bac2647798ff1a5d8c2a32387 : SUCCESS in 11m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c8ec9a86fad437d9c2109be4031ea04 : SUCCESS in 10m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/915f3017409a4595b7c63bb08095c2bf : SUCCESS in 7m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/581888d466cb441a9fbbad57cff4ec49 : SUCCESS in 6m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bf6929c92e5844e1a1da0b276ed403ff : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f0e730fe8284a5789fb7838303be108 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc20d5ca44f44572b2bfd2e0d1aa2f50 : SUCCESS in 6m 23s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/18b0996a8c7741a68d34fc240626243c : SUCCESS in 1h 17m 27s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/4764c3dee4ee4371a0530c1316cf2765 : SUCCESS in 1h 19m 36s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1a9a3188ee7847a6af7aa45a78956c23 : FAILURE in 1h 44m 36s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/d46cb499c92d4bbea769a3138056b1ac : SUCCESS in 1h 21m 55s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/800b3fdf16f04eaa925fe5dcc53dbcfb : SUCCESS in 7m 47s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/266236a2103341c1b24fd1d684d21259 : SUCCESS in 8m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/acaeeb3280c74a5b9f4b01d2ef0f4fc2 : SUCCESS in 7m 51s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3df70ceeabbf4da2a9778b6c4c2ff7d3 : SUCCESS in 59m 39s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/5bb113c0a85d49d6a3d3ffea559ba422 : SUCCESS in 1h 05m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2d9c4ba1bf1f44e3b0ad4f7260c5dd04 : SUCCESS in 43m 10s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/726f0dee51b747dea9196d1536dd5b3f : SUCCESS in 51m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/615d69d1f5d74d24a6ddc03903e61e71 : SUCCESS in 1h 35m 04s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/80aa31cecf0440859955b42d9fb63afe : SUCCESS in 1h 50m 06s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/2b2a0df1fdcd4dcca65fc43f32ae30ed : FAILURE in 20m 18s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/95455a7224f0464c8b7fb6be6c1413d7 : FAILURE in 28m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/529877924d1a43728aba699cf902b651 : FAILURE in 37m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7f1c3719c3a244c6aba097c60028d35c : SUCCESS in 1h 56m 10s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/979146ab057f482db43e313eda63c3bf : SUCCESS in 2h 02m 13s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4958f407c3ae4ee0b04c18b541d9c35c : SUCCESS in 1h 28m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cf36f612d8f94412a2de34afccdf7c31 : FAILURE in 1h 09m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c9a90af05d8b166240fdfdca2969260f3f2b7ca8","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-04-24 12:45:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"770095c0c461747b530e6dcddb68ca62629682e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 14:58:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b711930e02f44aa9c7050a69726c091 : SUCCESS in 8m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/24b2a4e32f45485d807b981e3ceb8c22 : SUCCESS in 9m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8eccfdfde913416ea936124ba6f6063b : SUCCESS in 6m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/042c36ba97694c4eba9c2bf5c54ee5a8 : SUCCESS in 4m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ef54052ba926442798507f10b85b38a6 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/410afa7f93ca48a7ab8b2526de70ee0e : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8cc1d52c2f449f6898c69a4d6bc925b : SUCCESS in 8m 44s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/350c3a6e196c4169be6eb8068e7e94d3 : SUCCESS in 1h 21m 43s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/ecb7b03f4b6c422799e65b84b2e003c9 : SUCCESS in 1h 21m 54s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1c98babf4cd448ff977ad23b687950aa : SUCCESS in 1h 25m 53s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/7b578dfed48542dfa6f47f30f0b48f68 : SUCCESS in 1h 24m 57s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3544dc60298342778f673cccd67f935e : SUCCESS in 6m 52s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/0dec2a4dc70b4c4aacc227b377354202 : SUCCESS in 8m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f805f28ab36d4d7882699a6df6dd3316 : SUCCESS in 6m 21s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d0447a9b00e24fa2b73f664804242fb6 : SUCCESS in 1h 07m 32s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/de825eaf96f1431780814c2b87719b45 : SUCCESS in 1h 03m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9b8d42ff77564bf59e7e9a8b004bec50 : SUCCESS in 45m 53s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c65cdd0ae17944268e7ce733057e3d7a : SUCCESS in 53m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9f7feb877ff449798d3b531fd5054bd7 : SUCCESS in 1h 50m 41s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4d8cac30d83a49b1ab3e4c4b99440a2a : SUCCESS in 1h 40m 24s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/5bdb3094cc9c41adbb889138c9af10b3 : FAILURE in 19m 32s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/73643c77a53047c695368db0781df9ac : FAILURE in 21m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c0085610242f4964b9e1f911795fcd18 : FAILURE in 35m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ddb9c33d2dbc4fa3ba315046a92baf79 : FAILURE in 1h 46m 33s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/8ce694aae9a64e7080b222fdf0f88407 : SUCCESS in 2h 06m 56s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bb90e8a71c2142fd845ce84efc45edb5 : SUCCESS in 1h 23m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f4a7bdb599a64e8daa4b956720ed0df1 : FAILURE in 52m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8e3f4cd6de223864947c23cbbe54bbb0a2e23f01","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-04-27 06:48:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4e590119592dbbe3e6073d3d25e1af40ae9fd30a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 08:42:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/abf51a1bc3c24e62bfb062b4153bec60 : SUCCESS in 10m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c038fffa763343f49a5ca90abe8703e2 : SUCCESS in 11m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d79fc15d826e4266a28fdb300db3f446 : SUCCESS in 10m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/350137756dd448a79c9acf7071a56ccc : SUCCESS in 9m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b86649b335df4b20a2a9c5315ffcae9a : SUCCESS in 9m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7402130470df4910b36d01a5804444ef : SUCCESS in 9m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae42be5127004cdeb39a89ee2d9a0340 : SUCCESS in 9m 56s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/b31fc95f6ac34d1497de999f37c33215 : SUCCESS in 1h 23m 23s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/dc8b6306d6524319a3c311ac1f479705 : SUCCESS in 1h 19m 43s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f4d3f8cc8a1b47b08d96598df69c28ad : SUCCESS in 1h 46m 36s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/4b399ab226334ab897df553423aa46cd : SUCCESS in 1h 25m 40s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3e70d5b61c574f958ef4e0c6003d9dee : SUCCESS in 9m 47s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/32839e54e11246cdbd64048acf5633bb : SUCCESS in 9m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b36481f76a8346e7a10b6a931e97fa01 : SUCCESS in 8m 41s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d3a1689864bb4b3fb117b02cde8f0bbb : SUCCESS in 58m 35s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/59ae4aac820a4e10a537ca814e2ac1ca : SUCCESS in 58m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/685624740d1740498bdfecf6b7fa03b1 : SUCCESS in 45m 12s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/5f9f6c044c3d48b8abf06566b540ba9b : SUCCESS in 49m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5b1efb9d308d4ff092d7e58bbbfaca9c : SUCCESS in 1h 11m 19s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/58f43e95be744e84ba7b399d20389139 : FAILURE in 1h 20m 52s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/4de5fe913c89462cbb15ed9e2e91b47f : FAILURE in 40m 40s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/ab68524fcba2420ea4dd527ce6f0aa91 : FAILURE in 20m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/623411b0590b4e7494580f804f80d18a : FAILURE in 32m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8fe35593de034ed69800e3da2f66bdce : SUCCESS in 1h 45m 25s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/fb748e61c66640c3a2cb581b1421cb9e : SUCCESS in 1h 51m 29s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3775908e0d1d4643968403a9b27f9946 : FAILURE in 1h 05m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1f2cd731a29d481f96a8218c9c69d415 : FAILURE in 1h 06m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"377c43a720014c3fa7bd50050097c97cfc98a8cb","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-04-28 07:41:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a97df2c3cebc8b3a6ee0b37a597871e17151093c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 09:20:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d51af3477f94e32b98dd63f27b181fa : SUCCESS in 8m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b410236685647b984853918065b8429 : SUCCESS in 11m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bad4c71ce92f4783a727b37edb62ea76 : SUCCESS in 10m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f94b7da5552a4a5ba5ff2a6370484165 : SUCCESS in 4m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/87d0736651384822876ec548e0fb85e1 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3f9101aa12b4b1489beb7babfa2ecc3 : SUCCESS in 7m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe96ac4a18f5419697bd4f139f019f60 : SUCCESS in 7m 43s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/ae87053b259b4945a1f4ae77cfb5c3fb : EXCEPTION\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/0bf06044d3c84381a981bd73455d6bb8 : EXCEPTION\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/736058f2fc2e4fd0b9cf1b580ee4c807 : EXCEPTION (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/ae536a3a0cb54ec7b2652b16c1e213df : EXCEPTION (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a74140e31678484fb8ac62773e4ad5ae : SUCCESS in 8m 49s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/d711264a7dd3403ca7d30ac0d2e3f96e : SUCCESS in 8m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7cb1f7ebb0cd466da67b926eef87b7fa : SUCCESS in 7m 30s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/004b546ae99143a1b21f4c50ae8e0f17 : EXCEPTION\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/bf5d99b171d8419a96cf8247fd7739c6 : EXCEPTION\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/21d643dd9b1e48bea421c96d73200aef : SUCCESS in 50m 02s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/f0c94ef6919847c8afb1a68cd404329e : SUCCESS in 40m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/59b3326063fe416f88c128889706cadb : EXCEPTION\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1925f8ad4dc54a989edc5e5949e56f89 : EXCEPTION\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/b2dfb5a5f405497f819b124eef7b3ceb : FAILURE in 39m 05s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/b101a8b7bab642e890147f80676c0aa2 : FAILURE in 33m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c1aba31f229424085e4e23bfc04313d : FAILURE in 34m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/977bac7e9f1144d2bbec0df8100e73f5 : EXCEPTION (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/33983187e0494a87806b95670b628687 : EXCEPTION (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8a8c80706387434a96e78e95694cb155 : EXCEPTION\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bf08b3527cea4b41ba846a6a486db08c : FAILURE in 54m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8a9e7751ffc246744953347b1bb44cddc1d325b2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-28 15:57:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c8bd3df1deaf9a66f4dfd0384f41046c1b3ab531","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 18:20:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab364259947844d79d96a8b1e804024c : SUCCESS in 10m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf5dfe6b8be34de6b88746552901f405 : SUCCESS in 12m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0726e9229ae454cbfd1319f572752c2 : SUCCESS in 10m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3b24536c82a94c14bb0fe70cd2108fd9 : SUCCESS in 7m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/dfc152e541b746bb9cd0ed8ba88c4e5b : SUCCESS in 8m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9471880167a4481bb3e8dcafcc60d8e : SUCCESS in 8m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d36dc985b4704f1ba24ed7d0d3497a4f : SUCCESS in 10m 44s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/d141a4d2f4344ed3a22af2ee46b96cc8 : SUCCESS in 1h 33m 44s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/b00b2fefc3834e27acad14b406fdcedc : SUCCESS in 1h 35m 16s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fe1d3f23b35f4a319a9a67c86c1b26f8 : SUCCESS in 1h 44m 52s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/fb987966ff474079ba28c4bc3545b587 : SUCCESS in 1h 31m 08s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f899f55e7fca4cadac778d948894cd2d : SUCCESS in 9m 06s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/3c349f4d2f9b4527a647dd55e6ecc99d : SUCCESS in 9m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3e5100a1421246f1afae67d463b74537 : SUCCESS in 10m 19s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4be750a9526e40e3b5845b55c3f3c7b8 : SUCCESS in 1h 04m 49s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/851d39573c794babb7fc9e8bc80d53d3 : SUCCESS in 1h 10m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a8548c1451654aaaa073bd83d0d4f295 : SUCCESS in 1h 00m 07s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/8119a7afe2aa41f4b5e5a1fdc60ffbf1 : SUCCESS in 49m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7403c0523a544fae966f4df1dc82805c : SUCCESS in 1h 53m 08s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8eff9f71a67949aca8c825f44fa8ed6f : SUCCESS in 2h 10m 58s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/3a85228a32cf484abbc2fdc42f7caa8a : FAILURE in 57m 25s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/e158626f6d9b4a7192feb5634665e57a : FAILURE in 28m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da731a08152646018ac42384a80c3829 : FAILURE in 39m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2d26455243724b17be1b680bb84abc53 : SUCCESS in 2h 08m 36s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/fddadb7d0d3e430da6f491430827279d : SUCCESS in 2h 13m 23s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fcfa9277c5574edb9b2b57e18dc7383b : SUCCESS in 1h 22m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1a0fbedfcaa2471fbdb8fc30a7878726 : FAILURE in 1h 03m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bdde20974ff8fd2bd55d3e15baea532d8ecf0777","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-28 19:23:37.000000000","message":"Patch Set 1:\n\nI\u0027m checking the xenial job with Ghanshyam from the QA team. Please don\u0027t recheck meanwhile.","accounts_in_message":[],"_revision_number":1},{"id":"4771df4c3ba5a648d2eeec363d64ebc786887bac","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-05-01 09:42:48.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a08f78a38d1d8b59574cd449c734b4bb6da59884","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 11:48:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60de4cdde954408abf8b92b4f77d481b : SUCCESS in 9m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69efa1fd6ca8403f94a441b82cc2c65e : SUCCESS in 9m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/651abda5100547e689629dc75983558a : SUCCESS in 6m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ea4ed78014d242d385abbb6a6e1f18fb : SUCCESS in 4m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cf92b55372234972a4135e2a03bf7381 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06587336a96d4298a5b40339477d5416 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a98693b9bfa44a93ac032dff5859fc97 : SUCCESS in 7m 32s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/ffb457b2cb4740688091dfc75ce23f66 : SUCCESS in 1h 16m 11s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/532a5ea74dac48be98ebd9597a233f2f : SUCCESS in 1h 20m 56s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6972c9f103d949ebbb74e97b0c31327a : SUCCESS in 1h 25m 27s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/a5f7ae725cf64d659558782522b82200 : SUCCESS in 1h 28m 52s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8fad95d5eec44ea790fde03dba8ac667 : SUCCESS in 7m 44s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/66a0e247a08d4c46aa1c3735aef0f685 : SUCCESS in 6m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/89c2185ba8e14150ba891047ec9a45a1 : SUCCESS in 6m 34s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3e58707cdd7b4d97828ca89d2d25acc2 : SUCCESS in 1h 00m 55s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/e4c5ada3585e4f0794e23e5152cc4da1 : SUCCESS in 1h 01m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e8c99ac639b0450cae5308bedd0c9b5e : SUCCESS in 43m 31s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/8dc66825b8214b03bd0a3082287ef718 : SUCCESS in 43m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/975f4df880894e1fa96a9397e70b24ff : FAILURE in 1h 13m 22s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ee71794e4a324cf691f15eb4a34f9487 : SUCCESS in 1h 46m 43s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/46be21d9b582492ca3e67ad7c6c82617 : FAILURE in 23m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/afeaf5caba704f38b0151447d0d9e79a : FAILURE in 35m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fb3c376e18104d75a815650bfaf39d97 : FAILURE in 1h 31m 53s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/e83cf01182f94220991a9386a7fa4d15 : SUCCESS in 2h 00m 16s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9778741d6b734dae9dfb9b942b3fa51e : FAILURE in 37m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c2a866a4fe334671afa198cd020267ba : FAILURE in 57m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e45420a3317172abda1494b2abf7221889c98eff","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-06-02 06:44:07.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"2a40014255c8e5b398e83007ab40f15912dd9cb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 08:14:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64b1b5bd49f04b3db50a4bab13c364fc : SUCCESS in 10m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f96506617716450799f7103cd21ec474 : SUCCESS in 11m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a15356bbc67941e3bd3b127025865cea : SUCCESS in 9m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/559400d8577041ad99ee1a273bb8e7bc : SUCCESS in 7m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e0b29400d7bf4104b773e6acc2ba2a60 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1e0d172eba641cd8a794c0dee6748b0 : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1228aea3a043490680915b1646bb7e85 : SUCCESS in 9m 34s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/bff078a8847b462db1998b2bbf8116b6 : SUCCESS in 1h 16m 48s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/d315a2b7c20c4ade8afb8d93993ea54d : SUCCESS in 1h 17m 05s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/608da0f4cba04f5d995a828bef389d69 : FAILURE in 1h 16m 32s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/3944fc2dbce94c178975ad69c0275bb8 : SUCCESS in 1h 24m 04s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/da47544b60a646cf9ac17b1885e2a3a8 : SUCCESS in 6m 56s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/dc80b8c27f04412bb29413a2c73412ba : SUCCESS in 9m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f50195794e89437f982185266c10a214 : SUCCESS in 10m 37s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/19925585ff8a4538ae693653501b1789 : FAILURE in 36m 54s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/d9a093328c0341ab929c0e5633d72169 : FAILURE in 35m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0d60bc9cebe6424fb68ce2fb72e4aa6b : SUCCESS in 46m 27s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/85ae9c3d34f84516a0372633ec3e8e2a : SUCCESS in 46m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c34d0a10023c41fc8774b2136e96d7a5 : FAILURE in 26m 47s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ddb9ef78a13e4338b9717bbf6d7a1043 : FAILURE in 33m 53s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/dbd04fab59624920bbf57b4a406814bd : FAILURE in 25m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/30e180f72fea474485c35708b8ba7b9b : FAILURE in 32m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2b4f283a02904438be511bd3c03dc7ef : FAILURE in 31m 42s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/fd3e1873859f4beab2ed4c9bd1710ad4 : FAILURE in 23m 49s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bf85538746234ca6a562283161c7fe02 : SUCCESS in 1h 08m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3cc63244af9f450eab9597777284502e : FAILURE in 31m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"daa0e7c885cee94beea41fc3e09ec4a8a0566a1f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-03 08:29:07.000000000","message":"Patch Set 3:\n\nrecheck\n\npy2 dib issue fixed in https://review.opendev.org/#/c/733008/","accounts_in_message":[],"_revision_number":3},{"id":"80416d6500f321f95d4a99e9748426be5b1b36b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 09:59:23.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5684c9ae9b6948c3af81c8f28e621c3e : SUCCESS in 9m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e1a8f4fc6684b1fa2336bfb66b7f5d1 : SUCCESS in 9m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69ef7da340734c9a97a4bceee660a4f3 : SUCCESS in 6m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/12f94a7d99d1468094016024ac5c0825 : SUCCESS in 5m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/11b81b84cbb34330801d5cc50a678973 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d990ca34a5644798982602307f3974f9 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8070c08ec5948aca0c41e75d341cff4 : SUCCESS in 8m 12s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/52f30a6abae948619dd7dd724a661254 : SUCCESS in 1h 22m 48s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/ecc52697907b4c61a5096c89ff2ef305 : SUCCESS in 1h 20m 34s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d9fb4cc795de40498159f21b4ba361ee : SUCCESS in 1h 19m 24s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/bcf10666dd094b8bbc04e130a41aa920 : SUCCESS in 1h 09m 44s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c6d0a330689244c5b5c314a6a0f30bd9 : SUCCESS in 8m 57s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/17713dd26e5e473da184fc552693c617 : SUCCESS in 8m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/128619915f08468488a1e3153e3c0b36 : SUCCESS in 8m 51s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1c88c9e21fa147f7bea3b12eeac63067 : FAILURE in 30m 45s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/5cbd826c734c4810847c5e8a4ae8a432 : FAILURE in 32m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/aaf0ff129c864d7ea374d56a00c937c5 : SUCCESS in 48m 20s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/e5f3542c90214364b2a1606dcbcadfd5 : SUCCESS in 45m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f067c17f6744400dbbbc9fca19d016ff : FAILURE in 32m 19s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5c7aadb8733f4aaaa0f639c6e5d8abe0 : FAILURE in 28m 05s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/6d703ac04a224db2ab63595d91e8a7c1 : FAILURE in 27m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3827395518d4b08bd07a72156cb429c : FAILURE in 28m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/71a7b3fbb3354c258e01755a6f53caaf : FAILURE in 32m 28s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/832fa60f277445a6a845a2af6837fa6d : FAILURE in 31m 11s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9f7d9440856e402591d47203b75135b6 : SUCCESS in 1h 23m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/467f4de542414c8f933bcf93c28806b8 : FAILURE in 21m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dd413522320cec556415d4f1f1056d02030aa0c7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-11 06:26:21.000000000","message":"Patch Set 3: Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"427345493bba45d0cd97dd87054cd7abf6d97246","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 09:07:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3c34e97d393442da2c8f948f72817a7 : SUCCESS in 12m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70d31baaf66f4599aacf08cda25523d2 : SUCCESS in 11m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4da53a894564618ac07d15d7d2ec1b0 : SUCCESS in 8m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8b88d665b78440788bc7a79d31f8e3d6 : SUCCESS in 8m 38s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/607f3cf62e904af4ab4a4eeae3fea436 : SUCCESS in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc78fd24e53c4100a3665076aa084d71 : SUCCESS in 10m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f30d79d9ad6149609592c37b4bd3d841 : SUCCESS in 10m 05s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/dc47e99a06164be89c08ff0a58939f26 : SUCCESS in 1h 20m 00s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/b1eee57449ef480f9707c7f95411bb1a : SUCCESS in 1h 14m 16s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/954da4bd68054db39eecba0e35deae51 : FAILURE in 34m 09s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/3d6b408479434b01b41e5bb924aed8ba : SUCCESS in 1h 30m 17s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/31c327d5163b4666b8c38533b812bad0 : SUCCESS in 11m 47s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/93a708d8d5444d49928030474cea552d : SUCCESS in 10m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b1b5efee6e0d44d2a7a7a925d571d06e : SUCCESS in 10m 39s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/185b8baa14aa4ab589eda894a93b2f48 : SUCCESS in 1h 00m 50s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/7953b5c5dbe04b86b1c376f6d1ad9d34 : SUCCESS in 1h 13m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3c3e64dec7db4407afac82c3320898fd : SUCCESS in 51m 38s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/4abd24236d9f4dbeb2c536cae7bd01cb : SUCCESS in 44m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f49ec54e318c4933b80f218df14ed1a7 : SUCCESS in 1h 40m 27s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/38228bd7ae0c4a8d8bebe4e4f12c85de : SUCCESS in 1h 30m 28s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/028061e5f559403e8e3b15d1203021c3 : FAILURE in 29m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a0c44db92b1e4624b5d3d95d9bbd1287 : FAILURE in 35m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0f751c8562684a23b19115af4a697473 : SUCCESS in 1h 57m 02s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/9a2a8482a58745d0b5c6eebe44bff053 : SUCCESS in 2h 06m 31s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b765c367ca094f26a1aa4f958d88ae51 : SUCCESS in 1h 26m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b8b32eb2090b4b72b40a456612583479 : FAILURE in 1h 05m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0ab4d5798f48ee23bf8d0c2eef055153701681f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 09:08:03.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ea3ae94cc2865744b19dc46f3130d5766d9ff13c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 10:56:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"b8bb291613ea08cd2edd395d6a964d0bd91cfc56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 10:56:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f8e853eca054b63b1904d2d5271fd84 : SUCCESS in 9m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96275b1a160440eca8ada7577087d1a2 : SUCCESS in 8m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dcad4d1a28174dee877320105127fa82 : SUCCESS in 6m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/17c8596a4291463ab5d1a8419dd5607a : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/139355e683c24997af0c638e1021d023 : SUCCESS in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03389bd590fe4ecd8806b219509e4813 : SUCCESS in 8m 26s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f387c2a7bbd04fc496cf29041d5f5f8e : SUCCESS in 12m 43s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/e10ec161f47f40e783df26430383ce75 : SUCCESS in 7m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a32de5ea1d6544699a97d33a535eaaa8 : SUCCESS in 12m 40s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/95b9550217d84bbda7ca07f14d2a3bdb : SUCCESS in 1h 00m 47s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/49f8c3c929834a37b7f715504fc29e34 : SUCCESS in 1h 07m 31s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/7432abade2b240b5b773e2fc30c477d2 : SUCCESS in 1h 28m 31s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/65f0e2178aee4c33bad0e22afff41c18 : SUCCESS in 1h 16m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b31071e73a2a49749336db021eaabc78 : SUCCESS in 42m 55s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c3eb0ba622a2400fb13900bbdfde4e47 : SUCCESS in 45m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8041773c5fe0489d8fe348c7c0d8cbb3 : SUCCESS in 1h 25m 02s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f62134a38c4741c498cc7456b1d9616e : SUCCESS in 1h 42m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2d371b82afd648759a399607b8238ed0 : SUCCESS in 1h 12m 31s","accounts_in_message":[],"_revision_number":3},{"id":"96c148ef11d38d3f37c072a88afcf806da651b19","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 10:58:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2399c833486f4763ad594b56ca2a7b3e : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ae03653afcbbe1ef74b9c2f603caf4c3f2071e2a","revisions":{"f5a86cf861d965c002d1747296dadb2ccb99394e":{"kind":"REWORK","_number":1,"created":"2020-03-25 08:08:34.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/55/714855/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/55/714855/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/714855/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/714855/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/714855/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/55/714855/1"}}},"commit":{"parents":[{"commit":"e3a95c9b0f1f59cea52097f03bbd8ab1a20e89c6","subject":"Remove the barbican \"Grant access\" from cookbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3a95c9b0f1f59cea52097f03bbd8ab1a20e89c6"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2020-03-06 11:20:50.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-03-25 08:08:34.000000000","tz":0},"subject":"Pick stale amphora randomly","message":"Pick stale amphora randomly\n\nAvoid getting the same amphora all the time if its loadbalancer\nstuck in PENDING state and could not be failovered.\n\nStory: 2007340\nTask: 38877\n\nChange-Id: I2f680a00be695c3828166e6803882765af79b79e\n(cherry picked from commit 0f7d03feca4e3a4796c9c1a5a55d8da2adc44ebd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f5a86cf861d965c002d1747296dadb2ccb99394e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f5a86cf861d965c002d1747296dadb2ccb99394e"}]},"branch":"refs/heads/stable/stein"},"98f75dc6a634b3bd0c858a888881629e19cf6132":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-01 09:42:48.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/55/714855/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/55/714855/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/714855/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/714855/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/714855/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/55/714855/2"}}},"commit":{"parents":[{"commit":"922c4987fb4db019212c7db322a65c3998f0c1a9","subject":"Remove xenial based jobs from stein gates.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/922c4987fb4db019212c7db322a65c3998f0c1a9"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2020-03-06 11:20:50.000000000","tz":240},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-05-01 09:42:48.000000000","tz":0},"subject":"Pick stale amphora randomly","message":"Pick stale amphora randomly\n\nAvoid getting the same amphora all the time if its loadbalancer\nstuck in PENDING state and could not be failovered.\n\nStory: 2007340\nTask: 38877\n\nChange-Id: I2f680a00be695c3828166e6803882765af79b79e\n(cherry picked from commit 0f7d03feca4e3a4796c9c1a5a55d8da2adc44ebd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/98f75dc6a634b3bd0c858a888881629e19cf6132"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/98f75dc6a634b3bd0c858a888881629e19cf6132"}]},"branch":"refs/heads/stable/stein"},"ae03653afcbbe1ef74b9c2f603caf4c3f2071e2a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-06-02 06:44:07.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/55/714855/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/55/714855/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/714855/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/714855/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/714855/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/55/714855/3"}}},"commit":{"parents":[{"commit":"5dd672c58c322e142b0e187da82267ffde606450","subject":"Do not run Tempest in octavia-grenade job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5dd672c58c322e142b0e187da82267ffde606450"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2020-03-06 11:20:50.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-06-02 06:44:07.000000000","tz":0},"subject":"Pick stale amphora randomly","message":"Pick stale amphora randomly\n\nAvoid getting the same amphora all the time if its loadbalancer\nstuck in PENDING state and could not be failovered.\n\nStory: 2007340\nTask: 38877\n\nChange-Id: I2f680a00be695c3828166e6803882765af79b79e\n(cherry picked from commit 0f7d03feca4e3a4796c9c1a5a55d8da2adc44ebd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ae03653afcbbe1ef74b9c2f603caf4c3f2071e2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ae03653afcbbe1ef74b9c2f603caf4c3f2071e2a"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
