)]}'
{"id":"openstack%2Foctavia~714995","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I5dec631d3e66e1dfb6b8741caea89938ee18a7d8","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"I5dec631d3e66e1dfb6b8741caea89938ee18a7d8","subject":"Stop \"dnf makecache\" in the amphora images","status":"MERGED","created":"2020-03-25 16:48:46.000000000","updated":"2020-03-25 21:36:13.000000000","submitted":"2020-03-25 21:33:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714995-1585172019201-046e596a","meta_rev_id":"9692406e0de093aaaa444112b4f25d9a84a485cb","_number":714995,"virtual_id_number":714995,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-25 21:33:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":2,"date":"2020-03-25 17:24:11.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":2,"date":"2020-03-25 16:51:07.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":1,"date":"2020-03-25 17:24:11.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-25 16:51:07.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-03-25 17:24:11.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"},{"updated":"2020-03-25 21:33:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"852ece78fd2ab8ca98dfb8c5847e5b7a0166e8d7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-03-25 16:48:46.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"49eb3507ee397c4033c1b33fee188a969e64a30e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-25 16:51:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"579ae63261d1b464933ae62a111bc84c222c4cca","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2020-03-25 17:24:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1d41d0c001898404b2b679919dfdbcbd233b1c65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 19:27:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d52d550a10d428fa37871c53e3849f1 : SUCCESS in 10m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7eeb1963e784f44b6f1a9c904310b9b : SUCCESS in 11m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf08705d3e524f69a6772ed58321b748 : SUCCESS in 6m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/56cfd563e04645c3bb5182c486a015fe : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8b225a9fa5347a5b133bf5e2af4e3f1 : SUCCESS in 5m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/15a2c1fcb4c14a7fa353fbb6e4bbc0e1 : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d4522cf8c5f442b845d7663a62bb1ee : SUCCESS in 11m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9d71f5ef22694c969e63773b10f6b21f : SUCCESS in 6m 58s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9366963208b046458b218ce7502abb9e : SUCCESS in 10m 32s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/bc78cdb7fa1944f18f100f2fa9e2f245 : SUCCESS in 6m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ce00da11a70645ab8301cbf06668737f : SUCCESS in 9m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/53d47756646348eeb7fc3c2a20df889d : SUCCESS in 41m 28s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/0b298f50f39c4190891969e0ab4b5036 : SUCCESS in 49m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4966f0b4bd16474387d1f6a863b91705 : SUCCESS in 1h 43m 50s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/03d835b2e99e40ba9f2975c90d730d5c : SUCCESS in 1h 44m 36s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3705be75401a470ab1e5302a0cf87dae : POST_FAILURE in 2h 31m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6e199957fedd4392b8341ee984b5ad57 : SUCCESS in 1h 30m 01s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/84192d70275d4fa29c83ca6440eb4ca4 : SUCCESS in 2h 07m 21s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9c1f7f83c4014564a9fc60be70c3186e : SUCCESS in 1h 16m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/26a345bd259b44b69a0ca4ca830b4fb7 : SUCCESS in 57m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f396dbf252b64d7985c173755ea76caa : SUCCESS in 1h 49m 48s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/92c75cb6738043be99ad7a7c33e79118 : SUCCESS in 1h 02m 30s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/178a7fbdf2ff48dc9b53300c47b87168 : SUCCESS in 1h 04m 23s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/24ff95e415564a1db8b9cf1301a86063 : POST_FAILURE in 1h 22m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b0cac1633c4d01a9b2bc20e178e33bdce2613574","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 19:28:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"67e5fdfb4f145f9eb5bdc9e7b9dfbf67178701fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 21:33:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/649c7befb14e4c5ca4054164723b1d8b : SUCCESS in 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7779496977c84af7ae45583de18e5972 : SUCCESS in 6m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/de94a7b4fc944ae180c27c1a1a6fff07 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae6298961da04719b14cb5c5d865511e : SUCCESS in 6m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/079aa5a8a24c4ee7a2d7e1af817e3bd4 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40faddd9488b4872934ed0202e493cd4 : SUCCESS in 12m 03s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1895f73ac3a04dd5b462d8485ab16c44 : SUCCESS in 10m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/72bf2ed3fb6d4b05901154c812d82644 : SUCCESS in 9m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/46b3c796f4d044fa81f5d129aefe3666 : SUCCESS in 44m 12s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/679fa0ce14334b56bac7c856e4dc00e6 : SUCCESS in 47m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ad74208f4b8f471996ad9983b7633798 : SUCCESS in 1h 49m 01s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e56d8a3157b34d178ea3b6a9de1a96b3 : SUCCESS in 2h 02m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f6f3f6f2277743faab574fa6095efabf : SUCCESS in 1h 20m 49s","accounts_in_message":[],"_revision_number":1},{"id":"e0005c9907d2d94407ea9875c29e9e59214bb671","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 21:33:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9692406e0de093aaaa444112b4f25d9a84a485cb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 21:36:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd64597ee7364ef3b4e06718e29586bb : SUCCESS in 2m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"26dc84f537f11307c1ed6f5f59edbff48bf933ac","revisions":{"26dc84f537f11307c1ed6f5f59edbff48bf933ac":{"kind":"REWORK","_number":1,"created":"2020-03-25 16:48:46.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/95/714995/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/95/714995/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/95/714995/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/95/714995/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/95/714995/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/95/714995/1"}}},"commit":{"parents":[{"commit":"cb55a8cf78605c3d0b9c483b922f0156a6b3286f","subject":"Remove the barbican \"Grant access\" from cookbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cb55a8cf78605c3d0b9c483b922f0156a6b3286f"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-03-10 23:40:19.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-03-25 16:48:46.000000000","tz":0},"subject":"Stop \"dnf makecache\" in the amphora images","message":"Stop \"dnf makecache\" in the amphora images\n\nOn base operating systems that use DNF there can be a systemd timer\nthat attempts to call out and update the dnf cache every hour.\nThis patch adds an diskimage-builder element that disables this\ntimer as most amphora will not be successful in updating this cache\nbecause they do not have network access or DNS services avaialable.\n\nChange-Id: I5dec631d3e66e1dfb6b8741caea89938ee18a7d8\nStory: 2007408\nTask: 39019\n(cherry picked from commit c8b044af4a82a670e1a6ee61fa9773e4f7fe5774)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/26dc84f537f11307c1ed6f5f59edbff48bf933ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/26dc84f537f11307c1ed6f5f59edbff48bf933ac"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
