)]}'
{"id":"openstack%2Foctavia~715094","triplet_id":"openstack%2Foctavia~master~I5f109a4e9ac2e31939ff28b655ffb00c1c02b417","project":"openstack/octavia","branch":"master","topic":"project-ptl-and-contrib-docs","hashtags":[],"change_id":"I5f109a4e9ac2e31939ff28b655ffb00c1c02b417","subject":"Ussuri contributor docs community goal","status":"MERGED","created":"2020-03-25 23:41:13.000000000","updated":"2020-03-30 11:10:25.000000000","submitted":"2020-03-30 11:08:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715094-1585566499980-93ed6149","meta_rev_id":"bed564927cf5e7fb741379f9aee59ef48c7169aa","_number":715094,"virtual_id_number":715094,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-30 11:08:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"recommended":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"all":[{"value":1,"date":"2020-03-27 18:56:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":2,"date":"2020-03-26 09:51:16.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-03-30 10:50:47.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":1,"date":"2020-03-26 20:55:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-03-30 10:50:47.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-26 09:51:16.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-03-26 20:55:06.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-03-27 18:56:59.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2020-03-30 10:50:47.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2020-03-30 11:08:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b4624df280a50493d4a4d837df1858464ef2d540","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-03-25 23:41:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7a8727e8dc08ab8100f4db90d74e38a04ea0532","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-03-25 23:42:19.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7d823159b516339087747ae44d4ed39dabe8d98b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-03-25 23:44:10.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"543b5b5f681adc5af5ccc694ccaaea748c36b839","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 23:55:00.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee7678e471a2457ca924c6886dd60657 : FAILURE in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f497e0381b6c4372bd884abc8167ce9b : SUCCESS in 10m 23s","accounts_in_message":[],"_revision_number":3},{"id":"5a87cd207a9b2e2b71e130ab1e76abf2c77f233f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-03-25 23:55:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b0131dbf116b64c5ae700c151d900bdeab19a44f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 00:10:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5cee91355354cda9a875978575f684e : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a56191cd31043069edf0677b0ca0735 : SUCCESS in 12m 53s","accounts_in_message":[],"_revision_number":4},{"id":"eb7ce6407268e72b4bcdd116c0e75be8a8232cd7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-26 09:51:16.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM. Thank you.","accounts_in_message":[],"_revision_number":4},{"id":"8b715c34b69a77b281bb0aa4a7a81d2ffce5a7aa","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-26 20:55:06.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"075f85fa67bb9aaa3064ed64d5421f59410afc48","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-03-27 18:56:59.000000000","message":"Patch Set 4: Code-Review+1\n\nLooks beautiful! Thanks Michael!","accounts_in_message":[],"_revision_number":4},{"id":"3a48a574c0ac7f80d3ec0a2c90ceb42adb8cb826","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2020-03-30 10:50:47.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"47c15a1ad5d3d7f1636bc70df23126297049c68b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 10:50:59.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"879ff9f54176d101fbd6ca56538316e877078842","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 11:08:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"5fac072e62ca8bb3ec5ceca3a73bfec9dbb1728d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 11:08:19.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16a5db17ab9f4a698f6e0829d794062b : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4bad41c36314f81b2ec4ba7d8869d5c : SUCCESS in 12m 17s","accounts_in_message":[],"_revision_number":4},{"id":"c088f9490ce7dfe3cabd14cc7d493e8f823197a1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 11:10:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08da09ceedaa4ac69368efd3be749ce7 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"399331b01a4d599674ad8f901b3ad530bc13df25","revisions":{"ba26fe1a65317a1df0cdfeefa9fef691cdaa0729":{"kind":"REWORK","_number":1,"created":"2020-03-25 23:41:13.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/94/715094/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/715094/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/715094/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/715094/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/715094/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/715094/1"}}},"commit":{"parents":[{"commit":"70e244fff58a5badf8a7002f3ae0b7676bf15ae3","subject":"Merge \"Pick stale amphora randomly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/70e244fff58a5badf8a7002f3ae0b7676bf15ae3"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-03-25 23:38:23.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-03-25 23:41:12.000000000","tz":-420},"subject":"Ussuri contributor docs community goal","message":"Ussuri contributor docs community goal\n\nThis patch updates the Octavia contributor documentation to follow\nthe guidelines of the Ussuri cycle community goal[1].\n\n[1] https://governance.openstack.org/tc/goals/selected/ussuri/project-ptl-and-contrib-docs.html\n\nChange-Id: I5f109a4e9ac2e31939ff28b655ffb00c1c02b417\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ba26fe1a65317a1df0cdfeefa9fef691cdaa0729"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ba26fe1a65317a1df0cdfeefa9fef691cdaa0729"}]},"branch":"refs/heads/master"},"1cdfaa35086cc2f60f4b4584f2c7b4731c4a94ed":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-03-25 23:42:19.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/94/715094/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/715094/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/715094/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/715094/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/715094/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/715094/2"}}},"commit":{"parents":[{"commit":"70e244fff58a5badf8a7002f3ae0b7676bf15ae3","subject":"Merge \"Pick stale amphora randomly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/70e244fff58a5badf8a7002f3ae0b7676bf15ae3"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-03-25 23:38:23.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-03-25 23:41:20.000000000","tz":-420},"subject":"Ussuri contributor docs community goal","message":"Ussuri contributor docs community goal\n\nThis patch updates the Octavia contributor documentation to follow\nthe guidelines of the Ussuri cycle community goal[1].\n\n[1] https://governance.openstack.org/tc/goals/selected/ussuri/project-ptl-and-contrib-docs.html\n\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nChange-Id: I5f109a4e9ac2e31939ff28b655ffb00c1c02b417\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1cdfaa35086cc2f60f4b4584f2c7b4731c4a94ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1cdfaa35086cc2f60f4b4584f2c7b4731c4a94ed"}]},"branch":"refs/heads/master"},"468151e4879e393a3a7744a9cd8321f4d6b94b16":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-03-25 23:44:10.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/94/715094/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/715094/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/715094/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/715094/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/715094/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/715094/3"}}},"commit":{"parents":[{"commit":"70e244fff58a5badf8a7002f3ae0b7676bf15ae3","subject":"Merge \"Pick stale amphora randomly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/70e244fff58a5badf8a7002f3ae0b7676bf15ae3"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-03-25 23:38:23.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-03-25 23:43:39.000000000","tz":-420},"subject":"Ussuri contributor docs community goal","message":"Ussuri contributor docs community goal\n\nThis patch updates the Octavia contributor documentation to follow\nthe guidelines of the Ussuri cycle community goal[1].\n\n[1] https://governance.openstack.org/tc/goals/selected/ussuri/project-ptl-and-contrib-docs.html\n\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nStory: 2007236\nTask: 38542\nChange-Id: I5f109a4e9ac2e31939ff28b655ffb00c1c02b417\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/468151e4879e393a3a7744a9cd8321f4d6b94b16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/468151e4879e393a3a7744a9cd8321f4d6b94b16"}]},"branch":"refs/heads/master"},"399331b01a4d599674ad8f901b3ad530bc13df25":{"kind":"REWORK","_number":4,"created":"2020-03-25 23:55:56.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/94/715094/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/715094/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/715094/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/715094/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/715094/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/715094/4"}}},"commit":{"parents":[{"commit":"70e244fff58a5badf8a7002f3ae0b7676bf15ae3","subject":"Merge \"Pick stale amphora randomly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/70e244fff58a5badf8a7002f3ae0b7676bf15ae3"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-03-25 23:38:23.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-03-25 23:55:48.000000000","tz":-420},"subject":"Ussuri contributor docs community goal","message":"Ussuri contributor docs community goal\n\nThis patch updates the Octavia contributor documentation to follow\nthe guidelines of the Ussuri cycle community goal[1].\n\n[1] https://governance.openstack.org/tc/goals/selected/ussuri/project-ptl-and-contrib-docs.html\n\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nStory: 2007236\nTask: 38542\nChange-Id: I5f109a4e9ac2e31939ff28b655ffb00c1c02b417\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/399331b01a4d599674ad8f901b3ad530bc13df25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/399331b01a4d599674ad8f901b3ad530bc13df25"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
