)]}'
{"id":"openstack%2Foctavia~715702","triplet_id":"openstack%2Foctavia~master~I06211ef20131c64deba135123e53d87f3b5356a2","project":"openstack/octavia","branch":"master","topic":"update-hacking","hashtags":[],"change_id":"I06211ef20131c64deba135123e53d87f3b5356a2","subject":"Update hacking for Python3","status":"MERGED","created":"2020-03-29 15:08:44.000000000","updated":"2020-04-02 13:51:49.000000000","submitted":"2020-04-02 13:32:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715702-1585834342485-87f092a5","meta_rev_id":"edf15ae527febbc266fb4bfe5412ce790b52b69d","_number":715702,"virtual_id_number":715702,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-03-31 13:21:12.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-02 13:32:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"date":"2020-03-31 02:40:28.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-04-02 11:45:01.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":1,"date":"2020-03-31 12:20:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2020-03-30 23:17:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-30 14:54:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-04-02 11:45:01.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-30 14:54:25.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-03-30 23:17:16.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-03-31 02:40:28.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-03-31 12:20:28.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-04-02 11:45:01.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2020-04-02 13:32:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"05386d1cff794bda1fc28c585f6821ff47e91e78","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-29 15:08:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6decbc72831c74447167f14777f16caf5c760e17","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-29 16:05:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3aa8950171536bde0232cbca2a3af4436cae9aff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-29 18:11:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d2a322a7576646128ef1fccf660911eb : SUCCESS in 5m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a3216d00b01b4db1a869f081f42f2f08 : SUCCESS in 2m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6362be7c3d704b479baa057bf1f65888 : SUCCESS in 7m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a0c187af830a4d228db4e82eecde3470 : SUCCESS in 8m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13548d445f464c488622df0da809daa2 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0281ea6a28f4276b410de7756bbfff9 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d39c70ee8e714671963cc94858f45c33 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b242373a2c4b4c1eac046fd38b6983b6 : FAILURE in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e63d54f7b05483587e7412032aeb665 : FAILURE in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82e87dd303b84ab39710e65741e43c67 : SUCCESS in 5m 41s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f536f37dbe684e6b83289c2ed6d24849 : SUCCESS in 4m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b21c109ac1624a95aaba23e4a63e8f42 : SUCCESS in 7m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/45eea98a97244df0a3b54508d0815d72 : SUCCESS in 5m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a336057ef2f74547bdbe2d31e610d594 : SUCCESS in 41m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1a10532c8ef8415fa643baa3db9eaef6 : SUCCESS in 1h 38m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9bacabd3545d4840b681339babdf87f1 : SUCCESS in 53m 22s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/cfed43e7e6f24566b415300408b22dcc : SUCCESS in 51m 23s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b9de0a56db1c4683aa80722395d35154 : SUCCESS in 57m 03s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f1a0611aee72465a80ec089c3526766e : SUCCESS in 1h 44m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a11ce793f54d452f95f31baaaada37c3 : SUCCESS in 2h 00m 45s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/affeb0547d744cc58afbda9f7abe24c1 : FAILURE in 1h 01m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8ef51f61a76048cc8c49a85127408434 : SUCCESS in 1h 45m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2ac61541bfcf4742c83348519ea172f7d0258a81","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-29 18:24:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4caa2dd31df37d2d1086e4dddc80ec808ab2d15b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-29 20:26:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e779310bc6ce4a2a930448e3823d63f9 : SUCCESS in 7m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c74a5b1e82c64119a729cd3921721981 : SUCCESS in 2m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47c6f4631c9340549ccd646f5e122951 : SUCCESS in 9m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12919a9e7ddd4a8a8a3a4dec0ef960fd : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/069f2c8ae36f429bafc3d4b7ef901615 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ab7e7f1cde649289fe464bb6c0c335a : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ddeacf7fbefb4b38a92bbd72628dfe52 : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65e79ce241684ac994fe5d2178e4943f : FAILURE in 7m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff3fb97b525044babb7baf7bb279bb0f : SUCCESS in 15m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9576805751e54b8c964417a6ee99aba5 : SUCCESS in 8m 10s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4f45d5622e284da3b7b3254d7d19d575 : SUCCESS in 6m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d4efab86245e4a51b27534c5e3f1f790 : SUCCESS in 9m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e73f7cd0cd3545609537ad0e975053b5 : SUCCESS in 7m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3b9cb0c0713e445683c1c5d2a0f980e3 : SUCCESS in 50m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/78e502fa6b6a4e8f8208e6c83fbe6256 : SUCCESS in 1h 52m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7337f5e2d465426f85be4696c85c1db7 : SUCCESS in 1h 06m 40s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/94db28f09a3040dbb0171e93339c4eac : SUCCESS in 56m 58s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4c94f23a9e79474abcd0a4e057d98bca : SUCCESS in 1h 15m 25s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4fce12bfc3164715a3255a639d9209df : SUCCESS in 1h 58m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/78abfa147cc249fcad893891944e4a98 : SUCCESS in 1h 13m 42s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4224c2a4b78b4c099bdfc187a6b67b01 : FAILURE in 1h 15m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1b7aebfb08a143319e7b48e211d7bb2d : SUCCESS in 1h 41m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c5f94570fac5f7c13c121f54b8b7604731f59af0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-03-30 04:16:25.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM, thanks Andreas","accounts_in_message":[],"_revision_number":3},{"id":"a7b0a3cfa4ee1d89fa941ec65572e8510901d950","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-30 10:03:07.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"ecab8174170084a63ca502111822c2174d37b1e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 12:51:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/225e0edb56204e5a85e6dd7a6f614090 : SUCCESS in 7m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ea5b360c67b746b0a2d35cf1323c3caf : SUCCESS in 2m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/789e3b37ebe441dfb6b762c80476515a : SUCCESS in 8m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65e1f8a242cd4e9ab95196a874377f79 : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63b8f2443a194f05804c4fd2545a3ae1 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76c43fa3a94e46bcbbdf041fd9c93b3e : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/87f1e8db5f8542dcae71609a07b3403b : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56b1ee767e484dbea6d436ae788b4dc0 : FAILURE in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c92c9b286bbb4c0bbbbcb18400830103 : SUCCESS in 12m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7bd29df8a8ac424497cf2c32d4b87819 : SUCCESS in 6m 50s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ea1aee400d894beea80c56cde0f2e2cc : SUCCESS in 6m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/44a9c779074645fd8d0a595d63575163 : SUCCESS in 9m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bda3d805708747aeaab846dff876b9a2 : SUCCESS in 6m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9f01438c79574388b8d5edb8488589ff : SUCCESS in 44m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b0b4aa72984b4238a150160e8eac24bb : SUCCESS in 1h 55m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/157072102674491691b7c7cbbbea11da : SUCCESS in 1h 01m 44s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b14029a2c06f4ae1b59544ba0361c8eb : SUCCESS in 1h 08m 32s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/79faab27de644e5e916a4075d832dfe4 : SUCCESS in 1h 15m 44s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dacb7dc470104ec58ecceaabd5254498 : SUCCESS in 1h 45m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3a7a92f4d5824d34b07761a1acf61722 : SUCCESS in 1h 15m 52s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2852a07c52424a7e8d365d7224caea7e : FAILURE in 1h 13m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ba03c4960d3449a1b3eec8d78521ac52 : SUCCESS in 2h 08m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b86448de61817365909eab7dd24136bef028f72e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-03-30 14:54:25.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e20fd1644a8cb418b40d962ae5c1949ea77860c6","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-30 23:17:16.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"daddc0a4a1bdf6da9d6ee77b04efab5d2394aa7e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-31 02:40:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"792be58b1b1e1c2a9b75a07125054e8fe7993a07","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-31 12:20:28.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)\n\nI\u0027ve concerns related to the version cap.\nElse LGTM","accounts_in_message":[],"_revision_number":4},{"id":"ea90d8ed37003f4a636cb91d67ee4cc1761f4763","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-31 13:21:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9eb101097a919b41ab311362dbf5b28cfe7a74c5","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2020-04-02 11:45:01.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"fb053f839b90603ec6683af5ab76f6d55a510976","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 11:45:18.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0b66ffd7f5cb0af19a234f53903600cbaf799bc4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 13:32:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"738176ff0d31941c46e1837e5e21c64a2cb14083","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 13:32:22.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cc92286f52744e2c814a86eba23bdf94 : SUCCESS in 12m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fc13acea62db43069521fdd3e0242df1 : SUCCESS in 4m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57326c716c514765ab6f77f3adcb0a69 : SUCCESS in 12m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ce2052efc1c4285bcded8523634f77e : SUCCESS in 7m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66d4b8ae636347c8b2b5b12666bfd89f : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/08c4f81b0b8f4334be1af58d6c802a57 : SUCCESS in 6m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecab42efd465423d8795e8b6ecccd300 : SUCCESS in 12m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a540d53685f3412f8d2d6a48b49f1ad4 : SUCCESS in 7m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e610d139f42940eba22b6412a2d6ecd3 : SUCCESS in 7m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/09067b1148c3407490b879a78b45c5ec : SUCCESS in 44m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d13b1abac9804cd2999844a611f59e82 : SUCCESS in 1h 39m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6db0d06deaec42a4add2d79f444fb087 : SUCCESS in 1h 01m 36s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ac452f2b68f0475d88b2bba6ab05a9e5 : SUCCESS in 1h 02m 07s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1aab2c81ebec4b12822b08f20dbbb31a : SUCCESS in 1h 09m 03s","accounts_in_message":[],"_revision_number":4},{"id":"d8e6f6a9eb4b3d4c89a76e4faf462ecb3e53e47c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 13:51:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/d9632055d0494a1c8701d72374913e8b : SUCCESS in 1m 40s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4cb17a7f10f428ba30e657903359c09 : SUCCESS in 1m 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d33472a09d646158bc4b5add341bcfe : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"79464c0e048e47e865d1744dc198e486bf448180","revisions":{"8c6ff5c87eaf33c8b1ff82346afdc9bcf1eee085":{"kind":"REWORK","_number":1,"created":"2020-03-29 15:08:44.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/02/715702/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/715702/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/715702/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/715702/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/715702/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/715702/1"}}},"commit":{"parents":[{"commit":"0475acaab13640cf8732f4ff66d6d6e127cfc0d9","subject":"Merge \"Fix TESTING.rst example\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0475acaab13640cf8732f4ff66d6d6e127cfc0d9"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-29 15:02:01.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-29 15:08:37.000000000","tz":120},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 2.0 which\nsupports Python 3.\n\nFix problems found.\n\nUpdate local hacking checks for new flake8.\n\nChange-Id: I06211ef20131c64deba135123e53d87f3b5356a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c6ff5c87eaf33c8b1ff82346afdc9bcf1eee085"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c6ff5c87eaf33c8b1ff82346afdc9bcf1eee085"}]},"branch":"refs/heads/master"},"96e7d90fcebc0fd0b3b9fce464245c829379b8bf":{"kind":"REWORK","_number":2,"created":"2020-03-29 16:05:41.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/02/715702/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/715702/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/715702/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/715702/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/715702/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/715702/2"}}},"commit":{"parents":[{"commit":"0475acaab13640cf8732f4ff66d6d6e127cfc0d9","subject":"Merge \"Fix TESTING.rst example\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0475acaab13640cf8732f4ff66d6d6e127cfc0d9"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-29 15:02:01.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-29 16:05:36.000000000","tz":120},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 2.0 which\nsupports Python 3.\n\nFix problems found.\n\nUpdate local hacking checks for new flake8.\n\nChange-Id: I06211ef20131c64deba135123e53d87f3b5356a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/96e7d90fcebc0fd0b3b9fce464245c829379b8bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/96e7d90fcebc0fd0b3b9fce464245c829379b8bf"}]},"branch":"refs/heads/master"},"8a03a0f4ae6999aa72a2ecef233cc421151ed4b0":{"kind":"REWORK","_number":3,"created":"2020-03-29 18:24:14.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/02/715702/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/715702/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/715702/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/715702/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/715702/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/715702/3"}}},"commit":{"parents":[{"commit":"0475acaab13640cf8732f4ff66d6d6e127cfc0d9","subject":"Merge \"Fix TESTING.rst example\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0475acaab13640cf8732f4ff66d6d6e127cfc0d9"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-29 15:02:01.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-29 18:23:35.000000000","tz":120},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 2.0 which\nsupports Python 3.\n\nFix problems found.\n\nUpdate local hacking checks for new flake8.\n\nAdd test-requirements.txt to doc building, hacking is now imported and\nneeded for autodoc.\n\nChange-Id: I06211ef20131c64deba135123e53d87f3b5356a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8a03a0f4ae6999aa72a2ecef233cc421151ed4b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8a03a0f4ae6999aa72a2ecef233cc421151ed4b0"}]},"branch":"refs/heads/master"},"79464c0e048e47e865d1744dc198e486bf448180":{"kind":"REWORK","_number":4,"created":"2020-03-30 10:03:07.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/02/715702/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/715702/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/715702/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/715702/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/715702/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/715702/4"}}},"commit":{"parents":[{"commit":"0475acaab13640cf8732f4ff66d6d6e127cfc0d9","subject":"Merge \"Fix TESTING.rst example\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0475acaab13640cf8732f4ff66d6d6e127cfc0d9"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-29 15:02:01.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2020-03-30 10:03:05.000000000","tz":0},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 3.0 which\nsupports Python 3.\n\nFix problems found.\n\nUpdate local hacking checks for new flake8.\n\nAdd test-requirements.txt to doc building, hacking is now imported and\nneeded for autodoc.\n\nChange-Id: I06211ef20131c64deba135123e53d87f3b5356a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/79464c0e048e47e865d1744dc198e486bf448180"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/79464c0e048e47e865d1744dc198e486bf448180"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
