)]}'
{"id":"openstack%2Foctavia~716358","triplet_id":"openstack%2Foctavia~master~I8626c0338d4ad7d11bf5d56b022136cb4513d37e","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I8626c0338d4ad7d11bf5d56b022136cb4513d37e","subject":"Add availability-zone to loadbalancer object docs","status":"MERGED","created":"2020-03-31 16:20:54.000000000","updated":"2020-04-09 13:07:13.000000000","submitted":"2020-04-09 13:04:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716358-1586437467957-49baf0dc","meta_rev_id":"a62918572106924c2a76a61f74db3354d263ab76","_number":716358,"virtual_id_number":716358,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-09 13:04:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-04-09 12:49:27.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-04-08 23:26:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2020-04-09 12:49:27.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":-2,"date":"2020-04-09 12:49:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-08 23:26:26.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-04-09 12:49:27.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-04-09 13:04:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dbfabde6ddaf0dde3a6597da9eeb5fb0810593d8","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-03-31 16:20:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45afd9c38a995aa8844f2abdf50d7802fc43bcc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 16:38:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/82fac2341a3e41178015235b222c8a96 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2af6cb1c17eb45b5b2354a0cd1a076c1 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f80cc5ef07774df1ad53a38e9f9051ba : SUCCESS in 13m 48s","accounts_in_message":[],"_revision_number":1},{"id":"bc2d67d4abf0b24670c1523e840e8a2d55fd30e2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-07 16:43:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nThe examples look ok, but the api-ref is broken. It has the wrong field name and description.","accounts_in_message":[],"_revision_number":1},{"id":"9c7a1f7cbf7ce031e3122121ba192c6eadf2e720","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-04-08 23:19:51.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"aac97cde27d0cc1aaf01d9c592a9f4900454cd94","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-04-08 23:24:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"879a6f70c007fa5caef805fde26fd1ad4cf3b5bd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-08 23:26:26.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"0b011cb6b173a3d77f595d74d058609070987462","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 23:43:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1e5b3cde24c2483e84c09db0b5c7b804 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fd230edcfbb4405b7fcccffabd7df1d : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79c6f9f31c26418582c569d3d35c0025 : SUCCESS in 13m 06s","accounts_in_message":[],"_revision_number":2},{"id":"78dcadd001940b3dbc2158745ec5491cfe8fef0f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-09 12:49:27.000000000","message":"Patch Set 2: Backport-Candidate-2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"941759124aeecd399c8d505f107d49461d3899ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 12:49:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2f0c8406a73353f71543d8de10b1953729ec6409","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 13:04:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6648f855f7ca4c5e9bed6ae5559c1a49 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/615c9814abfe41828a73d58d1478d4da : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e1d01f0999849d981e8daeef8eb9e66 : SUCCESS in 13m 39s","accounts_in_message":[],"_revision_number":2},{"id":"9b2874b4a16400eda8eea56568eee27e680670ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 13:04:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a62918572106924c2a76a61f74db3354d263ab76","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 13:07:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/b99629fe03aa41ddb49c9080fae5631a : SUCCESS in 2m 15s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42c3c345b630403eba2096ea333974d8 : SUCCESS in 1m 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4f4804259f8bbf26c9b700dac43c7fc0cb47fa57","revisions":{"e1b25e2f4c5840dcdccf6605e8e0f7785ec59eb8":{"kind":"REWORK","_number":1,"created":"2020-03-31 16:20:54.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/58/716358/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/58/716358/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/716358/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/716358/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/716358/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/58/716358/1"}}},"commit":{"parents":[{"commit":"cd4a52f50d5965d285d7de3f528dbbcd17932777","subject":"Merge \"Ussuri contributor docs community goal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cd4a52f50d5965d285d7de3f528dbbcd17932777"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-03-31 16:19:53.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-03-31 16:20:31.000000000","tz":-420},"subject":"Add availability-zone to loadbalancer object docs","message":"Add availability-zone to loadbalancer object docs\n\nChange-Id: I8626c0338d4ad7d11bf5d56b022136cb4513d37e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e1b25e2f4c5840dcdccf6605e8e0f7785ec59eb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e1b25e2f4c5840dcdccf6605e8e0f7785ec59eb8"}]},"branch":"refs/heads/master"},"4f4804259f8bbf26c9b700dac43c7fc0cb47fa57":{"kind":"REWORK","_number":2,"created":"2020-04-08 23:24:23.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/58/716358/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/58/716358/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/716358/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/716358/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/716358/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/58/716358/2"}}},"commit":{"parents":[{"commit":"d47f164a6058fa6487fbc03e8d4f7e180447a679","subject":"Add ability to specify TLS cipher list for pools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d47f164a6058fa6487fbc03e8d4f7e180447a679"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-03-31 16:19:53.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-04-08 23:24:11.000000000","tz":-420},"subject":"Add availability-zone to loadbalancer object docs","message":"Add availability-zone to loadbalancer object docs\n\nChange-Id: I8626c0338d4ad7d11bf5d56b022136cb4513d37e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4f4804259f8bbf26c9b700dac43c7fc0cb47fa57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4f4804259f8bbf26c9b700dac43c7fc0cb47fa57"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
