)]}'
{"id":"openstack%2Foctavia~717154","triplet_id":"openstack%2Foctavia~master~Iedb7774bfb8d70ea307d6a513248e1fe2389fa34","project":"openstack/octavia","branch":"master","topic":"pool_ciphers","hashtags":[],"change_id":"Iedb7774bfb8d70ea307d6a513248e1fe2389fa34","subject":"Add ability to specify TLS cipher list for pools","status":"MERGED","created":"2020-04-03 00:05:15.000000000","updated":"2020-04-08 20:44:20.000000000","submitted":"2020-04-08 20:39:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717154-1586378370135-441aa25d","meta_rev_id":"48641198c3ca9720fa172fae0f94d419c0cc44e5","_number":717154,"virtual_id_number":717154,"owner":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-08 20:39:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},{"value":2,"date":"2020-04-08 16:45:49.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-08 18:50:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-08 18:50:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},{"value":-2,"date":"2020-04-08 11:29:17.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-08 16:45:49.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-04-08 18:50:49.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-04-08 20:39:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d94ae4d2a3d4ecde4dc7a6c712ec513b75779b7a","author":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"date":"2020-04-03 00:05:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f6d5285cfe38502ef4b1cecc5bd02af48846080","author":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"date":"2020-04-03 00:39:28.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"42f2bd0d17921c2798be994726c2114a7778aaa3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 02:11:36.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2d340526f57a46519e9713740c54e9b6 : SUCCESS in 5m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d0c01777de042d480c7c022a0ac4e89 : FAILURE in 4m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aed15e85ea6544b4827a0b555093ae2b : FAILURE in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cad77e34e44e403dbdc9e4a36a09ca68 : FAILURE in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59dbdf2d2a484662af8ffd78b96bbe50 : FAILURE in 6m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/639169818bde49839af795a690bfa9e7 : FAILURE in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7601fcf578f4e38933c44ade200eef3 : FAILURE in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d219bc393824f14bd8d107862bbd2e4 : SUCCESS in 14m 41s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/93a83fcc7439493ca0aa50baabfefc46 : FAILURE in 5m 14s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/748f739f28424d2aa7b85dd95dd7b805 : FAILURE in 10m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1e768744382d473bba6bb28761556e12 : FAILURE in 6m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5b49e755da0d4489a0f57d7d0b48ae83 : SUCCESS in 42m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ec048b9dd3d1499b95583a87460645a3 : SUCCESS in 1h 55m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e55b615c51794d54b27bfbb50a782e84 : SUCCESS in 59m 08s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1bdf6283a4fb494f805aa38954d45653 : SUCCESS in 51m 49s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fd6f2dcd31c247d3963d8af134fbbc87 : SUCCESS in 1h 00m 23s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/11e004e5771a4840adabbe53d5fd605c : SUCCESS in 2h 00m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/57e38aa8db804eca94a265684e37c1e1 : SUCCESS in 1h 58m 04s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/54908e48ae0e4be5bd0634f37ff0b117 : FAILURE in 1h 07m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7129427ebed842c6bccb9d4accd3ac36 : SUCCESS in 1h 52m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fe9e85f403f993508305b554f9f27315dffb2dca","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-07 16:07:54.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\nNeeds release note.","accounts_in_message":[],"_revision_number":1},{"id":"850ab637614ef3a9c8d41dcc15185dd1d430b416","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-07 16:11:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9aa37c4cafeb8f63d4b1bf40fab693c0f09fc7f1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-07 16:17:49.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nNeeds a release note and the example config has the wrong value.","accounts_in_message":[],"_revision_number":2},{"id":"485778d96769a04a22f18c29a2aca14dfdc2700b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-07 16:19:29.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nPlease also see my comments in patch set 1. Good work!","accounts_in_message":[],"_revision_number":2},{"id":"ecbd05ead35db3d9d948b8e983f715c37da38b84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 18:51:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/34e0c4dad0794309b487aeae7b3f94d4 : SUCCESS in 9m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/860c118c07724d7fa1a467ac7f44a916 : FAILURE in 9m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73f89da081444efd9e355c27559cab25 : FAILURE in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e50d0c615c24d0a9cab63a368e965fe : SUCCESS in 9m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c14038f3239497a8e9b0bd7547bdd2a : FAILURE in 6m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b9aa5554abad496a8f442b19a65a00ff : FAILURE in 9m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/513cf77f525844d1a02465c56a3c171b : FAILURE in 9m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88f58df3088f48e7bda31ae7778c2fdb : SUCCESS in 15m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f24f9fe9774d466b92027c50c8e9d8f6 : FAILURE in 9m 48s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7cb264552e174a4594c6e54bad354151 : SUCCESS in 11m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a967daa732814a8f8bfe371a60c4c542 : SUCCESS in 12m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/55ed4546506b4aa3b71f074be6d477f2 : SUCCESS in 47m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0e80b07f94db477291ff9a3db85d1bea : SUCCESS in 1h 42m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1ab2268c92d9405c9bad6116a0eea507 : SUCCESS in 1h 07m 36s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0d2ef2a7669145a187966386a0d9f491 : SUCCESS in 55m 17s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f552beb4318f478bb0748f3274f3e1af : SUCCESS in 1h 09m 34s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1731ed25d543445b974cd430f90c5fee : SUCCESS in 2h 19m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7395c9a2afbd4ecbb1581d4fbf692594 : SUCCESS in 1h 59m 13s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/48714ae0b03d4f4799d36e9a78473cdd : FAILURE in 1h 08m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c61a1767cf254ab68fe5bae680df8d07 : SUCCESS in 2h 04m 48s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/tests/functional/api/v2/test_provider.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_base.py\n  Comments left for invalid file octavia/common/context.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_load_balancer.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_noop_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_driver_lib.py\n  Comments left for invalid file octavia/amphorae/backends/utils/keepalivedlvs_query.py","accounts_in_message":[],"_revision_number":2},{"id":"92a43749f3883527a68af70c6b425fff80f42680","author":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"date":"2020-04-08 01:37:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7389a6475b0274ef0c822b664869782389e6fc77","author":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"date":"2020-04-08 02:30:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8485ab1282177cc7b801e308ea3575f387ff71eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 04:35:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0e4d4215181a4039af42012d01d8aacd : SUCCESS in 8m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad0dc157cdc849a8af8ed32ceb0bb275 : SUCCESS in 9m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19067d0e7455476ea5351830746bd995 : SUCCESS in 9m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee94fa2c15cd4278913a2be237963dc1 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46fb32ebf3484fe2a04870ca5a1d01fe : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a15b51c547f04a64bea20cdb7fe9d2b4 : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed46eb302cd34e14a3c4aef7dc712ec3 : FAILURE in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d573243f40d54a9b9196f37aa00d06ee : SUCCESS in 11m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84b54a18d57a45909c848a2b73e1113f : SUCCESS in 7m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/cd7866cfef4a4e3c8d95217b7a4c1c3e : SUCCESS in 6m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/77ad01f02b0b4b62b96e91590f511211 : SUCCESS in 9m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f122c16a4caf459eb10c33975f25fd0a : SUCCESS in 7m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/250cee7fcda54937b841e9e1e774afb3 : SUCCESS in 40m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/de4d39781a38481ca0f8a0ff5b5955ba : SUCCESS in 1h 53m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4f70e81f41a042fba1fa6a46435ec680 : SUCCESS in 58m 47s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7354aba04bae46b5832b9d4f4fcb5ed7 : SUCCESS in 1h 00m 45s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a15a9a7f72354fb898811f7dc9627932 : SUCCESS in 1h 02m 55s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1b482f3cfe0a4131a8044e8f5fa6af63 : SUCCESS in 2h 00m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8bdf3c948d6742528803adbac09fbccb : SUCCESS in 1h 51m 31s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/237c10d44e1a46e99278712e62034454 : FAILURE in 1h 04m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/597025a272604557b8532f36533334b4 : SUCCESS in 1h 44m 11s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/tests/functional/api/v2/test_provider.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_base.py\n  Comments left for invalid file octavia/amphorae/backends/utils/keepalivedlvs_query.py\n  Comments left for invalid file octavia/common/context.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_load_balancer.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_noop_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_driver_lib.py","accounts_in_message":[],"_revision_number":3},{"id":"9b726d748fbd8947bf79eb02d92dc58562390f86","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-08 11:29:17.000000000","message":"Patch Set 3: Backport-Candidate-2 Code-Review+2\n\nLGTM. Thank you, Dawson!","accounts_in_message":[],"_revision_number":3},{"id":"5390e743ec1be4734342a19bf9d6276fdd4a2382","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-08 11:43:24.000000000","message":"Patch Set 3: Code-Review-1\n\nSorry, I need to revert my vote. I think we need to extend the split listener side too (octavia/common/jinja/haproxy/split_listeners/*). I missed this in https://review.opendev.org/#/c/711376/ too :/","accounts_in_message":[],"_revision_number":3},{"id":"37018f9b40140f4f34c446e79ea66c9a27b0d0f2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-08 16:45:49.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"fe2167e094cd6742a2e38cf2dbd8449d814107c7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-08 18:50:49.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"b79ae4bed0c3b9d09129ca49a92558632b5718bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 18:51:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f4a1c6b4e8008e7cd24c33a711232daf4b019b61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 20:39:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/91d40b6d40b4413d9f938626c46ef7ad : SUCCESS in 6m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bab4dfd9cf91486fac7f8cb6cbdf0a0e : SUCCESS in 10m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a7de0930a094dcea9b9d1cb14bb5fc1 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d0fe49b4da74371b96c5381258a4b85 : SUCCESS in 6m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/41c1d15e37be4794a5d78f861ccc5298 : SUCCESS in 8m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6554eb6540e44e2990c8d2476e17b570 : SUCCESS in 13m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9dda091672524126b1537394cd6386b7 : SUCCESS in 9m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/15e369adf23546ae903f67f0ef05a4e0 : SUCCESS in 9m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6985be6a3d47488e9b6f8d98ff5df55a : SUCCESS in 49m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/058438e6883f436392ebfa7cccd75c48 : SUCCESS in 1h 46m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/61e1ffdd9ff840e9a95b1fb29139c4e7 : SUCCESS in 1h 05m 50s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a58a26d273ed4bd8b77e41c253cbb926 : SUCCESS in 1h 10m 19s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b602f985c0bd44079efcb03743a4e92d : SUCCESS in 1h 10m 57s\n\nWarning:\n  Comments left for invalid file octavia/tests/functional/api/v2/test_provider.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_base.py\n  Comments left for invalid file octavia/common/context.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_load_balancer.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_noop_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_driver_lib.py\n  Comments left for invalid file octavia/amphorae/backends/utils/keepalivedlvs_query.py","accounts_in_message":[],"_revision_number":3},{"id":"d1b0fa5838d4f619227d0c73c91b5d2231c9190c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 20:39:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"9fdb0ecfcfeb7f495ca6b9526e3bf2678f92f915","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 20:44:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/5300c946996b4ef2a82168b584a4bb26 : SUCCESS in 3m 52s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f4372af3a7048a48e4ac914b33fb937 : SUCCESS in 4m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb1341df155b43c59ad0a5e564294a09 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d47f164a6058fa6487fbc03e8d4f7e180447a679","revisions":{"1c0e31889dc4f537717f713f286a0eb726b13f22":{"kind":"REWORK","_number":1,"created":"2020-04-03 00:05:15.000000000","uploader":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"ref":"refs/changes/54/717154/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/54/717154/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/717154/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/717154/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/717154/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/54/717154/1"}}},"commit":{"parents":[{"commit":"5a4fdeaa6c74dd6df10d59cb39c8b5e6088e2181","subject":"Add ability to set TLS cipher list for listeners","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5a4fdeaa6c74dd6df10d59cb39c8b5e6088e2181"}]}],"author":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-04-01 15:30:55.000000000","tz":-300},"committer":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-04-03 00:02:57.000000000","tz":-300},"subject":"Add ability to specify TLS cipher list for pools","message":"Add ability to specify TLS cipher list for pools\n\nPools can now be each be assigned an OpenSSL cipher string with the\nfield tls_ciphers.  A new configuration option, default_pool_ciphers,\nspecifies what cipher string to use for new tls-enabled pools\nif one is not explicitly specified at time of creation.\n\nChange-Id: Iedb7774bfb8d70ea307d6a513248e1fe2389fa34\nDepends-On: I77da6f14063877af0077f2c12df1aab5d5ead187\nStory: 2006627\nTask: 37172\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1c0e31889dc4f537717f713f286a0eb726b13f22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1c0e31889dc4f537717f713f286a0eb726b13f22"}]},"branch":"refs/heads/master"},"e9e4fd1dd81ae75fae1ca1ee77d35d1e56319c16":{"kind":"REWORK","_number":2,"created":"2020-04-07 16:11:12.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/54/717154/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/54/717154/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/717154/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/717154/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/717154/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/54/717154/2"}}},"commit":{"parents":[{"commit":"cd176e55c55441f306eae244a29dd19e9de8cc47","subject":"Add ability to set TLS cipher list for listeners","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cd176e55c55441f306eae244a29dd19e9de8cc47"}]}],"author":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-04-01 15:30:55.000000000","tz":-300},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-04-07 16:10:58.000000000","tz":-420},"subject":"Add ability to specify TLS cipher list for pools","message":"Add ability to specify TLS cipher list for pools\n\nPools can now be each be assigned an OpenSSL cipher string with the\nfield tls_ciphers.  A new configuration option, default_pool_ciphers,\nspecifies what cipher string to use for new tls-enabled pools\nif one is not explicitly specified at time of creation.\n\nChange-Id: Iedb7774bfb8d70ea307d6a513248e1fe2389fa34\nDepends-On: I77da6f14063877af0077f2c12df1aab5d5ead187\nStory: 2006627\nTask: 37172\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e9e4fd1dd81ae75fae1ca1ee77d35d1e56319c16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e9e4fd1dd81ae75fae1ca1ee77d35d1e56319c16"}]},"branch":"refs/heads/master"},"d47f164a6058fa6487fbc03e8d4f7e180447a679":{"kind":"REWORK","_number":3,"created":"2020-04-08 02:30:49.000000000","uploader":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"ref":"refs/changes/54/717154/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/54/717154/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/717154/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/717154/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/717154/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/54/717154/3"}}},"commit":{"parents":[{"commit":"78863cf7ee27b8d4932da162c5f87c96ed3005b8","subject":"Merge \"Don\u0027t register cli opts on import\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78863cf7ee27b8d4932da162c5f87c96ed3005b8"}]}],"author":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-04-01 15:30:55.000000000","tz":-300},"committer":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-04-08 01:59:56.000000000","tz":-300},"subject":"Add ability to specify TLS cipher list for pools","message":"Add ability to specify TLS cipher list for pools\n\nPools can now be each be assigned an OpenSSL cipher string with the\nfield tls_ciphers.  A new configuration option, default_pool_ciphers,\nspecifies what cipher string to use for new tls-enabled pools\nif one is not explicitly specified at time of creation.\n\nChange-Id: Iedb7774bfb8d70ea307d6a513248e1fe2389fa34\nDepends-On: I77da6f14063877af0077f2c12df1aab5d5ead187\nStory: 2006627\nTask: 37172\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d47f164a6058fa6487fbc03e8d4f7e180447a679"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d47f164a6058fa6487fbc03e8d4f7e180447a679"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
