)]}'
{"id":"openstack%2Foctavia~717885","triplet_id":"openstack%2Foctavia~master~Ifa5ad0194f7ab2df891a439e3d380cf783169a54","project":"openstack/octavia","branch":"master","topic":"test","hashtags":[],"change_id":"Ifa5ad0194f7ab2df891a439e3d380cf783169a54","subject":"Do not merge testing","status":"ABANDONED","created":"2020-04-07 00:30:03.000000000","updated":"2020-04-07 15:26:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5a25128ea08e11a372ead08cf3eace04afa13505","_number":717885,"virtual_id_number":717885,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-04-07 00:41:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-07 00:41:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2592f56baf1ae9e21344dcabc170f2a733cd1a74","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-04-07 00:30:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"880e9b2b1b6f69b429f50bff08662612833cd055","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 00:41:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ebfebd2de615456c828d4005f38a50fc : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6dd1e9e0b7c4752923f3480f547c5d1 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4b7740053d24b47be669864c06cc0e2 : SUCCESS in 11m 20s","accounts_in_message":[],"_revision_number":1},{"id":"8f488b2879c2cdca2c4a88a72bdc32eff8b8afef","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-04-07 15:26:11.000000000","message":"Abandoned\n\nThis ended up being a storyboard integration issue.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6f76a72cf4b6589d185045b43c6d2653a6658db0","revisions":{"6f76a72cf4b6589d185045b43c6d2653a6658db0":{"kind":"REWORK","_number":1,"created":"2020-04-07 00:30:03.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/85/717885/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/85/717885/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/717885/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/717885/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/717885/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/85/717885/1"}}},"commit":{"parents":[{"commit":"5cbe0453f6dbe37a30dc4d1fcd7f0cfdbc352be5","subject":"Merge \"Add F5 Networks provider driver by SAP SE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5cbe0453f6dbe37a30dc4d1fcd7f0cfdbc352be5"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2020-04-07 00:29:44.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2020-04-07 00:30:01.000000000","tz":-420},"subject":"Do not merge testing","message":"Do not merge testing\n\nThis is testing some odd gerrit behavior.\n\nChange-Id: Ifa5ad0194f7ab2df891a439e3d380cf783169a54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f76a72cf4b6589d185045b43c6d2653a6658db0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f76a72cf4b6589d185045b43c6d2653a6658db0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
