)]}'
{"id":"openstack%2Foctavia~718192","triplet_id":"openstack%2Foctavia~master~I814d4e1abf3cee686b9f4511373edd46d1f5f4e2","project":"openstack/octavia","branch":"master","topic":"devstack-lib-octavia","hashtags":[],"change_id":"I814d4e1abf3cee686b9f4511373edd46d1f5f4e2","subject":"Remove Neutron SDN-specific code","status":"MERGED","created":"2020-04-07 19:12:23.000000000","updated":"2021-03-10 11:21:50.000000000","submitted":"2021-03-10 11:07:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718192-devstack-lib-octavia","meta_rev_id":"f6c3f9fd473f70fe35e6469c14a84254e8689529","_number":718192,"virtual_id_number":718192,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-10 11:07:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2021-03-06 06:54:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2021-03-08 07:53:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-05 22:14:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2021-03-06 06:54:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-16 22:48:52.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-09-10 20:23:12.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-09-29 22:02:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-08 07:53:39.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"}],"messages":[{"id":"6731299f798976b9c477542bdea593ecc26f7d51","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-07 19:12:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9dcddb771ceec862bb5552db59404e30ed4b8662","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 20:24:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6767fac4d0c47e7afc162a77c553e39 : SUCCESS in 9m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7875b79448504f34b562c2b3c63792d6 : SUCCESS in 11m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/deb5cdf6264e4eb290b199399afebfab : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d4c77970a7c4a6ba2a9daa17e9ce247 : SUCCESS in 7m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a5e7b9ac3f944b3994435bf5d0034a90 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41a55982565040339f35a042d0d1616d : FAILURE in 6m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca776f83670e461b8e638e49e6d6b121 : SUCCESS in 10m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/15bf94b2df7e4fb2a5e0ee8be1464ec8 : SUCCESS in 7m 30s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/70ac135eee0f4e56b5f856d3e9118e64 : SUCCESS in 9m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/08019692cd634c70821dbb3d477ce228 : SUCCESS in 7m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5ea6825cf80d4a7a9d4223215449922f : FAILURE in 30m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/666b6bee6763474b9b97d1bdff984d6e : FAILURE in 12m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a199017ec2f64d69941089479fc062e1 : FAILURE in 12m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/da2801cab1c94fe385c56cd3fbfa4fcd : FAILURE in 11m 16s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6e67b2df6d8145f9a0dc89e0b9cd44cb : FAILURE in 13m 27s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ecc71f3e692e413a894fccfae6056c61 : FAILURE in 10m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a6e4a06a57c74f2abc9f99b47207511a : FAILURE in 12m 53s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4be69e4a7f95408b86024d676c4f3699 : FAILURE in 1h 02m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a852f8caf3a64269849896e3777f1e53 : FAILURE in 10m 13s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/tests/functional/api/v2/test_provider.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_base.py\n  Comments left for invalid file octavia/amphorae/backends/utils/keepalivedlvs_query.py\n  Comments left for invalid file octavia/common/context.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_load_balancer.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_noop_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_driver_lib.py","accounts_in_message":[],"_revision_number":1},{"id":"d0624ec8bb149d1a7aef48f2561ec14e58a90b24","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-08 15:33:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a10aa02c6a7547d8602b329494dab6bcb786b4a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 17:49:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8308a64045e74b73ae620f9d77c8f301 : SUCCESS in 11m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc939a418b194f16a2442b53ae632ecc : SUCCESS in 10m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a704a072e2c4ca697e141d8c6b50cf2 : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07e99e9e1339492d864dcb1b9a1bb1ac : SUCCESS in 6m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a62b2efef3fd4fca99fee8b32edcc72d : SUCCESS in 7m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61c7e7ab9f4a4e6ca668fb2887a432ce : FAILURE in 7m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/295bcc17cc874696b113e9d851b0f0e5 : SUCCESS in 16m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/05fc2b6f9ea6461a820ac5f3ebec5055 : SUCCESS in 8m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f6ef5ea4fa77446bbfaf95b5fd898795 : SUCCESS in 9m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b12fb676468e40afb97872fab4015f24 : SUCCESS in 9m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9dc0f9122d7645ce97ea48cc71e60d50 : FAILURE in 40m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ea05cb0a4820488cb01d75c570e1d747 : SUCCESS in 2h 07m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/30cc6e35ade9418d9613463888d7983f : SUCCESS in 1h 11m 22s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/df01edb5c32a4c998f3a033ca0bf3a78 : SUCCESS in 53m 31s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0ec0ae1c92be4140aca2bf9cb7f53630 : SUCCESS in 58m 11s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ecede16760c84420b26f0ef69763a40f : SUCCESS in 1h 47m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1b0a458570e54151be04e81d37f5c08f : FAILURE in 52m 54s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/313223ec8b59486b9d63403b4a3e7163 : FAILURE in 50m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/eacb648671064c64a31d0aabf4d48794 : SUCCESS in 2h 02m 51s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/tests/functional/api/v2/test_provider.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_base.py\n  Comments left for invalid file octavia/common/context.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_load_balancer.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_noop_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_driver_lib.py\n  Comments left for invalid file octavia/amphorae/backends/utils/keepalivedlvs_query.py","accounts_in_message":[],"_revision_number":1},{"id":"cfdf4b8ba50fb930010771847778546928393dd6","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-08 22:21:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"50f8970f022a9a66748cf809a5a88b62667b2145","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 00:48:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/454fcee38fb0413baee55ec990b1a141 : SUCCESS in 10m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62d07ecdcd964269b1c84e89d184bfa7 : SUCCESS in 11m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0c6f95003b648d09b7dda2c22b8eeaf : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36a026e534ad46898d7142d15afe8816 : SUCCESS in 7m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/92bc69d8398c4b03a27bae626c28a8f5 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/352cb51520604f60884f85769219809a : FAILURE in 6m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b439f59322bc470796ef2a1faf29ab72 : SUCCESS in 17m 17s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/69a4c417a0ca4abea5baeb13dc5eda28 : SUCCESS in 8m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7af77b20dc1b4ccfab715c61697d1726 : SUCCESS in 10m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/73927be00e0b4708beed731a04d13e82 : SUCCESS in 9m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/676b9c4de32547f99c6134f7ba108f59 : FAILURE in 33m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ade492b9155b471190ebf006e92bc35b : SUCCESS in 1h 39m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d29e858ed1b3457db0fe404036b84134 : SUCCESS in 53m 32s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8a2e67aeef614527bee8399965269deb : SUCCESS in 59m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b5d3f6a185da4931bb4ac10d1580eb61 : SUCCESS in 1h 08m 00s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9620f1e1324d420b9c9a3d0cb676c000 : FAILURE in 2h 17m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7d22e8dd8c62443d97eb5d589e524631 : FAILURE in 1h 01m 14s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4caeb4aeaf1444b492392ce799b7005f : FAILURE in 1h 01m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b8f1c05e843a48f9a1c57394ac0384ea : SUCCESS in 2h 06m 52s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/tests/functional/api/v2/test_provider.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_base.py\n  Comments left for invalid file octavia/common/context.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_load_balancer.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_noop_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_driver_lib.py\n  Comments left for invalid file octavia/amphorae/backends/utils/keepalivedlvs_query.py","accounts_in_message":[],"_revision_number":1},{"id":"518046ae40028a3ce7d8bd5b67187f0149e5e189","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-05 21:33:03.000000000","message":"Patch Set 1:\n\nrecheck\n\nneutron change has merged","accounts_in_message":[],"_revision_number":1},{"id":"694df84b4a8a014aa5e01c973c32d2bb476754dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 23:42:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96445a46d6084f35a9a02ee05c36fb6d : SUCCESS in 11m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78ffafddf2f24ea995a46abbc7984610 : SUCCESS in 9m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a023d80ea0d34febba205f1e8ae858ad : SUCCESS in 8m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/449fea18b745488eab0736ce76b772d4 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a212dfab08cb4b399c3481b88b61f0b0 : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96cce57b07de4912ae80db65305f1450 : SUCCESS in 15m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/021fb1cb0d194f5da8a18dfe86abffb7 : SUCCESS in 7m 19s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/539aab1ff47d49738cb18a98cffc0efc : SUCCESS in 10m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/15ee2ecf75e74db38d646b182e3a042a : SUCCESS in 9m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f171f911ddad42e6b373d3c180022357 : FAILURE in 28m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/942ca41d687849bba8a232d95d4aba32 : SUCCESS in 52m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1bb037afe1084948abf480538888e331 : SUCCESS in 1h 01m 51s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/fe1cbf70a9da48e3ac00a24411ed468a : SUCCESS in 53m 23s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/21599aac24d04190ba1034ebb0acb7c0 : SUCCESS in 41m 46s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6a9b6b3f927042b5959fa6616945560a : SUCCESS in 1h 54m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8cdd6e429cdd45c189a69bfd152d7c7e : FAILURE in 55m 18s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f15a5e1dfcf04446a5f1020f23874fa7 : FAILURE in 55m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ea081a0608dc4a4e91131209b1ce86f3 : SUCCESS in 2h 03m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4b341a2de47b940a3f499eabd5b273db8fd31e42","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-19 21:26:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e255413f996967478e1f2714990a06019592e0a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 23:40:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9182b8ed8cb842fabdf3f478a352870a : SUCCESS in 10m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b659b7813de4ca09dd8e1482db5e83f : SUCCESS in 11m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/950d7fd7264240e09e29a5af1c0fa588 : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6172962c30694a23b488fe62ca693576 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21f34f35378d4a60b2a1626c061333f9 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2a6149b648e42c6a3134b202dbb0bca : SUCCESS in 14m 24s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d05a39e16159488eb3f26b20fbdd50aa : SUCCESS in 6m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/642f6c0f959a4018a9fa895b7363a92f : SUCCESS in 9m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6e9e2433f7164cde9cc3c09b9b8596ea : SUCCESS in 8m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/04937f24062944e890fcc28433440946 : FAILURE in 43m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5fe83d9454104ff39169c8df76de6f17 : SUCCESS in 1h 47m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/260034b52d294074b60be25c89eb240e : SUCCESS in 1h 05m 48s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/36137a13e0104d9ead7d94dfb4f4529b : SUCCESS in 1h 04m 07s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/65327138e0664949b8c8978f19f9c2f3 : SUCCESS in 1h 14m 02s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c86f14f011544cea82e6d0c19a1090cc : FAILURE in 50m 00s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ce7e1f840e914d77bb8db14bdef89e84 : SUCCESS in 1h 53m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5caf99ec3e384385b1a828fb39d2bae1 : FAILURE in 53m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/18c99660fca74f939dc789878b3e3b0e : SUCCESS in 2h 06m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"df92459bf535e6d5bca8afba2529dcf1463bbb14","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-20 00:37:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f117fce6e1147c42fdc3cea1b13965140a15f90c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-13 20:50:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"581922979cf01487ccc1e2256736851790b02317","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 22:53:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/006f156ecb684f9898127185e5a775f8 : SUCCESS in 11m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e8468585d4f4f03a193da34ff722044 : SUCCESS in 11m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77a4ebe8a409442ca60339aa0e7a351c : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c045e769c40a43ad9d32c530dc11b71c : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed78f068b05941539f86a9ac73303980 : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b731c782a094628b729caff4a75b9dc : SUCCESS in 11m 56s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/56a9447176314d97920631ab4826a22a : SUCCESS in 6m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8f4e01c9239444839e1c502850659ee4 : SUCCESS in 6m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/19bc29a64c6f4f6a8127d9fe13b70539 : SUCCESS in 6m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1f0d7748a4d6437d94a201ff801b04d6 : FAILURE in 33m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/111486051a984a438919e175f39c90ea : SUCCESS in 1h 41m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/03d225cdeb7141ceb64e32a9275c50c5 : SUCCESS in 1h 00m 06s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a713c7becde347ec98e4671811041537 : SUCCESS in 55m 40s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b188b48d2a654f8da544c7d9c4490c11 : SUCCESS in 1h 09m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f3a9a892c0434f3db7492b11c35a2278 : SUCCESS in 1h 28m 27s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4729a5de9c484b9aafd5fbc307e85b32 : SUCCESS in 46m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2b477c0cfa044070af4229c992cb82ed : FAILURE in 53m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c1cb6c16eb934d459bd900e4c8b6e7a9 : SUCCESS in 1h 44m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/68b407581f71465f959ba074cc14679a : SUCCESS in 1h 57m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"617a18479da32784db0454a2f7b6cab42e3ec1d8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-16 21:23:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ee9261a7df8d94d2fc0012857d548a5eb21ba5c7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-16 22:11:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"94cc58d3b49c4b7768f3b6c1294cd00915ca69b6","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-16 22:42:09.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"399a4d66ee637f57c23d60c1c88ab40a035e7c5e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-16 22:48:52.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"878b7a935b8045c8a941b9d6152db8c167ec4e3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 00:26:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97cb04982e854976906f09aa4daee1d6 : SUCCESS in 10m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9ea8f69daf440d395b5d7c5af359493 : SUCCESS in 13m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b32810a7f7f546fc8d1cfbcab4bc9f94 : SUCCESS in 7m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cc050a337a74753a6785c267d16ab58 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0613831ef47464081040506e6ac1504 : SUCCESS in 7m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d88805adccb14fa0bbd53034e024c713 : SUCCESS in 12m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/838cd297ad204c9993f22395079a4b2d : SUCCESS in 7m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3b77cfc6337241b480917e1298f81eb5 : SUCCESS in 8m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a52a58b739b3412792ba42ade1aea3f3 : SUCCESS in 9m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8c897ef513b44ed098cd8e5a1b18fc5d : SUCCESS in 48m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/90671dc85e904c1c8402adc02d942248 : SUCCESS in 1h 08m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/72af964bf7a244d99d2f750ed53f6941 : SUCCESS in 33m 06s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/66e01af4d88a4701b193ffbdc63db8f3 : SUCCESS in 31m 30s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5dc43d8ba99248628c2bf08d7e819a4e : SUCCESS in 33m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4f84fd74bce146ff8cee5ae1926e1306 : SUCCESS in 1h 36m 26s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5c86ba7fb77d45178e27f9015ef84ba9 : SUCCESS in 46m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8a55ece08d5746038f4b9ad7934c2238 : SUCCESS in 51m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a2a4a234617e4bcbbf42ef59ebd6cf7d : SUCCESS in 54m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/697d4ff9caa64a719b54ec92e837894b : SUCCESS in 57m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/58e31ff4da1e43d29fe411e7949c0222 : FAILURE in 1h 16m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a041e761d89d9360f1fd6918007bdb3269fe083d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-17 14:29:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"82cc5831af70a7cd010b8c98af6a94d556841a77","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-17 15:37:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c15dc6456a7617d56f58d081cd814845eb222f79","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-17 21:12:20.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5cd5b05acd2396208ee126b3af9c0de6edc08ff9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 01:20:27.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f2e7d38ac314cd0baf8474c0dfc3744 : SUCCESS in 9m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06e59d9c68d24870ace5bfe936276b60 : SUCCESS in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12b030b3fe2249b7b22f87d5b18094df : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a04b1784f91946d391ab1afdabb73e00 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd81696123b84b309603ce4887b99799 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d4b595485ca4a65b2c1e3a0a92ae2b7 : SUCCESS in 12m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/32fb98fffff440ea863312239877eca9 : SUCCESS in 6m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7a8cf6742e47406facb43523d16c09ba : SUCCESS in 8m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4c299db115c042679d9ecbbfdd12d096 : SUCCESS in 8m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e14e489550584e55aef808aa4bdafb53 : SUCCESS in 45m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ec199ca1095b491c91ce8038c927075e : SUCCESS in 52m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5a0641ed49934b3cac8dbc03d9100df6 : SUCCESS in 39m 16s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7c335bc139f04daea62393b62271e588 : SUCCESS in 37m 48s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5188e110bb874b74834f2578ee2b2468 : SUCCESS in 38m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/70058de989ab4ede9e9595d3b13fc909 : FAILURE in 44m 38s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/918b3b5015cc4893939491188c0295b6 : SUCCESS in 1h 12m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/74f7cdc75a2d49cc884a6fa44404bd7b : FAILURE in 59m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/affeddb1e55d4b62acd18efad6848769 : SUCCESS in 1h 03m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9d507b698c574c2d9d2a152eb005e75f : SUCCESS in 56m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ad65e8874d9640fa9b018b488ee0a08a : SUCCESS in 1h 11m 17s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2e70afd0d3c6d371342d60af334b5e750d6341f1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-18 15:09:52.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"e5c475c690f2983564afcabfbdbfd7289e09f58b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 16:12:47.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54f2c1f4cb43417b902589054a062fa1 : SUCCESS in 8m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/268def4dce8e4f899491adcf050a1f1d : SUCCESS in 8m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f610b45f06634482bb819bf291e6d572 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bd80846cdd5421fb6d43cfffffa77fa : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e94d4db670d462ca93990ef252b7c40 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a4a4e104f11434b9eeb3a8caabc735a : SUCCESS in 10m 16s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/47f96e2713694492baffadea5b8a2159 : SUCCESS in 4m 29s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/84e00983effe4945a7ff6ca55ad769f7 : SUCCESS in 7m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ff24e34ee0e3404e83ddcf7b3b01af61 : SUCCESS in 11m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dc446dc171a345518a75ae06cb1a89bc : SUCCESS in 41m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9e6f5fcdf99b47f1977ecfd3619c9950 : SUCCESS in 57m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/972584c8ac7f409f93e52ca98fcc2c10 : SUCCESS in 31m 14s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/bc1b334625d34abcb11005f17b09bb09 : SUCCESS in 41m 43s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b947df4f2bd14aef8e36950d74fcbfd2 : SUCCESS in 29m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d564bc4ae1de47a29efb57ba79bf27ca : FAILURE in 48m 53s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ddc2129283b14cb0b569e8a25859dc3d : SUCCESS in 44m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9832f8afcf87421cacbca570c749c66f : FAILURE in 44m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3997ec6dacb54cd8bb8a1fc7896779e4 : SUCCESS in 43m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ae54e077bdd5472bbe61fd7dab470b14 : SUCCESS in 38m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a55ecee1936d4554865771bd7c8d869e : FAILURE in 33m 41s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f9fb38043abd9acf83110ed82afafc5a0f2b87ee","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-20 14:51:32.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"0c408571a782ec2029dc52dfd985601b4e9928cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 16:11:33.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c338131d5e57491c96715fe7ebacc7e5 : SUCCESS in 9m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0466b9a4ef21465594caf6c51d46fe7b : SUCCESS in 11m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d828d4c5e654c15a15a8c9ed37f59c6 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19e320bdd2b44e3aa92cc6739bf60e59 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57a228496ca645b8ac3c31463c16bc63 : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f4958038363484da92028cf5c74dbcd : SUCCESS in 13m 57s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/82e7fb3bffd3466090b2b05babac19cd : SUCCESS in 5m 51s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/436dd7d60afa41be965614aa79b4674e : SUCCESS in 8m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f0622c5784664204a263faa868f84cf6 : SUCCESS in 10m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/410233d5aefc44a880a791d769ab6054 : SUCCESS in 43m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4a951fcfb995443d83f8a2926e67b094 : SUCCESS in 52m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f2d92396a04e475086cdc04410912e9c : SUCCESS in 48m 09s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ce8bd3d857894002a525431db9d58718 : SUCCESS in 47m 27s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bdaae5e4ab184b17b94e20d6ace875fd : SUCCESS in 33m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a14f4cae1c3e4eb2b1c8c9a3fe3dcfef : FAILURE in 47m 03s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2ba85fd8d18c43fbb9ca2a157e6c5ba9 : SUCCESS in 56m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c8727de594e94efcbefa8d960742fde5 : FAILURE in 43m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/cf0fb37e036b4c96ab2d2fd8ce93f763 : SUCCESS in 44m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/54262df01e7f432aae1594def1ed1a88 : SUCCESS in 47m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2f5104cf991d46758130bc3bf805d612 : FAILURE in 37m 33s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"510a39a190a98c20a5f6ceeffa6bd2eb257db7f6","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-20 16:35:55.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"8d6d5f4c88f391f4b157593a16dc646f944500b4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-20 19:00:13.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"00b3ce38dcdeea3cfac0d575e358e5552e150059","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 20:37:20.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f7ad66792074306b881ee4dfcae6979 : SUCCESS in 8m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99e5dbefa5744bec9b315826c4ae90a2 : SUCCESS in 9m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42a571e6fd4744d9a29fa410fee8d666 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7475812bb90a4a449a6a783c54484d5f : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82d0ebd9e93b48cb8d4f83b9ce975956 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5330c36c3a334e87890507b03c1d45b0 : SUCCESS in 11m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1a6039df3c8e4a0aa3068c5534794093 : SUCCESS in 6m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3d61ce0dd6404154977ee983d9e4346c : SUCCESS in 6m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/194ebb0354574429b33c9f45d6533bea : SUCCESS in 6m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/be357bb7e59e4fb49e94735764d3299c : SUCCESS in 1h 01m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/936de23b3fba4657a5b5c4189f8f6eb7 : SUCCESS in 53m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a502cfe59b0345ab894d8310b6c699bf : SUCCESS in 42m 50s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9c9f0b84d2b340b6bcac6b716082068b : SUCCESS in 38m 43s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e6d3d154f6804a3a9d6f9209fa643c78 : SUCCESS in 39m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0d154dc99fe84725bf9e3f83839107ad : FAILURE in 50m 08s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5187553a9e0d41279683b978f694f443 : FAILURE in 10m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/be890b6b99b4420ea36a9f53240ae85c : FAILURE in 44m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/19a0661972014d6da97774d38fc09723 : SUCCESS in 43m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/051f4ae105834ac3a44462c84f64cfd0 : SUCCESS in 41m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/cbf5ec4deeb848948e610d12e913a389 : SUCCESS in 1h 31m 50s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"60c4c2c2ecf46c8cefb4ab2848c4365702ea4a2a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-20 21:02:56.000000000","message":"Uploaded patch set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"9f56ea047ea7b85409b771e754a5df2f4806c890","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 22:22:06.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84e8676c3f684377b51e77ab2e0c36e8 : SUCCESS in 8m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/428335d8e31645a8ba3ffe8fb14c0877 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ba7c728443a4464a0a8e9ae985fdd36 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c73081e7a4541e9b013ad8eddc7289b : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdf2bd593d4f4f5abb790b48a6b16c38 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d54d1b2426de4782966c5120b980cd61 : FAILURE in 7m 04s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/763d320306c64dcbbca52d2ef6df59d9 : SUCCESS in 5m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c9e73a03468243ea8fae62161c185549 : SUCCESS in 7m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/017f6408d07d4a72b9d3c6b77678776c : SUCCESS in 6m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4019408720c347db85cd4253acc78d94 : SUCCESS in 44m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/129785d0841d4356a8b8d6e36ee8787f : SUCCESS in 1h 05m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/395e61eafbcb40b8b683118ee650c101 : SUCCESS in 30m 22s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e73d73f1baac45cf9e07bc6480a9ac0a : SUCCESS in 39m 47s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/00ddac02a00f4cf6b7afd5401686eb9f : SUCCESS in 29m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/846b6a2e32b14b7f9180e1c02be9fc5e : FAILURE in 45m 49s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/27e811d6d2f44f319e312d273e38ee9a : SUCCESS in 1h 12m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/54aefee88fe74ba8b41d4fb9f5cd8a93 : FAILURE in 41m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a44f6a080c75434aa2fac2a0b37f2588 : SUCCESS in 39m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/76ecaf4e01d34646bffd8d8fc217a8f4 : SUCCESS in 41m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/533b86b8c1d74f04b9b67d9b9f6b0f20 : FAILURE in 36m 14s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f0d2f572c21272b73a2a6bd8cf830651ddea6c1f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-21 17:26:40.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"660bdabc76f626fc95e3e6f9b237751e5fe031d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 21:38:27.000000000","message":"Patch Set 8:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":8},{"id":"df1078d538e3fbf9381bfa80d0e134305c284851","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-21 21:38:30.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"168d7dd6e4ccd8297f1d3501fae957d513263733","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 23:13:18.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de78f48363d64baa821439953fee40d1 : SUCCESS in 9m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd58430992cb42168ffc7b5dd7561e7e : SUCCESS in 10m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/849e719c5b384a078945f89fae4abe59 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0ca9674197a4bc0ae121875ecf45793 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21bcdd18b46d44de8403537d3f36b9a1 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f679db3762fb4ed692e117ecfdea00d4 : FAILURE in 7m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9af3e72c251f463a88663e817a828c2e : SUCCESS in 5m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/00e3c0cf9a9042c386d7ea45f3456611 : SUCCESS in 8m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/97cf2a3bbd31463192aed853be5fa959 : SUCCESS in 7m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/19dbc3d7af5343138bb46dd1cdb304f4 : SUCCESS in 45m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b78d8d299c514880ba1a79e4fb169cef : SUCCESS in 1h 03m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/54e84bab2b7645a281eff578ea8826dd : SUCCESS in 46m 41s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/bd32505ee4154f58b370dda934873c86 : SUCCESS in 37m 05s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b10e1b553aa84cde86beadf191b3e6d8 : SUCCESS in 38m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1ebbdc48ab314e928f74d525ff3c32c3 : SUCCESS in 1h 31m 05s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b0e36069ee674ad5929687cd3b7359b9 : SUCCESS in 1h 06m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a9ffc95fb78946eb8ed444231c1ca553 : FAILURE in 1h 07m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/23dd44ed0423428a8281ed57fd69ca9f : SUCCESS in 1h 06m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b159911fa39548378753b5103a0fd7ce : SUCCESS in 44m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b1915cc11e5b4478906b667a8bd3e52b : SUCCESS in 1h 21m 13s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"a97aee71adeffe7d506d2ad51b428d27e9495bfd","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-22 01:10:57.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"8bd833eebb3116694fb48c432a1f347a0f4b0fb7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-22 01:11:34.000000000","message":"Uploaded patch set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"62bf49b938322d0f8450a55cc41f69c3b6fbf75e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 03:18:12.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2d6c600d02d494782c95d4c133005b8 : SUCCESS in 10m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b99e7b8caaf43b4990f12ad0cfa5afa : SUCCESS in 10m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ee2e67d4a8a4f989e29133052b63379 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d45d0d9dc1343168c241ef37da10276 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a0faccc21a64d818cd09736838629a3 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ebc37e663824260851bc42207764355 : FAILURE in 9m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b02b5bb6d1604bdaaf0098de95b42bb2 : SUCCESS in 4m 58s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8b2547269a7143859f7519b92913a0fe : SUCCESS in 7m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/78096c97fc53447b98382e38cf527b09 : SUCCESS in 7m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/de2256c326a3442d8770b2475dc368d4 : SUCCESS in 49m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8761a0d00dfc4cda87b5f0f255a12682 : SUCCESS in 59m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/121f76b4577d4bcc96f829b41bfd027c : SUCCESS in 53m 20s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/32d5e24f8fd142438d731f18286d4f51 : SUCCESS in 28m 48s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cd1cbb2b71d4410fa067318a2a57bc40 : SUCCESS in 31m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8756274c452d442186e73d8fdc5e60c7 : SUCCESS in 1h 55m 34s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0a3d8d441932414f840dd991382074ea : SUCCESS in 1h 11m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a9e814611be6402aae5362b0bd610c00 : FAILURE in 53m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/32fd630997a04933a65d1752929fffac : FAILURE in 47m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8620f01e08724a64a47ba060ea9a76e6 : SUCCESS in 40m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d4673e57e2e64c50befd205266a2874f : FAILURE in 37m 58s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"82f4012ecc7ebf1e4548c39167c28c9a5e0eaa37","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-22 19:51:38.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"8f1516fddfa23e3b1014a1fa7923f6e3098932a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 21:37:49.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c334f41f5d941d3a2dd2233d68c1ffa : SUCCESS in 9m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e95dc11798924cd0ad06f582dbe4d29a : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc1caa8e2d0b43d1984c5484067c3bd0 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17c7cc60cee04a9f9a89069b8795ceff : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54f26d7279ed49c3b41938d5c8a6cff4 : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/667b72a74ee04af0ae808dc11c78537f : FAILURE in 10m 28s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/14ae338b90ed4b429433e83319a484cb : SUCCESS in 6m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bb00dec4072040db84895f7f0878780a : SUCCESS in 7m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2feb2beccfbf4bd88c27506a13db8e2c : SUCCESS in 7m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9db7d9cc7a3b4186b151febfce78447d : SUCCESS in 48m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7deaf746a8a44109991d488326f6c648 : SUCCESS in 59m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e29f73ab2c794e949e5b6f5b0215937a : SUCCESS in 57m 48s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/12e0a527c9614197a201c639627423c3 : SUCCESS in 38m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/243e306f5d4447f1b613051f3162e558 : SUCCESS in 41m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/39cbe1bf2365456099290a90306f800d : SUCCESS in 1h 39m 23s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a49a8feb6c45410ea24e15b06c00d934 : SUCCESS in 52m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b32074e2a2ec48c38e65091a951ea809 : FAILURE in 50m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bb0e6b360ebd4668a564774e55f6876f : SUCCESS in 54m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fdc78969de524a9da5921141182d92aa : SUCCESS in 1h 12m 37s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3aa9061f90be41ff80e37aff99bdff6e : SUCCESS in 1h 28m 56s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"024cc9ecb40051618ebe7d3a432860ee62b52fc1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-26 21:08:51.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"d803ff69d0148868f16ad3c66df571c0e35bc0a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 23:21:38.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/129476ff646e45bc8b5e4156db5edc5f : SUCCESS in 9m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d081527ec3e4eebae235e39ed6609cb : SUCCESS in 9m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/776119c46c97465993217bb12db8209e : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4e05768ab564b57989c3dd7e24f09fc : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/513c4e76f9d44b228b826e8d17efbeb6 : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a38c654f95842f5adc60c08f9689bc5 : SUCCESS in 11m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b7b695ddcea94fcda12df0fc5d8a39d9 : SUCCESS in 5m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/82af34ea37ee43f1bca50a8dcd568ee6 : SUCCESS in 8m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/025fd5ab9af6465d8dd3862577d8ec78 : SUCCESS in 8m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9d7137a52c18463497f5b0345701fe17 : SUCCESS in 40m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d1cb66c6531547ca8db5e5a852510ff7 : SUCCESS in 45m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/44bfdd23b7634385b83c0b0d2a00672d : SUCCESS in 56m 01s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a7dc983b4add4166b5f8ed94f3492b43 : SUCCESS in 38m 48s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3a6ca7076e14402995ad72d201f9ecd9 : SUCCESS in 39m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ad6a9a4ecabb4a12b39b83c48e37e430 : SUCCESS in 2h 06m 47s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e46644df5d4448529d384ed695a970ef : SUCCESS in 1h 26m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/362cbc5603ff4516b5a7ece559de67bb : FAILURE in 45m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/56629d5f18cd45289f5c6b264e8c835c : SUCCESS in 1h 16m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a1ee9a53c7dc4222b8f32cb89759ea8e : SUCCESS in 45m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e89e0b97860d4c5689e67c888c99e116 : SUCCESS in 51m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d6b84d0ab7454c868d9850850a3f849f : SUCCESS in 1h 39m 08s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"190d9af13b1e7532733dbaba7fba87cefac4d9cb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-27 22:04:38.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"7910ccf28602734cdb1894eacbc52be3eb5831f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 22:38:25.000000000","message":"Patch Set 14: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1543162990e4e9fa2c2b3d550d5db72 : SUCCESS in 11m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50251bfc37da49e48360da4fac671170 : SUCCESS in 9m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/362bdabfa0624a54a5278cd17ea97203 : FAILURE in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5aa9b83979f4bc2af94fccd5a9d2742 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/839960dafba6411194501566d2b12f1e : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3fc778e505b41b982960a5bfdfd34e8 : SUCCESS in 11m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8857799a5c0d4ee3afebd6e56440f18d : SUCCESS in 7m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9e8252443e7f4c2fb783f943070e15e5 : FAILURE in 7m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/04014ab849354181a4681524b84a4f25 : SUCCESS in 7m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3ca593d6b5c6416f93170e2d33cf4430 : FAILURE in 9m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3b7372f91c0a45f1bda15ccd26e0b018 : FAILURE in 9m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0be90e494ff9402c92bb55685deaf3df : FAILURE in 10m 46s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6d3caf7113164e54baced6dca9827d0a : FAILURE in 10m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c0f8b6ddb6e945bfbe159d76624c1b66 : FAILURE in 7m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e55bbddcc743473fbab9cf204a0bf559 : FAILURE in 30m 35s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/700b640419764681b280a08ecab6ade9 : FAILURE in 8m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/35372a543f2b4e0ca8bc9cee4fc79779 : FAILURE in 10m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d343c312d21b421db03fcaf018ce2a0d : FAILURE in 9m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9bc514977efd41a7a0914b841a90852f : FAILURE in 8m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/08af3b9ada4e42a4809fadbb5655b582 : FAILURE in 8m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/01b38dacefd841a1a8c6d1db0512db60 : FAILURE in 8m 38s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"393efb634551fe1da4544d694e1f6ffb5baed85d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-28 17:32:39.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"13bb77034a72d3c709f8e3fe29404d21a8207c54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 19:26:28.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f1e367e77c94fbeb7f50c7fa928d901 : FAILURE in 11m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a336a3a16cd44536b4aa50f0e4001e12 : SUCCESS in 13m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c34597b522f4640a8daa859f8c1a281 : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64353a1d36774808b7a3d78888b9b1bf : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72afd83da59348bd95f9b2e536353974 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5c2211ee4344653bcfa95e6843a800f : SUCCESS in 17m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6b107c75c610424cbe2af3fd52c7de11 : SUCCESS in 5m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/658bbb9e319046ceb765dc82877fac31 : FAILURE in 11m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9407186f5250444c8756149fe43567a7 : FAILURE in 7m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9e31bc5eddfe417386a712fc8b7f950b : SUCCESS in 39m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/12d63e5f600d44e6a74cb257bcd9a827 : SUCCESS in 1h 04m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dadec98000dd47c1a3a45a55cbb86d27 : SUCCESS in 50m 36s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e44f35e65f9c4c229b6d54b2ffc5a21b : SUCCESS in 49m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4400bb8bc850437480339726110a0e37 : SUCCESS in 58m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/67929e6b2ba741eabacee21bb89a55ed : SUCCESS in 1h 48m 20s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0e44c51b7a49493f9d82c738fffb55c1 : SUCCESS in 1h 07m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4e55a3681b3a43f9886f2fa26f520874 : FAILURE in 1h 04m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7cca2ad45d3f4a6890ec97573aa2c455 : SUCCESS in 1h 16m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7388c3b912b74ad8b4d13879d0c5a2ed : SUCCESS in 1h 12m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/03070f7512b346cd97833bfb92efcc53 : SUCCESS in 1h 13m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4d66d5486c0b484b8709da40e0a7e08a : SUCCESS in 1h 19m 26s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"e21656129ffc724872bdb90bfe2f45f28508acff","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-01 00:43:08.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"c79acacd3b07fa259d6a94c46db6cacaec597302","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 02:22:52.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe9b14b46ed147acb1d931718dfc333f : FAILURE in 7m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a7add46e6ae4e44b7624e3ceb588ad6 : SUCCESS in 9m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/310e244ee8ea4da19dad5581cda2f72d : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63ab298b2a1a4efbb63b0e7df43e975c : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0adf2ca6079b4f879fd25b07924ad9e9 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/689c1b9ab6824b29846aef234814169c : SUCCESS in 10m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/246d54160c9842fcbacd4ec736b74c7d : SUCCESS in 4m 52s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b1024db60d224aa48697db8cdd691cb8 : FAILURE in 7m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/763c4a77b4c74a5d8285b422714a7f29 : FAILURE in 7m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/612b67b8b0524c3eb49affdb340268d6 : SUCCESS in 44m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/324d627c24894c13937f68ea4e3d5626 : SUCCESS in 58m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/77fccc8ac475455680cda5e5ffcff594 : SUCCESS in 31m 29s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3bfcdc19254645e3a4851813cbdc1837 : SUCCESS in 29m 01s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ec030c67828f4180b6be0450a4bccb29 : SUCCESS in 44m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b9afe92a2ef24cc9a9bf95660f9523f5 : SUCCESS in 1h 26m 32s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6780f942794b47198f042b4eb7fc4c14 : SUCCESS in 1h 02m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9242228d5f2747ec9eb84d4a26fe2225 : FAILURE in 40m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/54ac365dfdc94e00baefc0676671fec8 : SUCCESS in 51m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/61c0dfc5d0d5415eb91a7314e77428a7 : SUCCESS in 58m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5caefffed3da4ad4bc0cd3730d5f3b99 : SUCCESS in 1h 22m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/332efc86bb0c495a8038c84ba7c2b890 : SUCCESS in 1h 37m 06s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"293ca431f4d316945d05b3475984e54b9463ec5c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-01 16:32:26.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"8aba62764965616cf2f7cdc5e09c37cc2d30251c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 18:06:49.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c0f16c58d63473d8e45982e6d6ad80a : SUCCESS in 10m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b45a13bb5f4446879f4753a7f080fd4f : SUCCESS in 11m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2361d60d48624a28839ccd157a366842 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be67d8b2ff944119998dadeb73c73199 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46d219dff8bc49ce8355b54a48cfbf24 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5d0c62dbda24cccb859b550626fda3e : SUCCESS in 10m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b008270c44414a7491aba99011366328 : SUCCESS in 6m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7891e8b2b8f74b819327df957104a2d4 : SUCCESS in 8m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d3b7272c991747bdb6ab0e6ab1042e26 : SUCCESS in 7m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0eb09dfba3f44bebb03bae617dacd1e7 : SUCCESS in 44m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e8182388a51f47e489b3bb6ed9e36a9c : SUCCESS in 57m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/08f7ac48c0bb4449b147fd6ccf055226 : SUCCESS in 42m 19s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a76e4fff7cf9420a809a0b391f1850cf : SUCCESS in 31m 41s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/151cdd61083b4e91b1621455f8fa0b91 : SUCCESS in 32m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c2848435db1547748d416dc66011e66b : SUCCESS in 1h 29m 57s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/09af296a96e444f89c74878a766debff : SUCCESS in 1h 10m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/092429f83549469f915103029a30db86 : FAILURE in 1h 09m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3d3c2b403d36470db194c4f06da953d5 : SUCCESS in 1h 15m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1857246cdc994f929442518184222f7d : SUCCESS in 58m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/64ef137d5c8745b49f5e5476213ec5c8 : SUCCESS in 1h 05m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4150e3aac4e74a84b409511d629ec91f : FAILURE in 37m 38s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"d4b1600ee3eaf8936db560748dac78a2b1b52ece","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-09-10 20:23:12.000000000","message":"Patch Set 15: Code-Review+1\n\nOk, I think I get this. Basically, this function should now always exist so long as the two dependent patches merge, which means we should always be able to run this function. I am curious though -- is there a time when we DON\u0027T want this to run? Would it actually make sense at this point to remove the check entirely, and just try to run that function, and consider it a failure if it can\u0027t?","accounts_in_message":[],"_revision_number":15},{"id":"03e5264f97d835442a826fe263ffeb83d4dc6486","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-11 18:23:11.000000000","message":"Patch Set 15:\n\nAdam - I guess the check could go away next cycle as the function should always be present, as long as it \"dies\" if it doesn\u0027t.","accounts_in_message":[],"_revision_number":15},{"id":"b8e298484e505a2996881b7a910563ecb67828b2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-28 21:36:19.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"1782dc91cdfefb646556791ae4a4a6782f443e7a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-28 21:36:33.000000000","message":"Uploaded patch set 16: Commit message was updated.","accounts_in_message":[],"_revision_number":16},{"id":"283a7e0812373bdf5d31e2fc961f74c6652db22d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 00:14:18.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64514074a31b4b85b51e31deccb610b3 : SUCCESS in 9m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f40ece0ae1cf443e802549c7d28e9f49 : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29cc7a951f304458891f9424b4b3d964 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8489819d0a1949a7b4c9ddf016677da8 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3dc8e89148a5420e9e978da07f179703 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a6288b470904647bb6ea319a6ae5e2f : SUCCESS in 10m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/27824abf594f46a48a362ad1d4aac3c2 : SUCCESS in 6m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fc10f36fbcf24bf682d0f3a7c72f0076 : SUCCESS in 6m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/af41781b511a403aad93c08aa167b3fe : SUCCESS in 5m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ff2a45977b044017a47d6bf2975d167e : SUCCESS in 7m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dea473cb425c478cbe0f9909ce39e56a : FAILURE in 27m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/75372ee2ba7b490ab88fc1f44ce87418 : SUCCESS in 1h 30m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/26e97216beb14b8487f77aca67175af8 : SUCCESS in 46m 43s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/616fbda67ed94e3d92b5297d509777f3 : SUCCESS in 43m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c8ac72a12bc640cd87cbefd3e6864de7 : SUCCESS in 37m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8885830d3ffe4112912e48e7d57b4a7a : SUCCESS in 45m 24s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a7f3f8ae053046028e5002eb8fda698c : SUCCESS in 1h 44m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/192504aed7894bbfab36f0913506578d : FAILURE in 43m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c57fdec37f514aa58f52786ce0cdad6b : SUCCESS in 1h 24m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5ffd823711f844aa8a21b511e3f1a32c : TIMED_OUT in 2h 33m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1e3443cd7afd46d49715fd714457f3b1 : TIMED_OUT in 2h 32m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b00787add7024e76b1743d9caeb8fc99 : SUCCESS in 1h 57m 41s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"bcd16109b17a418256eae2a3075d6b5a36689552","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-29 18:36:08.000000000","message":"Patch Set 16:\n\nSo the noop API tests do need to enable the neutron plugin:\n\ndie \u0027Unknown network controller. Please define octavia_create_network_interface_device\u0027\n\nI\u0027ll change things back...","accounts_in_message":[],"_revision_number":16},{"id":"8e1235c36c479821209341a5ff427eed750ab2f1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-29 18:40:16.000000000","message":"Uploaded patch set 17: Commit message was updated.","accounts_in_message":[],"_revision_number":17},{"id":"c5c3b8fb102b4a530d721639ee165921ab7cd67c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 22:02:24.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a53d142dcdd4efea26e05de633b2e22 : SUCCESS in 11m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c86a135dedca4e0c8ce47fb5784e99be : SUCCESS in 12m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e18ba3ee0804ad89c297ae5cbcc2d52 : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b1fd300dd754cdfb70a0e7d5189bd09 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7473b5c8f7bd40749a3c52111b196231 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/888540ba09d84be3bb0dfc51f7f50b32 : SUCCESS in 12m 22s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/358e36b12812477bae02ea64c8a8de01 : SUCCESS in 5m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9387be5f8c0a4dd08526d1957a618375 : SUCCESS in 10m 20s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8ba5931edae2494b9920a2e34ec3e9d8 : SUCCESS in 9m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/06a12079b7d847f48b111224fc753454 : SUCCESS in 9m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c980be0c6e1d4bdd91e5e32114eab0db : SUCCESS in 1h 24m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/035f846d2b594cd6a53aa748796568b5 : SUCCESS in 2h 08m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3b0472cd3e764d54a12487c217cbdda1 : SUCCESS in 43m 53s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1ec0875f720f420ca505c8db3e640f17 : SUCCESS in 40m 28s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/da0255b50b85462fb4725a31fbdae711 : SUCCESS in 40m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/db4e5f2c0c1e440ebc6f2a071314e5d1 : SUCCESS in 1h 01m 13s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8ff48a93eff74790a6aff3ca8c9c8596 : SUCCESS in 2h 23m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4115a08ef50f4458aa301ce9136955d9 : FAILURE in 44m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0f28861f5f1348c48a90fc1da4cfd429 : TIMED_OUT in 2h 33m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/117b33bd7426486faab577ee8e5ed657 : TIMED_OUT in 2h 33m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/bfa139ad28fd4803b89de2251637afbe : SUCCESS in 1h 49m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/031f6a1465c542c9b494d9ace8c1392f : SUCCESS in 2h 26m 25s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"37862f2e3019765b8fa229195e3bb15267a4c816","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-03 20:01:07.000000000","message":"Patch Set 17:\n\nrecheck\n\nThis should be good to go, and was actually recently noticed on the -discuss list :-o","accounts_in_message":[],"_revision_number":17},{"id":"d04b485bd74299e457177ec19e10736beadbe080","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 22:49:52.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb10eac56cc3487c88decfa8376c9b0a : SUCCESS in 12m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13d61e122b9849ab95e29df8a5504e57 : SUCCESS in 11m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a0e043fc2ae4d0994dd5354612cefb2 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b40519be3d3a4584aeda0835be2ee5b0 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/594fd12590914547bff71e0415ff5f45 : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/488f9d74c68542e08a8df6c69f935af7 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e20a8ebf51342fd9c501a4000f26f62 : SUCCESS in 16m 01s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9e6f6c770a734a89b1c362ee21dfb759 : SUCCESS in 7m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/66dda01b07e2448db7a64a5569c0384a : SUCCESS in 8m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f4cac9401e1748b49a2990f4c8774324 : SUCCESS in 7m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5bfccb9b05be44b7bb28123926ee7e1e : SUCCESS in 10m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/24d45b54d97a4f598f57a05ad49980a3 : SUCCESS in 1h 16m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d553256bf98a48318a0d6e0cb8d17d0a : SUCCESS in 1h 49m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6f157fd64b0042d5ae2c66cc78d0c9fc : SUCCESS in 1h 01m 29s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f5ec76e497074aaeb0878b842a11536d : SUCCESS in 36m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ac7e3c984b51404699079cc8a91b8e1a : SUCCESS in 55m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cddd8e61d6ee48458d86b39b1d1952b0 : SUCCESS in 1h 07m 40s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/632efafed7d44e4e9870f9f40cfd92d6 : SUCCESS in 1h 31m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fa7d5577d0b34cb9891dfe27264da192 : FAILURE in 50m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/33ee8f7573654c1b8c2f3f6d37156244 : POST_FAILURE in 2h 25m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bc8a8b191ecb409eac0031527b3bff2b : SUCCESS in 1h 57m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7e7b215664384d62937c6ae54586b59f : SUCCESS in 1h 59m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ad153eb0a6f94f66b1fd7e78e95bf1cc : FAILURE in 36m 31s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"f5ca174f0f964d6d1fadfe53b37bddb0306c2720","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-05 22:14:31.000000000","message":"Patch Set 17: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":17},{"id":"ecdfef11f0a9136cd3b6b24906243f1312e82088","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-03-06 06:54:47.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"bf9e91c2115bc8d555f216cfd078c87cf61ce2e6","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2021-03-08 07:53:39.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"b7fb4cfd09a5aea697d67d590547166460093710","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 08:59:51.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"4e6a0ccd148e588f405322e0a8fab078b9b82f57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 11:07:10.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7d85f01dc614c6ba4a5668d06412444 : SUCCESS in 9m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d84404647f1f4270885196f262519c60 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fb0b98e3a49418aafcfed73babc73f0 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ab65671a70542a48372f14e2eb04a54 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab5d6d488c084dd1a566af8fc7102f8d : SUCCESS in 12m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fba46bb8b48b4d3fb6b408297abc483f : SUCCESS in 5m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3157440a6d5846259cd3586f9ec07ead : SUCCESS in 7m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/34bc020c21814cd380b1c9edc8eb98b2 : SUCCESS in 1h 34m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a6a537acb7484f14aa86a4462b214255 : SUCCESS in 1h 43m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b28c221c4ef648dd96310c5de72caa98 : SUCCESS in 37m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8c4eb9a0966f43ecbadd3ad36257f491 : SUCCESS in 36m 10s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ffc24d2ddabd412ebec4f5fa0ff77721 : SUCCESS in 33m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b654e220cd004014ba2ac060305895f2 : SUCCESS in 51m 31s","accounts_in_message":[],"_revision_number":17},{"id":"b56820f79ca7dded1a282f19822582a85b8ee9a8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 11:07:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":17},{"id":"f6c3f9fd473f70fe35e6469c14a84254e8689529","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 11:21:50.000000000","message":"Patch Set 17:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d3ed9117d114bdd92cecb576ad0408b : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"dcdf6315b31b2962fa8d0a5f080c369214bb222e","revisions":{"49d90b91e7506a56346ad4227af7743ce6512d5e":{"kind":"REWORK","_number":1,"created":"2020-04-07 19:12:23.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/92/718192/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/718192/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/718192/1"}}},"commit":{"parents":[{"commit":"8c6d64baa7c35f6c29ede2706f178dedb540bf37","subject":"Merge \"Add ability to set TLS cipher list for listeners\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c6d64baa7c35f6c29ede2706f178dedb540bf37"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-04-07 19:11:11.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-04-07 19:11:11.000000000","tz":-240},"subject":"Remove Neutron SDN-specific code","message":"Remove Neutron SDN-specific code\n\nMigrated to Neutron repository.\n\nDepends-on: https://review.opendev.org/#/c/717860\nChange-Id: I814d4e1abf3cee686b9f4511373edd46d1f5f4e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/49d90b91e7506a56346ad4227af7743ce6512d5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/49d90b91e7506a56346ad4227af7743ce6512d5e"}]},"branch":"refs/heads/master"},"988da564cba3077b04ea0e5d3a9557e81dcdbcf1":{"kind":"REWORK","_number":2,"created":"2020-07-13 20:50:48.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/92/718192/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/718192/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/718192/2"}}},"commit":{"parents":[{"commit":"8ae82a47e1928f99dd32f08785f876b3d1d9c3cb","subject":"Merge \"Stop to use the __future__ module.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ae82a47e1928f99dd32f08785f876b3d1d9c3cb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-04-07 19:11:11.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-07-13 20:49:52.000000000","tz":-240},"subject":"Remove Neutron SDN-specific code","message":"Remove Neutron SDN-specific code\n\nMigrated to Neutron repository.\n\nDepends-on: https://review.opendev.org/#/c/717860\nChange-Id: I814d4e1abf3cee686b9f4511373edd46d1f5f4e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/988da564cba3077b04ea0e5d3a9557e81dcdbcf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/988da564cba3077b04ea0e5d3a9557e81dcdbcf1"}]},"branch":"refs/heads/master"},"245061b7aa0fe7d5b7b82510ff1cb8ef2f1b9ba6":{"kind":"REWORK","_number":3,"created":"2020-07-16 21:23:50.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/92/718192/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/718192/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/718192/3"}}},"commit":{"parents":[{"commit":"8ae82a47e1928f99dd32f08785f876b3d1d9c3cb","subject":"Merge \"Stop to use the __future__ module.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ae82a47e1928f99dd32f08785f876b3d1d9c3cb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-04-07 19:11:11.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-07-16 21:23:22.000000000","tz":-240},"subject":"Remove Neutron SDN-specific code","message":"Remove Neutron SDN-specific code\n\nMigrated to Neutron repository.\n\nDepends-on: https://review.opendev.org/#/c/717860\nChange-Id: I814d4e1abf3cee686b9f4511373edd46d1f5f4e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/245061b7aa0fe7d5b7b82510ff1cb8ef2f1b9ba6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/245061b7aa0fe7d5b7b82510ff1cb8ef2f1b9ba6"}]},"branch":"refs/heads/master"},"29514f333f4af101235e3ea1f7335fa185265c2b":{"kind":"REWORK","_number":4,"created":"2020-07-16 22:42:09.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/92/718192/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/718192/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/718192/4"}}},"commit":{"parents":[{"commit":"8ae82a47e1928f99dd32f08785f876b3d1d9c3cb","subject":"Merge \"Stop to use the __future__ module.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ae82a47e1928f99dd32f08785f876b3d1d9c3cb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-04-07 19:11:11.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-07-16 22:41:50.000000000","tz":-240},"subject":"Remove Neutron SDN-specific code","message":"Remove Neutron SDN-specific code\n\nMigrated to Neutron repository.\n\nDepends-on: https://review.opendev.org/#/c/717860\nChange-Id: I814d4e1abf3cee686b9f4511373edd46d1f5f4e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/29514f333f4af101235e3ea1f7335fa185265c2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/29514f333f4af101235e3ea1f7335fa185265c2b"}]},"branch":"refs/heads/master"},"4b16ebd98234bbc55a9f48c3a089a75b9c187c1e":{"kind":"REWORK","_number":5,"created":"2020-07-17 21:12:20.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/92/718192/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/718192/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/718192/5"}}},"commit":{"parents":[{"commit":"8ae82a47e1928f99dd32f08785f876b3d1d9c3cb","subject":"Merge \"Stop to use the __future__ module.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ae82a47e1928f99dd32f08785f876b3d1d9c3cb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-04-07 19:11:11.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-07-17 21:08:35.000000000","tz":-240},"subject":"Remove Neutron SDN-specific code","message":"Remove Neutron SDN-specific code\n\nMigrated to Neutron repository.\n\nDepends-on: https://review.opendev.org/#/c/741719\nChange-Id: I814d4e1abf3cee686b9f4511373edd46d1f5f4e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b16ebd98234bbc55a9f48c3a089a75b9c187c1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b16ebd98234bbc55a9f48c3a089a75b9c187c1e"}]},"branch":"refs/heads/master"},"9396b5fefe099c6a0e70a27a911cb1777c2aa209":{"kind":"REWORK","_number":6,"created":"2020-07-20 16:35:55.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/92/718192/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/718192/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/718192/6"}}},"commit":{"parents":[{"commit":"8ae82a47e1928f99dd32f08785f876b3d1d9c3cb","subject":"Merge \"Stop to use the __future__ module.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ae82a47e1928f99dd32f08785f876b3d1d9c3cb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-04-07 19:11:11.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-07-20 16:35:41.000000000","tz":-240},"subject":"Remove Neutron SDN-specific code","message":"Remove Neutron SDN-specific code\n\nMigrated to Neutron repository.\n\nDepends-on: https://review.opendev.org/#/c/741719\nChange-Id: I814d4e1abf3cee686b9f4511373edd46d1f5f4e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9396b5fefe099c6a0e70a27a911cb1777c2aa209"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9396b5fefe099c6a0e70a27a911cb1777c2aa209"}]},"branch":"refs/heads/master"},"ebe85197f67303f1a4324daceb19b6f300f184d0":{"kind":"REWORK","_number":7,"created":"2020-07-20 19:00:13.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/92/718192/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/718192/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/718192/7"}}},"commit":{"parents":[{"commit":"8ae82a47e1928f99dd32f08785f876b3d1d9c3cb","subject":"Merge \"Stop to use the __future__ module.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ae82a47e1928f99dd32f08785f876b3d1d9c3cb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-04-07 19:11:11.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-07-20 18:59:56.000000000","tz":-240},"subject":"Remove Neutron SDN-specific code","message":"Remove Neutron SDN-specific code\n\nMigrated to Neutron repository.\n\nDepends-on: https://review.opendev.org/#/c/741719\nChange-Id: I814d4e1abf3cee686b9f4511373edd46d1f5f4e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ebe85197f67303f1a4324daceb19b6f300f184d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ebe85197f67303f1a4324daceb19b6f300f184d0"}]},"branch":"refs/heads/master"},"04fb50bb6c06c4ecdc9751249968c62820bca555":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2020-07-20 21:02:56.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/92/718192/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/718192/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/718192/8"}}},"commit":{"parents":[{"commit":"8ae82a47e1928f99dd32f08785f876b3d1d9c3cb","subject":"Merge \"Stop to use the __future__ module.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ae82a47e1928f99dd32f08785f876b3d1d9c3cb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-04-07 19:11:11.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-07-20 21:02:37.000000000","tz":-240},"subject":"Remove Neutron SDN-specific code","message":"Remove Neutron SDN-specific code\n\nMigrated to Neutron repository.\n\nDepends-on: https://review.opendev.org/#/c/741719\nDepends-on: https://review.opendev.org/#/c/742027\nChange-Id: I814d4e1abf3cee686b9f4511373edd46d1f5f4e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/04fb50bb6c06c4ecdc9751249968c62820bca555"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/04fb50bb6c06c4ecdc9751249968c62820bca555"}]},"branch":"refs/heads/master"},"26ec481d1535c58c972fd37199698af41ba7c1e2":{"kind":"REWORK","_number":9,"created":"2020-07-21 21:38:30.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/92/718192/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/718192/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/718192/9"}}},"commit":{"parents":[{"commit":"8ae82a47e1928f99dd32f08785f876b3d1d9c3cb","subject":"Merge \"Stop to use the __future__ module.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ae82a47e1928f99dd32f08785f876b3d1d9c3cb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-04-07 19:11:11.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-07-21 21:38:17.000000000","tz":-240},"subject":"Remove Neutron SDN-specific code","message":"Remove Neutron SDN-specific code\n\nMigrated to Neutron repository.\n\nDepends-on: https://review.opendev.org/#/c/741719\nDepends-on: https://review.opendev.org/#/c/742027\nChange-Id: I814d4e1abf3cee686b9f4511373edd46d1f5f4e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/26ec481d1535c58c972fd37199698af41ba7c1e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/26ec481d1535c58c972fd37199698af41ba7c1e2"}]},"branch":"refs/heads/master"},"1f738a50abaaf9fbeb588b9f381ee828d27798ba":{"kind":"REWORK","_number":10,"created":"2020-07-22 01:10:57.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/92/718192/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/718192/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/718192/10"}}},"commit":{"parents":[{"commit":"8ae82a47e1928f99dd32f08785f876b3d1d9c3cb","subject":"Merge \"Stop to use the __future__ module.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ae82a47e1928f99dd32f08785f876b3d1d9c3cb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-04-07 19:11:11.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-07-22 01:10:43.000000000","tz":-240},"subject":"Remove Neutron SDN-specific code","message":"Remove Neutron SDN-specific code\n\nMigrated to Neutron repository.\n\nDepends-on: https://review.opendev.org/#/c/741719\nDepends-on: https://review.opendev.org/#/c/742027\nChange-Id: I814d4e1abf3cee686b9f4511373edd46d1f5f4e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1f738a50abaaf9fbeb588b9f381ee828d27798ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1f738a50abaaf9fbeb588b9f381ee828d27798ba"}]},"branch":"refs/heads/master"},"f67af2c22b363bb4f46d7f550344656cf14fe39d":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2020-07-22 01:11:34.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/92/718192/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/718192/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/718192/11"}}},"commit":{"parents":[{"commit":"8ae82a47e1928f99dd32f08785f876b3d1d9c3cb","subject":"Merge \"Stop to use the __future__ module.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ae82a47e1928f99dd32f08785f876b3d1d9c3cb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-04-07 19:11:11.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-07-22 01:11:26.000000000","tz":-240},"subject":"Remove Neutron SDN-specific code","message":"Remove Neutron SDN-specific code\n\nMigrated to Neutron repository.\n\nDepends-on: https://review.opendev.org/#/c/741719\nChange-Id: I814d4e1abf3cee686b9f4511373edd46d1f5f4e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f67af2c22b363bb4f46d7f550344656cf14fe39d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f67af2c22b363bb4f46d7f550344656cf14fe39d"}]},"branch":"refs/heads/master"},"6bef34c5270129ea7410bb41f33496ad00044463":{"kind":"REWORK","_number":12,"created":"2020-07-22 19:51:38.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/92/718192/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/718192/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/718192/12"}}},"commit":{"parents":[{"commit":"8ae82a47e1928f99dd32f08785f876b3d1d9c3cb","subject":"Merge \"Stop to use the __future__ module.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ae82a47e1928f99dd32f08785f876b3d1d9c3cb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-04-07 19:11:11.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-07-22 19:51:04.000000000","tz":-240},"subject":"Remove Neutron SDN-specific code","message":"Remove Neutron SDN-specific code\n\nMigrated to Neutron repository.\n\nDepends-on: https://review.opendev.org/#/c/741719\nDepends-on: https://review.opendev.org/#/c/742027\nChange-Id: I814d4e1abf3cee686b9f4511373edd46d1f5f4e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6bef34c5270129ea7410bb41f33496ad00044463"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6bef34c5270129ea7410bb41f33496ad00044463"}]},"branch":"refs/heads/master"},"faabee1b59ce0ae6fc2eb6ab634f57faa700a0b2":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2020-08-26 21:08:51.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/92/718192/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/718192/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/718192/13"}}},"commit":{"parents":[{"commit":"9fb58eb9f4d2574106a48e3cd3c439d5345517f9","subject":"Fix invalid DOWN status when updating a UDP pool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9fb58eb9f4d2574106a48e3cd3c439d5345517f9"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-04-07 19:11:11.000000000","tz":-240},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-26 21:08:51.000000000","tz":0},"subject":"Remove Neutron SDN-specific code","message":"Remove Neutron SDN-specific code\n\nMigrated to Neutron repository.\n\nDepends-on: https://review.opendev.org/#/c/741719\nDepends-on: https://review.opendev.org/#/c/742027\nChange-Id: I814d4e1abf3cee686b9f4511373edd46d1f5f4e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/faabee1b59ce0ae6fc2eb6ab634f57faa700a0b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/faabee1b59ce0ae6fc2eb6ab634f57faa700a0b2"}]},"branch":"refs/heads/master"},"b520b90b90c34d2ee58fa8a71a294a6a7d7314ec":{"kind":"REWORK","_number":14,"created":"2020-08-27 22:04:38.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/92/718192/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/718192/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/718192/14"}}},"commit":{"parents":[{"commit":"9fb58eb9f4d2574106a48e3cd3c439d5345517f9","subject":"Fix invalid DOWN status when updating a UDP pool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9fb58eb9f4d2574106a48e3cd3c439d5345517f9"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-04-07 19:11:11.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-08-27 22:03:38.000000000","tz":-240},"subject":"Remove Neutron SDN-specific code","message":"Remove Neutron SDN-specific code\n\nMigrated to Neutron repository.\n\nDepends-on: https://review.opendev.org/#/c/741719\nDepends-on: https://review.opendev.org/#/c/742027\nChange-Id: I814d4e1abf3cee686b9f4511373edd46d1f5f4e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b520b90b90c34d2ee58fa8a71a294a6a7d7314ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b520b90b90c34d2ee58fa8a71a294a6a7d7314ec"}]},"branch":"refs/heads/master"},"cfc904d7d6e00a385aa8c6e20a8f58f9348345e1":{"kind":"REWORK","_number":15,"created":"2020-08-28 17:32:39.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/92/718192/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/718192/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/718192/15"}}},"commit":{"parents":[{"commit":"9fb58eb9f4d2574106a48e3cd3c439d5345517f9","subject":"Fix invalid DOWN status when updating a UDP pool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9fb58eb9f4d2574106a48e3cd3c439d5345517f9"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-04-07 19:11:11.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-08-28 17:32:28.000000000","tz":-240},"subject":"Remove Neutron SDN-specific code","message":"Remove Neutron SDN-specific code\n\nMigrated to Neutron repository.\n\nDepends-on: https://review.opendev.org/#/c/741719\nDepends-on: https://review.opendev.org/#/c/742027\nChange-Id: I814d4e1abf3cee686b9f4511373edd46d1f5f4e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cfc904d7d6e00a385aa8c6e20a8f58f9348345e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cfc904d7d6e00a385aa8c6e20a8f58f9348345e1"}]},"branch":"refs/heads/master"},"0f05f7e5a8c2cf479069210a33c396561489b129":{"kind":"NO_CODE_CHANGE","_number":16,"created":"2020-09-28 21:36:33.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/92/718192/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/718192/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/718192/16"}}},"commit":{"parents":[{"commit":"9fb58eb9f4d2574106a48e3cd3c439d5345517f9","subject":"Fix invalid DOWN status when updating a UDP pool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9fb58eb9f4d2574106a48e3cd3c439d5345517f9"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-04-07 19:11:11.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-09-28 21:36:02.000000000","tz":-240},"subject":"Remove Neutron SDN-specific code","message":"Remove Neutron SDN-specific code\n\nMigrated to Neutron repository.\n\nDepends-on: https://review.opendev.org/#/c/742027\nChange-Id: I814d4e1abf3cee686b9f4511373edd46d1f5f4e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0f05f7e5a8c2cf479069210a33c396561489b129"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0f05f7e5a8c2cf479069210a33c396561489b129"}]},"branch":"refs/heads/master"},"dcdf6315b31b2962fa8d0a5f080c369214bb222e":{"kind":"NO_CODE_CHANGE","_number":17,"created":"2020-09-29 18:40:16.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/92/718192/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/718192/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/718192/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/718192/17"}}},"commit":{"parents":[{"commit":"9fb58eb9f4d2574106a48e3cd3c439d5345517f9","subject":"Fix invalid DOWN status when updating a UDP pool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9fb58eb9f4d2574106a48e3cd3c439d5345517f9"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-04-07 19:11:11.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-09-29 18:39:40.000000000","tz":-240},"subject":"Remove Neutron SDN-specific code","message":"Remove Neutron SDN-specific code\n\nMigrated to Neutron repository.\n\nDepends-on: https://review.opendev.org/#/c/741719\nDepends-on: https://review.opendev.org/#/c/742027\nChange-Id: I814d4e1abf3cee686b9f4511373edd46d1f5f4e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dcdf6315b31b2962fa8d0a5f080c369214bb222e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dcdf6315b31b2962fa8d0a5f080c369214bb222e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
