)]}'
{"id":"openstack%2Foctavia~719922","triplet_id":"openstack%2Foctavia~master~I831b0b48d719397c14d80f8ebcbad997c50c7795","project":"openstack/octavia","branch":"master","topic":"719922","hashtags":[],"change_id":"I831b0b48d719397c14d80f8ebcbad997c50c7795","subject":"Fix py3 amphora-agent cert-rotation type bug","status":"MERGED","created":"2020-04-14 11:29:51.000000000","updated":"2020-04-27 11:48:57.000000000","submitted":"2020-04-17 22:54:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"719922-1587164086115-58e2f1df","meta_rev_id":"da805da0ef6faa2e2ec3cb5c6e2ef2b78e1bc349","_number":719922,"virtual_id_number":719922,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-04-15 20:08:45.000000000","_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-17 22:54:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"recommended":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2020-04-17 06:47:55.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-04-17 20:38:31.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-04-17 19:35:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2020-04-17 17:41:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2020-04-17 20:38:31.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-04-17 20:38:31.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2020-04-17 17:41:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-17 06:47:55.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-04-17 17:41:21.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-04-17 19:35:21.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-04-17 20:38:31.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-04-17 22:54:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"62d42dfe2f57ff49358662fec3c629f498b84b04","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-04-14 11:29:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb7ed8bafb4f402136b86a1aec30c3c7dfe82bce","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-14 11:57:50.000000000","message":"Patch Set 1: Code-Review-1\n\nNeeds a release note (\"fixes\" and \"upgrades\" calling out the need to manually rotate amps unless there\u0027s a better way to handle it).","accounts_in_message":[],"_revision_number":1},{"id":"077c6e05b3c6531fbd9bdab6c50e323566cd27c9","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-04-14 13:48:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5573a0ed0ec587db75a9796d1778aad66bec1b3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 16:26:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8514991b823446378afbed21ca640583 : FAILURE in 8m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0ebfd91c429434e9066eb386718d75a : SUCCESS in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9acf57ad1b044b6d883f4c107a899126 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f0a1ffc22dd4f099a41046c8bb20d38 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/18203e3ee2d34536ba9a03b99ef7d908 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0627dbf43bd04103b6169d952de909c2 : FAILURE in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43a75bae33c843e08cbcc6a47ffffedd : SUCCESS in 13m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1fb3d2da3c34443b2af3c05847a78b0 : SUCCESS in 7m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8da7e72de32443669a6bd405ed9594b9 : SUCCESS in 5m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/cf1e99b568ab4057aeb25e6ea6421fcb : FAILURE in 9m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/14dbcc6d0403485cada3b6751b918e1f : FAILURE in 6m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/23d1913f812e4fdebdf0b63234ad20f9 : SUCCESS in 47m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f408be7e74bf4b70ba1fbd25657a3927 : SUCCESS in 1h 35m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8559f359b51d4c6bb209094d5f6fdf92 : SUCCESS in 59m 09s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/32b0cb4cea5b4eb1a8a38fa3d21f23a8 : SUCCESS in 57m 39s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/153d46c7b2c840a89f47c4ae4573c0e8 : FAILURE in 1h 11m 18s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3b0e2cb547464d8e8c0173beeb12f317 : SUCCESS in 2h 10m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2ed52a5cd56444a4808b117af331c8b5 : SUCCESS in 2h 25m 46s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/473b731c0fc942d78c23914b2a08f4a2 : FAILURE in 1h 03m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/912040d1273b43faa733b4a3d8d25121 : SUCCESS in 1h 59m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8256ff0462770629e2bee406d6281018ff7e9c59","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-04-15 20:08:45.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"55648a4a5199fece72afff3998ca1faeae67f054","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 22:26:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1fb401c214b4be5ad655ef313f26c5f : SUCCESS in 9m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/753dd789c1c247c18f75db9735501240 : SUCCESS in 10m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d173b1bb0dbb4978aafe992679c02cef : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f417bc1232e94a7983712e67c5d2f490 : SUCCESS in 7m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ad081dddcc8345858558e9185d748aac : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee4654dc844c458a9853a32028495ae5 : FAILURE in 7m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09948c998bf3499aa51a8bce260d581c : SUCCESS in 12m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/215abf59e0c84b9b86c6dc84f136d638 : SUCCESS in 8m 54s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/161168e4ee7b40dd82f7433bbc581fe4 : SUCCESS in 7m 24s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/05933f4c2a32489395014a8521da82dc : SUCCESS in 9m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e726607d29874e218a81efc2169f39b5 : SUCCESS in 8m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8633e8c07aa34ed8927941b11a375c48 : SUCCESS in 46m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f26cb660becf4f0aba294e77de9de725 : SUCCESS in 1h 47m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2320c0f1e44e48f5a861fd0b516a1f6d : SUCCESS in 1h 00m 02s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/43903faa5e7a418fa2791b511b532d8e : SUCCESS in 56m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a13bf66873254264a75a32a111aa6014 : SUCCESS in 1h 04m 46s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5629ee3e3e6641699175430c21b0bff4 : SUCCESS in 2h 03m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/97bcaa43b3954363aecbb4308a991eb9 : SUCCESS in 2h 07m 59s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/16d06469da744f3c8af1154078118f9b : FAILURE in 1h 01m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/04ac759596ab46ce927f4af9475b54b5 : SUCCESS in 2h 11m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c8c9ac98af027b95295cbaf3d541a6e97aeb6798","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-04-17 06:47:55.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e4467c0266c9f1fe057ea242c52eba45076d0486","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-17 17:41:21.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"ac76c34366a350f62e2f25612dc254eb5461b521","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-17 19:35:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"13df1f8b8a73363ffa0cca0dda8745063ab3276c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-17 20:38:31.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1e8e408a8594b50925582e24b787647b59c6258d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 20:38:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2859408f645bca32a05374ca1006e3bf1692cf5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 22:54:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79585c4252d944fd808da5c64092d70c : SUCCESS in 10m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9770d1f046614d53993cabc1bd6a73bd : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f01f4e25f3f2491ba7cdccfe4fc43f0c : SUCCESS in 7m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/957a0db2e18e491e8782c7529a4d37f3 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f3225d8e34d4bbc8cd9323585447def : SUCCESS in 14m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9dabd827c0264687b26865a4d8871806 : SUCCESS in 6m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ad0b1993432045859cb1f5c42e9bccc6 : SUCCESS in 7m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1b56e645630c423c8d11697e04aef1d2 : SUCCESS in 58m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bc0386366d10483ebbcbda56ab7b3fd8 : SUCCESS in 2h 13m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/17465b0ad2ab4d34b8a19ed93dc9994e : SUCCESS in 58m 44s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/eb4e0f5c96ab4f9cb73efdaabac2b414 : SUCCESS in 52m 26s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/03316c62e5da4c448e5f007437763d93 : SUCCESS in 1h 20m 44s","accounts_in_message":[],"_revision_number":2},{"id":"cdf5e986d2af931e9e18591ae6628aa9c15571df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 22:54:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"5fc874a15f503d3e836dafa9539d73b924a45968","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 22:57:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b5683b72f1a4d0795afa3e8b7cc2987 : SUCCESS in 2m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c2341dfbf5e412da86f9045a8f36843 : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":2},{"id":"7f69e587789cf3ad5a9fa81b25eba13cb843e084","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-27 11:48:57.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 1cd2e64b6b04c1e83f4c8e0f31f88628e2780145","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"96a4482dff134b7be5c8d4da6c862b0a7f4a5a4b","revisions":{"9164b9fb3453c33a7b41ec3541f908c2b108bf85":{"kind":"REWORK","_number":1,"created":"2020-04-14 11:29:51.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/22/719922/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/22/719922/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/719922/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/719922/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/719922/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/22/719922/1"}}},"commit":{"parents":[{"commit":"c9e155155087931381d740e2ae6eef8dd7b7fafd","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c9e155155087931381d740e2ae6eef8dd7b7fafd"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-04-14 11:27:49.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-04-14 11:27:49.000000000","tz":-420},"subject":"Fix py3 amphora-agent cert-rotation type bug","message":"Fix py3 amphora-agent cert-rotation type bug\n\nFlask\u0027s stream always returns bytes, file write always takes string.\nThis causes py3 amps to return 500 on cert rotation AND wipe out the\ncertificate, so the amphora are no longer controllable and go to ERROR\nstate. Anyone running py3 amps prior to this patch will experience\namphorae breaking on a timer due to housekeeping cert rotation!\n\nChange-Id: I831b0b48d719397c14d80f8ebcbad997c50c7795\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9164b9fb3453c33a7b41ec3541f908c2b108bf85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9164b9fb3453c33a7b41ec3541f908c2b108bf85"}]},"branch":"refs/heads/master"},"96a4482dff134b7be5c8d4da6c862b0a7f4a5a4b":{"kind":"REWORK","_number":2,"created":"2020-04-14 13:48:54.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/22/719922/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/22/719922/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/719922/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/719922/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/719922/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/22/719922/2"}}},"commit":{"parents":[{"commit":"c9e155155087931381d740e2ae6eef8dd7b7fafd","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c9e155155087931381d740e2ae6eef8dd7b7fafd"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-04-14 11:27:49.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-04-14 13:48:43.000000000","tz":-420},"subject":"Fix py3 amphora-agent cert-rotation type bug","message":"Fix py3 amphora-agent cert-rotation type bug\n\nFlask\u0027s stream always returns bytes, file write always takes string.\nThis causes py3 amps to return 500 on cert rotation AND wipe out the\ncertificate, so the amphora are no longer controllable and go to ERROR\nstate. Anyone running py3 amps prior to this patch will experience\namphorae breaking on a timer due to housekeeping cert rotation!\n\nChange-Id: I831b0b48d719397c14d80f8ebcbad997c50c7795\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/96a4482dff134b7be5c8d4da6c862b0a7f4a5a4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/96a4482dff134b7be5c8d4da6c862b0a7f4a5a4b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
