)]}'
{"id":"openstack%2Foctavia~720033","triplet_id":"openstack%2Foctavia~master~I62e6940f3d36a1ede1535a3dbd889f2e6ae3f6e5","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I62e6940f3d36a1ede1535a3dbd889f2e6ae3f6e5","subject":"Add the \"redis\" extra for taskflow requirement","status":"ABANDONED","created":"2020-04-14 17:03:16.000000000","updated":"2020-04-15 16:11:21.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"26f67759ceb21b84ff963b956e2ddb7c80286354","_number":720033,"virtual_id_number":720033,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-04-14 19:24:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":-1,"date":"2020-04-15 15:14:40.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-14 19:24:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"960ef26ca5a54f52898d88adbe558c2f0d42ba4b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-14 17:03:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a51ad268dfdba1d8aaeec10b9bd5d77fc09790f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 19:24:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/150ef48b2adc499cb30fecf44e660af8 : SUCCESS in 3m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc260564a46347cf831bfb9966f91aa5 : FAILURE in 8m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21de4333d2024bdaa5ed05d2bb9d9b69 : SUCCESS in 11m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9a7f37eafd3468ebf625f49439815af : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70f99456aaa84db2b7963ad96b285cf1 : SUCCESS in 5m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/337957636f324d33b11eb222b4f71d19 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41b4720d05eb418d90b3643e99f21c79 : FAILURE in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fdd0a8c6aac4f7d9c43f92414002123 : SUCCESS in 12m 22s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1b87b8b270f341018edd956cfc54f8f9 : SUCCESS in 7m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f325d65208d244f9975da0880d648b30 : FAILURE in 9m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2d325f6363d2490f8dff62ecc95b1be7 : FAILURE in 8m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/63c54d7fd4ae43b79721f7cfe2508c62 : SUCCESS in 44m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d608ba8d351642348bab9d949ec7cedf : SUCCESS in 2h 05m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/90c8a5e8e8eb400e8c3a9d64e221b22d : SUCCESS in 53m 44s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/617c779590df4f28924ae38a8d30df3b : SUCCESS in 56m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/abca9d6fdcdb4ccf9f88be432c3f13c2 : SUCCESS in 1h 10m 50s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/87aba78fb062424389d6a78bb254cd6b : SUCCESS in 1h 59m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3f3db3b30ced4bf99e77c98207ad3685 : SUCCESS in 2h 07m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/046798d68290439eb7cd2ff47c3396b0 : SUCCESS in 2h 15m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1390dbb0ecf732ea164bf5834768d35a927e554e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-15 15:14:40.000000000","message":"Patch Set 1: Workflow-1\n\nThis is not a complete fix it seems","accounts_in_message":[],"_revision_number":1},{"id":"26f67759ceb21b84ff963b956e2ddb7c80286354","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-15 16:11:21.000000000","message":"Abandoned\n\nThis is a bigger issue than just the requirement, so abandoning for a different patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cbde0f825b65fd71876d0af90b54e09befcbacb2","revisions":{"cbde0f825b65fd71876d0af90b54e09befcbacb2":{"kind":"REWORK","_number":1,"created":"2020-04-14 17:03:16.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/33/720033/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/33/720033/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/720033/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/720033/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/720033/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/33/720033/1"}}},"commit":{"parents":[{"commit":"c9e155155087931381d740e2ae6eef8dd7b7fafd","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c9e155155087931381d740e2ae6eef8dd7b7fafd"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-04-14 17:01:00.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-04-14 17:03:15.000000000","tz":-420},"subject":"Add the \"redis\" extra for taskflow requirement","message":"Add the \"redis\" extra for taskflow requirement\n\nThe jobboard patch[1] always starts conductors which with the default\n\"redis\" driver, will require the taskflow \"extra\" requirement for\n\"redis\".\n\n[1] https://review.opendev.org/647406\n\nChange-Id: I62e6940f3d36a1ede1535a3dbd889f2e6ae3f6e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cbde0f825b65fd71876d0af90b54e09befcbacb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cbde0f825b65fd71876d0af90b54e09befcbacb2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
