)]}'
{"id":"openstack%2Foctavia~720041","triplet_id":"openstack%2Foctavia~master~I16513b0066c05cd6fc606da341df636094010a34","project":"openstack/octavia","branch":"master","topic":"stop-errexit","hashtags":[],"change_id":"I16513b0066c05cd6fc606da341df636094010a34","subject":"Don\u0027t inherit enforcing bash errexit in devstack plugin","status":"MERGED","created":"2020-04-14 17:47:41.000000000","updated":"2020-04-17 04:55:02.000000000","submitted":"2020-04-17 04:52:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720041-1587099177756-170bf913","meta_rev_id":"1a5e8a24d910b1ceca07f888651b16e021d22e2d","_number":720041,"virtual_id_number":720041,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"date":"2020-04-15 20:06:24.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-17 04:52:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-04-16 12:32:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-04-17 03:11:28.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":1,"date":"2020-04-16 07:20:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2020-04-17 03:11:28.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-16 07:20:07.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2020-04-16 12:32:33.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-04-17 03:11:28.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2020-04-17 04:52:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eadf35fc82e5642dc6a6154399a8215ca9b63d8b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-14 17:47:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"83448a47c7c80024bb61905618486bbc8fcb661c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 20:28:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03618716f5054e968970b146664b06c9 : FAILURE in 9m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/749a352afed44c9f870eb8c1c6c9044f : SUCCESS in 10m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/216dca44488648358c8f69420c4d7d77 : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6f82c63173d46b1a13cde680f5d7bbf : SUCCESS in 6m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c4ebfdf7f0924d8c91d1d9dc871380af : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b69015c6a2034cf7b19fb18a4e4e60cd : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4bd4bcb746346d8adb393bd57a1f4b5 : SUCCESS in 14m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8f49eb40edda4d44ab943cbcc802409e : SUCCESS in 8m 19s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9f8da251a3b64381b18017ba0e4a491f : FAILURE in 9m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4dc010093fa1493988e9de75608d76d2 : SUCCESS in 8m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/993bb0ae147040b29411c13190c2de55 : FAILURE in 23m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fbdb60e698e444f4b271b84dbc4a630e : FAILURE in 28m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e8b3aa5298194abc898e90fb079a67bb : FAILURE in 27m 18s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/243bad2083ae4923a659bb8c9a9441c1 : FAILURE in 26m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/31e72f9724954fa4822a0d5ecc29a36b : FAILURE in 22m 31s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/26e3ebd8323a4cf38ed2ad95823eeaaf : FAILURE in 24m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ea0f7e4420794ce38304972cf0806bc0 : SUCCESS in 2h 12m 28s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6ff12a0f0b0943a18e4b236258b52273 : FAILURE in 1h 04m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d11497d19a8748ea8e357c7834c95ed4 : FAILURE in 25m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6fbc9f0329404696e261d80043c9465df16ba94d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-15 14:13:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e2db5a2c228def0380db555102df93a5a059f056","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 16:36:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6aac982d806d413d9386ad8bac3a981a : FAILURE in 10m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a1a9bb5d91c4b4bb39232e3371864b1 : SUCCESS in 10m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/839e57daf8654df7a8f55e00af629bc3 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd9c9f3c96064ba6a45f76991feb3482 : SUCCESS in 7m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/15de81f59ade4e8eac5007dd1ea22822 : SUCCESS in 7m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ab3aa63c27f42dda60fc994c7db85b6 : FAILURE in 8m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5986071a6af94d45a3214cf8ca1b12eb : SUCCESS in 13m 46s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e827c75a6bc74e779a3b4fbdc9fdcca8 : SUCCESS in 8m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a9068c7a3a7742b589166055beacd6ac : FAILURE in 11m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fb9c013687914c95bc694a61b32a0d00 : FAILURE in 9m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/30fffe5edc8a43fabbfe7025c73f492b : SUCCESS in 46m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9d79df58ed2546d69e17ca31f5cdc5a0 : SUCCESS in 1h 44m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/58866bd5ed194cc4a117761df83d5357 : SUCCESS in 54m 32s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/aa6538bca4024c28a5c57728b26a5edc : SUCCESS in 1h 03m 14s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/29c3956d8e2d4db69b64c453935bd39b : SUCCESS in 1h 16m 57s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/eb8c088bbd354cd9919571597a497db0 : SUCCESS in 2h 10m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e4030325b4a340bcb2de58f7b05b5504 : SUCCESS in 1h 17m 08s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/333bfa1bb1644c19ab0bf84a1e7416b1 : FAILURE in 1h 05m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fc934c33039b4a2db5c97cb28574b3a1 : SUCCESS in 1h 59m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"adc8f177a5ed9023b985273f10692fce3787e7cf","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-15 20:06:24.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a05ee869c8f4257e59fb88f9b28ecd3556089341","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 22:21:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c647df0e619441896c8a78adfe36c23 : SUCCESS in 13m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eefb8b9faf1843a09a65fea5ae120d2c : SUCCESS in 12m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12470eb93b9f4416ad04f67d19863e68 : SUCCESS in 10m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcac2ffac805465ea41e440c74fdda4f : SUCCESS in 6m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cd40cfdf4f4b4dd1887f38c15818959f : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c295fed426344acfb55bb4d859a9b936 : FAILURE in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a133b0560f3644d281ae41994ee154a1 : SUCCESS in 19m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b3af9c0413ef478e8a6968a201151e93 : SUCCESS in 9m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e785bfa65b8b47c7a7e27346c9c00d1c : SUCCESS in 9m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0b8784252acd4977817fb1495788172e : SUCCESS in 9m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8fe77923515e48a581f6e766eaf2a3d9 : SUCCESS in 44m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ff6a026d26db4603a1de94b27f687a21 : SUCCESS in 1h 39m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8fd3b59b75ed4f79affc8e5f249a8b0e : SUCCESS in 56m 19s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9b568dd7a04547b8a05750acc374fe20 : SUCCESS in 1h 00m 54s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d6ae8c371c734c86a473db029a604a73 : SUCCESS in 58m 55s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/981991c45eda4b5f8a892c0dbe95d3f9 : SUCCESS in 1h 40m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/14c43f7614e1473f98769ed1460d0947 : SUCCESS in 2h 07m 42s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d1f127feb3744a3d86860e5709e29fcc : FAILURE in 1h 01m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f45566ee88af4ce7a968e871f445f38e : SUCCESS in 1h 37m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e061819c98d99d653f17ef59da448c36e7a128fc","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-04-16 07:19:57.000000000","message":"Patch Set 2: Backport-Candidate+1","accounts_in_message":[],"_revision_number":2},{"id":"4e4a6a5b4b81150b775a00d06a62ecafca37d604","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-04-16 07:20:07.000000000","message":"Patch Set 2: -Backport-Candidate Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3b71bc331da5d1174338997dcda18c337c739726","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-04-16 12:32:33.000000000","message":"Patch Set 2: Code-Review+2\n\nI\u0027m fine with this.","accounts_in_message":[],"_revision_number":2},{"id":"43e92d4ee3c356d11295bbc6e7ad782dd37b428c","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2020-04-17 03:11:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0468a48170af32e86286be8ceef2db5280d25bde","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 03:11:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"66dd9c7e5a90510f225dc39a6364522ca689fcc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 04:52:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b313b999de5e489abb6b262b8c620fe5 : SUCCESS in 11m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08d8fb972f1b4fe1aa477e2fd31b1b00 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/958b914caff54c36832928347d41170d : SUCCESS in 6m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/68d5bd37cd184b759d49a466a54d1c3d : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03b4cbf989a741099a1b85d84f9fe45d : SUCCESS in 13m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cc127b34314c4b3680ac10babd1d9352 : SUCCESS in 7m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cb5705884e7f47f7bf5548297735483c : SUCCESS in 48m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/924f83c093fd438594642fecbf256049 : SUCCESS in 1h 39m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a11843d35b504b069d7ac1eb3ffc4ad9 : SUCCESS in 56m 56s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1a1eef87ed35462f9b18092d60bad4a0 : SUCCESS in 1h 03m 43s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/63823c89a2594eefb8288dd591e9a1db : SUCCESS in 54m 54s","accounts_in_message":[],"_revision_number":2},{"id":"36ca393e0e0d6d2bf1bc2072e9f24f03b7d57047","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 04:52:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"0f5db79cf58481ae069518e3ebca236043bf5ecf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 04:55:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/994e4bd7250f4b479bd64b5b3812174b : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1c639f01a083e32111306f1f8d1a89df6aea2950","revisions":{"27b79b28fa65d039e4797b1d7fbf9466f0328ff0":{"kind":"REWORK","_number":1,"created":"2020-04-14 17:47:41.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/41/720041/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/41/720041/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/720041/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/720041/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/720041/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/41/720041/1"}}},"commit":{"parents":[{"commit":"c9e155155087931381d740e2ae6eef8dd7b7fafd","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c9e155155087931381d740e2ae6eef8dd7b7fafd"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-04-14 17:42:43.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-04-14 17:42:43.000000000","tz":-240},"subject":"Don\u0027t inherit enforcing bash errexit in devstack plugin","message":"Don\u0027t inherit enforcing bash errexit in devstack plugin\n\nThe devstack plugin code was sourcing a file that was\nenabling bash errexit, which was then inheriting it in\nlater operations that could cause the shell to exit\nunexpectedly.\n\nJust run the script instead, since it doesn\u0027t require\nany local variables to complete, and change both scripts\nto \u0027set +e\u0027 near exit so sourcing them manually doesn\u0027t\nhave issues either.\n\nChange-Id: I16513b0066c05cd6fc606da341df636094010a34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27b79b28fa65d039e4797b1d7fbf9466f0328ff0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27b79b28fa65d039e4797b1d7fbf9466f0328ff0"}]},"branch":"refs/heads/master"},"1c639f01a083e32111306f1f8d1a89df6aea2950":{"kind":"REWORK","_number":2,"created":"2020-04-15 14:13:43.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/41/720041/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/41/720041/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/720041/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/720041/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/720041/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/41/720041/2"}}},"commit":{"parents":[{"commit":"c9e155155087931381d740e2ae6eef8dd7b7fafd","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c9e155155087931381d740e2ae6eef8dd7b7fafd"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-04-14 17:42:43.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-04-15 14:10:34.000000000","tz":-240},"subject":"Don\u0027t inherit enforcing bash errexit in devstack plugin","message":"Don\u0027t inherit enforcing bash errexit in devstack plugin\n\nThe devstack plugin code was sourcing a file that was\nenabling bash errexit, which was then inheriting it in\nlater operations that could cause the shell to exit\nunexpectedly.\n\nChange both scripts to \u0027set +e\u0027 near exit so sourcing\nthem doesn\u0027t have issues.\n\nChange-Id: I16513b0066c05cd6fc606da341df636094010a34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1c639f01a083e32111306f1f8d1a89df6aea2950"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1c639f01a083e32111306f1f8d1a89df6aea2950"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
