)]}'
{"id":"openstack%2Foctavia~720706","triplet_id":"openstack%2Foctavia~stable%2Frocky~I96908e6cbdb523f95298aff463a749d15e98e1ad","project":"openstack/octavia","branch":"stable/rocky","hashtags":[],"change_id":"I96908e6cbdb523f95298aff463a749d15e98e1ad","subject":"Fix update API when barbican secret is missing","status":"ABANDONED","created":"2020-04-17 13:01:07.000000000","updated":"2020-07-24 11:42:05.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ebadedda8c72bf4e5777411c799b22355d325786","_number":720706,"virtual_id_number":720706,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"date":"2020-04-18 08:29:38.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-12 09:48:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-06-11 19:10:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-05-27 19:12:37.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-08 16:09:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2020-06-11 19:10:58.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-27 19:12:37.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-06-08 16:09:44.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-06-11 19:10:58.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-06-12 09:48:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1ca56f90dc00993428e0137b6106daef3cc36db0","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-17 13:01:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c67b18b20679a77841a68ab528a988c89d7142d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 15:00:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/59e2ed49ad374a238e553fbb6b5983c4 : SUCCESS in 9m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f72044160eb42d19734e6c9abdb789c : SUCCESS in 11m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e50886d033ae445eb119fdf5f2ea92cf : SUCCESS in 7m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b0c97ecffc1943d9b01428e20296b52e : SUCCESS in 9m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a2c7e3ff2c248bab0ee8aa50a642cc0 : SUCCESS in 8m 32s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/e4025a257eba4dc59abd6d3bc79ad998 : SUCCESS in 1h 14m 45s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/508c36b4f9fa49e88839acbd6e051d32 : SUCCESS in 1h 15m 45s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ecc3fc2b71934533936f6a70463d19e6 : SUCCESS in 1h 20m 11s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/60fe488d07a0410eacd05bd9107e127a : SUCCESS in 1h 17m 44s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a95320b529bc4d4d901ff9091521ff4f : SUCCESS in 7m 18s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/ffb49660415b4d458327da85623a6a9d : SUCCESS in 6m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/74ecb08b5caa402c828aa9cac671d990 : SUCCESS in 6m 39s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d929faf6cbe0440fb43026510c88b7be : SUCCESS in 1h 02m 16s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/ceda01f996e449eebc3a4d6ab45188bb : FAILURE in 1h 00m 04s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/e209078d7e9548759fcc83e49a314508 : FAILURE in 1h 31m 40s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/d64fc1327d7a4458a7e73b35de705047 : FAILURE in 1h 26m 17s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/bc2b01537bad44f6b2763fe6f0c7b375 : FAILURE in 28m 38s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fd3055280a9d4d5aaef3ac56c6473bef : SUCCESS in 50m 53s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/f7d26a1d17674e3586432a5cbe82fdff : SUCCESS in 50m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a11a703ae3cc4094a44c0a8f878b1764 : SUCCESS in 1h 34m 00s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/dc7698b0ce4c41439d3bf6631e1d31a7 : SUCCESS in 1h 50m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/081cbd7657e548fa9b0e333daf797809 : SUCCESS in 1h 15m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f216e90a2a6542f5a9467facc0e4a7df : SUCCESS in 12m 31s","accounts_in_message":[],"_revision_number":1},{"id":"166807d2a32479de15d9736bef4ba150e848f342","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-18 08:29:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7f881b1b3879db226d888322b55c5f82783a7f57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-18 10:18:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66d144fd59cf43938332c2cb0894a542 : SUCCESS in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6aafbf8064f451e9e9facec1fc02070 : SUCCESS in 8m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0b1ee1ffd526401cbd350824dbfc3068 : SUCCESS in 5m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/84d710bd1683417892bf49ac4e3de5b0 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba428df4902448458e6fba7e583835c0 : SUCCESS in 7m 20s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/b02ee6163c464249abb75b6d18f9c2a0 : SUCCESS in 1h 14m 49s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/f9fb10acf374444f8ea880f1cbb2cf96 : SUCCESS in 1h 16m 56s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c8c98cdac7b743de8b5b1a062549cbef : SUCCESS in 57m 39s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/cff2362a45ce4cf387acd3e12c60861a : SUCCESS in 1h 25m 56s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a52a668f65cf4adb93da61b4898508ba : SUCCESS in 6m 48s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/5ec5c0817fb64b7fb413b088f78f8178 : SUCCESS in 7m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7b319959fdf9400db28039d069dceb98 : SUCCESS in 5m 36s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/00edf1854ef24f4c8b7b321abe0855d7 : SUCCESS in 58m 06s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/fcac5ac751b94345b458571408cd987c : SUCCESS in 57m 30s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/d3597b5005ce4e3ea33eba773296c43e : FAILURE in 1h 17m 53s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/60091ca8ebd84c2892b7fcb79af5b243 : FAILURE in 1h 21m 56s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/d7acc5e3df654dfd8ce16f4ce7b0905f : FAILURE in 23m 10s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3393d3ba86f6436d8ca367a8f7dab723 : SUCCESS in 44m 00s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/21b027aa6a704e34bc044cf63cb377f2 : SUCCESS in 44m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/88bc02ba3bc24c5f8f3c1ae203eb7e65 : SUCCESS in 1h 24m 22s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/9ce5a2f81cd443e88521a01fe2b84e26 : SUCCESS in 1h 43m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a7d5c08e330843f69afb1a4ac89aed74 : SUCCESS in 1h 03m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0551bb3abffd4de1adcb5f08b713d508 : SUCCESS in 9m 44s","accounts_in_message":[],"_revision_number":1},{"id":"eb33c4e15182dceb07b617380d8bbe5c8b06c323","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-27 19:12:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"869a1fbc1f41cba93f6c811f27a7fe30afc11829","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-08 16:09:44.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"c904e956c1a8420f0078ac9fb7eb08b6b2e6e112","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-06-11 19:10:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4b3c62b6022a2251e564393283893ae8272be4e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 19:11:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c3d2382e35aea08a547a6a708ec0efb533679fcf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 22:11:13.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f26095ac1bf4a13935376145c102792 : SUCCESS in 9m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20e61cf87e09475db2fd9dce51e63ebe : SUCCESS in 9m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d5a84e085ed74bb88fb7800d47dd77c3 : SUCCESS in 6m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4bb10335fe6745c0a5d9bf8cb753b002 : SUCCESS in 8m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2ae45563ed34ba08fcb433d11b7741b : SUCCESS in 7m 31s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b06e7e321be4418b8b16d72870d7c225 : SUCCESS in 8m 25s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/04b39a0083364399ac5f447aa8f3adec : SUCCESS in 7m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0254457bf72f40f4906c1c054822e6d1 : SUCCESS in 7m 18s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3364b60a178d46acb7debb7298e08306 : FAILURE in 34m 48s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/4ea626a8c37b469a8acb3e20750bd545 : FAILURE in 36m 21s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/a11d458165ec4430bfc1827b87050019 : SUCCESS in 1h 21m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a7b024401cc04041bd1fa88e1b88df1c : SUCCESS in 46m 44s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/5a24c81e9f2a491a8f9334de5e815b71 : SUCCESS in 48m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8806b7d8faca4bbb9a178fca66bf3f3d : FAILURE in 31m 34s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/cb80db48c2ba42d0b1589f3fba23121b : FAILURE in 31m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fe23e909301942ec94969482034d0ea6 : SUCCESS in 1h 17m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25f03cc788ff49349ebf0209478fed37 : SUCCESS in 10m 42s","accounts_in_message":[],"_revision_number":1},{"id":"a24f7456c7bf91e326cbcaf5b5d3d03c00161159","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-06-12 08:10:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"67b7bc1494f1758331e289a6612c9548cbac5607","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 09:48:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f87931c037a42c899baceffc7d6a853 : SUCCESS in 8m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5e3f8624c0546a0bc5567e565003638 : SUCCESS in 6m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/49b7864fa1944023a20b9b0cd749f485 : SUCCESS in 5m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d3dc1d1551be4351b49425beb95ba704 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f54da3205c49492ba92a2fdac712053b : SUCCESS in 5m 57s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/8064d979dcc44c61baec3cb5922bfd8a : SUCCESS in 1h 19m 03s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/5ea7421e49ba4df19b7a3b7cdec0a073 : SUCCESS in 1h 14m 28s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/34ad0865b15346b3ac77b98a28970424 : SUCCESS in 1h 03m 58s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/d4de9f34a1b641b4b06487723a25dfbd : SUCCESS in 1h 32m 41s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/dd3518ef5c24447080f18be0a4f9480a : SUCCESS in 6m 38s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/0de65fa49b574b889171e6dce682051a : SUCCESS in 6m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/57f3002802284449b6ccede63d9d753b : SUCCESS in 7m 04s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/14d1b2c145d54eaa9dba301dc34f7c7e : FAILURE in 30m 09s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/959f8629d4044533a88cfc133b5e46ae : FAILURE in 31m 48s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/c8a80ff774c94ac38f1788fbfeac506c : FAILURE in 29m 43s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/bb3dc5c8f54d4736a39850b159dbf565 : FAILURE in 34m 17s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/cd75a1aba1534bd2a0c656d6c745ec22 : FAILURE in 33m 22s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0a839e6868e0469fa013f809991f46fd : SUCCESS in 42m 22s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/0b42d02d27c14395a9849f76dc7fa8e7 : SUCCESS in 40m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/82ce37b0c41f4324a7c37e3378ee7f14 : FAILURE in 31m 46s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/ac9b7ef221fe44058d6d5b8b5e4a1e5e : FAILURE in 26m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1fdf0bfabcb241b9a23e51a09715987a : SUCCESS in 1h 18m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4aeb9965dd074fdba28f381f9e43adec : SUCCESS in 8m 24s","accounts_in_message":[],"_revision_number":1},{"id":"ebadedda8c72bf4e5777411c799b22355d325786","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-24 11:42:05.000000000","message":"Abandoned\n\nRocky Octavia reached End of Life (EOL) state. This branch no longer accepts changes.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d159259d451ce4a893bc75618eb251cf92b04ca0","revisions":{"d159259d451ce4a893bc75618eb251cf92b04ca0":{"kind":"REWORK","_number":1,"created":"2020-04-17 13:01:07.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/06/720706/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/720706/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/720706/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/720706/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/720706/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/720706/1"}}},"commit":{"parents":[{"commit":"a0f75b10eda6280eaba76303056893d39b0a1416","subject":"Pick stale amphora randomly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a0f75b10eda6280eaba76303056893d39b0a1416"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-29 21:15:13.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-04-17 12:57:31.000000000","tz":120},"subject":"Fix update API when barbican secret is missing","message":"Fix update API when barbican secret is missing\n\nAPI update calls were blocked if the load balancer had a TLS\nterminated listener that the secret has been removed from barbican.\nThis patch corrects this problem allowing users to update the\ncertificate reference.\n\nChange-Id: I96908e6cbdb523f95298aff463a749d15e98e1ad\nStory: 2006676\nTask: 37322\n(cherry picked from commit 5af97a248b2c20cd12d743b124fec9848b1d965c)\n(cherry picked from commit db75e58e532b08852c3b5a71d835ea42dcf61526)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d159259d451ce4a893bc75618eb251cf92b04ca0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d159259d451ce4a893bc75618eb251cf92b04ca0"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
