)]}'
{"id":"openstack%2Foctavia~720777","triplet_id":"openstack%2Foctavia~master~Ib4944fb44fea53c8dd1908b7f264d45a84884639","project":"openstack/octavia","branch":"master","topic":"no-babel","hashtags":[],"change_id":"Ib4944fb44fea53c8dd1908b7f264d45a84884639","subject":"Remove Babel requirement","status":"MERGED","created":"2020-04-17 16:46:15.000000000","updated":"2020-06-05 11:37:56.000000000","submitted":"2020-06-05 11:33:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720777-1591356804244-13da08bf","meta_rev_id":"0234aea0855a198f18a1f87015afe42b204081af","_number":720777,"virtual_id_number":720777,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-05 11:33:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-06-02 16:32:27.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-06-05 09:36:36.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":1,"date":"2020-04-21 15:52:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-06-05 09:36:36.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":-2,"date":"2020-06-02 16:32:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-21 15:52:28.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-06-02 16:32:27.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-06-05 09:36:36.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2020-06-05 11:33:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fcfe367f8c1113b01720bb278f7cf88cb84b6430","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-17 16:46:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10e3bdce9798e106e135e67d26340ce31504775a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 20:25:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b364fc52bb074406b02f0f2eefecaca6 : SUCCESS in 5m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef1362868c9b4b168260a371f6e9d9a7 : SUCCESS in 8m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/107bb2c0a2be427297ea302c5fe436a2 : SUCCESS in 12m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62ed6b37f8a7467cabfcb6457fac92ce : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4554508484a646b58db09accc7df5947 : SUCCESS in 7m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/86c1765978994d409df00f927b9ebab3 : SUCCESS in 8m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/380536b9fbd74d3288beedb35b553448 : FAILURE in 9m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f71a34b46fac426583bf9edbca5e5c58 : SUCCESS in 15m 35s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c55308e844a74441a2cc36de31ae8caf : SUCCESS in 8m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b187358511f74e20abf6df9a73fe4673 : SUCCESS in 10m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a31a35a5e0d6495fb973bae7b7fbe0cc : SUCCESS in 8m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1acb9a7a37794a7d974f749759532381 : SUCCESS in 52m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4a53cd60d5c749f2a09a1a320cbbc434 : SUCCESS in 2h 06m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1bd07a1b1f304b38884febdda3b941e3 : SUCCESS in 56m 28s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b8eee86256e049a193ed0ff30d9a99b1 : FAILURE in 17m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a6d2f45853da45a2aa7256c0192c8b4f : SUCCESS in 1h 17m 27s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a3bc240ea623479f948636d0c5c9fc12 : SUCCESS in 2h 34m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/12007320198c4201966f2238133355a1 : SUCCESS in 1h 11m 46s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ae1c519e75b84d17abc3c27827b578d7 : FAILURE in 51m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b5d68c55169f4033a040fde563f062ba : SUCCESS in 1h 47m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"86dbf44427766157d5dec5f5a6e537be13006707","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-21 15:52:28.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9d6dd549cb30d218cabdb31384d36dbc0a91a36d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 18:22:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6e7431064e7c4c0b88895aa4a024d546 : SUCCESS in 3m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/024784fd46c84a38acd44fd29e2d7481 : SUCCESS in 12m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1afb5dfed38f4e3aa966343c3d0d26d5 : SUCCESS in 11m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7503443f43ee4997b3248fe081e6ac05 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0cb085aa0a5b4945a9c77a8d19911d82 : SUCCESS in 7m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c9ea512bf3d643588acb56632f2d6553 : SUCCESS in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45f102900a54424885901fc24da8c708 : FAILURE in 5m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/430c78bc6ed645a095031305c12a7bfd : SUCCESS in 14m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/40d567449a164d7d9acc0ce4758112bd : SUCCESS in 5m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/75aa2b22a97647fe902d9d47f069729a : SUCCESS in 9m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ee4ff2ade51e4a0f96d9f7a2fc7dfa7e : SUCCESS in 7m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c6739958cb794a5e8872226e9dd87afd : SUCCESS in 44m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/355b264b860346b2a1cb8e665f2e9fae : SUCCESS in 58m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a94318a3e3414364a2e1f59df6b08c9b : SUCCESS in 45m 54s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1e882b6279b34d1082599c7b512366db : SUCCESS in 44m 07s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9d8ee64079f34da485c0dddef1e1a0a5 : SUCCESS in 1h 14m 24s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/777695aada3f4cabada315c4b4e4f256 : SUCCESS in 2h 22m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/aa3a55248508417db75e244c7975c3d7 : SUCCESS in 2h 10m 13s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2e04de8c1f8943f6866c3b2776f16550 : FAILURE in 6m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4cec0a7849f54b56909bc3835a09aa3c : SUCCESS in 1h 52m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ec53bb2522482185fb46e702dab1be0eebb24750","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-02 16:32:27.000000000","message":"Patch Set 1: Backport-Candidate-2 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9bf76563074b1ff3fb5303b00b7852d423356247","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2020-06-05 09:36:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6f382a51178ca3be5408e1b566479d92c6dfa774","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 09:36:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fde34c9f8d7a1e036bd72d4d8ba97adf8c8a02ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 11:33:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/cefca9ff70c849a5bd79df0b82321824 : SUCCESS in 2m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/482a524c123647d8b4b05d43a8a79b5f : SUCCESS in 14m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/641be2475f98415b9ce794eba5f42a94 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88d60878551346b28b149b7853cebe40 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48fc6825a6334750ac3f2ad674dc6eca : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68efc33abda7446185549e3f6a1a6950 : SUCCESS in 14m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6928d6c06a974f5ab7b8813520a2fde4 : SUCCESS in 10m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ed32e838216f4abaaf096e3704d391ec : SUCCESS in 50m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c0053f288ae049679e7b733d89f58b13 : SUCCESS in 57m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4311d0eb22be40c09ed5d8dbdf8a883c : SUCCESS in 32m 01s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e3a04b6f7d0f4b0b980ca723979cfcfd : SUCCESS in 58m 31s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/36eca303bf6a494cb8658e6217c4cb64 : SUCCESS in 1h 05m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/428cb79534ce496aac0c25db16478e90 : SUCCESS in 1h 37m 00s","accounts_in_message":[],"_revision_number":1},{"id":"a16a4314fca7f4326fe75c4d734fd5498b4838be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 11:33:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"26df6f670c68e03fd7a168a9aa385b1cef29eb8c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 11:37:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ad6d23b4f26481ab9bb543a0f6fbfe2 : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a720f76c848255cdc7c47f7227c230a49e3326ef","revisions":{"a720f76c848255cdc7c47f7227c230a49e3326ef":{"kind":"REWORK","_number":1,"created":"2020-04-17 16:46:15.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/77/720777/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/720777/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/720777/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/720777/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/720777/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/720777/1"}}},"commit":{"parents":[{"commit":"0f52069be8263c3d86c9e2cde8fe2f96c36bd274","subject":"Merge \"Don\u0027t inherit enforcing bash errexit in devstack plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0f52069be8263c3d86c9e2cde8fe2f96c36bd274"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-04-17 16:45:08.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-04-17 16:46:14.000000000","tz":-420},"subject":"Remove Babel requirement","message":"Remove Babel requirement\n\nBabel is no longer a requirement for OpenStack projects.\nLet\u0027s remove it and save some disk space.\n\nSee also\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-April/014227.html\n\nChange-Id: Ib4944fb44fea53c8dd1908b7f264d45a84884639\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a720f76c848255cdc7c47f7227c230a49e3326ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a720f76c848255cdc7c47f7227c230a49e3326ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
