)]}'
{"id":"openstack%2Foctavia~721318","triplet_id":"openstack%2Foctavia~master~I9132e2d05bf67ddff4f2d7879d799c8dd25bd3c7","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I9132e2d05bf67ddff4f2d7879d799c8dd25bd3c7","subject":"Fix Octavia unit tests for python 3.8","status":"MERGED","created":"2020-04-20 17:57:22.000000000","updated":"2020-04-24 00:31:20.000000000","submitted":"2020-04-24 00:28:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721318-1587688130753-16b58591","meta_rev_id":"8ce301412b6f2fdec3ef2353384f358b06770d2e","_number":721318,"virtual_id_number":721318,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"date":"2020-04-23 22:55:30.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-24 00:28:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":2,"date":"2020-04-23 21:00:07.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":2,"date":"2020-04-21 02:27:03.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":1,"date":"2020-04-20 18:36:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":1,"date":"2020-04-23 21:00:07.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-20 18:36:06.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-04-21 02:27:03.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2020-04-23 21:00:07.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"},{"updated":"2020-04-24 00:28:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e2bdd9d2e170c7042c9b4886b0f6948e2adbcb03","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-20 17:57:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb539645f82b82adaec55bc9888d3fd84fe3f433","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 18:14:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e390420b26e74d03bcf6db8987069712 : SUCCESS in 8m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4fc27f743594054aa6af8d8f256c676 : SUCCESS in 8m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fbb62f9a61e4389b0b8078ccb32e5ec : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64442c1780494c8d9802de12230bbfff : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1f685696a054417fa1731dd2f895d710 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d24c8ce9033c4becaa3e7ab94093997a : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8042ac951d24b5d8b07e67238554a21 : SUCCESS in 11m 13s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bd118dd74f5d40f6a5efa32d859a7e9e : SUCCESS in 5m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/58b102e6ba3846f290405ec00fdddf59 : SUCCESS in 10m 30s","accounts_in_message":[],"_revision_number":1},{"id":"80c5160dc6b82db43a0b7963bc847bab1da6241d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-20 18:36:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"17e647810b50b9e3d64fc45fec77c43b6b63dabc","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2020-04-21 02:27:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d8ba6b3d201784338ca2a168c1ba1dae035bc902","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2020-04-23 21:00:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4c0b1e7bee1449424bfbef79bf349e42e1bec009","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 21:00:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"505b3aa3893cf8da5559b1806f0101000a84cc47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 21:33:10.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63f824d198394d3aa64cf27b6d921f20 : SUCCESS in 11m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cf6065f8ec04147801a31446e0f6c24 : SUCCESS in 9m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3cf04bdafdde4c63b8a415ee97a5f0dd : SUCCESS in 6m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ef5ffff2110e46cdac96083bbe0f64a8 : SUCCESS in 12m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a7fb659ab204fe0bc3003b0c253a3bc : SUCCESS in 12m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/59de7c98536a4557b9340352b3b8c19d : CANCELED in 30m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3e9c62378a3c48b8aaf5094ca0f95093 : CANCELED in 30m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ea5f9c52203b4fc492426074f98e4f92 : CANCELED in 30m 25s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/eeca7b988a884786bfeb081221071e79 : CANCELED in 30m 28s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f2dec6ed6a114ae9aa83eb84fde7d171 : FAILURE in 30m 09s","accounts_in_message":[],"_revision_number":1},{"id":"22772246cc38fc56edf0219e08a2f8e380b4bb56","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-23 22:55:30.000000000","message":"Patch Set 1:\n\nrecheck\nglance error","accounts_in_message":[],"_revision_number":1},{"id":"5766dad86d76559b454a9db7eb3480c95ba63db0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 23:09:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12e988955bbf4114bc06672ec7c0822b : SUCCESS in 9m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7502d8dc51f7411e954cb4452e51610d : SUCCESS in 9m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18628fa2712f47f382e20c3f2b7f3e70 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af5ea88828f2431abd1695008b16ecec : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/679e180b86444f348446dcbc73ebe7e9 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d6a3576ed954eab84d056dc6d8dec34 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2996c3d5dba4ad1b7194202377454ee : SUCCESS in 12m 31s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/78c1d88b37f447c4b8653bb8b81904ba : SUCCESS in 5m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a1d77079eae441dcbc52781e4eac1a24 : SUCCESS in 7m 02s","accounts_in_message":[],"_revision_number":1},{"id":"08d6ef05d567dcb29f632414170b9a77cdff1ac9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 23:09:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6d3dac5430cfe6c2173f2f24d03d364ea161c2f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 00:28:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a4fcb389295f86be4fabf8ba5fbba8761c7953e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 00:28:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e6f3c36db0b42b3902e8405a6605dcb : SUCCESS in 10m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c81874de3a174ea0aecd0ec0462058c0 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78cfaa13f10741b2b4a827ad18b5a7f0 : SUCCESS in 5m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/79af0c7ead9d4b3f95de1b3b78e8e117 : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ee91831a68245ecb8f772ddff9b5052 : SUCCESS in 17m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cfacd1674db540da9a09e18ef5b137ec : SUCCESS in 1h 03m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/28e8f991bac74e89ab9d0c08a9cf7a6f : SUCCESS in 1h 15m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bd5c1e3035e747db99a0fac92f6b06e0 : SUCCESS in 1h 04m 03s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5761367d599a4c37910078ca05ec8c17 : SUCCESS in 1h 02m 04s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9933090615004c67bcd803a944302ada : SUCCESS in 1h 03m 34s","accounts_in_message":[],"_revision_number":1},{"id":"8ce301412b6f2fdec3ef2353384f358b06770d2e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 00:31:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5de44ec935a8403ab5bcf78878e54a0b : SUCCESS in 2m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5ee26a20359a8aa4095ec5c102052efb976912fb","revisions":{"5ee26a20359a8aa4095ec5c102052efb976912fb":{"kind":"REWORK","_number":1,"created":"2020-04-20 17:57:22.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/18/721318/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/18/721318/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/721318/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/721318/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/721318/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/18/721318/1"}}},"commit":{"parents":[{"commit":"9d50c7918fd55559567463d68d2fcd314471130a","subject":"Merge \"Healthmanager opts aren\u0027t CLI-related\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9d50c7918fd55559567463d68d2fcd314471130a"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-04-20 17:54:22.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-04-20 17:57:21.000000000","tz":-420},"subject":"Fix Octavia unit tests for python 3.8","message":"Fix Octavia unit tests for python 3.8\n\nThe python 3.8 unit test jobs are failing on a\nmock.Mock(spec\u003dsecrets.Secret) call. Since this is legacy code and\nwe don\u0027t really need to be testing python-barbicanclient code here,\nI\u0027m removing the spec\u003d for these tests to stop artificial failures\nwhen running unit tests on python 3.8.\n\nChange-Id: I9132e2d05bf67ddff4f2d7879d799c8dd25bd3c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ee26a20359a8aa4095ec5c102052efb976912fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ee26a20359a8aa4095ec5c102052efb976912fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
