)]}'
{"id":"openstack%2Foctavia~721773","triplet_id":"openstack%2Foctavia~master~I480b7fb9756d98ba9dbcdfd1d4b193ce6868e291","project":"openstack/octavia","branch":"master","topic":"pool_tls_versions","hashtags":[],"change_id":"I480b7fb9756d98ba9dbcdfd1d4b193ce6868e291","subject":"Add TLS version configuration for pools","status":"MERGED","created":"2020-04-22 03:36:01.000000000","updated":"2020-06-04 07:07:19.000000000","submitted":"2020-06-04 07:04:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721773-1591254291820-dd03c814","meta_rev_id":"46ca01707686e04d1b04a5a989304cdbd551b9ee","_number":721773,"virtual_id_number":721773,"owner":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-04 07:04:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},{"value":2,"date":"2020-06-04 05:26:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-03 21:58:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},{"value":1,"date":"2020-06-04 05:26:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-01 19:50:43.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-06-04 05:26:24.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-06-04 07:04:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d118f648d9f7ecb00d8ab87140195c702f5a5709","author":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"date":"2020-04-22 03:36:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"682af3992465a029e84490c4407658d0de1579ba","author":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"date":"2020-04-22 03:36:31.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"be3d4de503e5301d95045b7b4b8ef4a15e6e8d2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 05:23:03.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b51033bfc01546458f425342c031a1a2 : SUCCESS in 7m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a924ea49a2344389a4b09eabb0d68c3f : SUCCESS in 11m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa4cb242139040b68b2ba5f39fbddb0b : SUCCESS in 9m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a26955439664364a9a7a9e793675caf : FAILURE in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0687fd20f2c4551ae4a589a42e36549 : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6307b5ac1f834e82b6b9bb7a5e58d0b3 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecb6eb295d094aa1844698df9ccb3b2f : FAILURE in 5m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd1717dc689c43218271255554a6ae06 : SUCCESS in 11m 16s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/44c2a90825214d1ab08faae49f82259f : SUCCESS in 5m 13s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b2874adcfc964aedac46ac794bd10063 : SUCCESS in 6m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b3664b2531674a2c931f373dea8b99d0 : SUCCESS in 6m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/11982a894007427b9f468c2eb7fd6e90 : SUCCESS in 43m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f963be45a7f842efb36592c95ffcf528 : SUCCESS in 1h 08m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/350f3e3d837c4d10b4d4c56eecce8529 : SUCCESS in 50m 54s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3c75f1a2dfb14ba59ae5bce3462b3905 : SUCCESS in 53m 16s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c9016c2ade734b6ba689ac0e68a0ef39 : SUCCESS in 59m 01s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5a0314ac232a4291ae9ab644faf074ba : SUCCESS in 59m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3f52bb1e118c487a9188ea8e43968ee9 : SUCCESS in 1h 45m 34s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/aa7226d3191544da8dc4d75c76fa7e2f : FAILURE in 1h 20m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d05b85e95a53485ea871527e29e89c14 : SUCCESS in 1h 10m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a8f35551b687d89c3cdcae3f124d782d9ff90eaa","author":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"date":"2020-04-22 16:54:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"111e3213d738f6c0344533ea78feb559a3f669b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 19:18:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9bf6a7161b164e61858105980efa67df : SUCCESS in 6m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec77f248f0ad4ec0b8b00a92766ad059 : SUCCESS in 8m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c0f8dfcc4cc49bc93ccec2b8971c4ff : SUCCESS in 12m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87c60bf6617f4e158ee2ae74d8ee1fce : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/217d9809190c4fa6a4191b49f422bbe3 : SUCCESS in 7m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f446af7b51dd40bf9c25588562f1eee9 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88e52f2a79014aea96c86a2f907e4400 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0ca5c4a3aa34c19b3961f579558a2d3 : SUCCESS in 12m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/dcdf5bdcda6d46dfad830ecdda8d4fa0 : SUCCESS in 6m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/85ffb4d738a84543bc41f4e76b894eab : SUCCESS in 8m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f74105d0bdbf4ce0b743b628f48054b2 : SUCCESS in 9m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bcb8ee71f61047ffbc5bf6844a00a850 : SUCCESS in 46m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fb42a90ad6474c3b87c9447944e98109 : FAILURE in 41m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/543ed04402184ad0a4b07a1be12b8df7 : FAILURE in 39m 47s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/721fdd6c9f214ac6b4c63440275842d3 : SUCCESS in 57m 43s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/af1b769766554ce3a9662bf6ba36653f : SUCCESS in 1h 13m 31s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/822681c777304815a47528ab22803650 : SUCCESS in 1h 10m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fd2ebe1bf3454152bf5fed27fec81032 : SUCCESS in 2h 15m 45s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/591b2c8b1f234d0cb74078889f66c845 : FAILURE in 1h 02m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/543b9c884cc6464e8b2b0857138410f6 : SUCCESS in 1h 46m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6929201aeb6793f7e5c40eee329c1f199fa9238c","author":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"date":"2020-04-30 05:43:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c1f580d4d1fc4e27ea51df454971f5a156bf3dc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 07:39:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/eea0f0e349dd47a0bd012d4879f955db : SUCCESS in 5m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cb9a145521f4419ab610a6bee2b2e0b : SUCCESS in 7m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66f1d3dfea684a759822c23105f19776 : SUCCESS in 9m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7ff68fdf4544e91bd8b3cee12f587ec : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba6c157f193040a18a8d5aa097465b88 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d31ff1a469a7411dba09336a9f8f1b64 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05ba127870d94e85a7149f2b708f465e : SUCCESS in 11m 17s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c64bc5bd8a264eeb9ada4a2c0225684b : SUCCESS in 5m 16s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3398fda44186411d91b88416333853b2 : SUCCESS in 8m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8d6924ce43984a88bb128f64abaa4b67 : SUCCESS in 6m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f3c6a95f6189450890a21342a95ce123 : SUCCESS in 43m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/01ca89409a644fa682941721bbaa7b27 : SUCCESS in 1h 42m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cfecbf821e184a87a93e6e133e16c882 : FAILURE in 57m 32s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/598e4638a01d43fc81fbf59ca0b1c187 : SUCCESS in 52m 14s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bd211df34b0442aabbb4f72a96410bb9 : SUCCESS in 1h 03m 23s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3a5b7d875db44b8eb312936b00309c15 : SUCCESS in 1h 52m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a63ceedc25304ba5a3be3fc70ba42227 : SUCCESS in 1h 52m 13s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6b99819c6fe549b9b0ad05496f861236 : SUCCESS in 1h 08m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0ab7ae21ee9a4e5c9647eb368a5dd407 : SUCCESS in 1h 39m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6c34a39659a899083b0e37313ada3ac25ff66320","author":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"date":"2020-05-10 08:17:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0ebce9dfb96a37303b7ab96c89f6217191245062","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-10 10:39:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4fb4f17e0e904e43ad54c146d58a03c9 : SUCCESS in 7m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9aed1d70bece41e883818f648138f426 : SUCCESS in 9m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/24c40d8d47ba4c43a803d4273ee55e6d : SUCCESS in 10m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08c5dd20d57c4211889b46b8dfd5462e : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f276e3ff78834ce49063f96d86a62fce : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4dbcb4d41ea34c03b13b5e95a46d31ae : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a51c55581f6e4010bf4256806918a4e4 : SUCCESS in 11m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/420d7a619c5c437698a35db65e0bd678 : SUCCESS in 6m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a48570a7035c4471a59805072fa510c5 : SUCCESS in 5m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5caa3cf06b0e469a9e8900cbe014d087 : SUCCESS in 7m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8e0f5366df3f4a14a141e162b0fc04a7 : SUCCESS in 7m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1677e910ff2a4a0caffb2d8829f2d69c : SUCCESS in 42m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/117e1f613f1d4e65b738fccd67286c30 : SUCCESS in 1h 43m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/25cfb46a6c1549128a6e1563074cdb07 : SUCCESS in 43m 02s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2b28cc0df5c34e10baffdb28d6c5572e : SUCCESS in 41m 04s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fa2fa79fc6a84a23a9430429e2f671a7 : SUCCESS in 1h 01m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/896c72762b8c471fadf926f2bdfac309 : SUCCESS in 1h 16m 18s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/22a17646d7484a8b95ec14608ba24765 : SUCCESS in 2h 01m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a4e6a6938f1e4348a4b33afa04f8f7c9 : SUCCESS in 2h 16m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/dab4f1dfdf294f8db2108859ac9c0812 : SUCCESS in 55m 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"40b53b6f7a89faccab83badff40e8a52c299cc8f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-18 22:03:43.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)\n\nMinor comment on the release note, otherwise looks good.\nThis patch probably also needs the validator for the [] case.","accounts_in_message":[],"_revision_number":4},{"id":"85f34e6fe5445e6ea760eed40b0af150ce0e3206","author":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"date":"2020-05-28 07:07:33.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a1ed3a32ffedb25b433f7e34f085bc421a72d2c0","author":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"date":"2020-05-28 07:12:46.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"001a817aa906325d420fcea6e63ae3fe4116cc95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 09:05:38.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/703827fc0dc14f3599d04c8ceb828e34 : SUCCESS in 6m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35bd9f207d734e64a1a5726caa11eeb3 : SUCCESS in 14m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd9dde02c7b144f8add279de75a3f957 : SUCCESS in 11m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8f41363e0c64011b513eb44635504a5 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2b1321445134377979a1cbc1a3b38df : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81f50c6cdbc04ad09bac77ee0abb4a7d : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87f36245f44446f688d43b9670c07f56 : SUCCESS in 15m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/126e3eda0e4a4630b37f2283ca4ffafe : SUCCESS in 8m 24s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b20e561895fd4b90851ad850659ad5c0 : SUCCESS in 5m 51s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/82a6781f5c024f56b3ddbff7f42b99db : SUCCESS in 8m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6e8ad8730c334de394741e0beb7a47f1 : SUCCESS in 7m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f2cb1098845c44cc9c4cf20b0a866b5f : SUCCESS in 43m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e16d1589c3484b71a0307f2d675a0965 : SUCCESS in 1h 32m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a6a55eb2faa9403fb82720a7c9593fc9 : SUCCESS in 49m 16s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/689403862de14c59a2414d6883bbceee : SUCCESS in 43m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7388c800cd15452c85278bb6391c1ff8 : SUCCESS in 44m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/06bae4221db641038bf3c93e61748db0 : SUCCESS in 1h 23m 58s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5f6af96f4eff476180c90f517544e887 : SUCCESS in 57m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/86eab5bc09dd48ad9c9b4c5d6d56275e : SUCCESS in 1h 04m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/00aea80a61af4097807bb086438d3c07 : SUCCESS in 1h 43m 38s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"fb76044f22615a2410cec6524f7c9a99b761b125","author":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"date":"2020-05-31 02:36:37.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5024953fdf28ca4c81884566527ed14d03d32f69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-31 04:01:31.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/30dedc6801ed48eb993de7d0a37021cc : SUCCESS in 9m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19d4dda05d8545e1b33ed4971a90e2ab : SUCCESS in 11m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56dc233bdc1c45c8af2c80453dfc070e : SUCCESS in 11m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40fb09365ed64b6ab46a5a0caf0794b3 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca3b02fcb6e34741bd8e9824788d7803 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ceca5d24c79c48dcbe6b71ea9f41a2e1 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81d062316e70437db89e7ed043585bdf : SUCCESS in 14m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2eed61a5bd3846e08d9d458f8fd516d0 : SUCCESS in 8m 27s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c0d2a75efda14040beb3a473bfc1429d : SUCCESS in 5m 26s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f2a08de40cc540009ea287f14762edc7 : SUCCESS in 10m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/35a78e2573c441ad9095a63227443f73 : SUCCESS in 6m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cffc1659d20f445aabef080510487b00 : SUCCESS in 50m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8a1db054f6b74bd9bc77fa31a6057f57 : SUCCESS in 52m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7b785e5364ca4b6bb036a72a6c9f50ed : SUCCESS in 49m 59s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d02e6a10367f4879a6edfd70dd30d7a0 : SUCCESS in 38m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/51d28bc16ce74bbe9b78c426667f47fb : SUCCESS in 47m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0eeee4e968044e8ca4d6063ec04c51d1 : SUCCESS in 1h 23m 50s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/34cab649df974b1b93b8124561bcf10a : SUCCESS in 1h 03m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e8c6dcfe485543cbb95d232380fa3e0c : SUCCESS in 1h 13m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/96b327d8b9dd4964bc1fd0fd988fcb22 : SUCCESS in 59m 39s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"57e37ce92ed6a80dc691adeb399b0490ca4dd47e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-01 19:50:06.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d966517af309dc78cb6d7816e24acbb86d102153","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-01 19:50:32.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"6aa0369329e8ace7ba18d4c1efe0e6664606d0d0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-01 19:50:43.000000000","message":"Patch Set 8: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":8},{"id":"3a01ba913da1bf4f5fdb6213a60075673f071f2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 21:56:46.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3ad459fc6c984901a3df71657fa6bc4d : SUCCESS in 5m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf09b428c6bc49e2b736a129e992c7af : SUCCESS in 10m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c828322790dd4163a57e3bec05c8e38e : SUCCESS in 10m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b50e579f663e44868b15dda22b9b2f68 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65bde8a4abda40b6957e6ace0cf818b9 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6966a85845b41f288743d38a2db636a : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96fc1da6fa6740faa6fb02525a4f9162 : SUCCESS in 11m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c60e4079805a4dd895706504a8fec168 : SUCCESS in 4m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bb0a68a3df7f48d2afaae0dc12678566 : SUCCESS in 5m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4f4b25e3ad284c29b4bb15f65c5ebec4 : SUCCESS in 9m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/439d3876f96c4d25b557f44c398db34d : SUCCESS in 6m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0f2bdb73a48d4d278d6e5b25f42aacaa : SUCCESS in 42m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/43f4c223f1574527a82d529f1ba8bbf9 : SUCCESS in 1h 56m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a20d0af95c9f4b7cae01b1ee95d27c4f : SUCCESS in 59m 04s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/99d19ddfd8fc49459b09729de4abe396 : SUCCESS in 51m 49s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/919e35c4452641db809004480b9f3526 : SUCCESS in 56m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/67eb4a9d914040978078b4c192304ff2 : SUCCESS in 1h 30m 36s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1babb856984b43048497173019aa745c : SUCCESS in 2h 00m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1fa03b306e674c2aadec243c11c56686 : SUCCESS in 1h 54m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/96f7560c45a74e7bae5cc3f2ed94c3ad : SUCCESS in 1h 40m 58s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"29e399a74d43acb6dc929fdbd6479bf22a0a3862","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-06-03 09:11:29.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)\n\nSo, are we opting to not keep the split_listeners config up to date? If that\u0027s the case, we should officially remove it, rather that just letting it be *wrong*. (I am actually in favor of removing it at this point, but IMO we can\u0027t just leave it and not update it.)","accounts_in_message":[],"_revision_number":8},{"id":"dfbed7123b58c247e484959812b2de7e06d8b75c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-03 21:58:47.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"405c064be54b38968de57ba2aaed539c5bbdf98a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 23:37:43.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a14338dca0944791b0a426deb9b44360 : SUCCESS in 8m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8722f495bd5c4cd1954fa56f48b87d4f : SUCCESS in 10m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78e11e1c995844b7b5acaf8795adc49b : SUCCESS in 11m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be07de25589848e69df848730f8ac0cf : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b03936ebd7a8405fa9a4ab7b8ee4fc99 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a55583fd76e44039cf7205729cac440 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0d05eb8f8154993b1703b6b98ce662f : SUCCESS in 12m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db9286873ddb482db77d83d5a60c7a8f : SUCCESS in 6m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d47e105d6b83462091449e550701267a : SUCCESS in 5m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/584ac311119a457abe484e35c4526650 : SUCCESS in 9m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/97916896b46745c18859c792712f8ad9 : SUCCESS in 11m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2230e64d40ec42daae13a4ac2c038b45 : SUCCESS in 1h 11m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ad306fc36cff4073b80d6b2f9efc2161 : SUCCESS in 1h 37m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/13fe11aa30b24697b93759183d20c8a7 : SUCCESS in 1h 15m 17s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2401acf7c4b54f4db186f618c011ffb6 : SUCCESS in 52m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eb2fe9e99b5d4973a3a7fd1651fb45e0 : SUCCESS in 1h 14m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1e7325630bcd43b5b5f5308bdfd2417d : SUCCESS in 1h 25m 49s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/00b8ffd93c7c47a7b8ccc47aa123eadf : SUCCESS in 1h 29m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/57b3ff48c89343f7b180c17e8c6874f6 : SUCCESS in 1h 11m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/08308687fbf94fe4b48fbf93cfd77321 : SUCCESS in 59m 57s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8d844ef8d6c4450c47785b370a35e2f6124417dc","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-06-04 05:26:24.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"8f397a8c1eaa600159f4e2eeb4a43c55374b9b2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 05:26:36.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"d9caabc34d378f1be5b93a4c476d78ecdd198a31","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 07:04:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"40a574a4ee2a88c6a6e606cde0e881cb973174c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 07:04:51.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/99b7d450755d4c3597e8253978290d23 : SUCCESS in 5m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50ff964d7bd84f9c9d47b42ad492b113 : SUCCESS in 9m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce87bf54fb8b4e898335875a634023f6 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/672636935fa74e629a951d1d9ddf5816 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5d23122d6ef46bc8d33a4e00013e8b7 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c87d8d590e894aaa85715b44ce8e92ec : SUCCESS in 12m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e600a145c0642c6b67751201791eeba : SUCCESS in 5m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b1c1f1ae8a284b869446dcb208bbb80b : SUCCESS in 7m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2086ee70968941d8bf89d584e12af734 : SUCCESS in 41m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/369e9d172f89471195702ef4d9b01909 : SUCCESS in 1h 37m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6a0c98ce93fe47649bbeb156e22260df : SUCCESS in 46m 48s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c246ed8f07ab4baa8d2a90bf2bebbcdc : SUCCESS in 53m 25s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ea7faf10342749eaaa17abdc595c7855 : SUCCESS in 47m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6f206c9b8ca44f44a53d91118c2deb8c : SUCCESS in 1h 22m 14s","accounts_in_message":[],"_revision_number":9},{"id":"a2eca20cd3ef9a2d40b691f6124b2d1997b13313","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 07:07:19.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/aaf214de55be48f7a5f59369cd20ac57 : SUCCESS in 1m 46s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71e72c8ad326427082cc756023fbc297 : SUCCESS in 2m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a1910d847f84d9fbd358d663a433e57 : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"9a6da86481f9fcbf6000b764fc4fc1f8c1341da4","revisions":{"d8e3607839c5d985908e33d83fad87edb7a850fa":{"kind":"REWORK","_number":1,"created":"2020-04-22 03:36:01.000000000","uploader":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"ref":"refs/changes/73/721773/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/721773/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/721773/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/721773/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/721773/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/721773/1"}}},"commit":{"parents":[{"commit":"13a8ee7710d24c7dc26decaadfa6162b033c40c5","subject":"Add TLS version configuration for listeners","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/13a8ee7710d24c7dc26decaadfa6162b033c40c5"}]}],"author":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-04-22 00:25:23.000000000","tz":-300},"committer":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-04-22 03:33:25.000000000","tz":-300},"subject":"Add TLS version configuration for pools","message":"Add TLS version configuration for pools\n\nAdd field tls_versions to pools for restricing TLS versions used.\nThis is a colon-separated string of versions to be used.\nAvailable values (as defined in octavia-lib):\nSSLv3, TLSv1, TLSv1.1, TLSv1.2\n\nAdd default_pool_tls_versions in octavia.conf\n\nTLSv1.3 is not explicitly supported, but will never be disabled if\npresent, see listeners patch for explanation.\n\nChange-Id: I480b7fb9756d98ba9dbcdfd1d4b193ce6868e291\nStory: 2006733\nTask: 37173\nDepends-On: Ic33d9b9a256490ae1b048cdfd2475d6340509fdb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d8e3607839c5d985908e33d83fad87edb7a850fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d8e3607839c5d985908e33d83fad87edb7a850fa"}]},"branch":"refs/heads/master"},"1352017f0da7fbc8e3101f3235af3d32aa42dcfd":{"kind":"REWORK","_number":2,"created":"2020-04-22 16:54:43.000000000","uploader":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"ref":"refs/changes/73/721773/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/721773/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/721773/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/721773/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/721773/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/721773/2"}}},"commit":{"parents":[{"commit":"13a8ee7710d24c7dc26decaadfa6162b033c40c5","subject":"Add TLS version configuration for listeners","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/13a8ee7710d24c7dc26decaadfa6162b033c40c5"}]}],"author":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-04-22 00:25:23.000000000","tz":-300},"committer":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-04-22 16:54:18.000000000","tz":-300},"subject":"Add TLS version configuration for pools","message":"Add TLS version configuration for pools\n\nAdd field tls_versions to pools for restricing TLS versions used.\nThis is a colon-separated string of versions to be used.\nAvailable values (as defined in octavia-lib):\nSSLv3, TLSv1, TLSv1.1, TLSv1.2\n\nAdd default_pool_tls_versions in octavia.conf\n\nTLSv1.3 is not explicitly supported, but will never be disabled if\npresent, see listeners patch for explanation.\n\nChange-Id: I480b7fb9756d98ba9dbcdfd1d4b193ce6868e291\nStory: 2006733\nTask: 37173\nDepends-On: Ic33d9b9a256490ae1b048cdfd2475d6340509fdb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1352017f0da7fbc8e3101f3235af3d32aa42dcfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1352017f0da7fbc8e3101f3235af3d32aa42dcfd"}]},"branch":"refs/heads/master"},"4b1e5c9724d3120d9be28515eb365ee67665c1b2":{"kind":"REWORK","_number":3,"created":"2020-04-30 05:43:14.000000000","uploader":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"ref":"refs/changes/73/721773/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/721773/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/721773/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/721773/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/721773/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/721773/3"}}},"commit":{"parents":[{"commit":"a60a24bab72f6f7cb0e40377f2cfff569282d7ac","subject":"Add TLS version configuration for listeners","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a60a24bab72f6f7cb0e40377f2cfff569282d7ac"}]}],"author":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-04-22 00:25:23.000000000","tz":-300},"committer":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-04-30 04:35:25.000000000","tz":-300},"subject":"Add TLS version configuration for pools","message":"Add TLS version configuration for pools\n\nAdd field tls_versions to pools for restricing TLS versions used.\nThis is a colon-separated string of versions to be used.\nAvailable values (as defined in octavia-lib):\nSSLv3, TLSv1, TLSv1.1, TLSv1.2, TLSv1.3\n\nAdd default_pool_tls_versions in octavia.conf\n\nNote: TLSv1.3 connections will use haproxy\u0027s default ciphers\ninstead of the listener\u0027s tls_ciphers field\n\nChange-Id: I480b7fb9756d98ba9dbcdfd1d4b193ce6868e291\nStory: 2006733\nTask: 37173\nDepends-On: Ic33d9b9a256490ae1b048cdfd2475d6340509fdb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b1e5c9724d3120d9be28515eb365ee67665c1b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b1e5c9724d3120d9be28515eb365ee67665c1b2"}]},"branch":"refs/heads/master"},"6109ccf746736cda95019ab7795667dfd6196fc6":{"kind":"REWORK","_number":4,"created":"2020-05-10 08:17:41.000000000","uploader":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"ref":"refs/changes/73/721773/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/721773/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/721773/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/721773/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/721773/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/721773/4"}}},"commit":{"parents":[{"commit":"274f69ca41c32c2e5f5b0bc6aef6c2a268f4e048","subject":"Add TLS version configuration for listeners","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/274f69ca41c32c2e5f5b0bc6aef6c2a268f4e048"}]}],"author":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-04-22 00:25:23.000000000","tz":-300},"committer":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-05-10 08:17:25.000000000","tz":-300},"subject":"Add TLS version configuration for pools","message":"Add TLS version configuration for pools\n\nAdd field tls_versions to pools for restricing TLS versions used.\nThis is a colon-separated string of versions to be used.\nAvailable values (as defined in octavia-lib):\nSSLv3, TLSv1, TLSv1.1, TLSv1.2, TLSv1.3\n\nAdd default_pool_tls_versions in octavia.conf\n\nNote: TLSv1.3 connections will use haproxy\u0027s default ciphers\ninstead of the listener\u0027s tls_ciphers field\n\nChange-Id: I480b7fb9756d98ba9dbcdfd1d4b193ce6868e291\nStory: 2006733\nTask: 37173\nDepends-On: Ic33d9b9a256490ae1b048cdfd2475d6340509fdb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6109ccf746736cda95019ab7795667dfd6196fc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6109ccf746736cda95019ab7795667dfd6196fc6"}]},"branch":"refs/heads/master"},"ba2ace659709d5ff64da2aad7cbc3ec62d7daf76":{"kind":"REWORK","_number":5,"created":"2020-05-28 07:07:33.000000000","uploader":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"ref":"refs/changes/73/721773/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/721773/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/721773/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/721773/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/721773/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/721773/5"}}},"commit":{"parents":[{"commit":"80e011f632676e456c65cf5ba224e8ec7fe1c834","subject":"Add TLS version configuration for listeners","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/80e011f632676e456c65cf5ba224e8ec7fe1c834"}]}],"author":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-04-22 00:25:23.000000000","tz":-300},"committer":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-05-28 07:07:10.000000000","tz":0},"subject":"Add TLS version configuration for pools","message":"Add TLS version configuration for pools\n\nAdd field tls_versions to pools for restricing TLS versions used.\nThis is a colon-separated string of versions to be used.\nAvailable values (as defined in octavia-lib):\nSSLv3, TLSv1, TLSv1.1, TLSv1.2, TLSv1.3\n\nAdd default_pool_tls_versions in octavia.conf\n\nNote: TLSv1.3 connections will use haproxy\u0027s default ciphers\ninstead of the listener\u0027s tls_ciphers field\n\nChange-Id: I480b7fb9756d98ba9dbcdfd1d4b193ce6868e291\nStory: 2006733\nTask: 37173\nDepends-On: Ic33d9b9a256490ae1b048cdfd2475d6340509fdb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ba2ace659709d5ff64da2aad7cbc3ec62d7daf76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ba2ace659709d5ff64da2aad7cbc3ec62d7daf76"}]},"branch":"refs/heads/master"},"f1e06813ddaaf00757e466cfafb0aee1761bb874":{"kind":"REWORK","_number":6,"created":"2020-05-28 07:12:46.000000000","uploader":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"ref":"refs/changes/73/721773/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/721773/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/721773/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/721773/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/721773/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/721773/6"}}},"commit":{"parents":[{"commit":"80e011f632676e456c65cf5ba224e8ec7fe1c834","subject":"Add TLS version configuration for listeners","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/80e011f632676e456c65cf5ba224e8ec7fe1c834"}]}],"author":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-04-22 00:25:23.000000000","tz":-300},"committer":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-05-28 07:11:42.000000000","tz":0},"subject":"Add TLS version configuration for pools","message":"Add TLS version configuration for pools\n\nAdd field tls_versions to pools for restricing TLS versions used.\nThis is a colon-separated string of versions to be used.\nAvailable values (as defined in octavia-lib):\nSSLv3, TLSv1, TLSv1.1, TLSv1.2, TLSv1.3\n\nAdd default_pool_tls_versions in octavia.conf\n\nNote: TLSv1.3 connections will use haproxy\u0027s default ciphers\ninstead of the listener\u0027s tls_ciphers field\n\nChange-Id: I480b7fb9756d98ba9dbcdfd1d4b193ce6868e291\nStory: 2006733\nTask: 37173\nDepends-On: Ic33d9b9a256490ae1b048cdfd2475d6340509fdb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f1e06813ddaaf00757e466cfafb0aee1761bb874"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f1e06813ddaaf00757e466cfafb0aee1761bb874"}]},"branch":"refs/heads/master"},"54d8e1453b0de6a9948d4a8d19f18560ace8b52b":{"kind":"REWORK","_number":7,"created":"2020-05-31 02:36:37.000000000","uploader":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"ref":"refs/changes/73/721773/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/721773/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/721773/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/721773/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/721773/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/721773/7"}}},"commit":{"parents":[{"commit":"5743fd44d0cff78ac44f41399bb0f68f564ed861","subject":"Add TLS version configuration for listeners","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5743fd44d0cff78ac44f41399bb0f68f564ed861"}]}],"author":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-04-22 00:25:23.000000000","tz":-300},"committer":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-05-31 02:35:50.000000000","tz":0},"subject":"Add TLS version configuration for pools","message":"Add TLS version configuration for pools\n\nAdd field tls_versions to pools for restricing TLS versions used.\nThis is a colon-separated string of versions to be used.\nAvailable values (as defined in octavia-lib):\nSSLv3, TLSv1, TLSv1.1, TLSv1.2, TLSv1.3\n\nAdd default_pool_tls_versions in octavia.conf\n\nNote: TLSv1.3 connections will use haproxy\u0027s default ciphers\ninstead of the listener\u0027s tls_ciphers field\n\nChange-Id: I480b7fb9756d98ba9dbcdfd1d4b193ce6868e291\nStory: 2006733\nTask: 37173\nDepends-On: Ic33d9b9a256490ae1b048cdfd2475d6340509fdb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/54d8e1453b0de6a9948d4a8d19f18560ace8b52b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/54d8e1453b0de6a9948d4a8d19f18560ace8b52b"}]},"branch":"refs/heads/master"},"1b587ff80ba4fba63c6eeaf3d9d64e627b48d206":{"kind":"REWORK","_number":8,"created":"2020-06-01 19:50:32.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/73/721773/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/721773/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/721773/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/721773/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/721773/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/721773/8"}}},"commit":{"parents":[{"commit":"5743fd44d0cff78ac44f41399bb0f68f564ed861","subject":"Add TLS version configuration for listeners","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5743fd44d0cff78ac44f41399bb0f68f564ed861"}]}],"author":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-04-22 00:25:23.000000000","tz":-300},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-01 19:50:30.000000000","tz":0},"subject":"Add TLS version configuration for pools","message":"Add TLS version configuration for pools\n\nAdd field tls_versions to pools for restricing TLS versions used.\nThis is a colon-separated string of versions to be used.\nAvailable values (as defined in octavia-lib):\nSSLv3, TLSv1, TLSv1.1, TLSv1.2, TLSv1.3\n\nAdd default_pool_tls_versions in octavia.conf\n\nNote: TLSv1.3 connections will use haproxy\u0027s default ciphers\ninstead of the listener\u0027s tls_ciphers field\n\nChange-Id: I480b7fb9756d98ba9dbcdfd1d4b193ce6868e291\nStory: 2006733\nTask: 37173\nDepends-On: Ic33d9b9a256490ae1b048cdfd2475d6340509fdb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1b587ff80ba4fba63c6eeaf3d9d64e627b48d206"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1b587ff80ba4fba63c6eeaf3d9d64e627b48d206"}]},"branch":"refs/heads/master"},"9a6da86481f9fcbf6000b764fc4fc1f8c1341da4":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-06-03 21:58:47.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/73/721773/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/721773/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/721773/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/721773/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/721773/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/721773/9"}}},"commit":{"parents":[{"commit":"6aad5d8b9f9414b729ffff5cb0524c1368a06d3b","subject":"Add TLS version configuration for listeners","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6aad5d8b9f9414b729ffff5cb0524c1368a06d3b"}]}],"author":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-04-22 00:25:23.000000000","tz":-300},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-03 21:58:47.000000000","tz":0},"subject":"Add TLS version configuration for pools","message":"Add TLS version configuration for pools\n\nAdd field tls_versions to pools for restricing TLS versions used.\nThis is a colon-separated string of versions to be used.\nAvailable values (as defined in octavia-lib):\nSSLv3, TLSv1, TLSv1.1, TLSv1.2, TLSv1.3\n\nAdd default_pool_tls_versions in octavia.conf\n\nNote: TLSv1.3 connections will use haproxy\u0027s default ciphers\ninstead of the listener\u0027s tls_ciphers field\n\nChange-Id: I480b7fb9756d98ba9dbcdfd1d4b193ce6868e291\nStory: 2006733\nTask: 37173\nDepends-On: Ic33d9b9a256490ae1b048cdfd2475d6340509fdb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a6da86481f9fcbf6000b764fc4fc1f8c1341da4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a6da86481f9fcbf6000b764fc4fc1f8c1341da4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
