)]}'
{"id":"openstack%2Foctavia~721816","triplet_id":"openstack%2Foctavia~master~Ia6d3f40ae1b9b352492162513c9262748ee67e6f","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ia6d3f40ae1b9b352492162513c9262748ee67e6f","subject":"Fix healthmanager not update amphora health when LB disable","status":"MERGED","created":"2020-04-22 11:02:11.000000000","updated":"2020-05-11 07:33:42.000000000","submitted":"2020-04-28 18:02:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721816-1588096980315-92884968","meta_rev_id":"d25c6b3a000bfafccdb5853104f9209648d9e8e4","_number":721816,"virtual_id_number":721816,"owner":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"date":"2020-04-28 13:32:17.000000000","_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-28 18:02:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"recommended":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-04-23 10:59:29.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-04-24 09:51:55.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-04-27 11:14:32.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":1,"date":"2020-04-23 13:39:10.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-04-27 11:14:32.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-04-24 09:51:55.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"date":"2020-04-23 21:21:43.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-23 10:59:29.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-04-23 13:39:10.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-04-23 21:21:43.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-04-24 09:51:55.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-04-27 11:14:32.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2020-04-28 18:02:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"de96e4eaace43d9152afbc6d1df0df37445d5428","author":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"date":"2020-04-22 11:02:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb48e4f3c49588da10f8b62dd20c8ff593054b62","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2020-04-22 11:02:20.000000000","message":"Patch Set 1:\n\nCongratulations, you\u0027ve proposed your first change in OpenDev.\n\nYour submission will now be tested automatically by Zuul, our gatekeeper,\nand reviewed by other friendly developers. They will give you feedback and\nmay require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches often take days (and sometimes weeks) to get reviewed, so be\npatient. Don\u0027t hesitate to ask for help, and answer questions about your\nwork promptly if you can. The more you get to know reviewers and get to be\nknown by them, the smoother the review and approval process will become. The\nfastest way to accomplish this is by reviewing other proposed changes\nyourself: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for participating in OpenDev, we look forward to seeing you\naround.\n\nWorkflow Guide: https://docs.openstack.org/infra/manual/developers.html","accounts_in_message":[],"_revision_number":1},{"id":"f0a8c3d4c672b6b41dd95dbea2612e85e24accaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 13:23:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b30c27efb1cd43fe9db9432c3e1cd001 : SUCCESS in 9m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dba09233e34e4958995106a087fe47d5 : SUCCESS in 9m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4212b25c301849c6b77ee51e539582ee : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44b2e9993d5243aea5e29f0c5d21e3f4 : SUCCESS in 6m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5907c9c5704c4a8d9760f24f6556a92e : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3266f329f7741b7a308e61cb9ec7d8f : FAILURE in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/397aa5ef9bef4da7bd2cc1fb24d214f8 : SUCCESS in 11m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b56c55302436475dbfada38c1abb4571 : SUCCESS in 6m 51s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/92043b201c824a7189b279d5e3a35801 : SUCCESS in 9m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c491662413774f109159a1896fa496e8 : SUCCESS in 8m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/31bc6c4ee9404254a2f8f37ede4ca44a : SUCCESS in 44m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/465f2dffedc6496ba3c72fc5d5e35b9e : SUCCESS in 2h 03m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/19b61a995ad4431eb2cb9e16f2311f0f : SUCCESS in 1h 00m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/803af16183c244ee9d1e428fdfbdfada : SUCCESS in 57m 17s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b75f28ee421e42f69998199104c0ac73 : SUCCESS in 1h 05m 43s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/da2d0df6efd74ff398b741e7a13943a9 : SUCCESS in 1h 49m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a47824260194453f895b3c7a3041001a : SUCCESS in 2h 10m 03s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a50267266bec4e5abe3caaaa12eb7775 : FAILURE in 1h 07m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5a7c28b0fdb049c8a361934f116dc790 : SUCCESS in 2h 11m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"679353d646560548b64318560186cf674ef3f85a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-04-22 15:42:24.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nThanks for your contribution, I see an issue that may affect LBs in PENDING status.","accounts_in_message":[],"_revision_number":1},{"id":"d6e23d374c1fe9abc511079ca6b41a85e846df20","author":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"date":"2020-04-22 16:19:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1ad94c580409cdf071bad4919ef975541a50a9cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 18:32:48.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/302b97de7807403982e45471b6bf8345 : SUCCESS in 12m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f5b5bae9aef447daa8647ed7ae34c0b : SUCCESS in 11m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6f730beeb724706833f9a1310855c7d : FAILURE in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/994e103a1912471192d12aa1167b8fda : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f247f45a7c5943b4a8e892a9cee2f232 : SUCCESS in 7m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cab2e414090241269bd733ca7af90bbc : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56f6159d5bd64b78a1d5966bc26677b3 : SUCCESS in 14m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/48490ed44edc4862a51caf4c6ab09276 : SUCCESS in 7m 26s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/65d44a3d67d54afbbd426fad83177c32 : SUCCESS in 10m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fb6ab906ec1543e3adc260978e79d994 : SUCCESS in 10m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1e4e06813f3b4302a6ef005509cebec0 : SUCCESS in 52m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/25077ed0d65547f291e6fa006c986981 : SUCCESS in 1h 40m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e06d256c647f49b885608690886c1c16 : SUCCESS in 56m 33s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/631f806da50f436eaad006f9dddf8904 : SUCCESS in 57m 47s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/19cd92c849d94f77b8c85ca9bbb42a51 : SUCCESS in 57m 01s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d13f9675c19b4ee69d53add33f21e4f4 : SUCCESS in 1h 48m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/691109df5f1b4647bb6092ca2ec9cc54 : SUCCESS in 1h 59m 55s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d340796c93554f07ab625dfa5d12cdb5 : FAILURE in 1h 01m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ab2fd6974d1e481ba2c2aaf3ef96ea94 : SUCCESS in 1h 44m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"997e68ce5d5d9ab35dd70c3ad42a6717a9bf84f3","author":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"date":"2020-04-22 21:27:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f9dbc1b6556b91f4002905eef43c74c00c3e2827","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 00:02:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57d2e768da8a48c2a6f1521e34070526 : SUCCESS in 9m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3414cf6831b7453a954e1c416c2b2e2b : SUCCESS in 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c57890b697d94a29aef9d2a49d64fa52 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9bb6005ee8e34281abcb219806187bfa : SUCCESS in 6m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5bcdb340ac1b4ed1bd9f4e90652139f8 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e87bdf2c64af450eb2137afd54170c19 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d4b874089f1445c96b6b922468435c2 : SUCCESS in 13m 27s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d6b4666b325b44a494745fe84faede24 : SUCCESS in 5m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/63b93c8f2e9d4218a376b8f787416af8 : SUCCESS in 9m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/03f2bde3531f4ea9a6b3c598dc3a5b0f : SUCCESS in 8m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8c95fa3fe70d4682b9e270cb5ac07467 : SUCCESS in 43m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/be0c2a55e2604d089d0d7102502abb7d : SUCCESS in 1h 12m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ab449b36146845d79e8b08a7689e4824 : SUCCESS in 57m 15s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8068dd10973d40b9b9d9368ac9524071 : SUCCESS in 52m 54s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a248c1ca503c4cae84f8a97afd5e863d : FAILURE in 44m 25s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d9d40c1b146c41e8a2cad1c55941e68b : SUCCESS in 1h 13m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4f15df1c9ad9454298f51b599c31b6d3 : SUCCESS in 2h 07m 33s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/faeb2798842c48c180119456c247bf03 : FAILURE in 1h 07m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c4f076f972de4dedaa43c5c1eb242289 : TIMED_OUT in 2h 33m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"768a2f4bb7208b93b3d1736ec55208967b3191f2","author":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"date":"2020-04-23 03:42:49.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"9d055c794a1d57b21199f6e46e0b6f21057699ae","author":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"date":"2020-04-23 04:03:53.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c979ab44a21e7e0c6704475d960b74e9b84cbe11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 05:59:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1e1bfa830e648439e69f40b64339c30 : SUCCESS in 7m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c617a325e8e42608a5aacf42c2cb20c : SUCCESS in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9943bf90dc8343738134162e8dfd1159 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef9b659dee524615b79f7265fe8ec9b8 : SUCCESS in 5m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a95f3bf233b4fe0ab1d6fac76e8dac5 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ba147606058406ebc39c605781a1603 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdb6a03dfc8846ce92ddd37af0c07dd0 : SUCCESS in 10m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/47896791b36945ea9e488e2a1112ec17 : SUCCESS in 4m 47s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/926c71d05fa847a69f47616d6e9e6e8f : SUCCESS in 7m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3b95fc6477c645dab0006d6bdb8a29e6 : SUCCESS in 6m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/da2cf0d3f1654c319624b872a0523927 : SUCCESS in 39m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a1534470dd6f4ebaae8d0c7451022921 : SUCCESS in 1h 45m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1775c7ceb7a64947b9f328dca66260b4 : SUCCESS in 53m 13s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5acd5bbd83cf48b5bbafbc01ab3512e4 : SUCCESS in 51m 38s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/191614521f8d42748a5abfd32cef81ad : SUCCESS in 50m 24s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9b2dfd25a2734865b7cf96ff18a9aecf : SUCCESS in 1h 17m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f5a546273797430494212aef04844315 : SUCCESS in 1h 11m 56s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/82b8b5bd1d684b0c9611679d52a7d55f : FAILURE in 1h 04m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ca9f5a5745444cfd9cacfcbca286705a : SUCCESS in 1h 49m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9e99a60e1a6f16ede47783551a85702cb8ad3ec7","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-04-23 06:24:31.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)\n\nThank you for your change!","accounts_in_message":[],"_revision_number":3},{"id":"9187c28274dc2214d15b62ec95a15875b5c58f70","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-04-23 06:35:23.000000000","message":"Patch Set 3: Code-Review-1\n\nI\u0027d add that this fix is worth a mention in the release notes (for example https://review.opendev.org/#/c/696944/ contains a release note for another fix)","accounts_in_message":[],"_revision_number":3},{"id":"8fb9468e38c926e9d152ded8b6204b7020f0ca91","author":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"date":"2020-04-23 09:11:04.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"91d5d517b2cf643363fc13f33eec394baa02311c","author":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"date":"2020-04-23 09:36:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"fb9894000b8c6cb503204f260a5d3386c5a7e08e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-04-23 10:02:46.000000000","message":"Patch Set 5: Backport-Candidate+1 Code-Review-1\n\n(3 comments)\n\nSome typos in the release note, otherwise patch looks good to me","accounts_in_message":[],"_revision_number":5},{"id":"50168bf44321e71a1b02f29b988a374a7d2cf54d","author":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"date":"2020-04-23 10:29:36.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"5ae195216a0f1f33afaa889271d90bb6fc916593","author":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"date":"2020-04-23 10:34:29.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"80fe1aeee837f65bf09bbeff8fbe12b80f4d3b95","author":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"date":"2020-04-23 10:36:26.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"83444b0d4da796a879bca28f8829a3eb17226ee0","author":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"date":"2020-04-23 10:38:53.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"d9a37ba2096fe71d5dddb5b24d59dea5a1e25e21","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-04-23 10:59:29.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"e489f5a3315a103bfe7bac56bb5e8d877f20a3ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 13:17:22.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cecbd9dbaf5e480b9663155011a99ea2 : SUCCESS in 10m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0fbebaa97ed48f1b323f6c320fe36c8 : SUCCESS in 11m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f277b368d6e408495f015f4b11c73cc : SUCCESS in 7m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5a2a28854794cfca7212e31d844bdb8 : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/65aca82cbe1c4801a36fcc12347f3104 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56445791594f4a338b3f9da042e4377b : FAILURE in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9768860e8e4c40199c2bff63542b5f38 : SUCCESS in 11m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db1794a6a0a44a09b48191738d868a5a : SUCCESS in 6m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ed7eb63de19d415ca41f1fc9f9bac0a5 : SUCCESS in 6m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ce011ca9219548c785a6f44c63ae42d7 : SUCCESS in 9m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3bc127f2bd2f4d4ba3f616f675b3d1e0 : SUCCESS in 7m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5b9708f0f9ac440896cd755d4d61d0bc : SUCCESS in 47m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/efe1fd74c4b44affa8fc7cf4a957ddbc : SUCCESS in 1h 40m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4d49af167624475498e389c13d57a9df : SUCCESS in 54m 54s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9e5406734e4848a99ab43a6395f1a78f : SUCCESS in 1h 07m 06s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8f19f744520540d4a3e2ff98dff186bb : FAILURE in 51m 03s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e6b8d198402547d9a94dd4602426e7b8 : SUCCESS in 1h 09m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a8efef847c964ccebfc49b1f2a4a170e : TIMED_OUT in 2h 34m 10s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/485b404187154122bf7e032db8510e59 : FAILURE in 1h 05m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/49b4ede969004a0e9f3748bdf1a6d1d6 : SUCCESS in 1h 38m 15s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"cc36f22b2a839cd852fb16fb39d4714ef201ebbb","author":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"date":"2020-04-23 13:24:22.000000000","message":"Patch Set 9:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"afa533dfc2e71666f9edf6ef78860012f828c9b7","author":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"date":"2020-04-23 13:24:47.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"b03f0b68e8d33496adb20a5e21cd17d311a227c6","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-04-23 13:39:10.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"ea9409ca440eacf89966756421876e2fb7596776","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 15:40:33.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8fcd5af2abc04fd1bbb7c8afc7e405be : SUCCESS in 13m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b47b8a7aab748fe9cc28ac1c29732ea : SUCCESS in 11m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c9ec325393e4677954f375e9f2feaca : SUCCESS in 7m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9147612506614baea7bab311c8552823 : SUCCESS in 7m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ef6e2aa6bdc54ccb99827d8132dfd18b : SUCCESS in 9m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8b110accc6a47859389e55581555b29 : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26feac12ce2b47619171b8fd77205597 : SUCCESS in 15m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a90b2e3fa3c4646886bcb0f36be1cb2 : SUCCESS in 9m 23s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e8b6b4268f26447ca8f035dad75a6d76 : SUCCESS in 9m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8a07f44390064fccb5e4ab023fee727f : SUCCESS in 11m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/740a2aa551934630bc4d9ddc9932617f : SUCCESS in 9m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3c0c718b64674cb0b17157d8a54ba334 : SUCCESS in 43m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b7bbdde8e7484615bf44a5f25e293d7a : SUCCESS in 1h 45m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c0c8fc10bd3b47a8a8cb077549845ec7 : SUCCESS in 59m 42s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ccf1552bef0e4368b9e99bbf67e840d1 : SUCCESS in 57m 50s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7b63ee2f88a344bb836b097401baed79 : SUCCESS in 1h 14m 09s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/262f9f8500144fe98471bad31056d02b : SUCCESS in 2h 04m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3ddc0e109d53499f88c2672c750f9d25 : SUCCESS in 2h 07m 20s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b891b40e306b4d0d8f4d6368a7b704c2 : FAILURE in 1h 21m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e0a7ef0cad82406697414dbb87aeeae5 : SUCCESS in 1h 14m 27s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d5707c8e77902cefa0ad11f9c5f95a9f6467ac8e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-24 09:51:55.000000000","message":"Patch Set 9: Backport-Candidate+2 Code-Review+2\n\nLooks good to me. Thank you for your contribution!","accounts_in_message":[],"_revision_number":9},{"id":"b334e461da2f407bbbff01698e5e6b02fae15af4","author":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"date":"2020-04-24 15:31:44.000000000","message":"Patch Set 9:\n\nhmm.why not merge this patch","accounts_in_message":[],"_revision_number":9},{"id":"feabb20983f3d3cad9df94c4886ee2c8605eeaf6","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-24 15:46:18.000000000","message":"Patch Set 9:\n\nhttps://docs.openstack.org/project-team-guide/review-the-openstack-way.html#code-review-2\n\n\"Projects are encouraged to require two +2 votes before a change can be merged [...] Voting +2 without Approving indicates that you’re happy for another core reviewer to Approve the change.\"","accounts_in_message":[],"_revision_number":9},{"id":"b2a416a7e1ac8cbfcfa27292779759309b732030","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2020-04-27 11:14:32.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"bfc11ab41135bb41683b7bedbacca40f095ead10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 11:14:44.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"42cf227d9381319e4798a00c688f491323114947","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 11:21:33.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ddd9f67f22df4281827efa0d14a878b1 : CANCELED\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/880017d9be5e409c8e51047cf8df1eac : FAILURE in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/726ab60ad98f4b5d9057ceb1e43e1514 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83b19a0b8e2e47419713bc26580d8637 : CANCELED\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df08163d24354eb3a2ec2c3e5da1f871 : CANCELED\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b01345206644e60a98cea86d2ac3c82 : CANCELED\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d2b8bcae6062489f8cfd67fa9ca3ce26 : SUCCESS in 6m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/058f4f6876c148c0b403ea7811529404 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1721b5965ca34376a6f7cf19209b3ebe : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/79d4c026c4e14b1799839bb15fd11268 : CANCELED\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ecf2d842b58341b2a801ad85fd6170c7 : CANCELED\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dd128218d54e41f3923f610c3deb3b5e : CANCELED\n\nWarning:\n  Comments left for invalid file octavia/api/drivers/utils.py","accounts_in_message":[],"_revision_number":9},{"id":"b47ef295124c4781b3341f5e390b8567d79585f6","author":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"date":"2020-04-27 14:57:10.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"e98a39127fa0891e9fef41346672e0f7499f4e8c","author":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"date":"2020-04-27 16:00:20.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"1fbe476f82e90d35c69718d89b015aee21de9dff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 17:41:03.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afe25afe7ebe4f4fa43c3ba9c781e5ea : SUCCESS in 9m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c559f5ce35b54f6c90dc5c342116c42f : SUCCESS in 11m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cc54075c61d4c40a1a1d43d1e79414b : FAILURE in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/563c6bc6411f42a3a6e7fb74ca10d6de : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/28f0e2e6dec04d16adebf8d17cf75c6b : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89a337a81d7444a8b6b5004bbb8e6cdb : SUCCESS in 17m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b72c155eb7f64a20a4e5045399fbb391 : SUCCESS in 9m 24s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fb699498482a4d919d9d1e0c551ac200 : SUCCESS in 6m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f0a41c090b6f45c5b7fed5036bcbb10f : SUCCESS in 9m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/89402218f9224f5ea58adf0f4d90121f : SUCCESS in 8m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5231b6bf31f94370b4a9fd5b43799003 : SUCCESS in 56m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0b5a1fc470fa4cb4b7f7ca6c18a23fe9 : SUCCESS in 1h 39m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/afe02011f78b40a2986d9dd2445355fe : SUCCESS in 55m 15s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/585ad05e7c414000bc1a1cfaf79fdeb6 : SUCCESS in 58m 28s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d82fbfec52d541ff9f4b8db758391578 : SUCCESS in 1h 09m 18s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ab738d21acc846fa89971c68a0794305 : SUCCESS in 2h 11m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/61d9ecdac04448dd884cce2f0d2e2ac6 : SUCCESS in 2h 28m 38s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/69e6b82803a944f18434c6e76fe9715a : FAILURE in 1h 18m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d25ec29c4e3547488a192b7b08fbf6a3 : SUCCESS in 2h 12m 30s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/api/drivers/utils.py","accounts_in_message":[],"_revision_number":9},{"id":"ab2bf14af365853aadbfbdb666b920ad3caf568b","author":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"date":"2020-04-27 19:03:01.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"404d8ba010afa48fd123f75d51248f43a1ac4290","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 21:11:59.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4dea047d9e9f4264984573fb17ed3f9c : SUCCESS in 12m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93c3ea2d4eb643208637d863fd1c8388 : SUCCESS in 9m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7c66a10863a49cb836541f4df67b282 : FAILURE in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b225a3e2510d4fcc934545c4efba421f : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8293a7700cfe43659cba595fef3286ac : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/750e20aee8164952b92c322f7aa517a4 : SUCCESS in 12m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9688d74780f5400e9348e04469039257 : SUCCESS in 8m 23s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/46a2efec406f42a5931471535b9619a9 : SUCCESS in 6m 19s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/11e4d5577bdb4711b5813662fedf2e85 : SUCCESS in 8m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f01b114287484e4a8e3598690ab81794 : SUCCESS in 8m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c22fa0b46a054b31983aeb0c3a888c83 : SUCCESS in 45m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6d70933e1f6a49b78641b43ae36e309f : SUCCESS in 2h 02m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fe5034a32e5c4679a11e6f2195c7dd69 : SUCCESS in 57m 50s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8e31ef274d5840438b12d452d6eac23e : FAILURE in 52m 21s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/893eaaa10d2e43909438fecd2ab8db32 : FAILURE in 39m 54s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/db5dc17a172240cfb7d05f675bc5e2f2 : SUCCESS in 1h 42m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/09a1f9c430ce4b32bc937b8dd4a5fb92 : SUCCESS in 1h 53m 20s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c9fbc25ad9ec4353a6f1e5b120abb518 : FAILURE in 1h 03m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2b3a0c176a4143a0a258083ef72271fd : SUCCESS in 1h 51m 43s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/api/drivers/utils.py","accounts_in_message":[],"_revision_number":9},{"id":"4db17b31acac3b8c6c433110669469d617751fb3","author":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"date":"2020-04-28 02:14:32.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"ecaf2b8315c1458157ae3f73ca0ea0db10637a69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 04:10:24.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81eb8a7ceb0f4ebd9621bc22a738cd4a : SUCCESS in 7m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29e53b06fba946779cb23f1125ab8f81 : SUCCESS in 8m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43cd45ee14ad489580c3297a317cd624 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/221b4ec8177e4cda8ec2fd8806cbbb7d : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f460e9b1ce944189a6fa4608f74aaee : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dd9d81a7ae44f068d913d1a876c3131 : SUCCESS in 11m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a58a5c656af41618ae0199d633090b7 : SUCCESS in 6m 31s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6a9ffe4b4018461eae9fbab2764edf61 : SUCCESS in 4m 42s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2d01f2f6d18a48f5be06054816681e53 : SUCCESS in 7m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4d1f052f94e4443fa4f1bc0bda56b5c5 : SUCCESS in 5m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/38dbcddc4859454bb34e14bc44ee473a : SUCCESS in 43m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/71e9572a1b89411c9f9afa2b71f65e8e : SUCCESS in 1h 38m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/86c107d7bb0d46d492c62eff25591cbe : SUCCESS in 57m 28s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7f00880a2d4a4279896facaeaf513543 : SUCCESS in 55m 03s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3d052a6bfa6d41c78b94b26092dc7be8 : SUCCESS in 1h 05m 30s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6910a38610c34636b401736e7daba4e9 : SUCCESS in 1h 46m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9861c9ee23584351b3a654354ee20ac4 : SUCCESS in 1h 50m 45s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/16980598318f4cfdbca9bfe6869a861b : FAILURE in 1h 01m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b64358eff8d34211a01f8e93401919da : SUCCESS in 1h 43m 42s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"7b75fdccf5ebdfae2bc8fac0532aa05525f88dfb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 04:10:35.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"4df1be6b4c13f0cdfe13d9ccd798a13714414f4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 05:10:21.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5186afc5bb184aa4bced4ec673afc575 : SUCCESS in 8m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3a27e93efe847fea69cc720c37d565d : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d237ff25afce4b4c99d919096b8b7970 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c8579d09b4e4e42bd09ae3a6898959f : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6afd3580b4044799f3973e844bee786 : SUCCESS in 10m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b31111029ad4161817020c2e222e9af : SUCCESS in 5m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/27da05c14fae4b77937b745d1982a527 : SUCCESS in 7m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f84999c4006540b2b7e6828afd0fda27 : SUCCESS in 41m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8eebac0c681a4867912caadabeed4ad9 : FAILURE in 52m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1ed190690e5e4e59bb8f139eb146be70 : SUCCESS in 39m 44s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b953e9e9d497474193b23ae7cfd1d0d8 : SUCCESS in 38m 19s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7053f3b5e8f04a51bebe690c2048b6a6 : SUCCESS in 41m 35s","accounts_in_message":[],"_revision_number":9},{"id":"2de4381fe58be5cf8e9fbbbf9d10e24edb07ed18","author":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"date":"2020-04-28 05:21:36.000000000","message":"Patch Set 9:\n\nwhy this Zuul gate keep unknow error and how to reslove it. any help","accounts_in_message":[],"_revision_number":9},{"id":"55e7261fed9d91a68cc5cc9bd6d203c3ebd7505f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-28 06:18:58.000000000","message":"Patch Set 9:\n\nrecheck\n\nThe pep8 job was failing yesterday due to a new release of astroid. It was fixed in https://review.opendev.org/#/c/723604/. The other failed jobs were due to issues with Glance and Nova timeouts.","accounts_in_message":[],"_revision_number":9},{"id":"a4fbca539c23b95bebaa0d8b609527aafeb470b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 08:09:26.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/219524ab20484288903a64cf2e7bf2be : SUCCESS in 12m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc60242b3c2e4c9f8b3350ef606473ba : SUCCESS in 14m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1530c7042a1d479ca6689180f489a1b1 : SUCCESS in 7m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b53027810954c668b6b90f16d411d38 : SUCCESS in 7m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f876a9338d6649308a45168c68085a7f : SUCCESS in 8m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e05d4fd8d3224d99995f08190a2706b4 : SUCCESS in 15m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8ae465e4d20428d92528c576bcc5e85 : SUCCESS in 7m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/01f8f37c930042028122344177d9232b : SUCCESS in 8m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5390d84449b643c692f8f792cee046cf : SUCCESS in 9m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/35261b8a32ea4686a6183318ca4d8f3f : SUCCESS in 9m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b2ffc809ddd7426eb08686da0cf3ade6 : SUCCESS in 45m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8e38ef7e084f415ca3f34bbf54ee1e3e : SUCCESS in 1h 41m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/398b546f96024cceaea1a5fff65b0622 : SUCCESS in 1h 00m 02s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8a7fbf31f1054d6d9a6d3bd845247376 : SUCCESS in 54m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1b9fd92e31d54957912113b39ef7f678 : SUCCESS in 1h 02m 22s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/324976c0c04d4c3eb9e0926d27f561ff : FAILURE in 1h 39m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5db2f64cb0144210ae2b9ec95f436133 : FAILURE in 1h 31m 12s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/791f22e85e94497a91f7c9c2988ff3f1 : FAILURE in 1h 11m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4d1ec1d529f6400a89d2f658e6cf6de5 : SUCCESS in 1h 33m 32s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"460e5717c851e7619f9fbf6c4a6cd0f80c18995a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 08:09:39.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"64dffa3111b815b112814940c8b9b919d1fd1c14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 08:54:50.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a9200adfc37f4e28bf9d233b97756e88 : SUCCESS in 11m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8368e090b6f44c696be775358c6b791 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d6b84d06cb147e6aee33b1bf8126c1f : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c800a7c5b7a48e69b16257c3b2f53b7 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8834ea1941134cedbcd3daf37a7fd61c : SUCCESS in 12m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e168c4f7a2cd4626b9d9314f0190f991 : SUCCESS in 6m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9c0001d338604a19b22b95cadf3e4bca : SUCCESS in 7m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c96b57402f494f019f35b32c1056ab0d : EXCEPTION\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/None : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/None : CANCELED\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/None : CANCELED\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b436a121b16d40129c72ce945725a02b : EXCEPTION","accounts_in_message":[],"_revision_number":9},{"id":"63bf8c99490aa30601898677a62a1b6d2d0a0201","author":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"date":"2020-04-28 09:01:08.000000000","message":"Patch Set 9:\n\nrecheck\nthank you","accounts_in_message":[],"_revision_number":9},{"id":"eef398e1bc532de47a43f3829d8889f36d901df6","author":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"date":"2020-04-28 11:56:50.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"32e74caa03c2ccfeec3bd37dfbac2fab0830f2bf","author":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"date":"2020-04-28 11:59:17.000000000","message":"Patch Set 9:\n\nim checking on zuul.openstack.org. it look like stucked by exception","accounts_in_message":[],"_revision_number":9},{"id":"f4bd78ac3f7b8a9dd42257960b80b4295720691a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-28 12:00:58.000000000","message":"Patch Set 9:\n\nZuul is having a bad day. It is impacting CI for every project. Announcement from OpenDev: \"Zuul is currently failing all testing, please refrain from approving, rechecking or submitting of new changes until this is solved.\"","accounts_in_message":[],"_revision_number":9},{"id":"234a4ca795a49cbb3b9781b4466cd4213a0ca99a","author":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"date":"2020-04-28 13:32:17.000000000","message":"Patch Set 9:\n\nrecheck\nZuul has been restarted, all events are lost, recheck or re-approve any changes submitted since 9:50 UTC","accounts_in_message":[],"_revision_number":9},{"id":"1f85ab63f5f4803045b5247399d056fc15e4d36b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 15:58:17.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/047520dc6d6f4b2b8aa89e17f27159b1 : SUCCESS in 8m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/148d1a73a5024fa2a8422fcc08d42d26 : SUCCESS in 10m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c06db318b74548b6b25d842fc0a8c7eb : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b80ad8b8c61141548201840e86bc8d3a : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1298837453634970a7a4fbffb6ffcb50 : SUCCESS in 8m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c08565eccb949ba9fe51d488e02e60c : SUCCESS in 11m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f5d1415091e408eb3af41afbb5c4829 : SUCCESS in 5m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/de8e29d3d3d24928b3c92226f2333b93 : SUCCESS in 8m 14s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d28c2c9120db4db6a4b5ffb118226015 : SUCCESS in 7m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e8a334e2fbad4f318cd3c4084ae239e5 : SUCCESS in 6m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/410219e4982c47d88aaee0ebbe6dc9ba : SUCCESS in 43m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/447c56e0fc1c4a669831b3f1fe691604 : SUCCESS in 1h 50m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7d1183135a01407e9d9d73b18a20d0a2 : SUCCESS in 59m 48s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4b2daaa1a65743168aa7013e493767cb : SUCCESS in 52m 17s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/39a45da7f2b941b7b1eb2030645fbef7 : SUCCESS in 1h 04m 27s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/65c6b5eee73d4ab3a8817af8b4c14a5a : SUCCESS in 1h 58m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d5d3ff5670224e2b83197a2875ef013d : SUCCESS in 2h 17m 53s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/22d49b3cf5464b6fa61120242cc1a1c4 : FAILURE in 1h 10m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0eea4ad121f54499ba0f32fdaf017927 : SUCCESS in 2h 06m 58s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"a379357527b78661920dcc2868771cd03dbabdc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 15:59:22.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"ac74ffcc0d88d2a70960eeb720452e80a626fb02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 18:02:59.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/010fa1f19bc0481d98c02a51e0dd2629 : SUCCESS in 11m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ee7fd93ba334c998666d84aaffa0eba : SUCCESS in 9m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ab28044e8bf47a08a13b0203766ed85 : SUCCESS in 9m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/453f63b249344547b1d5c58d866cd15a : SUCCESS in 7m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ae1eac9e0d647b68575d7167d625532 : SUCCESS in 19m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/041c211fea4e4c9aa0879a7cf4f7b0d6 : SUCCESS in 9m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f2401050c52949a6bd8d47ee1c1fd456 : SUCCESS in 11m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6819b86d2e094c22b08298a68b870af3 : SUCCESS in 46m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3435dda3bf2846e0b4e33afd49eed645 : SUCCESS in 1h 56m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9cbd8f6372b644af9222d0180ef6df12 : SUCCESS in 1h 00m 00s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/cd4f8d78011c4a96bc7f1d68f053443e : SUCCESS in 58m 27s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e9b3be41023f467494142c93d0bc8a0c : SUCCESS in 1h 08m 19s","accounts_in_message":[],"_revision_number":9},{"id":"40728096bc94f00b5d9207102b56d3bb2e585d1a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 18:03:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"f0a2a6f7f4f5182d9bf0e2ae94e120f2ae81dc9e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 18:07:46.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/589efef482224268b68e22bd6b8735c1 : SUCCESS in 1m 27s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2721036ec9946d8b4fd37a73bd7989a : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":9},{"id":"d32ef422190b3aab91ab6a8455793d600e040830","author":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"date":"2020-04-28 18:12:19.000000000","message":"Patch Set 9: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 734a87b472b8b8f347ab04602379b4d7c6f4a1ff","accounts_in_message":[],"_revision_number":9},{"id":"968f80a72fa97e8a1fa38890f0360720d0f28786","author":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"date":"2020-04-28 18:13:41.000000000","message":"Patch Set 9: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 8a6a0a071aeded9cca5199048092b4b5d644c04a","accounts_in_message":[],"_revision_number":9},{"id":"d25c6b3a000bfafccdb5853104f9209648d9e8e4","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-05-11 07:33:42.000000000","message":"Patch Set 9: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 5a14ea2f138dc1ff5a08a0ac5191bc7032a4daee","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"fe771a456d81b6600457b78210e21b88a3a9d83e","revisions":{"6a1d31489090c81c1fef4031282d4659f26575bc":{"kind":"REWORK","_number":1,"created":"2020-04-22 11:02:11.000000000","uploader":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"ref":"refs/changes/16/721816/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/721816/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/721816/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/721816/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/721816/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/721816/1"}}},"commit":{"parents":[{"commit":"9d50c7918fd55559567463d68d2fcd314471130a","subject":"Merge \"Healthmanager opts aren\u0027t CLI-related\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9d50c7918fd55559567463d68d2fcd314471130a"}]}],"author":{"name":"archiephan","email":"chungphan7819@gmail.com","date":"2020-04-22 10:59:20.000000000","tz":420},"committer":{"name":"archiephan","email":"chungphan7819@gmail.com","date":"2020-04-22 11:02:07.000000000","tz":420},"subject":"fix healthmonitor not update amphora health when LB disable Story: 2007587","message":"fix healthmonitor not update amphora health when LB disable\nStory: 2007587\n\nChange-Id: Ia6d3f40ae1b9b352492162513c9262748ee67e6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6a1d31489090c81c1fef4031282d4659f26575bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6a1d31489090c81c1fef4031282d4659f26575bc"}]},"branch":"refs/heads/master"},"30499584b9b286a02a8d0bc990856af778b572c0":{"kind":"REWORK","_number":2,"created":"2020-04-22 16:19:01.000000000","uploader":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"ref":"refs/changes/16/721816/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/721816/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/721816/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/721816/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/721816/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/721816/2"}}},"commit":{"parents":[{"commit":"9d50c7918fd55559567463d68d2fcd314471130a","subject":"Merge \"Healthmanager opts aren\u0027t CLI-related\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9d50c7918fd55559567463d68d2fcd314471130a"}]}],"author":{"name":"archiephan","email":"chungphan7819@gmail.com","date":"2020-04-22 10:59:20.000000000","tz":420},"committer":{"name":"archiephan","email":"chungphan7819@gmail.com","date":"2020-04-22 16:18:58.000000000","tz":420},"subject":"fix healthmonitor not update amphora health when LB disable Story: 2007587","message":"fix healthmonitor not update amphora health when LB disable\nStory: 2007587\n\nChange-Id: Ia6d3f40ae1b9b352492162513c9262748ee67e6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/30499584b9b286a02a8d0bc990856af778b572c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/30499584b9b286a02a8d0bc990856af778b572c0"}]},"branch":"refs/heads/master"},"ead6745364490615dd807529be0f5ece6ab1a502":{"kind":"REWORK","_number":3,"created":"2020-04-22 21:27:04.000000000","uploader":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"ref":"refs/changes/16/721816/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/721816/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/721816/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/721816/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/721816/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/721816/3"}}},"commit":{"parents":[{"commit":"9d50c7918fd55559567463d68d2fcd314471130a","subject":"Merge \"Healthmanager opts aren\u0027t CLI-related\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9d50c7918fd55559567463d68d2fcd314471130a"}]}],"author":{"name":"archiephan","email":"chungphan7819@gmail.com","date":"2020-04-22 10:59:20.000000000","tz":420},"committer":{"name":"archiephan","email":"chungphan7819@gmail.com","date":"2020-04-22 21:26:48.000000000","tz":420},"subject":"fix healthmonitor not update amphora health when LB disable fix pep8 Story: 2007587","message":"fix healthmonitor not update amphora health when LB disable\nfix pep8\nStory: 2007587\n\nChange-Id: Ia6d3f40ae1b9b352492162513c9262748ee67e6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ead6745364490615dd807529be0f5ece6ab1a502"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ead6745364490615dd807529be0f5ece6ab1a502"}]},"branch":"refs/heads/master"},"a44cb5a93e39289f11560382aa3f1b232acd152e":{"kind":"REWORK","_number":4,"created":"2020-04-23 09:11:04.000000000","uploader":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"ref":"refs/changes/16/721816/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/721816/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/721816/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/721816/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/721816/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/721816/4"}}},"commit":{"parents":[{"commit":"9d50c7918fd55559567463d68d2fcd314471130a","subject":"Merge \"Healthmanager opts aren\u0027t CLI-related\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9d50c7918fd55559567463d68d2fcd314471130a"}]}],"author":{"name":"archiephan","email":"chungphan7819@gmail.com","date":"2020-04-22 10:59:20.000000000","tz":420},"committer":{"name":"archiephan","email":"chungphan7819@gmail.com","date":"2020-04-23 09:10:51.000000000","tz":420},"subject":"Fix healthmanager not update amphora health when LB disable","message":"Fix healthmanager not update amphora health when LB disable\n\nWhen disable a loadbalancer have listener, the Heath Manager not update\namphora heath make it keep failover after heartbeat_timeout end time\n\nStory: 2007587\nTask: 39521\n\nChange-Id: Ia6d3f40ae1b9b352492162513c9262748ee67e6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a44cb5a93e39289f11560382aa3f1b232acd152e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a44cb5a93e39289f11560382aa3f1b232acd152e"}]},"branch":"refs/heads/master"},"d7d785853e217c2c4787a40384b47fd849937611":{"kind":"REWORK","_number":5,"created":"2020-04-23 09:36:49.000000000","uploader":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"ref":"refs/changes/16/721816/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/721816/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/721816/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/721816/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/721816/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/721816/5"}}},"commit":{"parents":[{"commit":"9d50c7918fd55559567463d68d2fcd314471130a","subject":"Merge \"Healthmanager opts aren\u0027t CLI-related\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9d50c7918fd55559567463d68d2fcd314471130a"}]}],"author":{"name":"archiephan","email":"chungphan7819@gmail.com","date":"2020-04-22 10:59:20.000000000","tz":420},"committer":{"name":"archiephan","email":"chungphan7819@gmail.com","date":"2020-04-23 09:36:35.000000000","tz":420},"subject":"Fix healthmanager not update amphora health when LB disable","message":"Fix healthmanager not update amphora health when LB disable\n\nWhen disable a loadbalancer have listener, the Heath Manager not update\namphora heath make it keep failover after heartbeat_timeout end time\n\nStory: 2007587\nTask: 39521\n\nChange-Id: Ia6d3f40ae1b9b352492162513c9262748ee67e6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d7d785853e217c2c4787a40384b47fd849937611"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d7d785853e217c2c4787a40384b47fd849937611"}]},"branch":"refs/heads/master"},"d4058679e2e57aa0201b015665babb3381a2c5f3":{"kind":"REWORK","_number":6,"created":"2020-04-23 10:29:36.000000000","uploader":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"ref":"refs/changes/16/721816/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/721816/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/721816/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/721816/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/721816/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/721816/6"}}},"commit":{"parents":[{"commit":"9d50c7918fd55559567463d68d2fcd314471130a","subject":"Merge \"Healthmanager opts aren\u0027t CLI-related\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9d50c7918fd55559567463d68d2fcd314471130a"}]}],"author":{"name":"archiephan","email":"chungphan7819@gmail.com","date":"2020-04-22 10:59:20.000000000","tz":420},"committer":{"name":"archiephan","email":"chungphan7819@gmail.com","date":"2020-04-23 10:28:22.000000000","tz":420},"subject":"Fix healthmanager not update amphora health when LB disable","message":"Fix healthmanager not update amphora health when LB disable\n\nWhen disable a loadbalancer have listener, the Heath Manager not update\namphora heath make it keep failover after heartbeat_timeout end time\n\nStory: 2007587\nTask: 39521\n\nChange-Id: Ia6d3f40ae1b9b352492162513c9262748ee67e6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d4058679e2e57aa0201b015665babb3381a2c5f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d4058679e2e57aa0201b015665babb3381a2c5f3"}]},"branch":"refs/heads/master"},"951a8c0926f7b6816bb0b6692974e7171395b28d":{"kind":"REWORK","_number":7,"created":"2020-04-23 10:34:29.000000000","uploader":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"ref":"refs/changes/16/721816/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/721816/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/721816/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/721816/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/721816/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/721816/7"}}},"commit":{"parents":[{"commit":"9d50c7918fd55559567463d68d2fcd314471130a","subject":"Merge \"Healthmanager opts aren\u0027t CLI-related\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9d50c7918fd55559567463d68d2fcd314471130a"}]}],"author":{"name":"archiephan","email":"chungphan7819@gmail.com","date":"2020-04-22 10:59:20.000000000","tz":420},"committer":{"name":"archiephan","email":"chungphan7819@gmail.com","date":"2020-04-23 10:34:18.000000000","tz":420},"subject":"Fix healthmanager not update amphora health when LB disable","message":"Fix healthmanager not update amphora health when LB disable\n\nWhen disable a loadbalancer have listener, the Heath Manager not update\namphora heath make it keep failover after heartbeat_timeout end time\n\nStory: 2007587\nTask: 39521\n\nChange-Id: Ia6d3f40ae1b9b352492162513c9262748ee67e6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/951a8c0926f7b6816bb0b6692974e7171395b28d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/951a8c0926f7b6816bb0b6692974e7171395b28d"}]},"branch":"refs/heads/master"},"58bf2b6930d227141b19ee68f66c1dda09750662":{"kind":"REWORK","_number":8,"created":"2020-04-23 10:36:26.000000000","uploader":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"ref":"refs/changes/16/721816/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/721816/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/721816/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/721816/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/721816/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/721816/8"}}},"commit":{"parents":[{"commit":"9d50c7918fd55559567463d68d2fcd314471130a","subject":"Merge \"Healthmanager opts aren\u0027t CLI-related\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9d50c7918fd55559567463d68d2fcd314471130a"}]}],"author":{"name":"archiephan","email":"chungphan7819@gmail.com","date":"2020-04-22 10:59:20.000000000","tz":420},"committer":{"name":"archiephan","email":"chungphan7819@gmail.com","date":"2020-04-23 10:36:12.000000000","tz":420},"subject":"Fix healthmanager not update amphora health when LB disable","message":"Fix healthmanager not update amphora health when LB disable\n\nWhen disable a loadbalancer have listener, the Heath Manager not update\namphora heath make it keep failover after heartbeat_timeout end time\n\nStory: 2007587\nTask: 39521\n\nChange-Id: Ia6d3f40ae1b9b352492162513c9262748ee67e6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/58bf2b6930d227141b19ee68f66c1dda09750662"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/58bf2b6930d227141b19ee68f66c1dda09750662"}]},"branch":"refs/heads/master"},"fe771a456d81b6600457b78210e21b88a3a9d83e":{"kind":"REWORK","_number":9,"created":"2020-04-23 10:38:53.000000000","uploader":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"ref":"refs/changes/16/721816/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/721816/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/721816/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/721816/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/721816/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/721816/9"}}},"commit":{"parents":[{"commit":"9d50c7918fd55559567463d68d2fcd314471130a","subject":"Merge \"Healthmanager opts aren\u0027t CLI-related\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9d50c7918fd55559567463d68d2fcd314471130a"}]}],"author":{"name":"archiephan","email":"chungphan7819@gmail.com","date":"2020-04-22 10:59:20.000000000","tz":420},"committer":{"name":"archiephan","email":"chungphan7819@gmail.com","date":"2020-04-23 10:38:41.000000000","tz":420},"subject":"Fix healthmanager not update amphora health when LB disable","message":"Fix healthmanager not update amphora health when LB disable\n\nWhen disable a loadbalancer have listener, the Heath Manager not update\namphora heath make it keep failover after heartbeat_timeout end time\n\nStory: 2007587\nTask: 39521\n\nChange-Id: Ia6d3f40ae1b9b352492162513c9262748ee67e6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fe771a456d81b6600457b78210e21b88a3a9d83e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fe771a456d81b6600457b78210e21b88a3a9d83e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
