)]}'
{"id":"openstack%2Foctavia~722121","triplet_id":"openstack%2Foctavia~master~Ifa0d695c2227772d6b37987a7857fe58ca660dc8","project":"openstack/octavia","branch":"master","topic":"min_tls_version","hashtags":[],"change_id":"Ifa0d695c2227772d6b37987a7857fe58ca660dc8","subject":"Add minimum TLS version option in octavia.conf","status":"MERGED","created":"2020-04-22 19:03:55.000000000","updated":"2020-06-29 22:49:12.000000000","submitted":"2020-06-29 22:47:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722121-1593470856504-ccd9e89b","meta_rev_id":"3ae67fb444fd9420a7c9e77c23af3405a89d2beb","_number":722121,"virtual_id_number":722121,"owner":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-29 22:47:36.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"recommended":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"all":[{"value":1,"date":"2020-06-05 06:29:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},{"value":2,"date":"2020-06-29 20:32:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-06-29 20:47:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2020-06-29 20:47:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-29 20:32:46.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-06-29 20:47:32.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-06-29 22:47:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7796cbe082d6ac2df67e0d6fad1bde8857e8ae8d","author":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"date":"2020-04-22 19:03:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d85f25b625f7a47ad579f2b22ad9a966da0434bf","author":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"date":"2020-04-22 19:04:09.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"cbc675d161d37096a6b72aa96945bd1374d0242a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 20:10:41.000000000","message":"Patch Set 1: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/463f440b58f84d0e8e3a425456405e17 : FAILURE in 4m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26af127574e041f493c76090d3978eca : FAILURE in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf877fc725f74cb69e9f8e2a7407bfe2 : FAILURE in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b5d1b796fe44498af5da3a27fcaa69f : FAILURE in 5m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe348369a7024e38a9ca10e6639b6dc6 : FAILURE in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42e6a19c5523446a92a40e16fcc396b8 : FAILURE in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87385398c7344ad5b19c8f1ed5feea1f : SUCCESS in 11m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdf3dcd771224899aefe64897a7210fb : SUCCESS in 8m 05s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/59178d2cc94349a396c835ca34c19132 : FAILURE in 5m 47s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b3b6bc9245a4421cb297551d17919199 : FAILURE in 7m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5239c3081cfe4033880cee79de71fc54 : FAILURE in 5m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bc12f5ec3c1e48f0a2496c93f1decdb3 : FAILURE in 39m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/562ea651e1224ee88c370c7f432d2ec3 : FAILURE in 34m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/81177a3fbe9b4c2197e8e413f9166313 : FAILURE in 39m 08s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4d087bd2ebe0455991c04e3c9271a25c : FAILURE in 39m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4c16f890b2e2426db2a44bf6ebe00bfd : FAILURE in 34m 47s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5d7f50e1faa3401cbc4ce4d9623bf3dd : FAILURE in 34m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/45d62c5c1ea4401d974bf79fbd35b9e0 : FAILURE in 40m 29s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fa4f8634e5fb425eb584ac8f4a23a9d3 : FAILURE in 1h 00m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4a5decac04e4442e91a6799e4056e847 : FAILURE in 33m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"736ed4e63d44f75addf4a5ce258fb314d378beea","author":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"date":"2020-05-10 20:47:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f900c75265eb259ed4b0de71f07a3b87feb96fed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-10 22:48:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5b694177d454e8b8bdf10627c8e40ce : SUCCESS in 9m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf5a5f13a5804db6a38f932f323112fd : SUCCESS in 8m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e70a5fb8d99401596ea968597bc7df9 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae969d071e9244cc8cf4725df1b0e536 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a4f5c7e78804ccba2f3205615cb0f03 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5c47a503f9c49da94b9a8cee7f16bbd : SUCCESS in 10m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e31f07082f674a37bd7f3424a6bd1d43 : SUCCESS in 6m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9d92aa9369bc4b02a6fabd72e251f576 : SUCCESS in 5m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3eb2a6c564ee454abc213b6a5dd8699e : SUCCESS in 8m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f0daaf406abe4627893a692fe463ad89 : SUCCESS in 6m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/aea82041a0eb4a6690ec59aa662311bc : SUCCESS in 47m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fb0511f4b4984db79f249e58051869a7 : SUCCESS in 1h 49m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/080abf77bb2b42ff81520e0a3577c001 : POST_FAILURE in 1h 00m 03s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/cf97ce68901247bea2610ced9f840ba9 : SUCCESS in 51m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e6693a2fc6634916833c9bfecee936ac : SUCCESS in 1h 10m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f7e9bcf8c9974f759be3bb58b7dcb4d6 : SUCCESS in 1h 28m 18s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6892fed810be4f2f843adb01bc420d65 : SUCCESS in 1h 50m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bd6b3b8b717a4b4dbeffa91df54e2d3a : SUCCESS in 1h 55m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6b7b41df864e40f392b8d3a02867475f : SUCCESS in 1h 42m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fd6afc37ad742e6f38cf7cf2b2e946f9acbbd6a5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-18 22:20:09.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)\n\nOne comment about adding TLSv1.3 back in, otherwise it is just a nit and a design question for the other reviewers.","accounts_in_message":[],"_revision_number":2},{"id":"33914b033467a8c9e6a1110fa14c31d19bc06143","author":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"date":"2020-05-29 06:31:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1f38bb17cae250721556fc7f12157923de5b2722","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 08:53:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44ebd02209834783bc949b4290283861 : SUCCESS in 11m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b1e899a6818497da396f0734303ebe9 : SUCCESS in 11m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26f6b1cf9f5f47738f5a8fb245621bfe : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6742917428bc4319ac2d57f1fe6dff13 : SUCCESS in 9m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ebf73dcdcb24cff90eb9225ae41aa83 : SUCCESS in 10m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4aa40218c2db41c79357c5d85cbd40b2 : SUCCESS in 17m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34c35bdd2c034879b699debfc0dbed82 : SUCCESS in 8m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7dd39a4765df4295898768485f2a34a3 : SUCCESS in 10m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/68f96b5fc1544ed585aa20c2207d3e9d : SUCCESS in 11m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4a2733cbc1774c4da9275c9aeab51697 : SUCCESS in 10m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/81e9c567a8244720b24b01ab6599a369 : SUCCESS in 46m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ac5a3077780b4f1ea0bf3b73f7df3e72 : SUCCESS in 42m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6322b1941ba24c299c8edcc3dc217d9f : SUCCESS in 59m 57s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d494f77289554589a5643a48c1083611 : SUCCESS in 57m 40s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/23b29836f48e4d279452dd1bdb08e8cb : SUCCESS in 1h 08m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a5d8b1caf042478fb053532d7b9638f7 : SUCCESS in 1h 35m 01s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/426c4d0cd4c44413b3050b86e323ab54 : SUCCESS in 1h 55m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/30822e3bb04242fa9474fe82f06f00ac : SUCCESS in 2h 14m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c3a1bac820184604b1286e806deda646 : SUCCESS in 1h 54m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0447a57c7deb71111649375a4e40619614597f2f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-29 21:13:17.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"21ae23f9ebb7e0a982a36fd2cd4580dd19d44422","author":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"date":"2020-05-31 03:02:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b5a4f5e15a6062cdd0eb21d2462269af5c0a0d96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-31 04:26:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8a71ebf8bb24715858868a3af37b7c7 : SUCCESS in 9m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25853fbe98ae41b1a3f765b967661c55 : SUCCESS in 8m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46729c1a30874944b22c5cdd46c6416d : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6e57a9b530248e0afcbad0fc813e6fd : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2184751140ba47f4b4d96353e8281273 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efbe69dd400f449483cd98b7a9f9cd19 : SUCCESS in 15m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd16daa832994248b1288d3f5832bd28 : SUCCESS in 4m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3bf20e068fb44ddd8e1423e2f334ebda : SUCCESS in 6m 35s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c9d817922638440cac73ceb43ca6e3e5 : SUCCESS in 7m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/78e54dd8b32c454ea41cead2da92960e : SUCCESS in 6m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/31fa17bb4d7d438b8325b8d0a0099f60 : SUCCESS in 40m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6c0a14cdcbb74ea3a6454a7a9334e561 : SUCCESS in 55m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b8beb3a3cd2c47a08d0c84bc138aeff5 : SUCCESS in 41m 21s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f046dbe4ac68475681ac254022c4aae3 : SUCCESS in 41m 34s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/611abd78ceec4b0c8331df5384e89daf : SUCCESS in 41m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dd166ccc37374b93ba160f262a1a48eb : SUCCESS in 1h 22m 38s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c485ae7eb19a484aa16e82f2778f4886 : SUCCESS in 56m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5a326f4e833c4586a2628405a66c4d9d : SUCCESS in 1h 04m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/011dc44469344fa99a134d23ae5bf8ac : SUCCESS in 45m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a9cf605eaf3e3e0a2aeac4175e41bbcaaac4cf7f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-01 20:32:31.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"19e5693ebb5ed3e8f051e8225a04b6a8870f0016","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-06-03 09:27:34.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)\n\n-1 primarily for the message translation question, but also using choices is much safer.","accounts_in_message":[],"_revision_number":4},{"id":"200e4d18612cc914ebb4fd4b7341527344e86d0a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-03 22:15:08.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"823886fa602fa7532c92fd19e21547da969e2fdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 23:37:15.000000000","message":"Patch Set 5: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f33c6179c5e4a119683033d774e2375 : FAILURE in 4m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/276fe044767242998b9dc20ad433e0c4 : FAILURE in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3badc244b9f54efa8891165fe27b6c33 : FAILURE in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29b15a71057d41eab664a3517dc9cacc : FAILURE in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5edec601aab4f6b8bca1ca52cf375ea : FAILURE in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0023daa4e30846d9a87185a3e83e8e30 : FAILURE in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/425f006ac7fa44419d763df5cc9d4288 : SUCCESS in 5m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f7f4c7aef1ca4aad9fa3a488cc3b8ce1 : FAILURE in 9m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f6b56365605f45ddbec65e72c92d2ecb : FAILURE in 11m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/59539ef529614be5badbbec7bee9be02 : FAILURE in 8m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2fb2d5e351114a5a93582e46bc472b24 : FAILURE in 51m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/08ff3bc9a85c470f9f742882fcf500b4 : FAILURE in 59m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/662022937b3d4d0eaa7c71858c07c0e6 : FAILURE in 1h 02m 33s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/602f470a43c04411b803d9a5e4cb82ed : FAILURE in 59m 46s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fc8f3e80bb6a46cc9cca5a0bda633653 : FAILURE in 1h 00m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/742e6cf6c4494a0ea3b53b053e2bb33b : FAILURE in 1h 19m 34s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/43f5d80d391e4829993c53d124544f7a : FAILURE in 1h 01m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3dcd3e77dcb74e019d868b468ef93acb : FAILURE in 1h 00m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9f3a2f8c7df9486b86247930d6e246e1 : FAILURE in 32m 20s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ba532d90cb51a70a7aa26acc36ed6adc61445ede","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-06-04 05:25:55.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4220eff1a8e2ab5c09da8d64d7165449fa7d6f9e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-06-04 05:27:36.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"b49058d72e81caeca8e9bb268672d61536445a00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 06:36:07.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b5164cae32c4c968c806140886d1416 : FAILURE in 4m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5804d1f164d4f0bbb172d42403cf4e3 : FAILURE in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06d85cbe63e841bbb017b807277b12b1 : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6308be6c31f4e07a412a99646387883 : FAILURE in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17984b6eb1b94b96a93b83f63cbe8d59 : FAILURE in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bd7404abb2349ee9c4742dd72c33de7 : FAILURE in 11m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d71c1d6e8f7d4bce8fd5b8350586ef9a : SUCCESS in 6m 25s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/89e3871a182e47d7b0c4a197719ee5bf : FAILURE in 6m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/61cf9d3f9739432a8cb1aae61147576c : FAILURE in 7m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/59d1500c46844971a9b70916d5de3fd0 : FAILURE in 3m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/77809898c7ec4f9a922454ec4b71cc7c : FAILURE in 27m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/949b197feed6401fbb970ddcdc2da6f7 : FAILURE in 33m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c530278c539f4f3099b8abc19cac9885 : FAILURE in 40m 17s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d76ca60eb2034409a29a546ea7b1054f : FAILURE in 33m 16s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/48fbfedf88114520a6ef210df1a8b4c3 : FAILURE in 36m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8a0a6a52d7e74a80956b10480da50ffc : FAILURE in 1h 03m 02s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/272770e4575441f9b550a85a99dfc41b : FAILURE in 38m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d01d24af9b8e406b8b0b84ea9646df57 : FAILURE in 38m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/054b36dc01744a0d8de46d4a5b3c1330 : FAILURE in 41m 15s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3d8142cccf014dba4f35ca30a697dcaf24a0ad24","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-06-04 07:18:30.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e6d748cf79bafc485c8ff3ec06ac7204731eb252","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-06-04 07:22:09.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b5e4ada044164c4ca6795a68de87e98a9fdf445b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 09:27:20.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/156d39c8d752476996c71c24bc3e78d9 : FAILURE in 9m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/74568eed17f4435e9174e39060ae88db : FAILURE in 9m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f467dd74e8e4c5899f57edcdbd2ee80 : SUCCESS in 11m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/517818d150784c31aa3be4af2df5aade : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5d9d405121f406ebcd05c01010d4b24 : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6f826459ff2491486ec8856ee54d4cd : SUCCESS in 14m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6641e46c957448239dad099d5c174ccf : SUCCESS in 6m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b507ce328d9844ce8b4d872f08aa345a : SUCCESS in 7m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6258e9948be84381bac7ee1c9c2be011 : FAILURE in 9m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eb8cf1fa81534ace955ecc9cb850860b : FAILURE in 8m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f684d1142d2b4a788d66b072191e9c70 : SUCCESS in 54m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0cedcb0c9bd34b8a9d0a38eca3885fa6 : SUCCESS in 59m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a70e95b1ad8b41ceb4801a08e3f1fcc1 : FAILURE in 55m 59s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b65d1db55cb04df7b4ac1d78fbe12cbc : SUCCESS in 1h 07m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4ea5f40f462745f28074aa3241575ca1 : SUCCESS in 47m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4eccc14873484054890b8dbd4bacd144 : SUCCESS in 1h 39m 40s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7cbf236bdd9c4178beb93f12761ad419 : SUCCESS in 1h 29m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a0c0bcaff1ea47a2957b672c1d099791 : SUCCESS in 1h 08m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3c12207b20ed490ca228f378764e037b : SUCCESS in 1h 59m 30s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b960f6f828434f64cfe25ebd3860cafb33698436","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-04 20:11:02.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"de214dcfc68343d4e06f90a3caf597ef784e480e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 22:15:04.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/edff67666c4f4e5498caa256d41ceab7 : SUCCESS in 9m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5759b271c21a4f7298d0de37bd2f20a5 : SUCCESS in 9m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb29cf4c918749b6bf3cdd6a9ac95d77 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5accbf3bd104f1c862a4057a7f82fa1 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a92231881be40da97d68296a21a7bac : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/090d4ac63c1e45308b3a55023ebaa138 : SUCCESS in 11m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11beefb34fcd487db686d33c6a99b566 : SUCCESS in 5m 41s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b14048baa7124edb857e28f0d35f6619 : SUCCESS in 5m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/349fe060b4624d3c976a77cdfd66db54 : SUCCESS in 9m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5471eecbd1ad486e99a320c2ab5b4800 : SUCCESS in 6m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7d67ea2344fd407892b8a398beffff11 : SUCCESS in 45m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c007c6893c0f4a839865fe72641650f2 : SUCCESS in 1h 58m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/29ca2f7c7ef0451ca0f51dd15f83d83f : SUCCESS in 1h 13m 53s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4137c63d0631453bbfca16983266350a : SUCCESS in 56m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c5f11ad444a44155ab0b7f0fa6797226 : SUCCESS in 1h 05m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/50a261b1bcec46f2b4d974976083f480 : SUCCESS in 1h 38m 10s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5abca5f94e054482adda28970b229416 : SUCCESS in 1h 48m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cebe0d07e49e4cdb9c995a400dc5ea4e : SUCCESS in 1h 12m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/db314d7b942d4860a32793bbf171e3e4 : SUCCESS in 56m 44s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f50449353345fcba76f59cba9be7143684fce81d","author":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"date":"2020-06-05 06:29:17.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"f045bf8ae59e60793ca78de5d7df00a0fd40d87f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-06-29 20:32:46.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"e7131c383069f3a5b08d4b511e937f44f819b8fa","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-29 20:47:32.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)\n\nLGTM","accounts_in_message":[],"_revision_number":8},{"id":"447feabc0d261a8686ce3f98e16a6d2cba0f548c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 20:47:44.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"2e9222a694e5769b6f522d7282cc7d2f2d5957c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 22:47:36.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62f1eb813cae4bab898d856af418a5a8 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b637d7cef5dd4edb861447376c91ae56 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20730ab145e946aab03cf1d084b057d7 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/610b3073f5c644ed954120c6845c809c : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/704b2c228e0e45bdb5e635f81cacd74e : SUCCESS in 9m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afff5f0881e548bfa29d202180850ad0 : SUCCESS in 4m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/647a788bed9a4dfe876bfa96e32779ea : SUCCESS in 5m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/45a9c4511d3549eeb82b7b2226589685 : SUCCESS in 32m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9f43540d33884d1abf6217bf89da69ff : SUCCESS in 57m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a1755d731c1e4a4eafa29726be5cd15e : SUCCESS in 53m 13s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/890ad28ecbdb434facdfd6588d0babb4 : SUCCESS in 30m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ffe089527ae84180860f9c7c9adb2a35 : SUCCESS in 57m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ef6728bc26734a448616accbb6b0de74 : SUCCESS in 1h 00m 38s","accounts_in_message":[],"_revision_number":8},{"id":"fb4ac1bbcab66527c767e2de2c2dcb3b9a9bd799","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 22:47:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"2e7a0c612a3b095ee7640cf44ee37e882da62265","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 22:49:12.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4920f85933e497ea4e728aae1506f0a : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fad28fa81d12466aaa2874c7a1948da6 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"270b973bf96c7a873f495ac6977b8188cf63bf72","revisions":{"0ccb8268f2e9bfcebb155fc9eb5fe7c0a5a519c7":{"kind":"REWORK","_number":1,"created":"2020-04-22 19:03:55.000000000","uploader":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"ref":"refs/changes/21/722121/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/21/722121/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/21/722121/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/21/722121/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/21/722121/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/21/722121/1"}}},"commit":{"parents":[{"commit":"1352017f0da7fbc8e3101f3235af3d32aa42dcfd","subject":"Add TLS version configuration for pools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1352017f0da7fbc8e3101f3235af3d32aa42dcfd"}]}],"author":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-04-22 19:02:00.000000000","tz":-300},"committer":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-04-22 19:02:00.000000000","tz":-300},"subject":"Add minimum TLS version option in octavia.conf","message":"Add minimum TLS version option in octavia.conf\n\nAdd new configuration option \"minimum_tls_versions\" to octavia.conf.\nListeners, pools, or the default values for either will be blocked from\nusing lower versions.\n\nChange-Id: Ifa0d695c2227772d6b37987a7857fe58ca660dc8\nStory: 2006733\nTask: 37171\nDepends-On: I480b7fb9756d98ba9dbcdfd1d4b193ce6868e291\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0ccb8268f2e9bfcebb155fc9eb5fe7c0a5a519c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0ccb8268f2e9bfcebb155fc9eb5fe7c0a5a519c7"}]},"branch":"refs/heads/master"},"65eeb423b05c8e3c2aa1f51f2dac2015c9bf4ef7":{"kind":"REWORK","_number":2,"created":"2020-05-10 20:47:03.000000000","uploader":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"ref":"refs/changes/21/722121/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/21/722121/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/21/722121/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/21/722121/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/21/722121/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/21/722121/2"}}},"commit":{"parents":[{"commit":"6109ccf746736cda95019ab7795667dfd6196fc6","subject":"Add TLS version configuration for pools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6109ccf746736cda95019ab7795667dfd6196fc6"}]}],"author":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-04-22 19:02:00.000000000","tz":-300},"committer":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-05-10 20:46:01.000000000","tz":-300},"subject":"Add minimum TLS version option in octavia.conf","message":"Add minimum TLS version option in octavia.conf\n\nAdd new configuration option \"minimum_tls_versions\" to octavia.conf.\nListeners, pools, or the default values for either will be blocked from\nusing lower versions.\n\nChange-Id: Ifa0d695c2227772d6b37987a7857fe58ca660dc8\nStory: 2006733\nTask: 37171\nDepends-On: I480b7fb9756d98ba9dbcdfd1d4b193ce6868e291\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/65eeb423b05c8e3c2aa1f51f2dac2015c9bf4ef7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/65eeb423b05c8e3c2aa1f51f2dac2015c9bf4ef7"}]},"branch":"refs/heads/master"},"a8e01cfda40687da1639353802668d6660b95a22":{"kind":"REWORK","_number":3,"created":"2020-05-29 06:31:25.000000000","uploader":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"ref":"refs/changes/21/722121/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/21/722121/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/21/722121/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/21/722121/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/21/722121/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/21/722121/3"}}},"commit":{"parents":[{"commit":"f1e06813ddaaf00757e466cfafb0aee1761bb874","subject":"Add TLS version configuration for pools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f1e06813ddaaf00757e466cfafb0aee1761bb874"}]}],"author":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-04-22 19:02:00.000000000","tz":-300},"committer":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-05-29 06:22:31.000000000","tz":0},"subject":"Add minimum TLS version option in octavia.conf","message":"Add minimum TLS version option in octavia.conf\n\nAdd new configuration option \"minimum_tls_versions\" to octavia.conf.\nListeners, pools, or the default values for either will be blocked from\nusing lower versions.\n\nChange-Id: Ifa0d695c2227772d6b37987a7857fe58ca660dc8\nStory: 2006733\nTask: 37171\nDepends-On: I480b7fb9756d98ba9dbcdfd1d4b193ce6868e291\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a8e01cfda40687da1639353802668d6660b95a22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a8e01cfda40687da1639353802668d6660b95a22"}]},"branch":"refs/heads/master"},"a262c7b9fd3934dbb512df93c0d73c6aba2121ac":{"kind":"REWORK","_number":4,"created":"2020-05-31 03:02:45.000000000","uploader":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"ref":"refs/changes/21/722121/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/21/722121/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/21/722121/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/21/722121/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/21/722121/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/21/722121/4"}}},"commit":{"parents":[{"commit":"54d8e1453b0de6a9948d4a8d19f18560ace8b52b","subject":"Add TLS version configuration for pools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/54d8e1453b0de6a9948d4a8d19f18560ace8b52b"}]}],"author":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-04-22 19:02:00.000000000","tz":-300},"committer":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-05-31 03:01:22.000000000","tz":0},"subject":"Add minimum TLS version option in octavia.conf","message":"Add minimum TLS version option in octavia.conf\n\nAdd new configuration option \"minimum_tls_versions\" to octavia.conf.\nListeners, pools, or the default values for either will be blocked from\nusing lower versions.\n\nChange-Id: Ifa0d695c2227772d6b37987a7857fe58ca660dc8\nStory: 2006733\nTask: 37171\nDepends-On: I480b7fb9756d98ba9dbcdfd1d4b193ce6868e291\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a262c7b9fd3934dbb512df93c0d73c6aba2121ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a262c7b9fd3934dbb512df93c0d73c6aba2121ac"}]},"branch":"refs/heads/master"},"4cd35a08e01c8ec72f76836f5367f4e93e3b677d":{"kind":"REWORK","_number":5,"created":"2020-06-03 22:15:08.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/21/722121/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/21/722121/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/21/722121/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/21/722121/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/21/722121/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/21/722121/5"}}},"commit":{"parents":[{"commit":"9a6da86481f9fcbf6000b764fc4fc1f8c1341da4","subject":"Add TLS version configuration for pools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a6da86481f9fcbf6000b764fc4fc1f8c1341da4"}]}],"author":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-04-22 19:02:00.000000000","tz":-300},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-03 22:14:50.000000000","tz":-420},"subject":"Add minimum TLS version option in octavia.conf","message":"Add minimum TLS version option in octavia.conf\n\nAdd new configuration option \"minimum_tls_versions\" to octavia.conf.\nListeners, pools, or the default values for either will be blocked from\nusing lower versions.\n\nChange-Id: Ifa0d695c2227772d6b37987a7857fe58ca660dc8\nStory: 2006733\nTask: 37171\nDepends-On: I480b7fb9756d98ba9dbcdfd1d4b193ce6868e291\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4cd35a08e01c8ec72f76836f5367f4e93e3b677d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4cd35a08e01c8ec72f76836f5367f4e93e3b677d"}]},"branch":"refs/heads/master"},"b39db7dd63ace950fadbac83f15a77c5170e053f":{"kind":"REWORK","_number":6,"created":"2020-06-04 05:27:36.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/21/722121/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/21/722121/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/21/722121/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/21/722121/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/21/722121/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/21/722121/6"}}},"commit":{"parents":[{"commit":"9a6da86481f9fcbf6000b764fc4fc1f8c1341da4","subject":"Add TLS version configuration for pools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a6da86481f9fcbf6000b764fc4fc1f8c1341da4"}]}],"author":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-04-22 19:02:00.000000000","tz":-300},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-06-04 05:27:32.000000000","tz":0},"subject":"Add minimum TLS version option in octavia.conf","message":"Add minimum TLS version option in octavia.conf\n\nAdd new configuration option \"minimum_tls_versions\" to octavia.conf.\nListeners, pools, or the default values for either will be blocked from\nusing lower versions.\n\nChange-Id: Ifa0d695c2227772d6b37987a7857fe58ca660dc8\nStory: 2006733\nTask: 37171\nDepends-On: I480b7fb9756d98ba9dbcdfd1d4b193ce6868e291\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b39db7dd63ace950fadbac83f15a77c5170e053f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b39db7dd63ace950fadbac83f15a77c5170e053f"}]},"branch":"refs/heads/master"},"4735be48fb403ee7521e9228d6a714de3dbc5044":{"kind":"REWORK","_number":7,"created":"2020-06-04 07:22:09.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/21/722121/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/21/722121/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/21/722121/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/21/722121/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/21/722121/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/21/722121/7"}}},"commit":{"parents":[{"commit":"0321c28588acf4e425dcfcc9f63353808e53247d","subject":"Merge \"Add TLS version configuration for pools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0321c28588acf4e425dcfcc9f63353808e53247d"}]}],"author":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-04-22 19:02:00.000000000","tz":-300},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-06-04 07:21:58.000000000","tz":-420},"subject":"Add minimum TLS version option in octavia.conf","message":"Add minimum TLS version option in octavia.conf\n\nAdd new configuration option \"minimum_tls_versions\" to octavia.conf.\nListeners, pools, or the default values for either will be blocked from\nusing lower versions.\n\nChange-Id: Ifa0d695c2227772d6b37987a7857fe58ca660dc8\nStory: 2006733\nTask: 37171\nDepends-On: I480b7fb9756d98ba9dbcdfd1d4b193ce6868e291\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4735be48fb403ee7521e9228d6a714de3dbc5044"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4735be48fb403ee7521e9228d6a714de3dbc5044"}]},"branch":"refs/heads/master"},"270b973bf96c7a873f495ac6977b8188cf63bf72":{"kind":"REWORK","_number":8,"created":"2020-06-04 20:11:02.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/21/722121/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/21/722121/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/21/722121/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/21/722121/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/21/722121/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/21/722121/8"}}},"commit":{"parents":[{"commit":"0321c28588acf4e425dcfcc9f63353808e53247d","subject":"Merge \"Add TLS version configuration for pools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0321c28588acf4e425dcfcc9f63353808e53247d"}]}],"author":{"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","date":"2020-04-22 19:02:00.000000000","tz":-300},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-04 20:11:01.000000000","tz":-420},"subject":"Add minimum TLS version option in octavia.conf","message":"Add minimum TLS version option in octavia.conf\n\nAdd new configuration option \"minimum_tls_versions\" to octavia.conf.\nListeners, pools, or the default values for either will be blocked from\nusing lower versions.\n\nChange-Id: Ifa0d695c2227772d6b37987a7857fe58ca660dc8\nStory: 2006733\nTask: 37171\nDepends-On: I480b7fb9756d98ba9dbcdfd1d4b193ce6868e291\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/270b973bf96c7a873f495ac6977b8188cf63bf72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/270b973bf96c7a873f495ac6977b8188cf63bf72"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
