)]}'
{"id":"openstack%2Foctavia~722162","triplet_id":"openstack%2Foctavia~stable%2Fqueens~I479019a911b5a1acfc1951d1cbbc2a351089cb4d","project":"openstack/octavia","branch":"stable/queens","topic":"validate-resource","hashtags":[],"change_id":"I479019a911b5a1acfc1951d1cbbc2a351089cb4d","subject":"Validate resource access when creating loadbalancer or member","status":"ABANDONED","created":"2020-04-22 22:26:14.000000000","updated":"2020-07-24 11:41:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"84df00fd291906e1f6c5955b51b8c49376ff0c5e","_number":722162,"virtual_id_number":722162,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"date":"2020-05-03 21:32:32.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-04-23 01:11:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-04-23 16:08:46.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-23 01:11:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-23 16:08:46.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"24d47453f0c63c00de2cf33a8779e0afa0b59e9a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-22 22:26:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15af3b6bfd18cfd12bff3100f6749379bdd5cfc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 01:11:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ee1ae8dea7a4e07b0f1933a07c321dd : SUCCESS in 14m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e9241a0abbed4cbaaf45dec18a86ecf7 : SUCCESS in 8m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e5fd698d87a5446fbc94d1bba1e36816 : SUCCESS in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/152dbb3a654c4657ad2627125f7c6677 : SUCCESS in 8m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6d0210e6ae84f6bb64d861d6fbefb40 : SUCCESS in 6m 56s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/5972680adae54a58b510b3e9de807d5b : SUCCESS in 1h 21m 20s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/883b37102f2c44e8b60054bea27a0d1d : SUCCESS in 1h 23m 33s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3c34c779bdfc4ebe98eafcc2b1e5d806 : FAILURE in 1h 34m 49s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/7ab5122b27b147db93d5871e48a4065a : FAILURE in 1h 18m 30s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22157f83105848248e97732de8f087bf : SUCCESS in 11m 19s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/48639f6e116f49538f4dfe530ccde1f4 : SUCCESS in 8m 44s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/1305b58c337c4414a1f7a0ca76e362de : SUCCESS in 10m 02s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c4a394718f854827a6a871b2a4094de7 : SUCCESS in 1h 04m 09s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/392c19ad406f4ba9b7d2f77db658e3a7 : SUCCESS in 1h 05m 06s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/f03690a0667e44158347acd46604b489 : SUCCESS in 1h 10m 29s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/eaef33e9bc6c44b1adc2194bce27f810 : SUCCESS in 1h 23m 50s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/e3f99129cffb46778ee40922b39042f6 : FAILURE in 33m 07s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/83b355294c3c48f6a0b40f2c994d01e8 : SUCCESS in 1h 23m 00s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/473388190af444259a1accd822e59d72 : SUCCESS in 59m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a5e537f93ecb4511875a869106c6b55a : TIMED_OUT in 2h 34m 42s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/3bf1cbff86ee41f8935e6c722c805c5d : FAILURE in 1h 22m 20s","accounts_in_message":[],"_revision_number":1},{"id":"4b17209b4eed7fce271b355abaa1584d6a202896","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-23 13:30:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"11c799cb355e6a86f225eb60b547ac167fe3118b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 15:16:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9874ab69254e4af39ceca70b583b0d1d : SUCCESS in 11m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6c5c54491d594925b796e7a8891c41a7 : SUCCESS in 8m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b2a841887820443ba87af9dbcce54ef6 : SUCCESS in 9m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5614a1400704800b1db294b8ec8efe7 : SUCCESS in 10m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09e37af468484cb58987419fff285495 : SUCCESS in 8m 29s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/ce5a2e01906c46c3a614b192df16d543 : SUCCESS in 1h 32m 11s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/abf5be8243f84580899e304ab5d6bfba : SUCCESS in 1h 33m 32s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6cda8f6df8a14cc69f819152c0e0397f : SUCCESS in 1h 22m 03s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/6e21e5e452ac4023a45ef04bce4f4df9 : SUCCESS in 1h 20m 20s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a276298e539749f68ec87499f9cf7489 : SUCCESS in 11m 10s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/034c4083edae4b748ef459d61f638947 : SUCCESS in 9m 09s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/d39f84c6684d4890b5112ab0da00dc15 : SUCCESS in 9m 56s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8999f49a4236412fa51dab15915060b7 : SUCCESS in 1h 04m 05s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/4aec04d4a0884caeaba8ef6809802004 : SUCCESS in 1h 02m 24s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/c08c4120fe0f45e08d213540e456d6d5 : SUCCESS in 1h 27m 44s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/9cf842696bc34f03a2816310f44e473f : SUCCESS in 1h 12m 33s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/074e632075d44cb0b7bef9ff2e4900a8 : FAILURE in 45m 59s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/459f54e3b81945799d28c3ce64325f7b : SUCCESS in 59m 00s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/dda12f15b63e45d6b656bb44e607cbbf : SUCCESS in 45m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0660ae9de9a14efea32c9eaeda596917 : FAILURE in 1h 08m 52s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/967158a91e8c47f4a486bcc958c6d0eb : FAILURE in 1h 10m 51s","accounts_in_message":[],"_revision_number":1},{"id":"17ab984085ea6068e28b4cc6a85e6e9367394f00","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-23 16:08:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d95214e120a8ca4b686648037c4b08f563015cb9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-30 15:25:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6664062906f688df278aa08406a13b939a0da74f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 16:54:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eda94a05dae54e0fb941c7482e0cb350 : SUCCESS in 9m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/82ff6ccf2e294fe7a5981fe010e2314a : SUCCESS in 6m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8cdd64eacef04d9489dce1d28d17db85 : SUCCESS in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f4edd46f9da4dde935c9016a60fcc4e : SUCCESS in 8m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50de25fced3a437e8718579cb14574ac : SUCCESS in 7m 00s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/9f40c478a77c4d3280a5ec1c9ad1dbe9 : SUCCESS in 1h 19m 31s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/5f6464c12cce4dc4ba4c591759323e15 : SUCCESS in 1h 21m 35s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0b0ec02f44494aa78cddd96ecfed7a1a : SUCCESS in 1h 14m 12s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/9bed28aa234a4d13a21ee74ed369b410 : SUCCESS in 1h 15m 59s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/967d04b3ed90486da063f6313eaf0ee4 : SUCCESS in 12m 16s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/aa9072f8e9174359949a6d67536fe9a4 : SUCCESS in 9m 06s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/93fdc5ab144d44dda48a554d865ebf86 : SUCCESS in 9m 13s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dc7eb85c211245f2893d55e6160d0b5e : SUCCESS in 1h 02m 34s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/ce3a7829aa7a46e786f86675de7abee0 : SUCCESS in 1h 06m 31s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/ed48778ac4a147aba6d52d20ba6846de : SUCCESS in 1h 22m 51s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/8d4a45c3201746d6873605f2758a06c1 : SUCCESS in 1h 23m 08s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/d27d054af708440e9f6075ff7eb2c097 : FAILURE in 30m 33s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3475ad4158c34d40a95d4b927bad475c : SUCCESS in 51m 13s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/1a70198e2a424f06b29bccd14fb92261 : SUCCESS in 56m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3ee1ba02c05e4987b6fb7c014ff5dca8 : FAILURE in 1h 11m 55s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/3efdf204592d4017a4c6d1a186c43b55 : FAILURE in 1h 09m 41s","accounts_in_message":[],"_revision_number":1},{"id":"837904fa96ac698d83514eef185a60bfaf79e7ce","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-30 22:26:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1f5906b6276f67210daf3d5e65709531f2ec391e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 23:50:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95fee85244a44e1baa6ac4748170bf57 : SUCCESS in 9m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b2b0681d85194868810480d6cfc12940 : SUCCESS in 8m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3e6854a9168142df83a9d0e2553f2f06 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc0a0fe7425a426b91a7af15c2b01f9a : SUCCESS in 7m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7951fbcd1b094ef08588af13c8349899 : SUCCESS in 5m 56s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/88ff1e43206a48e8b069ca0075357579 : SUCCESS in 1h 16m 11s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/d04a9f3f3fae4e508ab7fec23b9cf34d : SUCCESS in 1h 21m 36s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/da7fbf54b6a74f94b8d16613fbd8facc : SUCCESS in 1h 11m 39s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/792c0bd89e214d1a9748fb19931e1911 : SUCCESS in 1h 15m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a0633973c0244a5b6cf2b5f1f0f01ad : SUCCESS in 10m 32s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/70e1843aaad04659a07bb042544f7cee : SUCCESS in 8m 05s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/c38c97dbdfad4a97add0cd878f83aecf : SUCCESS in 6m 15s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9b53f7211c874945996c2d05eca71938 : SUCCESS in 56m 24s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/f47cc696995f48a09eb1cfbd544bcd92 : SUCCESS in 57m 04s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/a8012cad5fd449bfa6a82efc6a1e646e : SUCCESS in 1h 07m 14s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/63428164040a4acfa34e8c0c962b2904 : SUCCESS in 1h 06m 21s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/e73d9c03b81849cbaf39aa0842c86adb : FAILURE in 30m 22s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/854c7a5d123c42ad9c079730782f2048 : SUCCESS in 43m 30s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/649ea7b2391e44e08aff5bf5f9969440 : SUCCESS in 39m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f08c0eafd1f142aa9adae9255e4d306b : FAILURE in 49m 58s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/1519eb7cccb8493296eabb6e4f3ad98f : FAILURE in 1h 18m 11s","accounts_in_message":[],"_revision_number":1},{"id":"d3ff1ded36c835bc1d1b91edcfee41d96931711a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-01 15:03:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3213de89432519e6a8bc7d1efd59fd5064f95d4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 16:55:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b1cb5b9e3c145d3b8ea3524b47eb274 : SUCCESS in 9m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/be1d1a9d3d054baba8256343cbdddd06 : SUCCESS in 7m 38s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/49f6d46378764154af17ce197e0e4ba8 : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c688738892444f168853d967852ffb96 : SUCCESS in 9m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1b2329748f14865bbe1cee48e37643c : SUCCESS in 7m 59s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/b11b00c027134711b3739f58646e2091 : SUCCESS in 1h 13m 21s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/53ec42bfd02d43d7a0f1b53a92222b6d : SUCCESS in 1h 21m 41s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dce433f2726542a4a77ec00ed3e63796 : SUCCESS in 58m 52s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/664d795c457b4187b6206c6334d5dca2 : FAILURE in 1h 16m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd9d483dcfdb4ba0b638de470bc0b9cd : SUCCESS in 12m 07s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/eaa56dbc51de45f28847262e0aea50b0 : SUCCESS in 8m 29s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/8da31ca4d80e4eb0abc24be78ef8a598 : SUCCESS in 8m 35s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e9bef0eb0f81419c965d208fb2d0bb2c : SUCCESS in 1h 00m 27s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/8c181670807f42578d7357b82d9cfc26 : SUCCESS in 1h 02m 04s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/6144c59941ca4e32a55bcc495ac1da41 : SUCCESS in 1h 20m 34s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/a44ffb07e90940949341e94ab5f3a0ea : SUCCESS in 1h 12m 08s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/cb7244f9dd5f4f18bf5a30ca57b64a07 : FAILURE in 29m 11s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5413aa77cc3a4640880934689a82fd35 : SUCCESS in 44m 42s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/9bb60c11a91e48d396c52cf96ae80c1d : SUCCESS in 44m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/463a8b354e2d4700a5350d8719d65ded : FAILURE in 1h 37m 23s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/64506402e5734ae8b0d35208200cf429 : FAILURE in 57m 24s","accounts_in_message":[],"_revision_number":1},{"id":"c53ff9a35c3e965c78be39ec5f7345e6089a5d9d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-03 21:32:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"60d0736c5e41a265368ad8cf6b819ea3742dab6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 22:59:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1f106d9ce6242c69125973f2b08bc5d : SUCCESS in 9m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1e4a1f4125e042bf96f75a70fdf8839a : SUCCESS in 7m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/951d89b467f44c9c873568856f0c6f45 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/047ae69d024645ea80b13194bec4729e : SUCCESS in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71dc8c46b07e47098ee94a8d9b4b7ad0 : SUCCESS in 8m 25s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/8f39aa3b4da0434bb3b5a4da975e168a : SUCCESS in 1h 17m 34s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/441e5fdc050c4090b1d1847519a75813 : SUCCESS in 1h 22m 06s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cd7cfe5aa7d842a5a060c7bc46f2d498 : SUCCESS in 1h 13m 53s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/e1ba76ebed5442d2870f6ca10925cfb5 : SUCCESS in 1h 15m 05s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92a4622e295f47728788dbd70300418c : SUCCESS in 9m 47s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/348a08eb6c9740bfa60c717979e4b831 : SUCCESS in 6m 40s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/aef5368d997f4d5b8cf2fdc7d474769b : SUCCESS in 6m 25s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1202204838344577b40ab20a08623b78 : SUCCESS in 57m 34s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/dd9a243327c248aaa65535194d37cf2f : SUCCESS in 54m 47s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/8e430c2812e44dafb7b25f0e30d2ba15 : SUCCESS in 1h 05m 26s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/f5bc83db27574ec8b034c74695daeab6 : SUCCESS in 1h 07m 57s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/266a608724a34ea1a938e9e5deea3dd0 : FAILURE in 30m 33s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fc6cb3f746944dc5933cb88acb3b6d85 : SUCCESS in 44m 36s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/ad86cd21d9904e10a00a9dc5e1549a39 : SUCCESS in 45m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d957843c45164ea1a3f6c6238f0b305b : FAILURE in 51m 45s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/9dc533fb76df413288d7bf45126af797 : FAILURE in 1h 14m 12s","accounts_in_message":[],"_revision_number":1},{"id":"be29343fff5e40937c1b08b7f1064a712d08c47f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-05-04 06:57:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e3b7666116cfea5420be6c3acbb756e4d90f6d53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 08:34:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58b9c677c3ae4be7bb95bb1b3eb8c88f : SUCCESS in 9m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e5d09c1ac5d94c4f83e5d3ca2370f33b : SUCCESS in 5m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e53a747f96444fa2852f9f782e95aa2b : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2692468711e468ba5cec6b66a4ec492 : SUCCESS in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7f5f31069d94c3284a811b6fd0f7600 : SUCCESS in 6m 31s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/a6fbaaa5c4bf4bc4945a45a76e597d8f : SUCCESS in 1h 20m 52s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/de9f21dfe3ad4a24b918168d6b292db1 : SUCCESS in 1h 15m 54s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d22b3c7e951247f5b6040a72f1b40f1d : SUCCESS in 1h 33m 57s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/2e657883e79e470dab2383d6058a76fb : SUCCESS in 1h 34m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b17d6a18c30044289393c2b7834e509c : SUCCESS in 11m 00s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ed066d9e510f4196a7259d14a27da4b5 : SUCCESS in 6m 43s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/28969e66e2ef49c79a75494369a20358 : SUCCESS in 6m 56s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/57c7b6cb0c4e4e15a0644fc2fcf6e6fc : SUCCESS in 1h 01m 39s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/3bb3173285aa44238c3eab01ee1968a8 : SUCCESS in 54m 19s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/f1b4652904af485caafce157884c2232 : SUCCESS in 1h 11m 30s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/e12f7e27d9c247448c4e42a894443330 : SUCCESS in 1h 14m 53s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/3d5e40bb9d9a4e738d6031123270b1d0 : FAILURE in 33m 01s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/763e0eeff6014867aa8b1e52f4f93f33 : SUCCESS in 43m 59s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/22eedb7a066c4519b036af10568420f4 : SUCCESS in 48m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0b9015d54beb4bb0b302d3d5da08da78 : FAILURE in 1h 16m 05s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/29b4f58edef140e8b568dd20e2f263f3 : FAILURE in 1h 21m 37s","accounts_in_message":[],"_revision_number":1},{"id":"fc018934b9c44fa76f426b19111b959cce24d8bc","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-24 11:41:06.000000000","message":"Abandoned\n\nQueens Octavia reached End of Life (EOL) state. This branch no longer accepts changes.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"28b031054774da29e77f1fdbc6c799465da0054c","revisions":{"28b031054774da29e77f1fdbc6c799465da0054c":{"kind":"REWORK","_number":1,"created":"2020-04-22 22:26:14.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/62/722162/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/722162/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/722162/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/722162/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/722162/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/722162/1"}}},"commit":{"parents":[{"commit":"5528268e2f360c2eb921d665313a326387c2e1b7","subject":"Fix uncaught DB exception when trying to get a spare amphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5528268e2f360c2eb921d665313a326387c2e1b7"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-04-09 01:58:30.000000000","tz":720},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-04-22 22:26:13.000000000","tz":-420},"subject":"Validate resource access when creating loadbalancer or member","message":"Validate resource access when creating loadbalancer or member\n\n* Make sure the user has access to the subnet in the request for\n  creating or updating pool member.\n* Make sure the user has access to port or subnet or network for\n  creating load balancer\n\nStory: 2007531\nTask: 39339\n\nChange-Id: I479019a911b5a1acfc1951d1cbbc2a351089cb4d\n(cherry picked from commit a33d42fa59ada38dd5a5d3114a9a83f03a30d0b2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28b031054774da29e77f1fdbc6c799465da0054c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28b031054774da29e77f1fdbc6c799465da0054c"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
