)]}'
{"id":"openstack%2Foctavia~722849","triplet_id":"openstack%2Foctavia~master~I2372178351e961eeed5a819f39e75d54ba148798","project":"openstack/octavia","branch":"master","topic":"py38-tox","hashtags":[],"change_id":"I2372178351e961eeed5a819f39e75d54ba148798","subject":"Change default tox envs from python37 to python3","status":"MERGED","created":"2020-04-24 15:30:03.000000000","updated":"2020-06-08 06:49:44.000000000","submitted":"2020-06-08 06:45:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722849-1591598725100-ec5b561d","meta_rev_id":"a7576aa94d868667bd461f987faeef16cb7f6e7c","_number":722849,"virtual_id_number":722849,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-08 06:45:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-06-06 18:33:02.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-06-08 05:05:45.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":1,"date":"2020-06-05 19:09:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-06-08 05:05:45.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":-2,"date":"2020-06-06 18:33:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-29 22:05:58.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-05-01 18:30:26.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-06-05 19:09:44.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-06-06 18:33:02.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-06-08 05:05:45.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2020-06-08 06:45:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"66e4b540ad76eddd25fc2e5664f8b0d0893ea408","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-24 15:30:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8eb21bba2f6e288c0080d2194379925123b7190","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-24 16:28:01.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"31b4c92ac4907a6ee600eb8eab95b6d6cf5af8d8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-24 16:45:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e5d00d6de7c396af4abc88ce4b6eb804f6943ecb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-24 16:47:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"727d5b171882434614bfae38a2293a7190975a2a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-24 16:55:18.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nYes, I figured a script had done all these...","accounts_in_message":[],"_revision_number":2},{"id":"abe9f1ace0406bacd92777b43ff8ab6145161613","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 19:04:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/89230bd62b854b80b40ea46a9054e901 : POST_FAILURE in 8m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9dff64a0429049589b177a13dd4c9300 : SUCCESS in 14m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/74554d4defe74af7803a0da76da2a3a6 : SUCCESS in 13m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0db23e89b90e443cafc2ef8f787a0d66 : SUCCESS in 9m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86808ab556b4444abbd208dfc04c9705 : SUCCESS in 7m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2099769a40cc44ad8f931b996640f2c6 : SUCCESS in 8m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2719762cd7e34cc1832cc18000b57f4a : SUCCESS in 14m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6112a2bcf7504d46944f6783fa995ec9 : SUCCESS in 8m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b9543b790f1b4353951b4fbed77a5aa5 : SUCCESS in 10m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7e795424b9d4456eb3928aa29e8ca1b5 : SUCCESS in 11m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3155281b70fc464b9e40400ab013252b : SUCCESS in 8m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/da83ddd0649545a492b866b39e36e5ef : SUCCESS in 43m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b7d428c3a6ee412880968da8d15354bc : SUCCESS in 1h 49m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c0571b724d5d496596bb41357970cdbd : SUCCESS in 50m 13s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7d6f560617fa40e28bb2f421e0ed0d21 : SUCCESS in 40m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1300d2f3c54e414587aa80cf61e79a07 : FAILURE in 57m 44s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/beb62284fc7e491983ad0fe9115df229 : SUCCESS in 1h 54m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/72f34fa919ca410d90625e2846fe7ba5 : SUCCESS in 2h 05m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/99376992cfd34aab9fdc86e9641db30e : SUCCESS in 1h 46m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bdda80ee1372432892ccd0454ab21b23da4b809c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-29 21:58:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bd4eeb058d2434390c0ce0a0eabf2141c9f4a880","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-29 21:59:04.000000000","message":"Patch Set 3:\n\nJust fixed my nits to move it along","accounts_in_message":[],"_revision_number":3},{"id":"64a2a9d24d131569843345bf8ee96c60144fc49a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-29 22:05:58.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"46d0f135b9840127aa7fc1afe97665713565fc4a","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-04-29 22:29:15.000000000","message":"Patch Set 3: Code-Review-1\n\nWait, why are we dropping 3.7? Did openstack not commit to supporting 3.7? It seemed to be that 3.6 and 3.7 were the most widely supported versions, I don\u0027t know that dropping support for 3.7 this early makes sense?","accounts_in_message":[],"_revision_number":3},{"id":"af56ec9f7d5df2411b0b7319e31b98e75bb0d12e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-04-29 22:30:24.000000000","message":"Patch Set 3:\n\nNote: it says \"does not affect zuul jobs in any way\", but I see no `openstack-tox-py37` anymore... Does that mean the default openstack zuul templates removed this and I missed it?","accounts_in_message":[],"_revision_number":3},{"id":"c316bb7fc687ba27c4ccccac65e3d53d3fc83b3b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-04-29 22:32:02.000000000","message":"Patch Set 3:\n\nOk, to answer my own question... Apparently yes, py37 was dropped recently?! Interesting, as I thought that was going to be the version standardized for a while.","accounts_in_message":[],"_revision_number":3},{"id":"a9b5567b27bfefcc0103fadfc67c3d41c7452e7a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-29 23:48:26.000000000","message":"Patch Set 3:\n\n\u003e (1 comment)\n\nThat\u0027s up to this team, but a lot of other teams are just setting the highest version as the default since this only affects what is run locally be default. Zuul jobs ensure all required version are covered. It would just be an extra run for a developer to wait for when testing locally.","accounts_in_message":[],"_revision_number":3},{"id":"02e10279ae8602b63973613f7e353955282b66c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 00:05:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/92975fb90e754a60938cf89ad36e8650 : SUCCESS in 7m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9dcd0c46b6744ada8e9f72accd124cb : SUCCESS in 9m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96845d0d59014b87851316ff004d2c6c : SUCCESS in 10m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b111531d8da45c9b316d02f49878abd : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df14d76ceb60419ba3580afac2006f39 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb89a574d00d4b35ac8b7089f12c4358 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4434bd9e73c457991c8dc9c6eb33e1b : SUCCESS in 12m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f6a3fc2d11f46148f4772a24dca9122 : SUCCESS in 5m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0e4488714a1c41e990b4a6f0962ddb93 : SUCCESS in 6m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/79b7c3ee63fc48049b42d1c8e6a5578c : SUCCESS in 7m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c79a5fbd7469441f830c58727efce987 : SUCCESS in 6m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/188a7d3a63f94eb0a780ab09d0615771 : SUCCESS in 41m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e4427667014d47ce9d77aee3e0792855 : SUCCESS in 1h 38m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9d19d57f0b7f4806ac89023d80ccfc34 : SUCCESS in 59m 04s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/18b66b1101474b09977155499c7bb993 : SUCCESS in 52m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/25a704d976d64d1eae064ba9388c6e1d : SUCCESS in 44m 04s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/99f7dd19e2744f98935bc28922930c28 : SUCCESS in 1h 04m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d4f63b68bdd444518014bd6cc977e64c : SUCCESS in 1h 50m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6dcdc6fca47641d0b7066f5b6eb1cf9b : SUCCESS in 2h 01m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"485594c6f0c835ba49f359be20266f5afb47bfa8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-30 14:40:51.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nAdam - yes, the py37 unit test job is gone, that happened with https://review.opendev.org/#/c/722493/ but we can add it back manually in project.yaml if we want.","accounts_in_message":[],"_revision_number":3},{"id":"e4f984998d5cbb124838b5c0aef21fe1df4f0b31","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-05-01 18:30:26.000000000","message":"Patch Set 3: -Code-Review\n\nYeah, I see that now... That\u0027s very surprising to me that Victoria dropped py37. :/","accounts_in_message":[],"_revision_number":3},{"id":"2ef50d3e94d9a7a0099917334231f560cabc3572","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-29 17:32:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0f72f58ff98989e7baa359102a4ba47ff669279d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-29 18:56:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8a727d174d30b84be9625101ce4ceb9062cdc30b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-29 19:05:21.000000000","message":"Abandoned\n\nUp to the project if you have the expectation that they will run all environments by default locally. The PTI specifies what is needed to be run in gate. In most cases, running just one of those locally is enough to have a reasonable assumption that it will pass in the gate, with the gate being the protection to identify any differences that cause issues.","accounts_in_message":[],"_revision_number":3},{"id":"40bc92a7cdafb23e25ff59875bb405558c47ac79","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-29 19:16:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dc889909003414bd6d73eb2640a6e595ef1959f6","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-06-03 14:24:08.000000000","message":"Restored\n\nGoing with the ML suggestion, per agreement in PTG.","accounts_in_message":[],"_revision_number":3},{"id":"fc8c84613bfbe9363a2eed43c4124e7c47e96fbd","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-06-03 14:33:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5c6ee6b651a3c7c856339dda01c6f15c214654fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 17:10:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c010651a01a449bfb25edc835254ba0c : SUCCESS in 6m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/671f8dba69af428c96e3723e0a6f5a05 : SUCCESS in 9m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/197544a970b84dbebf991480cc02b79e : SUCCESS in 12m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2914c509a81744e2b8088b690c58b7d9 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/656dc80a03a84ae68ca5b8d3da4b8eda : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58b52d3ded8e4869bba774e7495721db : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f9940b07c554af1989b35e49784dea5 : SUCCESS in 13m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/673f611761ee4852bd7455c5db1b3f4f : SUCCESS in 8m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ff273d483d174f1f9fc4ef6314e779a2 : SUCCESS in 5m 35s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9f0cec86223b463e8ff50fbcbb3ba581 : SUCCESS in 9m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b4d314825cb649cd82b2caf302b2f679 : SUCCESS in 8m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c5dea09211814d38a51875a30c791fac : SUCCESS in 49m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3c6df9d9b23042feb11bb34f3d491dd8 : SUCCESS in 1h 03m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/59de7777995d43bbae5d2f5716b0830f : SUCCESS in 48m 07s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/bad8b12ae73e4fdaae53edd2696fb852 : SUCCESS in 45m 26s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/871bf0d8cf4841cfb192e5348ee088c9 : SUCCESS in 51m 09s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/56690555fe9d4445946d072f2f24e012 : SUCCESS in 1h 49m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c0452c8de7d04d6c84fe6d0f88c93656 : TIMED_OUT in 2h 32m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7ca8739c1dbd407e8ec5b7e9329d384d : SUCCESS in 1h 00m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f3dcb05e10a9e9fd6aa57df9e66b17fe5ba8e547","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-05 19:09:44.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"97ae712eb26aa0316e2759ca3424467c002a5e49","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-06 18:33:02.000000000","message":"Patch Set 4: Backport-Candidate-2 Code-Review+2\n\nI\u0027m okay with this change.","accounts_in_message":[],"_revision_number":4},{"id":"cf492ab2dfd8df238da4db972216aa1b1b810d5b","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2020-06-08 05:05:45.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"693e220acbcdb645d248382ed7139891885fa880","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 05:05:57.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3ca76f0944f059b29a2ba1d342ffdaeee858dc42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 06:45:24.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e6cd7bc4aa4c453e9eb532935cfcf82c : SUCCESS in 5m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3dcc14cc4e06468faa98849ce7eb9802 : SUCCESS in 10m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/008e85c3f3b048c8a29138252df350d4 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a22d6ee561544ff485327bc92d76a7b4 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a85d0714c7874b2a977aac4f48391499 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b78c48d2c144f1ab52df3763c90b611 : SUCCESS in 11m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4422773feb7f4673ac2053c82a612222 : SUCCESS in 4m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/847d4c89683a4fecbbc31a3e81dd76bc : SUCCESS in 7m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9cfb52beca6d4f5ab4be012e223a6334 : SUCCESS in 40m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cb2329deef8a44b2b7def0afcacc62b2 : SUCCESS in 54m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bdd2b2d8d59b4645940a44868ef1a931 : SUCCESS in 54m 24s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e79b40d00cfe4896b83c5d0259f94fcc : SUCCESS in 42m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9da6417ebf9f4c21a21ba2e98735ac5a : SUCCESS in 43m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f26b62f17ba1431c9c9defacf19c2e4f : SUCCESS in 1h 38m 35s","accounts_in_message":[],"_revision_number":4},{"id":"01b297a849ecb4fe273a9a81c6b73a361bef5dca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 06:45:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"315c99241c55fdabf3ce72fca39a26c703e65376","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 06:49:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/6a03e6fca7554c9091db80a064534c75 : SUCCESS in 3m 30s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e5401b632044abea907ed9c2e114ee7 : SUCCESS in 3m 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/edfa783e33de4e018f18d1fcc10220f0 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0deff25667f9bb529d1fed781e5cf7860a099399","revisions":{"22fb530307a999db5a945a6a1a27b8d1409f8569":{"kind":"REWORK","_number":1,"created":"2020-04-24 15:30:03.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/49/722849/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/49/722849/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/722849/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/722849/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/722849/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/49/722849/1"}}},"commit":{"parents":[{"commit":"c037a9b374223176248e0cbd7ce2ca382da21049","subject":"Merge \"Add Python3 master unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c037a9b374223176248e0cbd7ce2ca382da21049"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-24 15:25:47.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-24 15:25:47.000000000","tz":-300},"subject":"Bump default tox env from py37 to py38","message":"Bump default tox env from py37 to py38\n\nPython 3.8 is now our highest level supported python runtime.\nThis updates the default tox target environments to swap out\npy37 for py38 to make sure local development testing is\ncovering this version.\n\nThis does not impact zuul jobs in any way, nor prevent local\ntests against py37. It just changes the default if none is\nexplicitly provided.\n\nChange-Id: I2372178351e961eeed5a819f39e75d54ba148798\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/22fb530307a999db5a945a6a1a27b8d1409f8569"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/22fb530307a999db5a945a6a1a27b8d1409f8569"}]},"branch":"refs/heads/master"},"18d3108f786703b45a1e1130fad77c057d1a3c65":{"kind":"REWORK","_number":2,"created":"2020-04-24 16:47:55.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/49/722849/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/49/722849/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/722849/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/722849/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/722849/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/49/722849/2"}}},"commit":{"parents":[{"commit":"c037a9b374223176248e0cbd7ce2ca382da21049","subject":"Merge \"Add Python3 master unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c037a9b374223176248e0cbd7ce2ca382da21049"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-24 15:25:47.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-24 16:47:03.000000000","tz":-300},"subject":"Bump default tox env from py37 to py38","message":"Bump default tox env from py37 to py38\n\nPython 3.8 is now our highest level supported python runtime.\nThis updates the default tox target environments to swap out\npy37 for py38 to make sure local development testing is\ncovering this version.\n\nThis does not impact zuul jobs in any way, nor prevent local\ntests against py37. It just changes the default if none is\nexplicitly provided.\n\nAlso changed functional default target to only run one that uses\nwhatever is the default basepython version locally.\n\nChange-Id: I2372178351e961eeed5a819f39e75d54ba148798\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/18d3108f786703b45a1e1130fad77c057d1a3c65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/18d3108f786703b45a1e1130fad77c057d1a3c65"}]},"branch":"refs/heads/master"},"50761e9e659102fb0ac51babe983fd0d0d8905af":{"kind":"REWORK","_number":3,"created":"2020-04-29 21:58:44.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/49/722849/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/49/722849/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/722849/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/722849/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/722849/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/49/722849/3"}}},"commit":{"parents":[{"commit":"c037a9b374223176248e0cbd7ce2ca382da21049","subject":"Merge \"Add Python3 master unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c037a9b374223176248e0cbd7ce2ca382da21049"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-24 15:25:47.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-04-29 21:58:24.000000000","tz":-240},"subject":"Bump default tox env from py37 to py38","message":"Bump default tox env from py37 to py38\n\nPython 3.8 is now our highest level supported python runtime.\nThis updates the default tox target environments to swap out\npy37 for py38 to make sure local development testing is\ncovering this version.\n\nThis does not impact zuul jobs in any way, nor prevent local\ntests against py37. It just changes the default if none is\nexplicitly provided.\n\nAlso changed functional default target to only run one that uses\nwhatever is the default basepython version locally.\n\nChange-Id: I2372178351e961eeed5a819f39e75d54ba148798\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/50761e9e659102fb0ac51babe983fd0d0d8905af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/50761e9e659102fb0ac51babe983fd0d0d8905af"}]},"branch":"refs/heads/master"},"0deff25667f9bb529d1fed781e5cf7860a099399":{"kind":"REWORK","_number":4,"created":"2020-06-03 14:33:08.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/49/722849/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/49/722849/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/722849/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/722849/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/722849/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/49/722849/4"}}},"commit":{"parents":[{"commit":"2c7c7747b74e866e5ab41139e2f3f01d9dce15a5","subject":"Merge \"Workaround peer name starting with hyphen\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c7c7747b74e866e5ab41139e2f3f01d9dce15a5"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-24 15:25:47.000000000","tz":-300},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-06-03 14:28:27.000000000","tz":-420},"subject":"Change default tox envs from python37 to python3","message":"Change default tox envs from python37 to python3\n\nPython versions supported by OpenStack change over time, and for minor\nversions of Python 3 it is tedious to keep this file updated.\n\nSince this does not impact zuul jobs in any way, nor prevent local\ntests against py37, it should be safe to simply make this more easily\ncompatible for users that don\u0027t care about the specific Python versions\nand just need basic tests to run.\n\nThe *only* thing this does is changes the default versions tested if none\nare explicitly provided with `-e`.\n\nChange-Id: I2372178351e961eeed5a819f39e75d54ba148798\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0deff25667f9bb529d1fed781e5cf7860a099399"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0deff25667f9bb529d1fed781e5cf7860a099399"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
