)]}'
{"id":"openstack%2Foctavia~722920","triplet_id":"openstack%2Foctavia~master~I7ebf4137feb04827490dffc0dac3d6e4c8888075","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I7ebf4137feb04827490dffc0dac3d6e4c8888075","subject":"Make sure devstack aborts if DIB fails","status":"MERGED","created":"2020-04-24 15:52:05.000000000","updated":"2020-05-21 22:43:49.000000000","submitted":"2020-05-21 22:41:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722920-1590100907544-2e69b6b6","meta_rev_id":"b1bf81f4f26565a689fa7fa29bbfa8016769d467","_number":722920,"virtual_id_number":722920,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-21 22:41:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-05-21 20:50:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-05-19 16:52:34.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2020-05-21 20:50:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-30 15:28:49.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-05-19 16:52:34.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-05-21 20:50:47.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-05-21 22:41:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cb11f6784b45f72d334707dc15ff1c1b01eb12ae","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-24 15:52:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58353c7cf0e14fec33417d1b746369c755a152f9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-24 15:55:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"787fbb280b0eb22f05111cd2e99f3574d9eb856c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 18:27:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9993f34c08ed4b889cf34ad975d68dcc : SUCCESS in 11m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f5b49d160dd46e89796a629962ce62c : SUCCESS in 10m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55e0fd7544774b528e2730eb6da3b1c7 : SUCCESS in 8m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4186b5719f9b4141b03efc0946ae1171 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d74fb09abd4c450d8bc8fdbb0177280d : SUCCESS in 8m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c479601b3ba1468fa8f7f52d325d15a7 : SUCCESS in 13m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/58d3594ef5f445759e7e7844ba9b30e3 : SUCCESS in 7m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/086e9605daae44538cac085f235082d6 : SUCCESS in 10m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b54b7138c2f44d7ebcbab7df374af5fa : SUCCESS in 8m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f6a67c53d6a54d99948ebaad8b12c3bf : SUCCESS in 47m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/51d15d9be877420b88d84cdcd7af68dc : SUCCESS in 1h 35m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c28cf9a160ad4a85b2c39371ac6cdad7 : SUCCESS in 58m 40s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/07a32bfc23bb466da2aee4082a43c551 : SUCCESS in 53m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8fc680b2dda74266ba515562d8d63941 : SUCCESS in 56m 21s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3edd1db3a4014a09aed7c5d30bb1c42e : SUCCESS in 1h 37m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/22fd323471dd4dd3bd2890f51b98469f : SUCCESS in 2h 02m 36s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3f2286e6fd804d8e82ad14bb55ccbf02 : FAILURE in 59m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1bdb5757ce8c4e7c9016c43eb6eb5b62 : SUCCESS in 1h 43m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c40a54ed4c077318aa8d9f8b98102f687d63c453","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-04-30 15:28:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ac93a75815d5923f264873148ed509ad690b79fc","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-30 17:40:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b0685249d5f31ebfe65a9f81a312a4072041911d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-30 17:40:16.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"715a888f1f2054e1bf7aee4652c7a5e73168e6b7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-30 20:30:47.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"b87307624863f18700bd43d981c298fbd082f7c4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-30 22:57:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"97f56ddfba24e0cbbf34b281356514eacb7008e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 01:02:40.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b73b7cec1f5b4dfabaf5fb1652427b47 : SUCCESS in 13m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3173eda06eb34adf86bff7e926f08d0a : SUCCESS in 7m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9fed30e75f84a7b9834e0e1a9a179b9 : FAILURE in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e213fb1d5004aebabde5bf1f4f0c6dd : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/746acc47f4e84375987833f627782613 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07906708e7a54dfe8c7151631b1ea820 : SUCCESS in 12m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b4f15d511c1047b6b2d0208478fa2187 : SUCCESS in 4m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5227b9d74b8b4371aa5ab5f77a25aa5b : SUCCESS in 7m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/17fbbf1cf8434ebf8709a3f28934a193 : SUCCESS in 6m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/40e2592339844748a3af1b27d1ae245c : SUCCESS in 43m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0bd394f5c6a249a49fae4f20a203ba0e : SUCCESS in 1h 43m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b1448cbf29e3460f8d23964dbfed4b19 : SUCCESS in 54m 52s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/85508f65aae843f081ce6999539b1842 : SUCCESS in 52m 25s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ead3e1513f1e42418b6949abc0d49a0f : SUCCESS in 1h 00m 59s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/45b1b4dce7c24d08a350090f5c028f64 : SUCCESS in 1h 44m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/085e3f1c79f24442b1d617a58bd69eb4 : SUCCESS in 1h 59m 01s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0d4bbc458fd74e5e979dec50e94a62ed : SUCCESS in 1h 08m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e269bb6fc7f141b1869d895a38bbc989 : SUCCESS in 1h 52m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"09d73fe43cd3d65b900c72783b181a7e4b703801","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-19 16:26:59.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"7ea919aa83e007a20bd51ea49ea852cd124cda84","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-05-19 16:52:34.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8a283dfbbbb525b4977640a493f6ebd873df9478","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 18:34:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f70eb25e5ee40e5b9b34d175d05e27b : SUCCESS in 11m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/240b985ef9da4846855bb4195e8e27e5 : SUCCESS in 9m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49e2eeaab0994eb18e25c07dbb22fd37 : SUCCESS in 8m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c05652159b540c392af9a19979db0f8 : SUCCESS in 7m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/646985763ff6457f9120df387fee4d8b : SUCCESS in 8m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f8a3c9838bd4636b747945bfef156f4 : SUCCESS in 16m 13s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e46d6c62c80d425eae528d9514accb27 : SUCCESS in 7m 03s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/762e107f0bf7445fa65066990ed8faf2 : SUCCESS in 12m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/caf8f9ead6e04b49bcd03b5f392611b5 : SUCCESS in 11m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7146c772870746efa1db185928915cc5 : SUCCESS in 51m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/de9a945d2bb84df7b004e3030f9618c8 : SUCCESS in 1h 54m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dc627089881a4bbe8b66e9d0de9a366e : SUCCESS in 47m 06s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e72a5b9315a8448c9acd8855cd27c6d2 : SUCCESS in 1h 00m 16s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7d9dcb1305a74bb0893848a69fd0ba8c : SUCCESS in 1h 35m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6b65c8ccd32542469a5b3a19aef18d3e : SUCCESS in 1h 29m 52s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d98601a9216f4abd9f600055b43eb0ac : SUCCESS in 1h 58m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8d629a83bdf74081aa22bc6f043803b4 : SUCCESS in 1h 58m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/db414bda2ae14ff29df8e159f0824208 : SUCCESS in 2h 01m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f87c2f7706196cbbef2e745732577ba6dd6c47ba","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-05-21 20:50:47.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ebecc0c43346b0e66588f2c3f98faf5a67ca7588","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 20:50:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bd525c9f775c2ff29731b44ecdc3cc83c477925c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 22:41:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"6289ec740c9ea10e2450bb66882a3134d707e166","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 22:41:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d706fa600364cb09fa45ace29a5ca6f : SUCCESS in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/669d0e6dcd6a4ccfadc03a9ea72a26d5 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/556edee2dcee43dbb61c19f858f5cea3 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4df328f72844859993f4c420eb28a76 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bd002c874fd4abf91dfd470a9639834 : SUCCESS in 11m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/75e4ab1d3f2141d2aa124425655c1313 : SUCCESS in 8m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1b8200917cf54a3d8e45a1805a3cd8e8 : SUCCESS in 53m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/44883011500840bdb021de08b2df7ea5 : SUCCESS in 1h 50m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3f223e7d32874c4a9efe4c0558a64f3e : SUCCESS in 55m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/eb0db602d1654637abf3c79509cc7e38 : SUCCESS in 46m 23s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1b644f5d0faf40039a311b3afac939b5 : SUCCESS in 59m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/02351ff7f8c74dc09321ff5181a00069 : SUCCESS in 1h 33m 52s","accounts_in_message":[],"_revision_number":3},{"id":"b1bf81f4f26565a689fa7fa29bbfa8016769d467","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 22:43:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec8690b9ca8d400d88aff747cadb34ff : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d0f0233061243f284f05843ebdd504f3052049f6","revisions":{"2ce16d9a52e97efc023c42daffb8edb43bbcafae":{"kind":"REWORK","_number":1,"created":"2020-04-24 15:52:05.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/20/722920/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/20/722920/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/722920/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/722920/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/722920/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/20/722920/1"}}},"commit":{"parents":[{"commit":"c037a9b374223176248e0cbd7ce2ca382da21049","subject":"Merge \"Add Python3 master unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c037a9b374223176248e0cbd7ce2ca382da21049"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-04-24 15:50:19.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-04-24 15:52:04.000000000","tz":-420},"subject":"Make sure devstack aborts if DIB fails","message":"Make sure devstack aborts if DIB fails\n\nThere are cases where DIB can fail to create an image but devstack\ndoes not abort. This leads the gate job to run all the way down to\nstarting the tempest test before the job will fail out.\nThis adds a simple check for the image file and will abort early\nif the image is not present.\n\nChange-Id: I7ebf4137feb04827490dffc0dac3d6e4c8888075\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2ce16d9a52e97efc023c42daffb8edb43bbcafae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2ce16d9a52e97efc023c42daffb8edb43bbcafae"}]},"branch":"refs/heads/master"},"70425e0d0d3c8cb95cf5c6d8c61f261b8e384600":{"kind":"REWORK","_number":2,"created":"2020-04-30 22:57:13.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/20/722920/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/20/722920/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/722920/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/722920/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/722920/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/20/722920/2"}}},"commit":{"parents":[{"commit":"07a93de755c0ec461f51b94e3465708700ff1549","subject":"Merge \"Fix healthmanager not update amphora health when LB disable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/07a93de755c0ec461f51b94e3465708700ff1549"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-04-24 15:50:19.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-04-30 22:57:12.000000000","tz":-420},"subject":"Make sure devstack aborts if DIB fails","message":"Make sure devstack aborts if DIB fails\n\nThere are cases where DIB can fail to create an image but devstack\ndoes not abort. This leads the gate job to run all the way down to\nstarting the tempest test before the job will fail out.\nThis adds a simple check for the image file and will abort early\nif the image is not present.\n\nChange-Id: I7ebf4137feb04827490dffc0dac3d6e4c8888075\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/70425e0d0d3c8cb95cf5c6d8c61f261b8e384600"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/70425e0d0d3c8cb95cf5c6d8c61f261b8e384600"}]},"branch":"refs/heads/master"},"d0f0233061243f284f05843ebdd504f3052049f6":{"kind":"REWORK","_number":3,"created":"2020-05-19 16:26:59.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/20/722920/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/20/722920/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/722920/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/722920/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/722920/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/20/722920/3"}}},"commit":{"parents":[{"commit":"07a93de755c0ec461f51b94e3465708700ff1549","subject":"Merge \"Fix healthmanager not update amphora health when LB disable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/07a93de755c0ec461f51b94e3465708700ff1549"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-04-24 15:50:19.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-05-19 16:26:57.000000000","tz":0},"subject":"Make sure devstack aborts if DIB fails","message":"Make sure devstack aborts if DIB fails\n\nThere are cases where DIB can fail to create an image but devstack\ndoes not abort. This leads the gate job to run all the way down to\nstarting the tempest test before the job will fail out.\nThis adds a simple check for the image file and will abort early\nif the image is not present.\n\nChange-Id: I7ebf4137feb04827490dffc0dac3d6e4c8888075\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d0f0233061243f284f05843ebdd504f3052049f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d0f0233061243f284f05843ebdd504f3052049f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
