)]}'
{"id":"openstack%2Foctavia~723389","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I831b0b48d719397c14d80f8ebcbad997c50c7795","project":"openstack/octavia","branch":"stable/train","topic":"719922-stable/train","hashtags":[],"change_id":"I831b0b48d719397c14d80f8ebcbad997c50c7795","subject":"Fix py3 amphora-agent cert-rotation type bug","status":"MERGED","created":"2020-04-27 11:48:57.000000000","updated":"2020-05-04 19:07:26.000000000","submitted":"2020-05-04 19:05:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723389-1588619138756-6bb723e0","meta_rev_id":"10a9083f41350e40b543de535bdc554d1fc6cd34","_number":723389,"virtual_id_number":723389,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"date":"2020-04-29 18:47:00.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-04 19:05:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-05-03 22:37:08.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":1,"date":"2020-04-30 01:10:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":2,"date":"2020-05-04 16:28:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":1,"date":"2020-05-04 16:28:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2020-04-29 11:39:28.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-27 11:48:57.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-04-30 01:10:14.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-05-03 22:37:08.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2020-05-04 16:28:32.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-05-04 19:05:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d39a990cf11e9111fade1d679c6d1a5dbd96ee20","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-27 11:48:57.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8cd32ce9dcf48a36c17395384bf7fb34ab183348","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-27 11:49:09.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit d2270fad058385a461376ebcb484c1fc9a8d908a","accounts_in_message":[],"_revision_number":1},{"id":"bd2f902b26b57cb00318c20d81c5373b688db930","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 14:30:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/689ddb12a7784582af1109dd31edd93c : SUCCESS in 10m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1275aa7d6e564c89a7d8a8d316165a19 : SUCCESS in 9m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bf336bc0fed43a7a1d007f4562dd8f7 : FAILURE in 7m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/198b758935ed47f986bd5d3bab52ae88 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbdf9e6ae8784f929be0dc8ac9547eb8 : SUCCESS in 6m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/895810dd10c24faf9ca622713e921b38 : SUCCESS in 7m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d293222ca21140be91fad2fb8f85ed71 : SUCCESS in 12m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/830f610131434bb28d86e067e2c97ab1 : SUCCESS in 7m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b7099d4020884d1aa55488bc5c93a0b1 : SUCCESS in 7m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0f9f86d60b684490b7c5487920353e2c : SUCCESS in 9m 14s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ae20cee2b8b34c06894515804d81fcc8 : SUCCESS in 8m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a3e0b5ffdb9f4f18933dc3a9c0c3356b : SUCCESS in 8m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d1fc682ec2fa4b229c48583d4ff0de5b : SUCCESS in 45m 53s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b03f7c1683b3401287df1314b567f88c : SUCCESS in 43m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2e84a6e860374035b51edf8fd51ecfa0 : SUCCESS in 1h 44m 00s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f230398aa757444ca225b3252864d5d7 : SUCCESS in 2h 00m 42s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/dadfff1079d9455eb8c118f2722ed338 : TIMED_OUT in 2h 36m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b99ebe5af038420c8eb9c9a6d3c86dbd : FAILURE in 2h 18m 38s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/77070d585bba4326ba7df2dbf96d0cbc : SUCCESS in 2h 08m 44s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/96d4f0339076416eafd3f829472efc65 : SUCCESS in 1h 21m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ed5950ec92c54c6eb14eccb5ce80dc8a : SUCCESS in 1h 00m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2684d2b60723467d82abf91e210340f1 : SUCCESS in 2h 00m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6503a2f2ee764b1cb95601b3ce063293 : SUCCESS in 1h 09m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5c059b18358b495a93e069aa7a7e34f6 : SUCCESS in 1h 03m 59s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8c023032bf7c4058aea009883a42daf2 : POST_FAILURE in 1h 26m 44s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/api/drivers/utils.py","accounts_in_message":[],"_revision_number":1},{"id":"380ebba697515d482cc3a4aec468004e8f3a76e3","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-04-27 22:24:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"983c9e2220335e6f4fe796f5a31d2e8086fa1eae","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-29 11:39:28.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"cd6acb822801f22baf86891e4f68cd9446d5c998","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 14:09:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7761692b5e6429591d68239dec59e76 : SUCCESS in 10m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab1c326396054b5cb06627b223011ab9 : SUCCESS in 10m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d99d0394a8574188a43a9e04d292046a : SUCCESS in 7m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d8a1a6962e6f48c8ad22ed11f80a067a : SUCCESS in 7m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84023b6096ef432d9266e8afff9d9216 : SUCCESS in 7m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e8f21d926024e16a62295384764608f : SUCCESS in 8m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1fbf18fcd154c3f8b708685cdded128 : SUCCESS in 14m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d2cdc14b18e4a81855ae161f62297df : SUCCESS in 6m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/91e1e052147b4af28bdfb2f9179427dd : SUCCESS in 7m 27s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/084b54133d66419b9599e8d40d3de35d : SUCCESS in 8m 42s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/dbc509b91d6c44efa04bdf79a56847b4 : SUCCESS in 9m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/65eb6dd1238a4589a8d8c36d68b46a4e : SUCCESS in 9m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/23ddd25d80944297be0fa5ad39c95bc1 : SUCCESS in 45m 02s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/9cfbfe0207ee462bbcc6632816efa254 : SUCCESS in 51m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7de82394b97c418da540ceaf35109e2b : SUCCESS in 2h 01m 37s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e33ce93fa71645e184531872a5d98365 : SUCCESS in 1h 58m 19s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/30d113340e884153ba87435e4d0b3338 : EXCEPTION (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/72af6ce8b5384356bda5d7ef0eef1b82 : SUCCESS in 2h 10m 23s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/742bb93a691447fbb21c10fbb0837f39 : EXCEPTION (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/43d819a0f6a24b4396153dd8dcdcf842 : FAILURE in 1h 34m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d91dae4819e2444a8b0eef42a527af2e : SUCCESS in 1h 02m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c41869e3a9ad44cf9c6568c1c9817728 : SUCCESS in 1h 46m 15s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5117abb66f464e5c857a98c90e3c6174 : SUCCESS in 1h 01m 34s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/9b1857ef74774ffea5adf8e3e4896c2c : SUCCESS in 1h 06m 58s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/acd6413a57e342d781b22c4d725d2a0b : POST_FAILURE in 1h 40m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"58ff66c1e3094a149d64fb3adf15e45367dfdede","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 17:04:39.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3fbd8b2b4d0f48598a48a4ec3e0cf723 : SUCCESS in 12m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff0402b6b2db462099ccd721f2631bb1 : SUCCESS in 11m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/589fc660278f499cbb9fba5e4ede8ad0 : SUCCESS in 7m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/113a96fc4711415b8481eb565d87ff31 : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d39305da31254b86843f6cdfbd89f255 : SUCCESS in 7m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5ccdfe229aff4cca86c5d2d21edb4fbd : SUCCESS in 9m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf07abaa7a194d7f9586bb26a151957b : SUCCESS in 15m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/321074786358469298293aeeb9e276c9 : SUCCESS in 9m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/522526f5fa634fa38e1a6b9633d759e2 : SUCCESS in 8m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/cc004464c2624221a06e4d5407977bbe : SUCCESS in 11m 32s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9ea2296592eb4f08b19d326b491c7334 : SUCCESS in 8m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/100cfb08b52b4cb991eff280c7c9d3c7 : SUCCESS in 9m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6b563730634049c88f1c4affcb63a698 : SUCCESS in 48m 27s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/601aaaeb5aca4bfaa808df66308d9836 : SUCCESS in 46m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d65af3973847490aada0deab5c4f0403 : SUCCESS in 1h 47m 32s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/091451844f5b4bf688f36382208333b9 : FAILURE in 1h 10m 11s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/ce3d4a8c46e24c01a09f304fd497cca3 : TIMED_OUT in 2h 34m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5b29a206db03455d91b5f3ab4c9f8229 : SUCCESS in 1h 51m 45s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/20acd1897c2a4b78a0b646a35785288a : SUCCESS in 2h 10m 05s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/54406802011c462eac87428762c6e9fc : SUCCESS in 1h 25m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/85ad4d58652649c19dfdc697aa57c640 : SUCCESS in 1h 09m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/357442c7314d46168920c21bdcbc7d2f : SUCCESS in 1h 50m 32s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/151adc1a3bc3405687c5e6cb7f55b95c : SUCCESS in 1h 13m 42s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8c85f82e9dbd47969bf391a26052c190 : SUCCESS in 1h 17m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/603d9a4eb7ac4d26a56ba6ae2bf0c758 : POST_FAILURE in 1h 21m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"232e030eb5033579a1c86d6f8d56e472c5bd86f4","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-29 18:47:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"da90e4a5d1ab4a28fcf4ab23551c5ca0b746c6ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 21:21:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e117e896afcc41eea076a69603a59e44 : SUCCESS in 9m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27ba14724f0742d989c1b6bd0f801506 : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f35b11bd4814cd7912115b348c2df20 : SUCCESS in 8m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7ed27a3178a4449299a61b196600a458 : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29b511f9a0e047caa5eb98813264e20f : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2bf47c21d23a4c7c8aa78c80916f7802 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b27ace8e6d94748abe951d316bee351 : SUCCESS in 10m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b74ca38cf324ffc8addd2a103ea847d : SUCCESS in 9m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f5e70888ea804262b348bf1d170eb6a7 : SUCCESS in 6m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9d511bf442d54ec2acef2a5c5878b597 : SUCCESS in 8m 48s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7ebac0267cbb48c88a986db239ddf0f6 : SUCCESS in 7m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a0bc8fbf090e49a1929c09eae8cdaf65 : SUCCESS in 7m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c52b125b0b2a4c789e16bb84879325ef : SUCCESS in 45m 07s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/6c5b7fdf717c47e3b86a6bfa164894d3 : SUCCESS in 52m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/88046a6f66774eaab6357811f3311cc9 : SUCCESS in 2h 05m 35s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b50751834feb4e6c9dfe2ee7f07ca147 : SUCCESS in 2h 00m 09s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/881c70ca910545b98666257398776550 : POST_FAILURE in 2h 22m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f93d0cc447ff445290861c58e77859fd : SUCCESS in 1h 06m 51s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/e16c67c03ca44659925505d551fccfdc : SUCCESS in 1h 57m 14s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/023e0b65cf7741d9bf5d9f15bc51b3e4 : SUCCESS in 1h 25m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/27da8a05302544f998f22945738ac472 : SUCCESS in 1h 05m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6746e5b263494e94af449e37bae11854 : SUCCESS in 1h 47m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ee8c6d8747db438fa59a58cc7d4a8418 : SUCCESS in 58m 48s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a23b5faa7ed84ccf80f96eb4a76def65 : SUCCESS in 47m 26s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/b1d1b648af0d4167ac15cb8479255fb2 : POST_FAILURE in 51m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3381db6527ff39e87f090fa313de47a67f72d5e1","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-04-30 01:10:14.000000000","message":"Patch Set 2: Code-Review+1\n\nthanks. We are affected.","accounts_in_message":[],"_revision_number":2},{"id":"3ac8d8222dbca0d14392232966a26b64a35199b4","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2020-05-03 22:37:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7a33ef99f0c3549a0643943b539f5b28cc1077e5","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2020-05-03 22:37:16.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"b44fd6fe5902053e01eb5407d93c514c0ded97f2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-04 16:28:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"7667bf928c6510f65e55d410d4213ba88f95173a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 16:28:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"af16a9dd8c5750c2d4ca8da6be5b5cb1e3e92073","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 19:05:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d4ef4d56d79c6816143b939159f1068bab489b12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 19:05:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56353064d9594205a6fa618d9664a258 : SUCCESS in 9m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee7ae3edef1c418897d55345b3f925f6 : SUCCESS in 6m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/25218a45b0a6441e897b6934531fc0fe : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22109eede5144a4591ae32ffcbd253af : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dc431209841e4cf28164034fe5bd5d84 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81189f79c7a74ca09ec948b442e64485 : SUCCESS in 10m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84ee75eff1ed4249aad9bb6bbf3b653d : SUCCESS in 7m 20s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/98e06716c69b42819d47c103157dbf5a : SUCCESS in 6m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/15aa1460c0484dc2abacfcb5d1eda4d4 : SUCCESS in 7m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4cf8a20b0d3544e897fd38bf57b04803 : SUCCESS in 43m 52s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/1b0f508585394c3d8fe4908fdcb75c39 : SUCCESS in 46m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e4fd7e5573194110a889cc4f7ff783b7 : SUCCESS in 2h 01m 12s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3d77451bda03449d94b77986d83faf47 : SUCCESS in 2h 29m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/de2f79644d7e4eb9a2e3c786091d1fc2 : SUCCESS in 1h 15m 53s","accounts_in_message":[],"_revision_number":2},{"id":"58392dbc759f9fd9227f14681eb9024298fa5cdc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 19:07:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff0b60f828b74a9f928d633679493b60 : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7dba0f99eee4ad1bbadf1049fc1c931 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"43a3897d3828f3c162a335d784b258f96a72f9b6","revisions":{"1cd2e64b6b04c1e83f4c8e0f31f88628e2780145":{"kind":"REWORK","_number":1,"created":"2020-04-27 11:48:57.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/89/723389/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/723389/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/723389/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/723389/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/723389/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/723389/1"}}},"commit":{"parents":[{"commit":"872dbdda95b95482c62cedb40145b00ddbab1bb9","subject":"Merge \"Fix load balancer update with provider filtered params\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/872dbdda95b95482c62cedb40145b00ddbab1bb9"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-04-14 11:27:49.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-04-27 11:48:57.000000000","tz":0},"subject":"Fix py3 amphora-agent cert-rotation type bug","message":"Fix py3 amphora-agent cert-rotation type bug\n\nFlask\u0027s stream always returns bytes, file write always takes string.\nThis causes py3 amps to return 500 on cert rotation AND wipe out the\ncertificate, so the amphora are no longer controllable and go to ERROR\nstate. Anyone running py3 amps prior to this patch will experience\namphorae breaking on a timer due to housekeeping cert rotation!\n\nChange-Id: I831b0b48d719397c14d80f8ebcbad997c50c7795\n(cherry picked from commit 96a4482dff134b7be5c8d4da6c862b0a7f4a5a4b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1cd2e64b6b04c1e83f4c8e0f31f88628e2780145"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1cd2e64b6b04c1e83f4c8e0f31f88628e2780145"}]},"branch":"refs/heads/stable/train"},"43a3897d3828f3c162a335d784b258f96a72f9b6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-29 11:39:28.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/89/723389/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/723389/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/723389/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/723389/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/723389/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/723389/2"}}},"commit":{"parents":[{"commit":"7fa2f933a2c99e7ef0b2cb8d4cbcc1e9f9c0c79f","subject":"Disable catching-non-exception in pylint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7fa2f933a2c99e7ef0b2cb8d4cbcc1e9f9c0c79f"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-04-14 11:27:49.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-04-29 11:39:28.000000000","tz":0},"subject":"Fix py3 amphora-agent cert-rotation type bug","message":"Fix py3 amphora-agent cert-rotation type bug\n\nFlask\u0027s stream always returns bytes, file write always takes string.\nThis causes py3 amps to return 500 on cert rotation AND wipe out the\ncertificate, so the amphora are no longer controllable and go to ERROR\nstate. Anyone running py3 amps prior to this patch will experience\namphorae breaking on a timer due to housekeeping cert rotation!\n\nChange-Id: I831b0b48d719397c14d80f8ebcbad997c50c7795\n(cherry picked from commit 96a4482dff134b7be5c8d4da6c862b0a7f4a5a4b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/43a3897d3828f3c162a335d784b258f96a72f9b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/43a3897d3828f3c162a335d784b258f96a72f9b6"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
