)]}'
{"id":"openstack%2Foctavia~723393","triplet_id":"openstack%2Foctavia~stable%2Ftrain~Ia9fec4e72c00f7086489b245d9dc50ed9c27f12a","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"Ia9fec4e72c00f7086489b245d9dc50ed9c27f12a","subject":"Fix padding logic for UDP health daemon","status":"MERGED","created":"2020-04-27 11:51:36.000000000","updated":"2020-05-06 13:58:27.000000000","submitted":"2020-05-06 13:56:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723393-1588773391119-71cafbf4","meta_rev_id":"3b9e8c394d34709015adfff87645c8127c8a1750","_number":723393,"virtual_id_number":723393,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-06 13:56:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-05-06 11:48:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-29 18:05:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2020-05-06 11:48:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-29 18:05:41.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-05-06 11:48:15.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-05-06 13:56:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e9a60876f38e996732514a2ea3a6488ebe3befb8","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-27 11:51:36.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"90fc560e4db2ce61fc8100a9cb22e8971c935c54","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-27 11:51:43.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 5cec11cf647c74570cb2de6246b22da5e2c0401e","accounts_in_message":[],"_revision_number":1},{"id":"e924ab5f674838a53834709e4cea394cbc52e4a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 14:24:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/153a32f2eaf840af9b8357c6a58963fa : SUCCESS in 9m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9199b2e833a7437aa61f5b6c3593f3c6 : SUCCESS in 10m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93c2cb8559a1485d9cbeafcf6d3e0136 : FAILURE in 8m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7ee8619836dd44fd834b2094deb1be99 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8cde6b5c26364acc8ae454ca39de940a : SUCCESS in 5m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1fe8eb9c9bf04537a33f3bc16c126255 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9dd94f6f6984e5480d3b953b82dac78 : SUCCESS in 10m 36s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a789cb327fe84cb99d87d42c9c81496f : SUCCESS in 6m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3a937d97a53c4cedb2cc68537af18111 : SUCCESS in 7m 58s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2b93a885ae0648ec8a28f941b2f45b99 : SUCCESS in 6m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/390c15c4f6e546d182d3c52bf0f4e90b : SUCCESS in 7m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7feca117b6894b9aa9a83e94c5f64c78 : SUCCESS in 45m 20s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/a070da97a336497981b8f667ba64ec2c : SUCCESS in 48m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d519b94dba7a4b76990296e6f8d1fd65 : SUCCESS in 1h 54m 10s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/abd71d6034ca45429dad1a94c7523490 : SUCCESS in 2h 15m 23s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/169f7a7a2fcb4cad85813f33e6f2c3f2 : FAILURE in 43m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2d4137a772024067bd75724b15defc27 : SUCCESS in 2h 17m 48s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/3d68b04705c34359bf8467be9c6ea89f : SUCCESS in 2h 25m 47s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f70fa77e79564bee84c7a94914652969 : SUCCESS in 1h 26m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/17ee38145b7b42a0a27f052751001407 : FAILURE in 37m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d64d4387dce74b399a9af29a298e0fef : SUCCESS in 1h 58m 35s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5e961e7746a44877ae7e5aa73854b47c : SUCCESS in 1h 08m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f6460827987b40578eb26d2973cd19cf : SUCCESS in 1h 22m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/866cfeef355f4919a9f6a0c0ded6cffd : POST_FAILURE in 1h 35m 01s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/api/drivers/utils.py","accounts_in_message":[],"_revision_number":1},{"id":"d7a16e5864c46e34e728e3465201cb83233da599","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-29 11:39:38.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"439fa8d61fd46d01714952244fd8a6f1f1054b34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 14:09:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61ba194192e1495183b7b5d0499a50a4 : SUCCESS in 11m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d70cb59dec4b462eaa65417a38666c0a : SUCCESS in 10m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b456614b60f84bd580249d1824304916 : SUCCESS in 8m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a69d2b3caf884033abacdd3c73753eec : SUCCESS in 9m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d46e06ac6db247baa9b0b90fdbe229d3 : SUCCESS in 5m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b6bc35ac23a148719f2a048983d45a8b : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd34c3eeefed4c59a67fb1aa82b66e0d : SUCCESS in 12m 57s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d04b06e4a67e45c6b7aef38dec76fbb4 : SUCCESS in 5m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6d87eec9c6bb42f29517840a474013ba : SUCCESS in 8m 20s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1f4ebcbecfa940d09edb88764a7d8761 : SUCCESS in 8m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/65e8a01af67842f6ae1636d35ce942bd : SUCCESS in 9m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/27a0ad871f454874ad7e40edeb9105c9 : SUCCESS in 45m 01s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/7e99882f5527417dba9eee2bad6a66ff : SUCCESS in 45m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fdbc7da343b742d8bd651f0677c28f55 : SUCCESS in 1h 45m 32s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/032a75aac7d64ece8eb00146ab8b1ebe : SUCCESS in 2h 00m 31s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/b96ddf706bdf41c3bc6746e746aef55b : EXCEPTION (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a15ffdf5f7274a5c97723809a6b178e9 : SUCCESS in 2h 09m 03s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/adfd81d5a47440329ed4ca12028f33ca : SUCCESS in 2h 11m 48s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7a34a8aa723b49a0a275b553a5a85a33 : SUCCESS in 1h 14m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0da75c7dd17d46d386a62c74a05efd7a : SUCCESS in 59m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/48cb5c75ed3841c58909e03bd35ea020 : SUCCESS in 1h 03m 08s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cfab73af6ae242ebace003c600e5461f : SUCCESS in 1h 06m 36s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3251b559edc9409bbd4eb890801d7bf2 : SUCCESS in 1h 06m 25s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c167c0f336c14956a7ed2dd8ba5218bc : POST_FAILURE in 1h 28m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7f898663f7b0f743856226b88295c1097649f8af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 17:07:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/febe3e6e6f324a69a21bb714af1685c9 : SUCCESS in 10m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9674b21679484e9495b733c19cc4a206 : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58349c2ee4574557887adfbfe674227e : SUCCESS in 7m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/56d7d34c89c749c69c1067ee76834e45 : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/493d362319814d42b6b60cc23421e83c : SUCCESS in 6m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7d2abd6987874c0bbf4e4e9cd5cb0477 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87fcca69dafa4bb28a55fcfe88adbb5e : SUCCESS in 13m 10s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9b2a7848915546c5951bc52a96e9a4eb : SUCCESS in 8m 26s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/016168f670a34843bf4d90ee3b0e5172 : SUCCESS in 8m 54s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1a98e71fc3ed4c8f88773a3b9581f241 : SUCCESS in 8m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/637746f8df714f15a448332c766e497e : SUCCESS in 11m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/74441432813e4ff3a191d636ab4dfee8 : SUCCESS in 44m 48s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/02cd2eb20e0d4c609246f37698b6fbd7 : SUCCESS in 44m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/809d1c18077a496eb2ce6af2ecfbd731 : SUCCESS in 1h 51m 44s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7f3c59957908460784f33829a6c6fada : SUCCESS in 2h 02m 28s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3cdaa90ca4754f928cf021adde442b51 : POST_FAILURE in 2h 14m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/04d0901e79484bb4aede38d4c985d88e : TIMED_OUT in 2h 36m 38s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/601d5f05b3fa4433ade6bb9e73ce0d5e : SUCCESS in 1h 26m 58s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4a634ae32e324eb98687f59970f31813 : SUCCESS in 1h 20m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d1ce6c3fe22a4276a2beefb24413b1a9 : SUCCESS in 1h 15m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/66ee04fd5b9e4147a43c1611c8ce4f62 : SUCCESS in 1h 02m 00s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1bab84f9695a4bb4be7ed72e69dee2ee : SUCCESS in 1h 01m 59s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d6ff3c07372142089297dd406debf7c3 : SUCCESS in 58m 49s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/62889c08af1c47a492e64dbf125aaeb9 : POST_FAILURE in 1h 00m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"41584892d9124276c157ae0171414c3b5a153ecc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-29 18:05:41.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"b22828bfbf7542f4669b88d40a32c2c7f0278258","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-05-06 11:48:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"02257fa1c2bd47014725fe198024f26f52df9299","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 11:48:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f9bed1de12b3c3097ce9c0e81f2532ba7b15467f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 13:56:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c69a023717594199b8d285792abc1c4c : SUCCESS in 10m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73529d0e0ee24cb0bec0b3ed6b78e197 : SUCCESS in 7m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5ad4a537b9b84a0a89e8a06a945288ca : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5490a6dcb9b4e4e8d7c2029415d5147 : SUCCESS in 6m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f9f865978e404e65a1792a4fcd4719a3 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af698dcd26764b53b53ac96f14f48707 : SUCCESS in 11m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d27b7f2d73664891bee7c726987effc9 : SUCCESS in 7m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/efde9db8fcf64299a345a27fd29f2217 : SUCCESS in 9m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f45bd9db4bf7410c953ecd3a735dd76d : SUCCESS in 46m 11s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b4642dd8b50347afb63ac825586e3874 : SUCCESS in 52m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0ac226d16d1d41f581d5b5f222718eb0 : SUCCESS in 2h 02m 00s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7e147862e443490abbe6be0e9c9ce4ee : SUCCESS in 1h 48m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dcd0836dd5074664b32421eea6fb556a : SUCCESS in 1h 21m 09s","accounts_in_message":[],"_revision_number":2},{"id":"e09dd36a565b2b5ffe3e0b2bdc2ec910aec233fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 13:56:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3b9e8c394d34709015adfff87645c8127c8a1750","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 13:58:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5d9e3f7e4974b588e1459ebe27f756c : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fb8ec7a38168e8fa1381136a468037161abd6157","revisions":{"49d58bb81e097ef87f5ae525620695242ee9f312":{"kind":"REWORK","_number":1,"created":"2020-04-27 11:51:36.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/93/723393/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/723393/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/723393/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/723393/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/723393/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/723393/1"}}},"commit":{"parents":[{"commit":"872dbdda95b95482c62cedb40145b00ddbab1bb9","subject":"Merge \"Fix load balancer update with provider filtered params\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/872dbdda95b95482c62cedb40145b00ddbab1bb9"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-03-20 23:51:44.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-04-27 11:51:36.000000000","tz":0},"subject":"Fix padding logic for UDP health daemon","message":"Fix padding logic for UDP health daemon\n\nShould have done \"pad to 8 characters\" on the hex conversion, but it was\ninstead hardcoded to pad a single `0`, which is right in a lot of cases\nbut not all.\n\nFor example:\n\u003e\u003e\u003e ip1 \u003d ipaddress.ip_address(\u002798.136.140.23\u0027)\n\u003e\u003e\u003e ip2 \u003d ipaddress.ip_address(\u002710.1.1.1\u0027)\n\u003e\u003e\u003e \"%X\" % ip1._ip\n\u002762888C17\u0027\n\u003e\u003e\u003e \"%X\" % ip2._ip\n\u0027A010101\u0027\n\nChange-Id: Ia9fec4e72c00f7086489b245d9dc50ed9c27f12a\n(cherry picked from commit d27ee3f0ee98934e6ab354704edc47f61fffdcda)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/49d58bb81e097ef87f5ae525620695242ee9f312"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/49d58bb81e097ef87f5ae525620695242ee9f312"}]},"branch":"refs/heads/stable/train"},"fb8ec7a38168e8fa1381136a468037161abd6157":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-29 11:39:38.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/93/723393/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/723393/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/723393/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/723393/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/723393/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/723393/2"}}},"commit":{"parents":[{"commit":"7fa2f933a2c99e7ef0b2cb8d4cbcc1e9f9c0c79f","subject":"Disable catching-non-exception in pylint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7fa2f933a2c99e7ef0b2cb8d4cbcc1e9f9c0c79f"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-03-20 23:51:44.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-04-29 11:39:38.000000000","tz":0},"subject":"Fix padding logic for UDP health daemon","message":"Fix padding logic for UDP health daemon\n\nShould have done \"pad to 8 characters\" on the hex conversion, but it was\ninstead hardcoded to pad a single `0`, which is right in a lot of cases\nbut not all.\n\nFor example:\n\u003e\u003e\u003e ip1 \u003d ipaddress.ip_address(\u002798.136.140.23\u0027)\n\u003e\u003e\u003e ip2 \u003d ipaddress.ip_address(\u002710.1.1.1\u0027)\n\u003e\u003e\u003e \"%X\" % ip1._ip\n\u002762888C17\u0027\n\u003e\u003e\u003e \"%X\" % ip2._ip\n\u0027A010101\u0027\n\nChange-Id: Ia9fec4e72c00f7086489b245d9dc50ed9c27f12a\n(cherry picked from commit d27ee3f0ee98934e6ab354704edc47f61fffdcda)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fb8ec7a38168e8fa1381136a468037161abd6157"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fb8ec7a38168e8fa1381136a468037161abd6157"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
