)]}'
{"id":"openstack%2Foctavia~723604","triplet_id":"openstack%2Foctavia~master~I66b2d0687f4edfbcbff99c29c9c5d539e4e7ea7f","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I66b2d0687f4edfbcbff99c29c9c5d539e4e7ea7f","subject":"Remove deprecated exception classes","status":"MERGED","created":"2020-04-27 16:27:16.000000000","updated":"2020-04-28 06:36:32.000000000","submitted":"2020-04-27 22:59:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723604-1588028391200-39316be2","meta_rev_id":"ba5e17e0e4d5677d360950aedf0f097db782e921","_number":723604,"virtual_id_number":723604,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 22:59:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-04-27 19:48:29.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":2,"date":"2020-04-27 19:38:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2020-04-27 19:48:29.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-27 19:08:37.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-04-27 19:38:46.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-04-27 19:48:29.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"},{"updated":"2020-04-27 22:59:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"241de508070fe83b5df8aaebc08918dbd6bef265","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-27 16:27:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53586931c98d7e09b7f497841d1cba5d6cdc1cee","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-27 16:44:53.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"aef5bedd3f1839a338c5cdbb718906bd1eeedc1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 18:42:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20b32be280a6424ba623cc69e4a597ef : SUCCESS in 10m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84b60929443b454481818d565994070c : SUCCESS in 10m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f52cc3251304de8af71d6e5598f9bc3 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67ef03bdaf25495ca3dfa1286f6fe969 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f40cc9a7cbf467493bec9c07eb9b316 : SUCCESS in 8m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41846117a1c7409291f22daa9708b868 : SUCCESS in 11m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c39d33a57ed04e64a6ed88a22b4b1ee1 : SUCCESS in 5m 53s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/cb27f9fff82241359f609ebba46b30d6 : SUCCESS in 8m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e1a057846075474abff75e3418321a4e : SUCCESS in 8m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c9c4d2e758da45059070253a8dfa8e71 : SUCCESS in 50m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/304f3b1a761243eb88c623165660f775 : SUCCESS in 1h 02m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5326f37d21344d6d89be22c5433440ae : SUCCESS in 54m 46s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b7f87503818c49f1bee8f9a608c2bbc9 : SUCCESS in 51m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1ecca2fa0f7c4e778a56bc4c55447daa : FAILURE in 1h 39m 31s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fb790320df0a44528e8e08e82785eadb : SUCCESS in 1h 48m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e37ebff5f49b4e789c5a015ad68059dc : SUCCESS in 2h 02m 04s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/02daef4e8dbb4a5a8ad396ad3499add5 : FAILURE in 1h 07m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/434fa8e8582440d8b3e4ff048d13c752 : SUCCESS in 1h 47m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b01859ace0cc37bb2cc776c5705f6ef4e401d5a3","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-04-27 19:08:37.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nnit -1 because you need to recheck anyway","accounts_in_message":[],"_revision_number":1},{"id":"6ff4c61350fdf081cd9a9591f1d9ec1d73f32ee8","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-27 19:22:11.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3d2b0925486e0b87b0d9b0ce8f15c0a4fe864224","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-27 19:38:46.000000000","message":"Patch Set 2: Code-Review+2\n\nStill looks good to me","accounts_in_message":[],"_revision_number":2},{"id":"a7181a4f2ac5be947b23c6d46724259587146519","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2020-04-27 19:48:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4941ec650599b481221afe84c42b690ed6b34577","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 21:54:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2423065959449b68cfdc52cf0b7badc : SUCCESS in 12m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f9502d58dad0439fa404c5eed45a06ad : SUCCESS in 12m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7376d80759f54cc9a72b7f3eeaf090ab : SUCCESS in 8m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc344e01757746c4818b3f80ecea7794 : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5142451d0c684f2f9c765279718de101 : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d4b80785abe4b95ab4a3d3a06641dd8 : SUCCESS in 27m 11s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/596d777d52854e81b02b76a44e7ad7a3 : SUCCESS in 9m 30s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ae8dd37035ae4c7ca99d3a3d611af16b : SUCCESS in 8m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b706e9ce01534fb79f4288626464e807 : SUCCESS in 9m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9cefcc7c3cec4f0397685ad77e074f5f : SUCCESS in 48m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7d17a940df6f4f3cb477d6c27af61e11 : SUCCESS in 1h 07m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a36628d463724e2cb7bf396e4e93eb03 : SUCCESS in 55m 03s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d2e892f0e9ff413db18467a71b6ed0a0 : SUCCESS in 54m 25s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/11195d56a02a4684aebfab01b1bcbfb8 : SUCCESS in 1h 15m 11s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5a850fea660a4551aab864f936ff4e31 : SUCCESS in 1h 13m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c55ef6134fd24222974d32f829c88fcc : SUCCESS in 2h 06m 45s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5416ffc15ab34aee9d3bb455f437e3e2 : FAILURE in 1h 14m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e8a648fc9e974b17b5284953a4a1bb47 : SUCCESS in 2h 24m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2aa78251dd038effc74635ffa31e855642629cb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 21:55:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d917e5957619aa29f0d4ccc6c9d5e13fb3bc226f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 22:59:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13c03df327a44c41b8ca8729b93144f7 : SUCCESS in 8m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79a2408f00024993a91cfce4fc1d8732 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46db682c64614cdd98b493c280e4829a : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57b02a3011494e1a933955fb4823677b : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f0c94dc6a2e4a97948ea74df72d435a : SUCCESS in 12m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0537231c4992409fb534df04de323ec4 : SUCCESS in 9m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7e29b50e722442528b94e103c2e88403 : SUCCESS in 38m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e7c4ddca53594df8a2cf4bcb9db95849 : SUCCESS in 54m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6746bd61e31f4d8e860459d3c12d6e01 : SUCCESS in 1h 03m 36s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6a40955c79d04f099278b9aaad42f351 : SUCCESS in 48m 30s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9113dfe49ad649a68b75e8e59d99c982 : SUCCESS in 39m 25s","accounts_in_message":[],"_revision_number":2},{"id":"823c8aa44ef673677f6dc608802697371156c16a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 22:59:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"9792c4ae3c5d79576a1fcbb7525a09877c75533d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 23:01:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89b1d4ea49ae4547bbed3927167fa244 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":2},{"id":"ba5e17e0e4d5677d360950aedf0f097db782e921","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-28 06:36:32.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 0fb5dd53e024dbc45ea5dc43b01d5d15cf807376","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0056b5175f89070164849501ec6d960549b95e34","revisions":{"7f93f6c6ade0c96185de2a38a7b035c72a9c8bec":{"kind":"REWORK","_number":1,"created":"2020-04-27 16:27:16.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/04/723604/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/04/723604/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/723604/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/723604/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/723604/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/04/723604/1"}}},"commit":{"parents":[{"commit":"c037a9b374223176248e0cbd7ce2ca382da21049","subject":"Merge \"Add Python3 master unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c037a9b374223176248e0cbd7ce2ca382da21049"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-04-27 16:20:20.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-04-27 16:20:20.000000000","tz":120},"subject":"Remove deprecated exception classes","message":"Remove deprecated exception classes\n\nThese classes were deprecated in Stein and marked for removal in Ussuri.\nBy removing these classes, we fix pep8 issues (catching-non-exception)\nwe started seeing at the gate with the release of astroid 2.4.0.\n\nChange-Id: I66b2d0687f4edfbcbff99c29c9c5d539e4e7ea7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7f93f6c6ade0c96185de2a38a7b035c72a9c8bec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7f93f6c6ade0c96185de2a38a7b035c72a9c8bec"}]},"branch":"refs/heads/master"},"0056b5175f89070164849501ec6d960549b95e34":{"kind":"REWORK","_number":2,"created":"2020-04-27 19:22:11.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/04/723604/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/04/723604/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/723604/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/723604/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/723604/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/04/723604/2"}}},"commit":{"parents":[{"commit":"c037a9b374223176248e0cbd7ce2ca382da21049","subject":"Merge \"Add Python3 master unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c037a9b374223176248e0cbd7ce2ca382da21049"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-04-27 16:20:20.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-04-27 19:22:06.000000000","tz":0},"subject":"Remove deprecated exception classes","message":"Remove deprecated exception classes\n\nThese classes were deprecated in Stein and marked for removal in Ussuri.\nBy removing these classes, we fix pep8 issues (catching-non-exception)\nwe started seeing at the gate with the release of astroid 2.4.0.\n\nChange-Id: I66b2d0687f4edfbcbff99c29c9c5d539e4e7ea7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0056b5175f89070164849501ec6d960549b95e34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0056b5175f89070164849501ec6d960549b95e34"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
