)]}'
{"id":"openstack%2Foctavia~723810","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I66b2d0687f4edfbcbff99c29c9c5d539e4e7ea7f","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"I66b2d0687f4edfbcbff99c29c9c5d539e4e7ea7f","subject":"Remove deprecated exception classes","status":"ABANDONED","created":"2020-04-28 06:36:32.000000000","updated":"2020-06-12 19:53:18.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d2c89449cda19ae19bab11d9b70547e953cf90a8","_number":723810,"virtual_id_number":723810,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"date":"2020-05-18 13:00:48.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-05-18 15:04:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":1,"date":"2020-05-18 12:18:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":-1,"date":"2020-05-18 19:35:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-18 12:18:03.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-05-18 15:04:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-18 19:35:54.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"391e8eb521ce56035cc4b801f6e13eff81965ff9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-28 06:36:32.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"eac8f9fd6c1ad14498fd89a4af1d93501991163c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 08:46:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee76c463827547ad905e99b4dbf2caea : SUCCESS in 11m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7fef4be2579f4217b61439fc4988f43a : SUCCESS in 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fae72c0fbd24d7b8475e37a6c74eb7e : SUCCESS in 8m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b265156ef27f4e7eac7199dda9a5f22e : SUCCESS in 7m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8bd1a686a7004242bd8433a7dd68d6d0 : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45a025fd08d0426abca1502306bb703a : SUCCESS in 14m 54s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e2e85d41d1e84bd089f809de69631a77 : SUCCESS in 6m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5a1d32b7e7dd4b53a4481e2b88076283 : SUCCESS in 9m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/04f38d28ca004b7fb2d26e8e9e62b039 : SUCCESS in 9m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/07be1d373b1146c8a9b73fec40ecee3e : SUCCESS in 42m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/445ffd27989a4b1596295a19d84e14c5 : SUCCESS in 1h 04m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bf68c317e2464762892735f7cff7fb02 : SUCCESS in 58m 21s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3d50ed9b38eb4bf092735898b5c098dc : SUCCESS in 56m 39s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a5c241fbee0b44ebb81ea191e3739f87 : SUCCESS in 1h 01m 28s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5151712e8ed64da6ad3cbd35f094601b : SUCCESS in 1h 51m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e77da21548ad4383b6a0d62f419ea0ba : SUCCESS in 1h 51m 39s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0d0159a07f774491aa64d8ff00d960e0 : FAILURE in 1h 07m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c3d8333a21a0415291d754b3dcf78c1d : SUCCESS in 1h 48m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e46ff6e0d1d6d417f8578ff704899943cfd392ec","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-05-18 12:18:03.000000000","message":"Patch Set 1: Code-Review+1\n\nHope pep8 won\u0027t fail here","accounts_in_message":[],"_revision_number":1},{"id":"53a083c265a6492f825c687cc262a7d45fa9152a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-05-18 13:00:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1c85077be788a564591b3e6ab45b6f13b8bac052","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 15:04:31.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/386d43ba812645a9890e232c7805031f : SUCCESS in 9m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9226b266d4944f36996d74972c6391d4 : SUCCESS in 10m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4dc29462be04e11861924e704404e10 : FAILURE in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca4a7d4afb0f4443ab9253bde7800b61 : SUCCESS in 5m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1cfcb0556b684c3db5db6c115d0ec17e : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8031414f3dbc4f5eb1c6edf9d5674e7c : SUCCESS in 11m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/691576ee8fd147219b5139842b1f539b : SUCCESS in 6m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e0dfac8912534503819296f88f06f8ca : SUCCESS in 7m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/07db19aac2f74d4c901f80eaa630f96f : SUCCESS in 7m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a922664865014351928c0b22b1c3313d : SUCCESS in 52m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/742658b533ca4bb88d26eba39b1c22d7 : SUCCESS in 1h 35m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cc82e8841e5543e1a5ec4b9e00729392 : SUCCESS in 1h 00m 29s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7b46757f663c4c728d324bb0a75f6d14 : SUCCESS in 57m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6d346ab6dcee4d48a0b0b27c4305c946 : SUCCESS in 1h 06m 54s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/53ea3d2275f04a97932b80f5a1a788b2 : SUCCESS in 1h 50m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c495ae916efe485db19e5378ed39d2f3 : SUCCESS in 1h 57m 56s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c53221066af94452be985f7218dd3f94 : FAILURE in 1h 08m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7493b73851a84614b37fa449a58c0b28 : SUCCESS in 1h 54m 19s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/tests/functional/api/v2/test_listener.py\n  Comments left for invalid file octavia/api/v2/controllers/member.py\n  Comments left for invalid file octavia/tests/functional/db/test_models.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_quotas.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/loadbalancer.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/database_tasks.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_member.py\n  Comments left for invalid file octavia/amphorae/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/network/drivers/neutron/allowed_address_pairs.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_health_monitor.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_l7rule.py\n  Comments left for invalid file octavia/db/models.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_pool.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_amphora.py\n  Comments left for invalid file octavia/api/v2/controllers/load_balancer.py\n  Comments left for invalid file octavia/controller/worker/v2/controller_worker.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/rest_api_driver.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/database_tasks.py\n  Comments left for invalid file octavia/api/v2/controllers/health_monitor.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_l7policy.py\n  Comments left for invalid file octavia/api/v2/controllers/pool.py","accounts_in_message":[],"_revision_number":1},{"id":"137bd23153d7a631720274e5394ba40c57c513d9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-18 19:35:54.000000000","message":"Patch Set 1: Code-Review-1\n\nI squashed this into https://review.opendev.org/#/c/728777/ since we need both to fix the gate.","accounts_in_message":[],"_revision_number":1},{"id":"d2c89449cda19ae19bab11d9b70547e953cf90a8","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-12 19:53:18.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0fb5dd53e024dbc45ea5dc43b01d5d15cf807376","revisions":{"0fb5dd53e024dbc45ea5dc43b01d5d15cf807376":{"kind":"REWORK","_number":1,"created":"2020-04-28 06:36:32.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/10/723810/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/10/723810/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/723810/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/723810/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/723810/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/10/723810/1"}}},"commit":{"parents":[{"commit":"0e5372fb3a2e1a34e598da1991a41739547035f5","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0e5372fb3a2e1a34e598da1991a41739547035f5"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-04-27 16:20:20.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-04-28 06:36:32.000000000","tz":0},"subject":"Remove deprecated exception classes","message":"Remove deprecated exception classes\n\nThese classes were deprecated in Stein and marked for removal in Ussuri.\nBy removing these classes, we fix pep8 issues (catching-non-exception)\nwe started seeing at the gate with the release of astroid 2.4.0.\n\nChange-Id: I66b2d0687f4edfbcbff99c29c9c5d539e4e7ea7f\n(cherry picked from commit 0056b5175f89070164849501ec6d960549b95e34)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0fb5dd53e024dbc45ea5dc43b01d5d15cf807376"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0fb5dd53e024dbc45ea5dc43b01d5d15cf807376"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
