)]}'
{"id":"openstack%2Foctavia~723864","triplet_id":"openstack%2Foctavia~master~Ie29137f2a22467f110725b790debfd2a6448ebc8","project":"openstack/octavia","branch":"master","topic":"starfish","hashtags":[],"change_id":"Ie29137f2a22467f110725b790debfd2a6448ebc8","subject":"Alternative Distributor for L3 Active-Active, N+1 Amphora Setup","status":"NEW","created":"2020-04-28 10:45:04.000000000","updated":"2021-07-29 22:15:15.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":84,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"40789bec2c31092a7191da4677c10cee59893b53","_number":723864,"virtual_id_number":723864,"owner":{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":31814,"name":"XiaoYu Zhu","email":"z503755743@gmail.com","username":"z503755743"},{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-11-18 03:26:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31814,"name":"XiaoYu Zhu","email":"z503755743@gmail.com","username":"z503755743"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":31814,"name":"XiaoYu Zhu","email":"z503755743@gmail.com","username":"z503755743"},{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31814,"name":"XiaoYu Zhu","email":"z503755743@gmail.com","username":"z503755743"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},{"_account_id":31814,"name":"XiaoYu Zhu","email":"z503755743@gmail.com","username":"z503755743"},{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-28 10:45:19.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2020-04-28 10:52:19.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-04-28 10:53:50.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-04-28 10:55:13.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-04-28 12:07:50.000000000","updated_by":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"reviewer":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"state":"REVIEWER"},{"updated":"2020-05-12 07:55:39.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-05-13 02:48:08.000000000","updated_by":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"reviewer":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"state":"REVIEWER"},{"updated":"2020-05-15 03:16:48.000000000","updated_by":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"reviewer":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"state":"REVIEWER"},{"updated":"2020-06-04 22:55:52.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-08-17 12:49:01.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-11-04 17:26:00.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-11-16 02:21:38.000000000","updated_by":{"_account_id":31814,"name":"XiaoYu Zhu","email":"z503755743@gmail.com","username":"z503755743"},"reviewer":{"_account_id":31814,"name":"XiaoYu Zhu","email":"z503755743@gmail.com","username":"z503755743"},"state":"REVIEWER"},{"updated":"2020-11-18 03:26:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-24 09:16:18.000000000","updated_by":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"reviewer":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"state":"REVIEWER"},{"updated":"2021-07-29 22:15:15.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REMOVED"}],"messages":[{"id":"626bf4de22721e2f833efe2183e4c2406fc4cab9","author":{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},"date":"2020-04-28 10:45:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6944bf90cd136d911b402af807aeb9091c990794","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2020-04-28 10:45:19.000000000","message":"Patch Set 1:\n\nCongratulations, you\u0027ve proposed your first change in OpenDev.\n\nYour submission will now be tested automatically by Zuul, our gatekeeper,\nand reviewed by other friendly developers. They will give you feedback and\nmay require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches often take days (and sometimes weeks) to get reviewed, so be\npatient. Don\u0027t hesitate to ask for help, and answer questions about your\nwork promptly if you can. The more you get to know reviewers and get to be\nknown by them, the smoother the review and approval process will become. The\nfastest way to accomplish this is by reviewing other proposed changes\nyourself: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for participating in OpenDev, we look forward to seeing you\naround.\n\nWorkflow Guide: https://docs.openstack.org/infra/manual/developers.html","accounts_in_message":[],"_revision_number":1},{"id":"6ee9fe3b7f011d139e86bb17a127d748799172bb","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-28 16:40:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7259a095da5a17aeba7bde3e4a0558dedb2cc554","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 17:56:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e10dd7f16809471cb32572be48982df4 : FAILURE in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6aeaf4693824963a1cf6b5bc862eeee : SUCCESS in 14m 26s","accounts_in_message":[],"_revision_number":1},{"id":"7a9ee55521465eea3fb8fb75fe35324289f73d83","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-28 18:25:55.000000000","message":"Patch Set 1:\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6880ca709cd62506f07ab00c707e5f510ddb096b","author":{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},"date":"2020-04-29 09:30:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9ce1d47b64b5068737925a8e6105b06c8bea387f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 09:43:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/065607c517a24760a0f9da9be2f74fa0 : FAILURE in 9m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/434b678056fd41fab803cc95bebda31d : SUCCESS in 12m 07s","accounts_in_message":[],"_revision_number":2},{"id":"b0f160afe091bbd862a1230f066ccb80b1095a3b","author":{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},"date":"2020-04-29 10:11:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f4cd032ba99a99fea76595b9dcc57fb0079843cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 10:26:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/693506e5283c46a78ce91bfd0dee2730 : FAILURE in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4d799a827f44114b22f21abacb4f0ee : SUCCESS in 12m 10s","accounts_in_message":[],"_revision_number":3},{"id":"1f432f25af984783719ceef98804aac6add6979b","author":{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},"date":"2020-04-29 10:42:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"364f6fd54b09a861b0222c3b6c6ecfe062c71f4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 10:59:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7195eebdda374b52915f0af69195ff66 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/849e478c5ad541539497a18f1eb8f060 : SUCCESS in 11m 33s","accounts_in_message":[],"_revision_number":4},{"id":"d09d581e2d49cd8abd6b1e191f1032b8aba8d725","author":{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},"date":"2020-05-02 02:15:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"db8f523671af7cd4cc099e56a9dbe9b6ef5d8c61","author":{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},"date":"2020-05-02 02:18:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a07e5fc23dffafb83850a490486a89d2ab3abb63","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-07 20:24:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"700043ffd57ee1e070d56c947c2bf076b9d04c60","author":{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},"date":"2020-05-08 02:52:50.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c63e9dc32f2b4ef53719c282e40ceefcbed67ac5","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2020-05-13 02:48:08.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"cb4ca52666a2777e7eeefa878a2dd95ddf7f71e3","author":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"date":"2020-05-15 03:16:48.000000000","message":"Patch Set 4:\n\nHi JayLiu,\n\nAs I understand, you are using virtual router as a distributor. It means that It can be used only with self-service network mode. \n\nHow about provider network? Should we consider about the provider network model.","accounts_in_message":[],"_revision_number":4},{"id":"43fc67cca160ced774b4150a3f8e9a711bd8450a","author":{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},"date":"2020-05-27 08:48:41.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"85a8cbdf0ec2860d258c1ab34566b1f25a61bad4","author":{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},"date":"2020-05-27 08:56:48.000000000","message":"Patch Set 4:\n\nHi sapd, I\u0027m sorry to reply so late. We want to use qrouter as distributor, and of course we should be able to control the router in the network. I don\u0027t have idea to handle the situation you refered...","accounts_in_message":[],"_revision_number":4},{"id":"f463bcaff5dd2ddb2de3bcc0d40c089ba6f8c7bc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-04 22:15:20.000000000","message":"Patch Set 4:\n\nrecheck\nrendered version has expired","accounts_in_message":[],"_revision_number":4},{"id":"1053a9a242fb20bdea69052d1ac795ad706c6398","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 22:29:34.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2ec075019f543bca5f68daeda7a5331 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c283b52492b4d41ab55a70e4f15e933 : SUCCESS in 13m 10s","accounts_in_message":[],"_revision_number":4},{"id":"cad100c77d2b697912d83a26078049a7f4c009c6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-04 22:55:52.000000000","message":"Patch Set 4:\n\n(16 comments)\n\nOk, this is my first pass review. I think fundamentally this is an ok approach. It is basically taking the functionality we were using OVS VMs for and configuring it directly on neutron. This is nice as it removes one VM hop and was where we wanted to go anyway.","accounts_in_message":[],"_revision_number":4},{"id":"ed479a0960dda8accfb3be5860fa5f9737e2503d","author":{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},"date":"2020-06-10 02:18:51.000000000","message":"Patch Set 4:\n\nJohnsom, sorry to reply so late. First of all, your comments inspire me with more idea, and pointed out some sensitive issues. We take your comments very seriously. Because I want to express our thoughts more explictly, and make it clear with some details, I need some time to discuss with my colleague and find out how to resolve it. If you could give me time to prepare my response, I\u0027d appreciate it. I will reply again and modify this spec before next week.","accounts_in_message":[],"_revision_number":4},{"id":"c94883cae4b1deaf0e913340294303477139372d","author":{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},"date":"2020-06-12 03:16:50.000000000","message":"Patch Set 4:\n\n(13 comments)\n\nSorry I spent a lot of time confirming the details and looking up a lot of relevant information. Many of your comments are very helpful. I will be making major changes to the spec later.","accounts_in_message":[],"_revision_number":4},{"id":"34b19ec689e98c5ad889a8e6241ccdad0fc7be8d","author":{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},"date":"2020-07-15 10:11:11.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2de623d0528076288fcda66ac430dca27593f6ea","author":{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},"date":"2020-07-15 12:03:53.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"bb82716c13765768160cb443ce32b7db3f26b618","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 12:22:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c228f4670aee4a4397708157bd460fc0 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14aaa63a29334e4a835efbb3b04b6c00 : SUCCESS in 13m 51s","accounts_in_message":[],"_revision_number":5},{"id":"7b81b595ccdadd74c4f6c2f28b2faa8a7bde2056","author":{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},"date":"2020-07-29 07:03:43.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e045dd414ecda61695e0b18ba05bd1f0a6b41e07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 07:17:49.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3657bc0c72c8445a93012bae0ed72e64 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9302a3a3a47149749430cd7253a3d1a0 : SUCCESS in 11m 55s","accounts_in_message":[],"_revision_number":6},{"id":"4f04223498a3a958887a75b035e427e8b2c187be","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-17 12:49:01.000000000","message":"Patch Set 6: Code-Review-1\n\n(6 comments)\n\nThank you for sharing another alternative for L3 active-active load balancing!","accounts_in_message":[],"_revision_number":6},{"id":"f632b51799d9f42131c5442da25f690cd94fe87f","author":{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},"date":"2020-08-19 11:57:31.000000000","message":"Patch Set 6:\n\n(4 comments)\n\nTo Carlos:\nThanks for your reivew!","accounts_in_message":[],"_revision_number":6},{"id":"436a159df79f675c3f98bc2cbc4403dbe3b2b4be","author":{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},"date":"2020-10-28 11:46:04.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"d31342ccf927fd60a58c31a7a27b9b1097d67cd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 12:03:47.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/befe37e58589427da4d4737ab6cdafeb : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21cf2e1016ae47fab02c8cdecb1ebf17 : SUCCESS in 13m 03s","accounts_in_message":[],"_revision_number":7},{"id":"9feda19755a26f5d751aa27d1bef79791237f549","author":{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},"date":"2020-11-04 08:38:03.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"e574721b9adb680bf8095b7b68880e1cd208625d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 09:06:05.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bad670c4bb24a04b853c45e2609165c : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c61a594dc4d4762b0580eb06c32de58 : SUCCESS in 20m 04s","accounts_in_message":[],"_revision_number":8},{"id":"826435d8b713b01f931074929b8d4bd13c9a1913","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-04 17:26:00.000000000","message":"Patch Set 8:\n\n(28 comments)","accounts_in_message":[],"_revision_number":8},{"id":"0a208008d883090fec2be18f6cef1cebe58fbb3c","author":{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},"date":"2020-11-13 09:08:32.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"fa346f395cf19857a95abc3d4dbf8edba224556b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 09:29:20.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b44db54590bd4b1e9016dcc016721334 : FAILURE in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d569396f1ac4ebe8d6f98d1c3572b4f : SUCCESS in 16m 53s\n\nWarning:\n  Comments left for invalid file octavia/controller/worker/v1/controller_worker.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/database_tasks.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/amphora_driver_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/controller_worker.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/rest_api_driver.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/server.py\n  Comments left for invalid file octavia/network/drivers/neutron/allowed_address_pairs.py\n  Comments left for invalid file octavia/cmd/house_keeping.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_availability_zones.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_provider.py","accounts_in_message":[],"_revision_number":9},{"id":"9a53bc8983532cd975f8ceeea35f9bb3b139c58d","author":{"_account_id":31814,"name":"XiaoYu Zhu","email":"z503755743@gmail.com","username":"z503755743"},"date":"2020-11-16 02:21:38.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"ec2300f590b0728e963014bd7d489ce2a5a5a145","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 02:37:45.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9aef5197480047c6b8862fa4092ad7b8 : FAILURE in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c5446c6c3864c649eb31ceac8c98b06 : SUCCESS in 15m 11s\n\nWarning:\n  Comments left for invalid file octavia/cmd/house_keeping.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/server.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/rest_api_driver.py\n  Comments left for invalid file octavia/controller/worker/v1/controller_worker.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/amphora_driver_tasks.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/database_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/controller_worker.py\n  Comments left for invalid file octavia/network/drivers/neutron/allowed_address_pairs.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_provider.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_availability_zones.py","accounts_in_message":[],"_revision_number":9},{"id":"b0d173a771b9f605c2dbcbace04606afd121c4a6","author":{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},"date":"2020-11-17 07:34:28.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"8cc08074ac9c249c64ae1374c865ce7bb16ff3e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 07:55:39.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdb514b15e4e453b876f2fa15947f30a : FAILURE in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae343b84b09b4cdfa85120ad14dfd250 : SUCCESS in 13m 18s","accounts_in_message":[],"_revision_number":10},{"id":"a238b2b8d17098e2aaca0f4ab291280388740b45","author":{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},"date":"2020-11-18 02:50:18.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"6e801a3430754eb7370f9f98c1d0e3596ad59741","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 03:26:13.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4922b3f89d6a44439a352470a2e0bf76 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36f7a1b2ee284dfcb6c538d5c8f38565 : SUCCESS in 29m 54s","accounts_in_message":[],"_revision_number":11},{"id":"40789bec2c31092a7191da4677c10cee59893b53","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-29 22:15:15.000000000","message":"Removed reviewer Brian Haley.","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"e68b1594b3c88835b18e085db36f43cbfe28b368","revisions":{"bcd8f7d800bc046bb70e0cf94df4d03f9cd52dab":{"kind":"REWORK","_number":1,"created":"2020-04-28 10:45:04.000000000","uploader":{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},"ref":"refs/changes/64/723864/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/723864/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/723864/1"}}},"commit":{"parents":[{"commit":"0056b5175f89070164849501ec6d960549b95e34","subject":"Remove deprecated exception classes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0056b5175f89070164849501ec6d960549b95e34"}]}],"author":{"name":"JayLiu","email":"1299870737@qq.com","date":"2020-04-28 10:12:36.000000000","tz":480},"committer":{"name":"JayLiu","email":"1299870737@qq.com","date":"2020-04-28 10:12:36.000000000","tz":480},"subject":"Altinative Distributor for L3 Active-Active, N+1 Amphora Setup","message":"Altinative Distributor for L3 Active-Active,\nN+1 Amphora Setup\n\nOctavia need to support active-active feature. The community has\npreviously had a scheme for implementing active-active feature using L3\ndistributors. This solution uses the BGP/ECMP approach. We would like\nto provide an alternative solution that utilizes the L3 network to\nimplement this feature.\nThis solution is architecturally simple, however, it requires Neutron\nsupport.\n\nChange-Id: Ie29137f2a22467f110725b790debfd2a6448ebc8\nStory: 2007602\nTask: 39585\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bcd8f7d800bc046bb70e0cf94df4d03f9cd52dab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bcd8f7d800bc046bb70e0cf94df4d03f9cd52dab"}]},"branch":"refs/heads/master"},"04a4dd2a8c12cfde06c3a9e5afd647eaf4db5bc4":{"kind":"REWORK","_number":2,"created":"2020-04-29 09:30:06.000000000","uploader":{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},"ref":"refs/changes/64/723864/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/723864/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/723864/2"}}},"commit":{"parents":[{"commit":"0056b5175f89070164849501ec6d960549b95e34","subject":"Remove deprecated exception classes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0056b5175f89070164849501ec6d960549b95e34"}]}],"author":{"name":"JayLiu","email":"1299870737@qq.com","date":"2020-04-28 10:12:36.000000000","tz":480},"committer":{"name":"JayLiu","email":"1299870737@qq.com","date":"2020-04-29 09:27:09.000000000","tz":480},"subject":"Alternative Distributor for L3 Active-Active, N+1 Amphora Setup","message":"Alternative Distributor for L3 Active-Active, N+1 Amphora Setup\n\nOctavia need to support active-active feature. The community has\npreviously had a scheme for implementing active-active feature using L3\ndistributors. This solution uses the BGP/ECMP approach. We would like\nto provide an alternative solution that utilizes the L3 network to\nimplement this feature.\nThis solution is architecturally simple, however, it requires Neutron\nsupport.\n\nChange-Id: Ie29137f2a22467f110725b790debfd2a6448ebc8\nStory: 2007602\nTask: 39585\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/04a4dd2a8c12cfde06c3a9e5afd647eaf4db5bc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/04a4dd2a8c12cfde06c3a9e5afd647eaf4db5bc4"}]},"branch":"refs/heads/master"},"97cf308ae36ffae372047cee396139b4eeb2f94e":{"kind":"REWORK","_number":3,"created":"2020-04-29 10:11:08.000000000","uploader":{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},"ref":"refs/changes/64/723864/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/723864/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/723864/3"}}},"commit":{"parents":[{"commit":"0056b5175f89070164849501ec6d960549b95e34","subject":"Remove deprecated exception classes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0056b5175f89070164849501ec6d960549b95e34"}]}],"author":{"name":"JayLiu","email":"1299870737@qq.com","date":"2020-04-28 10:12:36.000000000","tz":480},"committer":{"name":"JayLiu","email":"1299870737@qq.com","date":"2020-04-29 09:58:26.000000000","tz":480},"subject":"Alternative Distributor for L3 Active-Active, N+1 Amphora Setup","message":"Alternative Distributor for L3 Active-Active, N+1 Amphora Setup\n\nOctavia need to support active-active feature. The community has\npreviously had a scheme for implementing active-active feature using L3\ndistributors. This solution uses the BGP/ECMP approach. We would like\nto provide an alternative solution that utilizes the L3 network to\nimplement this feature.\nThis solution is architecturally simple, however, it requires Neutron\nsupport.\n\nChange-Id: Ie29137f2a22467f110725b790debfd2a6448ebc8\nStory: 2007602\nTask: 39585\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/97cf308ae36ffae372047cee396139b4eeb2f94e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/97cf308ae36ffae372047cee396139b4eeb2f94e"}]},"branch":"refs/heads/master"},"1203689124980ec127826165b30e3a47a4f46dbb":{"kind":"REWORK","_number":4,"created":"2020-04-29 10:42:31.000000000","uploader":{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},"ref":"refs/changes/64/723864/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/723864/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/723864/4"}}},"commit":{"parents":[{"commit":"0056b5175f89070164849501ec6d960549b95e34","subject":"Remove deprecated exception classes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0056b5175f89070164849501ec6d960549b95e34"}]}],"author":{"name":"JayLiu","email":"1299870737@qq.com","date":"2020-04-28 10:12:36.000000000","tz":480},"committer":{"name":"JayLiu","email":"1299870737@qq.com","date":"2020-04-29 10:37:51.000000000","tz":480},"subject":"Alternative Distributor for L3 Active-Active, N+1 Amphora Setup","message":"Alternative Distributor for L3 Active-Active, N+1 Amphora Setup\n\nOctavia need to support active-active feature. The community has\npreviously had a scheme for implementing active-active feature using L3\ndistributors. This solution uses the BGP/ECMP approach. We would like\nto provide an alternative solution that utilizes the L3 network to\nimplement this feature.\nThis solution is architecturally simple, however, it requires Neutron\nsupport.\n\nChange-Id: Ie29137f2a22467f110725b790debfd2a6448ebc8\nStory: 2007602\nTask: 39585\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1203689124980ec127826165b30e3a47a4f46dbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1203689124980ec127826165b30e3a47a4f46dbb"}]},"branch":"refs/heads/master"},"01c17877783594e49494b41e4d36700bc84e9286":{"kind":"REWORK","_number":5,"created":"2020-07-15 12:03:53.000000000","uploader":{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},"ref":"refs/changes/64/723864/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/723864/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/723864/5"}}},"commit":{"parents":[{"commit":"8ae82a47e1928f99dd32f08785f876b3d1d9c3cb","subject":"Merge \"Stop to use the __future__ module.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ae82a47e1928f99dd32f08785f876b3d1d9c3cb"}]}],"author":{"name":"Jay Liu","email":"1299870737@qq.com","date":"2020-07-10 13:41:37.000000000","tz":0},"committer":{"name":"Jay Liu","email":"1299870737@qq.com","date":"2020-07-15 12:02:58.000000000","tz":-420},"subject":"Alternative Distributor for L3 Active-Active, N+1 Amphora Setup","message":"Alternative Distributor for L3 Active-Active, N+1 Amphora Setup\n\nOctavia need to support active-active feature. The community has\npreviously had a scheme for implementing active-active feature using L3\ndistributors. This solution uses the BGP/ECMP approach. We would like\nto provide an alternative solution that utilizes the L3 network to\nimplement this feature.\nThis solution is architecturally simple, however, it requires Neutron\nsupport.\n\nChange-Id: Ie29137f2a22467f110725b790debfd2a6448ebc8\nStory: 2007602\nTask: 39585\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/01c17877783594e49494b41e4d36700bc84e9286"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/01c17877783594e49494b41e4d36700bc84e9286"}]},"branch":"refs/heads/master"},"1a2c9c289ed2cdec0a5884bb5fde36731bd07134":{"kind":"REWORK","_number":6,"created":"2020-07-29 07:03:43.000000000","uploader":{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},"ref":"refs/changes/64/723864/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/723864/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/723864/6"}}},"commit":{"parents":[{"commit":"c78a1b2c299bc5a7bde58e98262ece6e4700b6e8","subject":"Merge \"Add quota support to octavia\u0027s l7policy and l7rule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c78a1b2c299bc5a7bde58e98262ece6e4700b6e8"}]}],"author":{"name":"Jay Liu","email":"1299870737@qq.com","date":"2020-07-10 13:41:37.000000000","tz":0},"committer":{"name":"Jay Liu","email":"1299870737@qq.com","date":"2020-07-29 07:02:22.000000000","tz":-420},"subject":"Alternative Distributor for L3 Active-Active, N+1 Amphora Setup","message":"Alternative Distributor for L3 Active-Active, N+1 Amphora Setup\n\nOctavia need to support active-active feature. The community has\npreviously had a scheme for implementing active-active feature using L3\ndistributors. This solution uses the BGP/ECMP approach. We would like\nto provide an alternative solution that utilizes the L3 network to\nimplement this feature.\nThis solution is architecturally simple, however, it requires Neutron\nsupport.\n\nChange-Id: Ie29137f2a22467f110725b790debfd2a6448ebc8\nStory: 2007602\nTask: 39585\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a2c9c289ed2cdec0a5884bb5fde36731bd07134"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a2c9c289ed2cdec0a5884bb5fde36731bd07134"}]},"branch":"refs/heads/master"},"1c81cf49576487b3b276bf5f8baa89574ed227f6":{"kind":"REWORK","_number":7,"created":"2020-10-28 11:46:04.000000000","uploader":{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},"ref":"refs/changes/64/723864/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/723864/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/723864/7"}}},"commit":{"parents":[{"commit":"71440bdbdb6e66f154fe69eb43ce116d131506a6","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/71440bdbdb6e66f154fe69eb43ce116d131506a6"}]}],"author":{"name":"Jay Liu","email":"1299870737@qq.com","date":"2020-07-10 13:41:37.000000000","tz":0},"committer":{"name":"Jay Liu","email":"1299870737@qq.com","date":"2020-10-28 11:43:43.000000000","tz":-420},"subject":"Alternative Distributor for L3 Active-Active, N+1 Amphora Setup","message":"Alternative Distributor for L3 Active-Active, N+1 Amphora Setup\n\nOctavia need to support active-active feature. The community has\npreviously had a scheme for implementing active-active feature using L3\ndistributors. This solution uses the BGP/ECMP approach. We would like\nto provide an alternative solution that utilizes the L3 network to\nimplement this feature.\nThis solution is architecturally simple, however, it requires Neutron\nsupport.\n\nChange-Id: Ie29137f2a22467f110725b790debfd2a6448ebc8\nStory: 2007602\nTask: 39585\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1c81cf49576487b3b276bf5f8baa89574ed227f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1c81cf49576487b3b276bf5f8baa89574ed227f6"}]},"branch":"refs/heads/master"},"6fae2f9e8f25435436b821fdcc957a19df1c7cb1":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":8,"created":"2020-11-04 08:38:03.000000000","uploader":{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},"ref":"refs/changes/64/723864/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/723864/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/723864/8"}}},"commit":{"parents":[{"commit":"198980639cbe02820f90783fa04d5239717ed80d","subject":"Merge \"Use run_flow for failover functions in v2 control worker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/198980639cbe02820f90783fa04d5239717ed80d"}]}],"author":{"name":"Jay Liu","email":"1299870737@qq.com","date":"2020-07-10 13:41:37.000000000","tz":0},"committer":{"name":"Jay Liu","email":"1299870737@qq.com","date":"2020-11-04 08:33:18.000000000","tz":-480},"subject":"Alternative Distributor for L3 Active-Active, N+1 Amphora Setup","message":"Alternative Distributor for L3 Active-Active, N+1 Amphora Setup\n\nOctavia need to support active-active feature. The community has\npreviously had a scheme for implementing active-active feature using L3\ndistributors. This solution uses the BGP/ECMP approach. We would like\nto provide an alternative solution that utilizes the L3 network to\nimplement this feature.\nThis solution is architecturally simple, however, it requires Neutron\nsupport.\n\nPatch for this spec: https://review.opendev.org/#/c/746688/\n\nChange-Id: Ie29137f2a22467f110725b790debfd2a6448ebc8\nStory: 2007602\nTask: 39585\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6fae2f9e8f25435436b821fdcc957a19df1c7cb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6fae2f9e8f25435436b821fdcc957a19df1c7cb1"}]},"branch":"refs/heads/master"},"a3779eb7ea7a77487889076a17e6884af66e4d58":{"kind":"REWORK","_number":9,"created":"2020-11-13 09:08:32.000000000","uploader":{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},"ref":"refs/changes/64/723864/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/723864/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/723864/9"}}},"commit":{"parents":[{"commit":"198980639cbe02820f90783fa04d5239717ed80d","subject":"Merge \"Use run_flow for failover functions in v2 control worker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/198980639cbe02820f90783fa04d5239717ed80d"}]}],"author":{"name":"Jay Liu","email":"1299870737@qq.com","date":"2020-07-10 13:41:37.000000000","tz":0},"committer":{"name":"Jay Liu","email":"1299870737@qq.com","date":"2020-11-13 09:02:37.000000000","tz":-480},"subject":"Alternative Distributor for L3 Active-Active, N+1 Amphora Setup","message":"Alternative Distributor for L3 Active-Active, N+1 Amphora Setup\n\nOctavia need to support active-active feature. The community has\npreviously had a scheme for implementing active-active feature using L3\ndistributors. This solution uses the BGP/ECMP approach. We would like\nto provide an alternative solution that utilizes the L3 network to\nimplement this feature.\nThis solution is architecturally simple, however, it requires Neutron\nsupport.\n\nImplementation: https://review.opendev.org/#/c/746688/\n\nChange-Id: Ie29137f2a22467f110725b790debfd2a6448ebc8\nStory: 2007602\nTask: 39585\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a3779eb7ea7a77487889076a17e6884af66e4d58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a3779eb7ea7a77487889076a17e6884af66e4d58"}]},"branch":"refs/heads/master"},"d4d9d1d4ed423dd5944fcfc855e710eff780a6d3":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2020-11-17 07:34:28.000000000","uploader":{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},"ref":"refs/changes/64/723864/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/723864/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/723864/10"}}},"commit":{"parents":[{"commit":"f96b0986cda925d471073620047346c70b73927a","subject":"Merge \"Replace assertItemsEqual with assertCountEqual\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f96b0986cda925d471073620047346c70b73927a"}]}],"author":{"name":"Jay Liu","email":"1299870737@qq.com","date":"2020-07-10 13:41:37.000000000","tz":0},"committer":{"name":"Jay Liu","email":"1299870737@qq.com","date":"2020-11-17 07:23:49.000000000","tz":-480},"subject":"Alternative Distributor for L3 Active-Active, N+1 Amphora Setup","message":"Alternative Distributor for L3 Active-Active, N+1 Amphora Setup\n\nOctavia need to support active-active feature. The community has\npreviously had a scheme for implementing active-active feature using L3\ndistributors. This solution uses the BGP/ECMP approach. We would like\nto provide an alternative solution that utilizes the L3 network to\nimplement this feature.\nThis solution is architecturally simple, however, it requires Neutron\nsupport.\n\nImplementation: https://review.opendev.org/#/c/746688/\n\nChange-Id: Ie29137f2a22467f110725b790debfd2a6448ebc8\nStory: 2007602\nTask: 39585\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d4d9d1d4ed423dd5944fcfc855e710eff780a6d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d4d9d1d4ed423dd5944fcfc855e710eff780a6d3"}]},"branch":"refs/heads/master"},"e68b1594b3c88835b18e085db36f43cbfe28b368":{"kind":"REWORK","_number":11,"created":"2020-11-18 02:50:18.000000000","uploader":{"_account_id":31843,"name":"Jay Liu","email":"1299870737@qq.com","username":"JayLiu"},"ref":"refs/changes/64/723864/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/723864/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/723864/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/723864/11"}}},"commit":{"parents":[{"commit":"f96b0986cda925d471073620047346c70b73927a","subject":"Merge \"Replace assertItemsEqual with assertCountEqual\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f96b0986cda925d471073620047346c70b73927a"}]}],"author":{"name":"Jay Liu","email":"1299870737@qq.com","date":"2020-07-10 13:41:37.000000000","tz":0},"committer":{"name":"Jay Liu","email":"1299870737@qq.com","date":"2020-11-18 02:47:47.000000000","tz":-480},"subject":"Alternative Distributor for L3 Active-Active, N+1 Amphora Setup","message":"Alternative Distributor for L3 Active-Active, N+1 Amphora Setup\n\nOctavia need to support active-active feature. The community has\npreviously had a scheme for implementing active-active feature using L3\ndistributors. This solution uses the BGP/ECMP approach. We would like\nto provide an alternative solution that utilizes the L3 network to\nimplement this feature.\nThis solution is architecturally simple, however, it requires Neutron\nsupport.\n\nImplementation: https://review.opendev.org/#/c/746688/\n\nChange-Id: Ie29137f2a22467f110725b790debfd2a6448ebc8\nStory: 2007602\nTask: 39585\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e68b1594b3c88835b18e085db36f43cbfe28b368"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e68b1594b3c88835b18e085db36f43cbfe28b368"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
