)]}'
{"id":"openstack%2Foctavia~724149","triplet_id":"openstack%2Foctavia~stable%2Fstein~Ic43797e01b1cd74e9cba1fb4af26015bb301ac2b","project":"openstack/octavia","branch":"stable/stein","hashtags":[],"change_id":"Ic43797e01b1cd74e9cba1fb4af26015bb301ac2b","subject":"Remove xenial based jobs from stein gates.","status":"MERGED","created":"2020-04-28 20:19:06.000000000","updated":"2020-04-29 20:11:58.000000000","submitted":"2020-04-29 20:07:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724149-1588190842275-276f548f","meta_rev_id":"7b974ad6bb35d9991ba4dee0e971eab72d5e8653","_number":724149,"virtual_id_number":724149,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-29 20:07:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-04-29 06:39:31.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-04-29 09:44:35.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-04-29 09:44:35.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-29 06:39:31.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-04-29 09:44:35.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2020-04-29 20:07:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fbd2b63ddd8d04a74250f612408e6dd5cefcf605","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-28 20:19:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7df96e2e2ba6a30be120dd78bdcfbbf53b1bb774","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 22:41:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/025b864208a04d8594dc45883463fba6 : SUCCESS in 11m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5fb80177ddd64cafa5e01dec8d2ca613 : SUCCESS in 11m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10ce8f36d3c6465e9959f29863a600fc : SUCCESS in 8m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/08195680690a4cda85cf9aa85d245f56 : SUCCESS in 6m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/679c231c38b047809aaa03bbc6c6d1ab : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8a558b04e2a42318bda465d1e3d311b : SUCCESS in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a3c34dd4fd9483f936b7ac219149ffb : SUCCESS in 9m 05s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/4eebe4f13fae45ea9fa1ad24498952f4 : SUCCESS in 1h 24m 12s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/b5e05d874d694f72afa4a111b5787701 : SUCCESS in 1h 23m 29s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0b2d2454cd704c39a117ac0c87a0e0b9 : SUCCESS in 1h 19m 38s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/d6863043a8f9453881c1128009811c5c : SUCCESS in 1h 24m 50s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ccb353cfda5f4b0a8791bc34ddfff50c : SUCCESS in 11m 52s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/752d63fc36b04bb9b54e943e96efe095 : SUCCESS in 11m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a155201fa17a4398bf910c9f5cbbbb9c : SUCCESS in 9m 07s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/812a2c67d6d64472b06798fe0048bb65 : SUCCESS in 56m 02s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/766857260d8a4c9fa9f74752f6d1e23f : SUCCESS in 1h 07m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e934a0da1f6947beaed1a0e3663b8a73 : SUCCESS in 46m 09s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c2ddd071fae64ee4bfeb877cbd97900b : SUCCESS in 54m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b5b1206227e7496d93c5cce8344adeb8 : SUCCESS in 1h 41m 51s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/9a932750dce9402eba9b9d92a7719f62 : SUCCESS in 1h 42m 37s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/10eab700511546c8a3d996d3f32a8a20 : FAILURE in 21m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4dafbfe225a74e06898b5f4368667fc4 : FAILURE in 34m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f0617d3edb8d4ad5b93162fa394d27b0 : SUCCESS in 1h 57m 41s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/9717ce55f91f49d581d18b329ac89680 : SUCCESS in 2h 14m 52s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0b44d4547cfe4c59b47013f8415ee95e : SUCCESS in 1h 19m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/11bfe8041525415599a2b416e227fd76 : FAILURE in 1h 12m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e56b2ac0974b5dc25c7dfc7bab21ff17c9ea5052","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-29 06:39:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c539a7a7e087f8b836fe9b1dc6df34b6b2ba9dc4","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2020-04-29 09:44:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bc0be3eff44ab525bb710ffab0b09a684e1b72c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 09:44:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4088366087dc5ab9c32334451af2ae18f773f0c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 11:12:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bcb74b035ffb49ae8227fe4fb51ada60 : SUCCESS in 9m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a55e63df6d94155a4efd2023ac8a775 : SUCCESS in 7m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/06503590429343118b846b54836c2573 : SUCCESS in 5m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ccaf75efb02f4d57b9b7287c17031988 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c65861668d1465ba197a8fefbb2bed5 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce5674f8c1bb4c49b006bf61d35da96b : SUCCESS in 8m 18s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/dbd438ae8dd745b695f224cdde61388b : SUCCESS in 10m 14s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/89809968c46a42089e18cacd55bac294 : SUCCESS in 9m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f82eadc0f5364f8b8ba0096715b5014e : SUCCESS in 6m 30s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/293d98331f214f69977f304d3928e100 : SUCCESS in 1h 01m 37s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/73495ed9b1f2439b97db9fd9fbe7e5d8 : SUCCESS in 59m 00s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/d305dda9a07e498cbfa59a1db43b037e : SUCCESS in 1h 15m 34s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/81d54abfa3a14c11a116350748f17153 : SUCCESS in 1h 14m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/868f9cdd6c66444e89661c97d963fa45 : SUCCESS in 44m 02s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/fe0067becf0e4ee9beffbcd6159e19f4 : SUCCESS in 45m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2a24366cc0d24f7a8bd8c6fe2c819dbb : FAILURE in 1h 03m 07s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/9d6fe66a4f514c72b6d5607b364494ec : SUCCESS in 1h 00m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7c98c086ebd14e80b2971aeb4e241b2e : SUCCESS in 1h 20m 13s","accounts_in_message":[],"_revision_number":1},{"id":"41094e84049b6f8341ecb6d495d441ff1b4861dc","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-29 11:20:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"aa828f5a38cb795b994c480459ee04d50c887e47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 13:37:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06f920f69b3342c7a34682d43cb7f647 : SUCCESS in 11m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4fa9f9c6ea44967a3a1d360d3f202e6 : SUCCESS in 10m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa8a7b42d2e543c78e9852ab2f1dbebd : SUCCESS in 7m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/61a76f14860f4fb3ae5708bce3481227 : SUCCESS in 6m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/72d464cc03b442729b11ac91bc506ef5 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16e2f1d98b08411e8567185a4e697e83 : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2463fd1951849a0a54b6352dc767c6e : SUCCESS in 8m 17s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/4028788342064ad18918955880aef88b : SUCCESS in 1h 44m 10s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/9a715037b9ae4dd2bf38f385b06189a4 : SUCCESS in 1h 21m 57s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6664f5adba184f35a09d8d915c263c66 : SUCCESS in 1h 42m 05s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/3f33b21f748e4a0a9949855e0bc0da42 : SUCCESS in 1h 28m 57s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/017f134514d542abb740d5c2fd23d735 : SUCCESS in 7m 45s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/d6b88ef9ebb54942a319c7beb04df6be : SUCCESS in 7m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8a7f396eed9a4a34ab174c8eaf41e4f4 : SUCCESS in 11m 04s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/257c6570e2f24bc3a17616f7d785ffce : SUCCESS in 1h 02m 56s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/9528d8177e08478fb02c5b49c7912dca : SUCCESS in 1h 00m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8e8c5b3d3c794bbcabf8c1a3c8a378d2 : SUCCESS in 46m 41s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/e1a06299e54f49d89ef345defe9c80fe : SUCCESS in 44m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c7af35822a7d4157b13e13ad5ec0441e : SUCCESS in 1h 39m 13s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/dec9dc738b76409db296f8bac3c64328 : SUCCESS in 1h 55m 59s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/1244effaf7fa4e78a0e91586506ea61d : FAILURE in 20m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ca36e0893fd4013950923edc11f6e1c : FAILURE in 36m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0192ed2641d54afdb1a455878b5c568e : FAILURE in 2h 12m 17s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/c20f394990d247e3966b2c87198d146c : SUCCESS in 1h 55m 51s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/92d505c296d84975bb72ca4010f21d77 : SUCCESS in 1h 19m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8a099e0784544a3cadac491b231d93e6 : FAILURE in 59m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ab3bbc8b7fdb2cbcd3b812b99d9aa3bd125cda06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 13:37:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6c6c0ef37dcc28054555c2e9edc19fc25fa3fb83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 14:16:11.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f397c5f1f50147f5bccac246c5a59075 : SUCCESS in 8m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3a7f5d533e34b57b10cf11bb44a3cee : SUCCESS in 6m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/08f572418ea148128b7bedcb864d8496 : SUCCESS in 5m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4be4dd4bbbe2409e9f2bc0c2db609a48 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ff615d47edf47ceaf5edccdb9122199 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4b94d83e03d40d480e9b274b0b23395 : SUCCESS in 6m 28s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/627f3335aa794fc1863eb515c4db4dbb : SUCCESS in 9m 23s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/ce1895deb90a4644a9346c8c0c80a9ac : SUCCESS in 7m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5d3bb22eff6d40048682e7b93124780a : SUCCESS in 7m 58s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/43df3d8d34814100b74c61ba6885d09f : EXCEPTION\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/4cb9a9f68e0e434ebbbcf11b13fd3e48 : EXCEPTION\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/d1d61e959ee0433dbc78636ab3f0504a : EXCEPTION\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/f697b86b74b341b5b8f99065f0006423 : EXCEPTION\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5b9b4873791442eaa6b54eb1b684202d : EXCEPTION\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/433e5702ad1d41fb97a5c564eb8ab3a3 : EXCEPTION\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6111fe083fac497d9845e713eddeed4e : EXCEPTION\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7b1e13c66f2d411990940aad9648be08 : EXCEPTION\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c2982b78eec54734ae4aaf5d54cd1e11 : EXCEPTION","accounts_in_message":[],"_revision_number":1},{"id":"a83eee070e60565d4238a63e1eadedb38bb41d96","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-29 16:08:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"609fca4df50da0bcb40e53c94cf5fa2af2d0c669","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 18:25:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba9901d924af4b3186a6b2b6931293a6 : SUCCESS in 11m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c502990ee564c29a8d49b825bbacb23 : SUCCESS in 12m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdc4535983dd423ba2a8f155ff257f22 : SUCCESS in 9m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/76ec738ae4d94b04a1ea92047a0e0b25 : SUCCESS in 6m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7eaaecb37aaf4d6db9e4592362cefcfe : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e12e7218a694873b376bc1750b292bc : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1262c29965c34a0a957b03cbb8d42e07 : SUCCESS in 9m 27s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/2e15e6efac314d199d50858c3821660b : SUCCESS in 1h 16m 03s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/72bc1835db844d5baca0389b6307c369 : SUCCESS in 1h 17m 29s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/15890b3cf7184864bda967398b904dab : SUCCESS in 1h 35m 39s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/e2bfc0ec063b4942bdacc1275b39772d : FAILURE in 1h 27m 14s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3f3dbeb1dd83404ea02cf450597d70f3 : SUCCESS in 9m 46s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/8b158a124a3b4faa92646c6b83befa3c : SUCCESS in 9m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/137bc7172a6f4036b0a67a5cc6c0b878 : SUCCESS in 9m 29s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a671a0618b6b49b78d9fc78cd1674e3d : SUCCESS in 1h 04m 05s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/5ea51b42017a421fa9ab407585a086bf : SUCCESS in 1h 02m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/72c1836187774bfab42ee5fcd7ab5c62 : SUCCESS in 43m 35s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/8cb9cb972e3a4084be631bcf4ee65555 : SUCCESS in 41m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b351349096104c4c8548d60d9f4d96f7 : SUCCESS in 1h 32m 07s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/64d44444b0554a4a909642749f2586f1 : SUCCESS in 1h 37m 51s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/74a0e852d121427ea05c94f22eef345a : FAILURE in 36m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/31cf6aa38d60479f8c07c82ee3659cc8 : FAILURE in 37m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/782cab3c315247d7bb5f9afcd41c45da : SUCCESS in 1h 42m 05s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/fa75f20fe59a4946b7e22c823fcfcd6b : SUCCESS in 2h 01m 12s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4c1a8bd4fab24d72babb478345f114fd : SUCCESS in 1h 23m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/32bd62ca41da48809e8ea0bebfd6bf20 : FAILURE in 56m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f370a25b294e2684d80be4a370064c1ace16c025","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 18:25:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"004758b8975ad492a9665541b482e5a33249337e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 20:07:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"eaf35da7b1711b2340cac22010de28de43177d7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 20:07:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0d7b289c9274b8da3bb28c2e7630c39 : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a7494ff170d449eb06773ebfb0111cf : SUCCESS in 7m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f2ee6c57da4d4f0aa7ae10f9468d296d : SUCCESS in 7m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e5087f20de914f03abaea6f079969db1 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5686c82049348e3b2b23699b850a979 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d80f2b77ef954f2a820d914b24023d9a : SUCCESS in 7m 39s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2e4ec27c4f914d50aede2f6c8e2e698b : SUCCESS in 7m 05s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/e97a4476bff9440ba6bcaf383dd4d3b1 : SUCCESS in 10m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6a679937095b4384b131731ee0be01db : SUCCESS in 7m 23s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1784b21cabb7473a8e88787c6ed9606e : SUCCESS in 1h 09m 13s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/05aac183f16f4435a68822ff46b690da : SUCCESS in 1h 04m 09s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/c9e0ee853c6f427684636b1551be56c6 : SUCCESS in 1h 14m 02s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/795d45ca385342a3b1f0b425ab48eb5f : SUCCESS in 1h 21m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/628e030e3f454c4790243f00f64cfcde : SUCCESS in 48m 52s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/194eea3263f14539817dee294bab4dfe : SUCCESS in 45m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/11297b40d9ce4a63affe2ef79e94bd73 : SUCCESS in 1h 39m 57s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/51ea4798150045679e55c67a44e6f5a6 : SUCCESS in 1h 37m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/18762ade72a848e080868abc380091b3 : SUCCESS in 1h 08m 17s","accounts_in_message":[],"_revision_number":1},{"id":"7b974ad6bb35d9991ba4dee0e971eab72d5e8653","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 20:11:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3efc53c2dc3417c8b502288bf2a59ef : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"922c4987fb4db019212c7db322a65c3998f0c1a9","revisions":{"922c4987fb4db019212c7db322a65c3998f0c1a9":{"kind":"REWORK","_number":1,"created":"2020-04-28 20:19:06.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/49/724149/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/49/724149/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/724149/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/724149/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/724149/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/49/724149/1"}}},"commit":{"parents":[{"commit":"e3a95c9b0f1f59cea52097f03bbd8ab1a20e89c6","subject":"Remove the barbican \"Grant access\" from cookbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3a95c9b0f1f59cea52097f03bbd8ab1a20e89c6"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-04-28 20:18:00.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-04-28 20:19:05.000000000","tz":-420},"subject":"Remove xenial based jobs from stein gates.","message":"Remove xenial based jobs from stein gates.\n\nXenial was not supported (long past) on Stein so we should not be\nrunning Stein jobs on xenial.\n\nChange-Id: Ic43797e01b1cd74e9cba1fb4af26015bb301ac2b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/922c4987fb4db019212c7db322a65c3998f0c1a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/922c4987fb4db019212c7db322a65c3998f0c1a9"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
