)]}'
{"id":"openstack%2Foctavia~724890","triplet_id":"openstack%2Foctavia~master~Ib82014f92eecbcfd96a4219f2de307e5631486cb","project":"openstack/octavia","branch":"master","topic":"remove-status_update_threads","hashtags":[],"change_id":"Ib82014f92eecbcfd96a4219f2de307e5631486cb","subject":"Remove deprecated status_update_threads option","status":"MERGED","created":"2020-05-01 16:36:46.000000000","updated":"2020-06-30 16:28:38.000000000","submitted":"2020-06-30 16:23:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724890-1593534190265-72369fb8","meta_rev_id":"63f3c29c0462955770f64530562eef078e3df250","_number":724890,"virtual_id_number":724890,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-30 16:23:10.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-06-30 14:16:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-06-30 14:22:42.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2020-06-30 14:22:42.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":-2,"date":"2020-06-30 14:22:42.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":-1,"date":"2020-06-30 01:51:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-30 14:16:14.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-06-30 14:22:42.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-06-30 16:23:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d46e8709ebeb6ac01a690f59388ed80361aded73","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-01 16:36:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"babd47419ec045dc3c6cc53068bb77ac861a67fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 18:58:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8038a2af107348d2be433dea33804760 : SUCCESS in 8m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49232e8c43f842b79a257823d6958442 : SUCCESS in 11m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad26ac4a8b86421ea1cc105af7a22609 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3a1860535c845cfafd200d0e6e7a9fb : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d746fc705fe482399782fd9d11deb6d : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0946021e8fb842a39bc5b7a1f7b07db2 : POST_FAILURE in 13m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ab0bc5d9ca944bd811b8e99c544e363 : SUCCESS in 6m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8f27cac4f3b747eb9d036fc7a545dc7d : SUCCESS in 5m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/51880f97646c47c59ceadae35c2cb29e : SUCCESS in 8m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7908d9e518bd4dbf9aa11d961fbbb2ef : SUCCESS in 7m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2fd1c51e222649bc8c291e9b913d2f34 : POST_FAILURE in 47m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/21788f7c50f741af98c0ca3cf0d07fb1 : POST_FAILURE in 2h 05m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/51b3c69cda564a4fa42178bc7ac518d2 : POST_FAILURE in 55m 25s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f0893abe42fe4dddbda4055f473f3a5c : POST_FAILURE in 54m 17s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5da444b8a8ee4e22977ada0c943632ee : POST_FAILURE in 52m 59s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7cd903b5f8124802b08d4865c21d3190 : POST_FAILURE in 1h 55m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8dd26c69951447dab7de785b07d4815d : POST_FAILURE in 1h 58m 13s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1298c769a443496ba27f04f7dccebe07 : POST_FAILURE in 1h 20m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/76597549cfce483596830bed62b473e5 : POST_FAILURE in 1h 52m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d98607eddf1c9f70d7ee38854e594392951fd2cd","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-05 21:33:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"683686ddb4f31fd040c08e93d16b52fcb6585efd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 23:36:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/404661a3089b41f7a07d3728de536054 : SUCCESS in 13m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04e691960f584586a40d90e21a292071 : SUCCESS in 11m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b0e693a60174a48863241dbfd11b386 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40f0583cd8184d74b87a988eede7f99f : SUCCESS in 8m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11660c1fdcc34678867b1fa26130cc68 : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e103e73a91e44b5daa5a1e20aae50221 : SUCCESS in 13m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c64435f64aa451aa9bfb77f2b1b7a45 : SUCCESS in 6m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3080dd70f69b4bbfb250a12505f99df1 : SUCCESS in 4m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/33b8c4c76a274ee08b453ad9ce95aa59 : SUCCESS in 7m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6941842c51334c3eb030897233533399 : SUCCESS in 11m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dcfb59e10ac343fdb816407d89ab872f : SUCCESS in 46m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2c649820d36b4af29d66f07b902c90b8 : SUCCESS in 57m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3c461b485c1a4b22944241026ae40f86 : SUCCESS in 45m 10s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e358eb5601d348eb9b37bcb71b1c7b5c : SUCCESS in 1h 03m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/be3208dc6c8c43a69f17e6b46e8048ca : SUCCESS in 1h 12m 07s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ebd90e658add4894a211a7c192046592 : SUCCESS in 2h 00m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e94d75a21c9b4690bff7cc90203845bd : SUCCESS in 1h 57m 49s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e6e7631f314e4289ba0d6fbc5bf4cb76 : SUCCESS in 1h 12m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f486e2036cde46bf97d748ca7b403177 : SUCCESS in 57m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eecf4c96c61ba337e250a8382109f37a62196dfb","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-23 13:46:20.000000000","message":"Patch Set 1: Backport-Candidate-2 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"651dc824e0ace903c747604a780a5c45f7ec6af6","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-06-29 22:25:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0c4e8c7ba9eeb378c6e2362fbd0eb71b37716bf4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-30 01:47:52.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nLooks like this needs a rebase too","accounts_in_message":[],"_revision_number":1},{"id":"cf50bf93c4a2e0b9134e1960d1acd77cc41e0708","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-30 01:51:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"028dd4a562d850e9f61f7e556e895e05db18f48b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-30 01:51:35.000000000","message":"Patch Set 2: Backport-Candidate-1","accounts_in_message":[],"_revision_number":2},{"id":"b29bae0f47dc496bbfd4cc8ad639b8001f79cac7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 03:45:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35ab21cf8312479db47c6fc9e4f3ca25 : SUCCESS in 9m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cfdf2af2a03b4118913f7498cd3d53fe : SUCCESS in 10m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/600c181ea9214a09be1bb549492343ae : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20bc5499bea04cfb9dec466f765b6868 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f173a1992334cca98eafe82aad8abdb : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3b11e7a3cfb4652b0b5ef4a41f852fc : SUCCESS in 11m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c689d6855864f6d959e9fc14b1907c3 : SUCCESS in 4m 25s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/05d8f9662cc445488e3ff24ed8915dec : SUCCESS in 5m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5b208860f83b43a8b76151c738aad19c : SUCCESS in 7m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9056ca7c85044e96be7777cac33c9961 : SUCCESS in 6m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ccc90e3303634deaa64acdc025cc7e68 : SUCCESS in 45m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1d6894ab95b340fcb07410162d6d7256 : SUCCESS in 1h 37m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c343f7d394ea49d6b4bd038295483ba2 : SUCCESS in 56m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8ce5cc0c9b2448dc88ddcc28d45ee8fb : SUCCESS in 51m 44s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9ab73e3a040648d1a499856f5d288de7 : SUCCESS in 1h 00m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/55b736bea3d84570bab010400ae6382e : SUCCESS in 1h 32m 24s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/103ae31ae48741f097ae82b57ef5fe66 : SUCCESS in 1h 46m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/71297fb3f61a40fabc39e1925d7dd027 : SUCCESS in 1h 49m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8892252d64a049e3b384e1a9edb2810f : SUCCESS in 1h 47m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"186cdad03542cb80f833725035fdcbe0599be639","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-06-30 14:15:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"951c90f5ac49334b06528816debad7223e69a547","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-06-30 14:16:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"332c802fc9e7a2bd53a497e77332cde462e1047f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-30 14:22:42.000000000","message":"Patch Set 2: Backport-Candidate-2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ea081d84542894cf874f0eb0e3c393c52ac1ccfa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 14:22:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8a9f58e634e25e810a9ae6b29f96fb98b6327c31","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 16:23:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"91ab1b84192afa2974669320c4a847714e7a697f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 16:23:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a962164277e8403996a03e5269ddb398 : SUCCESS in 12m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4255825b45e04a3988b6e11b526f8cfe : SUCCESS in 9m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18e6c760360443519af3b395c97515f1 : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/142cff22573e4c61a08378d95aa53ce2 : SUCCESS in 9m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c25c21498f664fbeb7be9b003cac69dd : SUCCESS in 12m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0753cd049799494a94c3da6bb0e120f4 : SUCCESS in 8m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b42a0ca99807426d9ea200e0e45433e0 : SUCCESS in 7m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/52606e7b47ae485f842466bf27af1d71 : SUCCESS in 48m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/35c8622c9f7147fe9b419d61d16ebcaa : SUCCESS in 1h 54m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/92da971022294effb73635f7b3c2d4d5 : SUCCESS in 1h 04m 36s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f869c15c8b444bc0b394456bdef0a17a : SUCCESS in 52m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/48d6a33cf1b84f989e4148ae27222270 : SUCCESS in 1h 00m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d97082375a0b49f1869acbfd649d5ee5 : SUCCESS in 1h 34m 06s","accounts_in_message":[],"_revision_number":2},{"id":"bf07922dbd2b51cddccb097d738f52d3aac87130","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 16:28:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b35db760a0644c85851eb00969dc6220 : SUCCESS in 4m 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f9cdf01a09a497fa8f66b1f827f0388 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f6a66d86d44827c7c77984d5a591f65158678b60","revisions":{"31efaaef18a4785c24672fc9a18d2c4a39babacc":{"kind":"REWORK","_number":1,"created":"2020-05-01 16:36:46.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/90/724890/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/724890/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/724890/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/724890/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/724890/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/724890/1"}}},"commit":{"parents":[{"commit":"07a93de755c0ec461f51b94e3465708700ff1549","subject":"Merge \"Fix healthmanager not update amphora health when LB disable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/07a93de755c0ec461f51b94e3465708700ff1549"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-05-01 16:33:34.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-05-01 16:34:07.000000000","tz":-240},"subject":"Remove deprecated status_update_threads option","message":"Remove deprecated status_update_threads option\n\nIt was marked for removal in the Stein cycle.\n\nChange-Id: Ib82014f92eecbcfd96a4219f2de307e5631486cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/31efaaef18a4785c24672fc9a18d2c4a39babacc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/31efaaef18a4785c24672fc9a18d2c4a39babacc"}]},"branch":"refs/heads/master"},"f6a66d86d44827c7c77984d5a591f65158678b60":{"kind":"REWORK","_number":2,"created":"2020-06-30 01:51:18.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/90/724890/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/724890/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/724890/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/724890/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/724890/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/724890/2"}}},"commit":{"parents":[{"commit":"ec2f89830daf6f31b120bc0d10f6f41b2ea092ba","subject":"Merge \"Fix UDP pool transform in LVS driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ec2f89830daf6f31b120bc0d10f6f41b2ea092ba"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-05-01 16:33:34.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-06-30 01:51:01.000000000","tz":-240},"subject":"Remove deprecated status_update_threads option","message":"Remove deprecated status_update_threads option\n\nIt was marked for removal in the Stein cycle.\n\nChange-Id: Ib82014f92eecbcfd96a4219f2de307e5631486cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f6a66d86d44827c7c77984d5a591f65158678b60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f6a66d86d44827c7c77984d5a591f65158678b60"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
