)]}'
{"id":"openstack%2Foctavia~725337","triplet_id":"openstack%2Foctavia~stable%2Fstein~I2ede4bacd0aa4f97b2e50a8cefda45317289d86c","project":"openstack/octavia","branch":"stable/stein","hashtags":[],"change_id":"I2ede4bacd0aa4f97b2e50a8cefda45317289d86c","subject":"Do not run Tempest in octavia-grenade job","status":"MERGED","created":"2020-05-04 16:54:40.000000000","updated":"2020-06-02 00:19:36.000000000","submitted":"2020-06-02 00:17:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725337-1591057020244-560d1e1e","meta_rev_id":"2d8eb9bd4ae4a1350237b68a0dc92d1e13b1547e","_number":725337,"virtual_id_number":725337,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-02 00:16:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"recommended":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2020-05-19 07:11:12.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-06-01 03:08:35.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":1,"date":"2020-05-27 18:54:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2020-05-27 16:26:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"date":"2020-05-04 18:58:47.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-06-01 03:08:35.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2020-05-09 14:02:45.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-09 14:02:45.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-05-19 07:11:12.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-05-27 16:26:36.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-05-27 18:54:02.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-06-01 03:08:35.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2020-06-02 00:16:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e6dba9c2de53f3215cc0bfd3fa0d7829d0dd3ebe","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-05-04 16:54:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40d46b0ac5f7253dc260f64cf912778f71c3d29f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-05-04 16:54:56.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"fec1c25122badb0d34c920e79c3d625fa8b2a78a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-05-04 18:58:47.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"6f198ffedc07ed9751a5f2b0f73be7968dd7f141","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 19:16:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a84bd14954c493db80d6bc7ca8826dc : SUCCESS in 7m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4511b69d061342be96846888418bb8af : SUCCESS in 10m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da9b5de6cdd34af78f41049bcb8762bc : SUCCESS in 6m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d92251e932f245999516e1e4ed5b2694 : SUCCESS in 5m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bdcb7d8dc8774a0aa4b3cab7d94fd398 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0bd75964e5c460f869b3ebadb47f4f1 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6165982b5a5a418ca63acf83a2759eaa : SUCCESS in 7m 44s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/d889ceb008aa4522a504e7e63de95222 : SUCCESS in 1h 20m 02s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/94c954d80db04d9e838f517841156e1c : SUCCESS in 1h 24m 57s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/098f7af6a01f47039f2ee8eb97daa55f : SUCCESS in 1h 21m 12s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/39530addb8a5443598fcbfa3635cf214 : SUCCESS in 1h 24m 18s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a745974a9a84477fa8a2344e67d522f9 : SUCCESS in 7m 25s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/7f34683698f940d6a72ba6f779d2fd05 : SUCCESS in 8m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9170ed5478d148088034a87846babcb6 : SUCCESS in 7m 35s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/218c3d77e4574a6a9a2cd8c107fdbf75 : SUCCESS in 58m 49s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/5eda81a0050b4eb7956cae8c3ef4348f : SUCCESS in 1h 03m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c2ea4a9854ae41e3a9ddb59e0ae2e2a6 : SUCCESS in 42m 27s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/e185dafc9ace45fdaaa73429904d0fa9 : SUCCESS in 41m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/49d7dfba56704de5b98d2c019488281e : SUCCESS in 1h 40m 49s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0e522436aa2a4072ba00046f0f8b7150 : SUCCESS in 1h 16m 12s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/fc8df4401fd146659f0c004cd22f5014 : FAILURE in 25m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/131839a2dcb344f7b1f8e2299e1e52ab : FAILURE in 48m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bcd3352a8fb849dfa5637132e041501c : SUCCESS in 1h 37m 53s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/cfe655b9ebd545c4a1794c899b97a633 : SUCCESS in 2h 03m 59s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ff07c0da15594db697921b2709440319 : SUCCESS in 1h 11m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e7dd4a971c9e44a78cf0fc8e1c6884d2 : FAILURE in 55m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"857cbd69d855c7f7c620ff524aee36f66453d462","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-05-19 07:11:12.000000000","message":"Patch Set 1: Code-Review+1\n\nThis should fix gates for stable/stein?","accounts_in_message":[],"_revision_number":1},{"id":"f51a2b22de1977023b607df155f94d6c564603f6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-27 16:26:36.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM, tests are part of the grenade job here.","accounts_in_message":[],"_revision_number":1},{"id":"7b247b7c98ed25e8e054a04e59739e9f5916aae6","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-27 18:54:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"250c34a891bfb62a81ef565be9a06ddaf5265b57","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2020-06-01 03:08:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2993ca56b6c296df0b97ba42b33ff83ed38b4703","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 03:08:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"48574f803b7c24996edfa9f2958a248493a70bbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 04:30:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77a2bfe7ecb24fbfa48c1862f617e276 : SUCCESS in 8m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05697a371f64409fb41f40c0a0e27566 : SUCCESS in 5m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0f6ac53c63cf486bbfe03bdf3e3502f5 : SUCCESS in 4m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/00c7a01181d04845940a6cf86d31621c : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ac28b0600a24dc3a18a767c98b65408 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9764ad4239354744bbacfa4f6962d373 : SUCCESS in 7m 38s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/397a4b6e742f4ff5a43acc87e22e4942 : SUCCESS in 6m 22s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/4d3f3b8b9ce34f0b927f8bc1f1aa02bb : SUCCESS in 7m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bc2e8c6c1c9b4f04a357c763dfac8404 : SUCCESS in 6m 25s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a99f82ae31484b18868ceaa2bb5b6553 : SUCCESS in 50m 18s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/f39ee957a6434b23af3ab6d2b9110710 : SUCCESS in 59m 43s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/cc58a33cab3b418b85952d6d540ea4bd : SUCCESS in 1h 10m 38s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/f1f58a4891ee4cec9822b8b11c2c6c31 : SUCCESS in 1h 20m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/887dca4afb444169b6d4fbf7b242fdbf : SUCCESS in 41m 57s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/14c4a4ca09e749908991aced8670db42 : SUCCESS in 36m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/64d22f8bba27415b871016a74f77ef95 : FAILURE in 52m 58s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d14ba9d09f2549d4b3fce72e997654d9 : SUCCESS in 53m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4de75538def1423ca364f9a97be2ed88 : SUCCESS in 1h 03m 50s","accounts_in_message":[],"_revision_number":1},{"id":"bf5de99908f70ee07b1feaa62ae7ee1107339c34","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-06-01 06:23:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b717a0274cfafdc759c6173442e9005a5271f6e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 08:22:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5135b17b582e44b18a9d794826708bf1 : SUCCESS in 9m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c41967dfb76341379fcbc3983b21f477 : SUCCESS in 10m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b08d09be583a40659b6511c4167edd4e : SUCCESS in 8m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1d897e1b4da74782b60eddf70c5dba95 : SUCCESS in 5m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d36b0dbf3f9a4ac58f03524205895ecc : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba5d3aa02aff4a74baff626a35c3ba65 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af337e70de2749ec8d7d352fbb9207e5 : SUCCESS in 8m 39s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/c398a476bef4491ab30a8f05bc422dc5 : SUCCESS in 1h 20m 20s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/eeadfc7a63b541c79a4126e5259d5f4f : SUCCESS in 1h 20m 55s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/17349c36d5a641c6961fb3bca822c14b : SUCCESS in 1h 28m 18s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/1df291f60e914c3ea887e179d458de18 : FAILURE in 1h 22m 37s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9b5e28e061e3413b866b437ea1555643 : SUCCESS in 7m 37s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/2c0849569a0343f8975e0159f891c16a : SUCCESS in 7m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f24da49af52b4ae486b107113c94fc50 : SUCCESS in 7m 59s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aa50868ff7f7472d9d7a6debd46cab3a : SUCCESS in 1h 02m 06s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/2ada9579b4b447f5a99527b3b175fa62 : SUCCESS in 59m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ae5ff2dc430046939134b14f4f3cc489 : SUCCESS in 46m 58s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/8604b801e2fc4b2ba1732284071f2831 : SUCCESS in 44m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6ffbb123a91c4dc59c62e1f70c13f50c : SUCCESS in 1h 32m 09s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d3799f5a88b745b7acecff8e417d4a7b : SUCCESS in 1h 00m 15s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c3295ede654249d89a1479e1bcfed3a4 : FAILURE in 18m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a555295c83c4497c97692590ad906f48 : FAILURE in 30m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2b7d1b09a76548598d6e970038fd422b : SUCCESS in 1h 52m 05s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/b3df490b688247c181d6e8e02c1ede04 : SUCCESS in 1h 50m 27s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bf9678d273d04ac59d3894355c75d88d : SUCCESS in 1h 27m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/83d97ab5114849f098208d41275f13b6 : FAILURE in 1h 11m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"01bc1164a2c17f526b740e71d16340dc09aa1a63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 08:22:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"924ac02411ac5a35230dae408960923eb1b8cec5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 09:59:35.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dfce933d0e9f431ab384cdccfa41ae41 : SUCCESS in 10m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc2d0d32c852479092b7e0cd2668ea2d : SUCCESS in 5m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0a1896eda78f4bf7863dafe5e52aed3e : SUCCESS in 6m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1e9b30e64f9d4df39dd339fe66022cc0 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/561265faae754ac0ac13915074652a5a : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5d952a5b0214e009769d471a0030c1f : SUCCESS in 7m 08s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2cc8f5fb5e0a42e2910ab0c7edffa7cd : SUCCESS in 8m 00s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/1516ae0a5e144c64b6243e1ab62bdd56 : SUCCESS in 6m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5eb0b5b65fb9481f80379de88865f3d0 : SUCCESS in 6m 52s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ed11c39366f047308560bad633b64e8f : SUCCESS in 1h 01m 07s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/1a9e19f34f10412492a16966c8af28e9 : SUCCESS in 56m 34s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/f10a75f398824f0aa2b3c8404292a3ef : SUCCESS in 1h 26m 22s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/9471a17ec8fc4cb29c66d7e46e33ad19 : SUCCESS in 1h 17m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7c048a4b814f4e9f90e1be1c33afe4d9 : SUCCESS in 45m 52s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/928d0bf06bc248cfb4df0fe597be96e0 : SUCCESS in 41m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f27339ec0d65484ba8c2665e9ef1515b : FAILURE in 1h 20m 36s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d67a76b2bc77436f8dd5a8c8b0243bbe : SUCCESS in 57m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ea568ceb3ed24a2ea09cc405119a22a0 : SUCCESS in 1h 35m 13s","accounts_in_message":[],"_revision_number":1},{"id":"f519b132907d527f51178b65338a15be3416f96f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-01 20:38:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a1be2c6b19f3b2d5f93c1ff1d870678d0e9ea5f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 22:34:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a398b1d564304d86a23cee4e85c0b9f0 : SUCCESS in 13m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/293e23be6e8749939aec0d53d416a118 : SUCCESS in 8m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bbf0e20b9d14dbb9b9efd4f06de0659 : SUCCESS in 7m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6fcd9325168545bf9a198054f8d81706 : SUCCESS in 7m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e5cfcfc8d6604ca494f7d319b1550a96 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/680d5594a6404d129b40281e36f8380c : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85912f372abe4b3580cdbc336d0362a9 : SUCCESS in 5m 57s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/e8e43b2cbd434072a50a0e73925356fe : SUCCESS in 1h 17m 32s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/a08b704d25244d2fb60e1d70ebdde903 : SUCCESS in 1h 19m 30s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6b3a82eab62349828ebbf4956b8a2b2f : FAILURE in 1h 53m 52s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/e6d6db0d3c094904b45016aa0cbc44b8 : SUCCESS in 1h 12m 11s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6cf53cf7fd84470d84a1a734eebf33c3 : SUCCESS in 8m 58s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/6dd82f12bf954c009707605acebeeebd : SUCCESS in 6m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/21c27add90284aeab0a14be8841d40cf : SUCCESS in 8m 25s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0214603023fd49ddba2064dd4d54e46d : SUCCESS in 1h 03m 15s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/cb71b68afb8041baa29513eaaddf2b2a : SUCCESS in 52m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/76b706877a6840cf996577c98d5feb16 : SUCCESS in 42m 47s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/f6ce8f442c354906a0666f872e124863 : SUCCESS in 46m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4cab6f7119b241ada425b522c5756121 : SUCCESS in 48m 48s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/dfd4231aa8ac42cab254b126cc86b580 : SUCCESS in 1h 04m 05s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/72aa6c217ea84ae5bf74631ee7ecb53c : FAILURE in 16m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66711a6831f9447093db43df7a3be1d3 : FAILURE in 31m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d657284ccecb47c2a56dda0dcae31f03 : SUCCESS in 1h 00m 43s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/d0c491ab740a452db2fe3e712b5fef37 : SUCCESS in 1h 08m 49s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b87f5c8c20044e019dd78e26871d07b2 : SUCCESS in 1h 16m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/72350dc5e9dd43239fc832edb3fa7af8 : FAILURE in 47m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"24b2681d04245d034d6921115b4e5fe17db73b3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 22:34:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"70545aea1cea1491862dfb4329e16c6c383eb9d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 00:16:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0abbb77f3be4def83751ebc5052cf32 : SUCCESS in 9m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff070f5d1fa44a07a15d8b7d198eb29b : SUCCESS in 6m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/839899a49a934aedbc8e8c63e462cd2a : SUCCESS in 5m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8d49ceedc3e0473dbcb569c3ecf0dd12 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1682ffa8c23f420893f6be639dfb8a0f : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ebc03f3a01946289ab25d2f8b096705 : SUCCESS in 8m 15s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2a164314c2414a62a103c3ea27b94835 : SUCCESS in 7m 48s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/451cba1ba9494df0962f3f3724018e14 : SUCCESS in 8m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/087c167b83884c339998c0bbd820560e : SUCCESS in 6m 54s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/26aed25125d84adba00f498105744771 : SUCCESS in 1h 05m 24s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/98f84a984f7f4aeea0c93fc430c9873d : SUCCESS in 36m 26s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/2a09bcd7c6354c8cbfe32be120831f38 : SUCCESS in 1h 15m 44s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/1099dcf8ec064b70b458b0108fcf58be : SUCCESS in 1h 22m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bb5660ca32994c669c4613d4fdebab93 : SUCCESS in 57m 20s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/23dd7fe9e3274b7f8f13e60ee51296f0 : SUCCESS in 44m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e88ed3b418b747ab869de87e70fd7591 : SUCCESS in 1h 36m 30s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ecea82ff2d1449a6bde927dbf4279e32 : SUCCESS in 1h 42m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fb2cbb827a554aabb756cda8d5e3d172 : SUCCESS in 1h 16m 30s","accounts_in_message":[],"_revision_number":1},{"id":"6742ec714140751db4f3770f051e2ccf7057c9e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 00:17:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2d8eb9bd4ae4a1350237b68a0dc92d1e13b1547e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 00:19:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5b752a6ad8242e48c0f571c4e08c852 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5dd672c58c322e142b0e187da82267ffde606450","revisions":{"5dd672c58c322e142b0e187da82267ffde606450":{"kind":"REWORK","_number":1,"created":"2020-05-04 16:54:40.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/37/725337/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/725337/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/725337/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/725337/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/725337/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/725337/1"}}},"commit":{"parents":[{"commit":"922c4987fb4db019212c7db322a65c3998f0c1a9","subject":"Remove xenial based jobs from stein gates.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/922c4987fb4db019212c7db322a65c3998f0c1a9"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-05-04 16:53:00.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-05-04 16:53:00.000000000","tz":120},"subject":"Do not run Tempest in octavia-grenade job","message":"Do not run Tempest in octavia-grenade job\n\nChange-Id: I2ede4bacd0aa4f97b2e50a8cefda45317289d86c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5dd672c58c322e142b0e187da82267ffde606450"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5dd672c58c322e142b0e187da82267ffde606450"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
