)]}'
{"id":"openstack%2Foctavia~726042","triplet_id":"openstack%2Foctavia~master~I05a541a77f254a77ad5036e1062b61c8ce93b754","project":"openstack/octavia","branch":"master","topic":"fix-trust-token","hashtags":[],"change_id":"I05a541a77f254a77ad5036e1062b61c8ce93b754","subject":"Fix getting user neutron client","status":"MERGED","created":"2020-05-07 05:33:13.000000000","updated":"2020-05-21 21:51:40.000000000","submitted":"2020-05-21 00:21:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726042-1590020467573-cc49b14e","meta_rev_id":"9a53bf08f790eccd6e3a4077e7d36056ab8527e9","_number":726042,"virtual_id_number":726042,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"date":"2020-05-20 22:11:01.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-21 00:21:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"recommended":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-05-20 10:22:53.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-05-20 10:22:53.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2020-05-20 10:22:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2020-05-20 22:16:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"date":"2020-05-20 10:22:53.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2020-05-20 10:22:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-07 16:16:24.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-05-12 07:48:31.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-05-12 12:25:07.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2020-05-20 22:16:46.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-05-21 00:21:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9f1f8a627234ea40dddfbb8c77e3761e075d6f59","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-07 05:33:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5a08db29508bd6b7904adac12b182b23640d50e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-07 05:37:11.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"da9a8bedaa2529505f88183c176bababef59a7cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 08:13:35.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6515fc42d812463584b59317dc6c3f0d : SUCCESS in 9m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa95ae98cb5d434883065645722b6e26 : SUCCESS in 10m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/100cfc2cae724f1cb5f5f19f42426cb2 : FAILURE in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e94ebb92566423194b45335c51dcef8 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b23da04916a4a308df0f1d9652ba29d : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/842e35f974d247c899105288c4ac19b8 : SUCCESS in 12m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/076d3e6158844834ba47a310383b4d46 : SUCCESS in 5m 25s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/45c46817cc414fe8b2b59c4ffce91022 : SUCCESS in 8m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/116574e216bd4d459c8dfa776b91e5b1 : SUCCESS in 7m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1e45426d85da49ac830a02f227e6d7ea : SUCCESS in 48m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/26e393dad82c4affb59ed79413640362 : SUCCESS in 2h 00m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/36c6423aff424b17a43f5ed1dcabe57c : SUCCESS in 1h 08m 54s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0b523c4e65dd40ef94a825df8dcf1073 : SUCCESS in 56m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f18b510be04b458997755e32d13e0a99 : SUCCESS in 50m 24s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1ed8b792c49b4dc18692cb7b04e55185 : SUCCESS in 1h 10m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a326ce7997454abda4c706c5d55f3120 : TIMED_OUT in 2h 35m 23s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0956d1fa130f42d893245b1eb18d50a9 : SUCCESS in 1h 18m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/51ae83d30be04856a10c4158c1feeabb : SUCCESS in 58m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bd8422c4207f16426ca11bf4c46434a5d270882b","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-07 10:18:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ee69e06fdfdad6da03e2301c3d28f211df41696d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 12:40:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1913eda965ec4664ac79ec0d607fea37 : SUCCESS in 9m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89c00a6db9bd4073b75dbd3db59b344b : SUCCESS in 10m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9aaf8be740549e38594ff5c0d0a4ccb : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ae52346ead14b839d30b73e7bb8c2c1 : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/885479da82d84dddb6a7e54a6fad1112 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87c03f9b67ff44dba2aaf9a55144584d : SUCCESS in 13m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4a7b0c4c8f1f40e392cebfd46afaad27 : SUCCESS in 6m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7e4b1fea2a2a449b8f79bfa41c6097aa : SUCCESS in 8m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/179695c4578a40c3bf43c82ec8d6e186 : SUCCESS in 12m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5261b77366914b59b3f1cfaf0bfd6675 : SUCCESS in 51m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3bd3618ce1774c98bb593bed2835cce4 : SUCCESS in 1h 23m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5baa0649190f4b598337237389c3668b : SUCCESS in 49m 34s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0e3ff5e90b894e60b8b0c54ad8420fc9 : SUCCESS in 1h 02m 37s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c62954ba82c84e02b9a39c15961730bb : SUCCESS in 1h 31m 52s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/945414e6c89841a3b7de73bc9adb45b5 : SUCCESS in 2h 14m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a35e3f4d1a8a4046b5ca6ec2bdc08be5 : SUCCESS in 1h 25m 44s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/09709a8295c9499fa8ae07f72fe1a7be : SUCCESS in 1h 18m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ac6701a8d3ed42c1aed7b98be4a72bd0 : SUCCESS in 1h 28m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"89f6b5b4c0e8183500f5428d1e186737be7aba1d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-11 21:15:02.000000000","message":"Patch Set 3: Backport-Candidate+2 Code-Review+2\n\nLGTM\nWe should investigate switching this for the barbican client as well. Adding a task to the story to track that.","accounts_in_message":[],"_revision_number":3},{"id":"70bf0eec255c33a095703c1ffef377cb3364dbdc","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-05-12 07:48:31.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ee5c71708db29b737e236a9fb7e6b9f9d8fe6807","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2020-05-12 12:25:07.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"12eae54ee26d5353187fce817a3dbebeae57cd06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 12:25:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c3f1dbcb504ed4b06efd26f7477b4ef69bdb03f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 12:35:25.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/350eb7b1373c4dbf951104f674d78b73 : CANCELED\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/767be9606d0e494b9e721c4daf8ab22e : FAILURE in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa046ca030f04f0cb02a1b16a15b9686 : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66b8f3c2dbaa4904852224f9ae0cb503 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfb6770a68c04f469afd369eb0d7f0a3 : CANCELED\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7c7ef254294942e693a0e84df64b1131 : SUCCESS in 7m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6d5e4a6b638e4a6fb5a2311336a3e82d : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/61891195b714454b8cfc6863930783df : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1a4c08b410d74aa98148c777ac749d0a : CANCELED\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c8e541041bf44d4caf583c4af325dfe0 : CANCELED\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bcc1376f76d54ea99a50d655555dc2e0 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/33a9832a5ff345b4afd6af52b703d985 : CANCELED\n\nWarning:\n  Comments left for invalid file octavia/tests/functional/api/v2/test_pool.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_l7rule.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/database_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/database_tasks.py\n  Comments left for invalid file octavia/network/drivers/neutron/allowed_address_pairs.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_load_balancer.py\n  Comments left for invalid file octavia/amphorae/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/loadbalancer.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_l7policy.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_health_monitor.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_quotas.py\n  Comments left for invalid file octavia/api/v2/controllers/pool.py\n  Comments left for invalid file octavia/db/models.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_member.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_amphora.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/rest_api_driver.py\n  Comments left for invalid file octavia/controller/worker/v2/controller_worker.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_listener.py\n  Comments left for invalid file octavia/api/v2/controllers/member.py\n  Comments left for invalid file octavia/api/v2/controllers/health_monitor.py\n  Comments left for invalid file octavia/tests/functional/db/test_models.py\n  Comments left for invalid file octavia/api/v2/controllers/load_balancer.py","accounts_in_message":[],"_revision_number":3},{"id":"4fb422dfe538a2fe67bcfb4a254cefbbd9043f0e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-12 12:52:59.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"047fe3531d6ca08535b7a1ef7fa756abfb46bb66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 15:21:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2e26c6009ee4007a2923615268e6fd3 : SUCCESS in 10m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c583ae7366c400ca9b4ca158ec4f05c : SUCCESS in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcde85fe7cc9420bb64044cf6daab60f : FAILURE in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd975dbc5003432cb2d090fa1f0bd48e : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2eacad66966c49e5b47dda1280279a49 : SUCCESS in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b348a513f5f94d49a877f98ad1394001 : SUCCESS in 12m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7304c28af30a4fe48d9342d1af4784f7 : SUCCESS in 6m 58s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d6d4208bb20e4b9ea377e86ac6d69de7 : SUCCESS in 8m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b10e912aadce436393a9aead47700594 : SUCCESS in 8m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fe695d652ffd408ea852a345a8d08770 : SUCCESS in 44m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/09de855c6cbe454c8d88dca0ee568414 : SUCCESS in 1h 50m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fa9ac0e4ed9a421e94bedae4fef6ada8 : SUCCESS in 1h 03m 49s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/07eea6728c9842afa25b317148f3c312 : SUCCESS in 44m 03s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/342215780cae402ba15122c11c6d4629 : SUCCESS in 59m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/18e48954e03342a5a73d9271bc7eb078 : SUCCESS in 1h 45m 00s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ab464781da41482f979f610bef854051 : SUCCESS in 2h 19m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/27c9c01cf41e436e9739cfd689f00f5d : SUCCESS in 2h 06m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/af46c1caabbd4b99b4e2fc18cbaf8da6 : SUCCESS in 2h 11m 32s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/tests/functional/api/v2/test_member.py\n  Comments left for invalid file octavia/api/v2/controllers/health_monitor.py\n  Comments left for invalid file octavia/api/v2/controllers/load_balancer.py\n  Comments left for invalid file octavia/db/models.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_l7rule.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/rest_api_driver.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_load_balancer.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_quotas.py\n  Comments left for invalid file octavia/network/drivers/neutron/allowed_address_pairs.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_listener.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/loadbalancer.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_pool.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/database_tasks.py\n  Comments left for invalid file octavia/amphorae/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_l7policy.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_health_monitor.py\n  Comments left for invalid file octavia/api/v2/controllers/pool.py\n  Comments left for invalid file octavia/controller/worker/v2/controller_worker.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/database_tasks.py\n  Comments left for invalid file octavia/api/v2/controllers/member.py\n  Comments left for invalid file octavia/tests/functional/db/test_models.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_amphora.py","accounts_in_message":[],"_revision_number":3},{"id":"a45c64383cb9bc2a8275a4aa9934ad0048c66958","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-12 21:51:40.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"10ddcee41c938073b2cd0122498574b6bf3191d4","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-12 21:51:53.000000000","message":"Patch Set 4:\n\nrebased on https://review.opendev.org/#/c/726972/","accounts_in_message":[],"_revision_number":4},{"id":"27c4781bfc70204f7e7b5486e7f9650a80382a1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 00:23:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d589c66bf2c5480d911609dd24e30bf5 : SUCCESS in 9m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93e93b2a4a10489ca4660528b509f75b : SUCCESS in 10m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6e9e9d527d34f41adb35fe18b20c4f5 : SUCCESS in 7m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9b476d2758040cb9b58ab3d2a03d564 : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b8beb5e2ff54b91ae4121ee6f05f669 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d460d7dd1e1c443196409b3f96b80390 : SUCCESS in 11m 50s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/eb448457f24847108f35c1567f66fe50 : SUCCESS in 8m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6be29dde9f7d417d872452c4ba640c7e : SUCCESS in 7m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a60eb14069374444b280b920ea6eaf37 : SUCCESS in 10m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bf9de683e632427c9caafbf56f5449f0 : SUCCESS in 46m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/757f68e5501a4b0daf4c63bcea619292 : SUCCESS in 1h 35m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8087777fc0b04a43b41a3400dc5c329b : SUCCESS in 54m 45s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/23c9442c544b483cb48f61fe1b88e849 : SUCCESS in 55m 14s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c57e0c0e547a4277a53a8876d14fcdae : SUCCESS in 1h 01m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/08830ba271f84592a3729e56ec6810c3 : SUCCESS in 1h 23m 58s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7535423e1a24415fba83ad6ab357208c : SUCCESS in 1h 42m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/59fd16a188be42209eb57d8d1a68179a : SUCCESS in 1h 55m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e89a7cbc85904f0da5a3abccb975e5ba : SUCCESS in 1h 39m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4eb5ac7bf76ae956e2fb27bb56ba0900c0938886","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2020-05-13 05:32:53.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"1ea030d86c34c3fa846b9290fb1200b4ce0f5170","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 14:45:00.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"998815cad09bfe5b9d5f15757328a302e1ca3347","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 16:14:27.000000000","message":"Patch Set 4: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"30e07d1115d0b52c384be7311a7f43fccf2511b3","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-05-19 07:08:49.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"76e34cd83804abe3ed89f9eb3b0e4d16b374a00e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 09:15:26.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16aa8f6b73d34e728846969051869822 : SUCCESS in 11m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04f1a10f6ae344b884514cdcce6d14b7 : SUCCESS in 11m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d8a59b202b44dc29b5f871acb038e46 : SUCCESS in 7m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e1c8566fc8d417892dacbaddc4d1132 : SUCCESS in 9m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecf08f99a4fb42419aae1b0813119c67 : SUCCESS in 7m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e96c58d2c4e4994a851e414cf9b7936 : SUCCESS in 13m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f75ac7f426ea46938cf3506aa84c2bf1 : SUCCESS in 7m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3b25f39600894f41be574896541d5ecb : POST_FAILURE in 17m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8e36b9d6d9fe4b11af4ab19d5b615905 : SUCCESS in 8m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/416d19550f9b43d994a24edc55df5bd8 : SUCCESS in 44m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/54a7efd8e2b44846957b4206a68db7d1 : SUCCESS in 1h 35m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/06feb23950df4f12aff0dc1438d3edd7 : SUCCESS in 58m 00s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a94fe8dfcef8470eb4091d2c1298d00c : SUCCESS in 57m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e4a02380465149d79f70872a58f528ec : FAILURE in 53m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/20bfb69da00f4555b144d9fc34c1bdef : SUCCESS in 1h 36m 18s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c0bdf774ee644e9199f2ddd4ab8cb019 : SUCCESS in 1h 53m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8cc397bfe45d4964b65f3524ab08656f : FAILURE in 1h 42m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2a0d4020d0cb4694b021632b6e44e2fa : SUCCESS in 2h 01m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a559420d2d6f84902ce29c8773f756508350a3b4","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-20 10:22:53.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"37c84b34989e55f94e1f455a93b5758f9ed64127","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 12:44:13.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8a17a3cf47b4c999adb7771afefa49d : SUCCESS in 11m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ebd8c4008d224e17ba9fb1b3885fb01f : SUCCESS in 10m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd430f6f6455418a98b49b624a482e03 : SUCCESS in 7m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3752ac571a04518a3ea5131729fb457 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9a0e2be3f88486b9475be0141497e90 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfdf2571910548e5aa65ba98d23cf488 : SUCCESS in 12m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a2ac017863614ab8850d6261d29796fb : SUCCESS in 6m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d98ca6cd9b13482886175856407da53b : SUCCESS in 11m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/38479c408e334a5b8dd144f3696c413e : SUCCESS in 8m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1c695d620cee4514b96f1c3a83434375 : SUCCESS in 47m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f0c0189464d5447eb013728ea78f2146 : SUCCESS in 1h 38m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4bb56c1fd30f43debad9af414af20672 : SUCCESS in 1h 03m 01s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/eb726beae7a64fabb8f1f16f276edac6 : SUCCESS in 53m 25s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1bdc3d11634543a89763da4a23da066d : FAILURE in 51m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/08eaf0d7e06340bca11106eb5f732b0c : SUCCESS in 1h 33m 20s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/be4b1165e35d4738b859668cf00aac85 : SUCCESS in 2h 17m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0824da6c58d0455c9a5d2b83864c7bbc : SUCCESS in 1h 43m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/16abbb628d914f01814c273f81c9f443 : SUCCESS in 1h 02m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9d6b4fb770e2302b7155f6ef466eea3443073ea1","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-20 19:29:25.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"35b2284caa7fd8663633e8e02e524eddceb03ce0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 22:02:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2472018c05d8432abef5e55bf46d911e : SUCCESS in 10m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e43f9f2b648840e6af10b13303f85961 : SUCCESS in 11m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d43974b004494b0f9f4fa1e203d98b40 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1113e5e31014b8c94e712c4e76fb136 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9984b8b590c40109d409e1c387241d7 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b42d1b19d9ab4b48b9831d7873f0ba7a : SUCCESS in 13m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c37a034a6c164b64a38273bb2cdb3e5c : SUCCESS in 7m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/05a18569ccbe4a5fb9c484092b84f01d : SUCCESS in 13m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b9aac1f5890d411ba7bd8bf22b03897b : SUCCESS in 7m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/eb860270a7e445d1b32c43c5befec39e : SUCCESS in 50m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3769373d83bd481b914a98b3e5321b74 : SUCCESS in 1h 32m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/39c8a4f2733f4a3c9d9e7caf78895d47 : SUCCESS in 55m 57s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/428292d00147432bb21b0511a379448a : SUCCESS in 50m 44s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/915527756df843739e2c3e04b3c61f0b : SUCCESS in 58m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/146f8e153c8f494aad948cee6ebec424 : SUCCESS in 1h 35m 21s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e23c8f06bceb4a9b87d9c49faa8a16a0 : SUCCESS in 1h 00m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fd6801399fb44fd09d696b9ebf9f98a3 : FAILURE in 2h 14m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/75d5852f5bc04b6189a83402bd1fb7a1 : SUCCESS in 1h 32m 35s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f5d38485128446324728c7782eb61ebf8be2cca8","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-20 22:11:01.000000000","message":"Patch Set 5:\n\nHi can we try again to merge?","accounts_in_message":[],"_revision_number":5},{"id":"819fd2f2ce22f2dd5e1d8cdc65a68be7cabe1347","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-20 22:16:46.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"262af1c54afaa34486345b21352a0101a735a63f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 22:16:58.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c08726dad5a5cda76a18409f241e9f0f3af145f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 00:21:07.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66b9baa11b7645a884c702f44b808e96 : SUCCESS in 10m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ffcfb8f7d394661acfbeb7de2b4e263 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4a143935eb847b7826a50c15de87e33 : SUCCESS in 7m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e9bbbfcfc994b9595ccf806a64cc7d0 : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54838d6ba8fe4aae9eb36b90e7cb1f2b : SUCCESS in 11m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0b9ee08a970a4549bae803d7c5375146 : SUCCESS in 8m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/86037f12d4124c4c99c0a82eed255f22 : SUCCESS in 44m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6e950268fb54468e80e44f9fe22aa56d : SUCCESS in 1h 58m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/39c0c1b2e72044dea69204b597b4e178 : SUCCESS in 41m 33s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3fd28d1223f64b9bb0be70c2dce7b6d2 : SUCCESS in 1h 01m 06s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cfdaab139d0e433280eafbb22c1702c9 : SUCCESS in 1h 00m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b8bc283bd49e49519bac39670dcf7b68 : SUCCESS in 1h 39m 10s","accounts_in_message":[],"_revision_number":5},{"id":"52db94d2637b46781669079321431bd417b2c60d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 00:21:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"daa850144c78336f3d167c4ca5abf6c71f13595f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 00:22:47.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17911af3a6a4469d95b445ef50d5c2cc : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":5},{"id":"7610b4c07b3c0bd4c86b520cd79c5ede8c514a39","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-21 21:51:20.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 09e863eb539e29c8950ddfce12d602c0e77c1b7e","accounts_in_message":[],"_revision_number":5},{"id":"b17062c756553ce21b21aeec4a0bfb79112003ae","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-21 21:51:40.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 90defc9bf91baae3b1b5ad65d211c2c09ea6a3db","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"b54f373cac62ec170562d7a2d84e66773b9a2deb","revisions":{"1e3bad0b97db54b24e728b00830f93326fc9987a":{"kind":"REWORK","_number":1,"created":"2020-05-07 05:33:13.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/42/726042/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/42/726042/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/726042/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/726042/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/726042/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/42/726042/1"}}},"commit":{"parents":[{"commit":"837b7bf94092438e486d3e18c5b9a21bb1298f98","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/837b7bf94092438e486d3e18c5b9a21bb1298f98"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-07 04:31:38.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-07 05:32:11.000000000","tz":720},"subject":"Fix getting user neutron client","message":"Fix getting user neutron client\n\nUse token and endpoint URL to initialize neutron client for the\nrequest user.\n\nStory: 2007619\nTask:  39641\n\nChange-Id: I05a541a77f254a77ad5036e1062b61c8ce93b754\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1e3bad0b97db54b24e728b00830f93326fc9987a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1e3bad0b97db54b24e728b00830f93326fc9987a"}]},"branch":"refs/heads/master"},"78a8954e53101de650cb0f434368e30eb5cf9857":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-07 05:37:11.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/42/726042/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/42/726042/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/726042/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/726042/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/726042/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/42/726042/2"}}},"commit":{"parents":[{"commit":"837b7bf94092438e486d3e18c5b9a21bb1298f98","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/837b7bf94092438e486d3e18c5b9a21bb1298f98"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-07 04:31:38.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-07 05:37:06.000000000","tz":0},"subject":"Fix getting user neutron client","message":"Fix getting user neutron client\n\nUse token and endpoint URL to initialize neutron client for the\nrequest user.\n\nStory: #2007619\nTask: #39641\n\nChange-Id: I05a541a77f254a77ad5036e1062b61c8ce93b754\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78a8954e53101de650cb0f434368e30eb5cf9857"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78a8954e53101de650cb0f434368e30eb5cf9857"}]},"branch":"refs/heads/master"},"195c1a46be0fb63731504fd71a6f1ad4e03f3df8":{"kind":"REWORK","_number":3,"created":"2020-05-07 10:18:34.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/42/726042/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/42/726042/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/726042/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/726042/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/726042/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/42/726042/3"}}},"commit":{"parents":[{"commit":"837b7bf94092438e486d3e18c5b9a21bb1298f98","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/837b7bf94092438e486d3e18c5b9a21bb1298f98"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-07 04:31:38.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-07 10:18:23.000000000","tz":720},"subject":"Fix getting user neutron client","message":"Fix getting user neutron client\n\nUse token and endpoint URL to initialize neutron client for the\nrequest user.\n\nStory: 2007619\nTask:  39641\n\nChange-Id: I05a541a77f254a77ad5036e1062b61c8ce93b754\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/195c1a46be0fb63731504fd71a6f1ad4e03f3df8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/195c1a46be0fb63731504fd71a6f1ad4e03f3df8"}]},"branch":"refs/heads/master"},"9f13913ce66db80cf949f5ab5316a1983b35b78f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-05-12 21:51:40.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/42/726042/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/42/726042/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/726042/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/726042/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/726042/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/42/726042/4"}}},"commit":{"parents":[{"commit":"ffdbe8942ef43fb171bebed8699d7436ef448ef9","subject":"Fix E741 pep8 errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ffdbe8942ef43fb171bebed8699d7436ef448ef9"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-07 04:31:38.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-12 21:51:40.000000000","tz":0},"subject":"Fix getting user neutron client","message":"Fix getting user neutron client\n\nUse token and endpoint URL to initialize neutron client for the\nrequest user.\n\nStory: 2007619\nTask:  39641\n\nChange-Id: I05a541a77f254a77ad5036e1062b61c8ce93b754\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9f13913ce66db80cf949f5ab5316a1983b35b78f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9f13913ce66db80cf949f5ab5316a1983b35b78f"}]},"branch":"refs/heads/master"},"b54f373cac62ec170562d7a2d84e66773b9a2deb":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-05-20 10:22:53.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/42/726042/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/42/726042/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/726042/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/726042/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/726042/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/42/726042/5"}}},"commit":{"parents":[{"commit":"bcef33a3ff4b5cf67964919c9826514f8ddceca1","subject":"Merge \"Add TLS cipher blacklist to octavia.conf\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bcef33a3ff4b5cf67964919c9826514f8ddceca1"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-07 04:31:38.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-20 10:22:53.000000000","tz":0},"subject":"Fix getting user neutron client","message":"Fix getting user neutron client\n\nUse token and endpoint URL to initialize neutron client for the\nrequest user.\n\nStory: 2007619\nTask:  39641\n\nChange-Id: I05a541a77f254a77ad5036e1062b61c8ce93b754\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b54f373cac62ec170562d7a2d84e66773b9a2deb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b54f373cac62ec170562d7a2d84e66773b9a2deb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
