)]}'
{"id":"openstack%2Foctavia~726084","triplet_id":"openstack%2Foctavia~master~Ic94d226ad6e25323b94bc85f3ad847937b3be218","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ic94d226ad6e25323b94bc85f3ad847937b3be218","subject":"Add retry for getting amphora VM","status":"MERGED","created":"2020-05-07 10:21:38.000000000","updated":"2021-03-05 14:39:26.000000000","submitted":"2021-03-05 14:35:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":27,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"726084","meta_rev_id":"cc463a5313e5cd0390712cc26d2a238b4ce1511f","_number":726084,"virtual_id_number":726084,"owner":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-05 14:35:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-03-05 12:39:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2021-03-05 12:36:58.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2021-03-01 17:01:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-03-05 12:39:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-23 13:05:05.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-09-23 23:47:24.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-09-25 12:02:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-25 21:16:23.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-02-11 13:31:17.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2021-03-05 12:36:58.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"8f32c66eb214d767908328844a82c402d5902c45","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-05-07 10:21:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d2c35534b72efd00a0cad282fb6e028cf9b22636","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 12:48:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9768988d70643baa2a47c6eab06ca43 : SUCCESS in 10m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dfc9d2a31d5e46f28dc3fc2fde799170 : SUCCESS in 14m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4105d091f7624e8886484f3ff587cfc2 : SUCCESS in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9071a5256d544b45a6dd49984f7f809c : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5add512d7ef34a6284f281c554b41626 : SUCCESS in 9m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19835227b64f462f871a9f29ece3f31b : SUCCESS in 16m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/98c8143533974a85bea823d38c21aea4 : SUCCESS in 7m 16s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d4da9c4f34ab4fe58b8215b8ff375ff9 : SUCCESS in 10m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9dbe816f224f4cc28608a5586f6569fe : SUCCESS in 7m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/128f8949b63e4739bce29e06374267b2 : SUCCESS in 43m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d59f62ddb7054f62b6b1fa6b7853d90b : SUCCESS in 1h 54m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4793e9ecd18145e3aaab003b242637b4 : SUCCESS in 1h 05m 10s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/302d38f034d4462ab091fee923064711 : SUCCESS in 55m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6c2acc3f14584851b5120cd5427936a2 : SUCCESS in 59m 55s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/94ac97fca9a74ab79fe20a2c86468865 : SUCCESS in 1h 12m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/19ed0cbd7a454039a97ea3ba61c8ef28 : FAILURE in 2h 19m 52s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9b3945d1be0a4edaa1692a06f9559a8f : SUCCESS in 1h 23m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/67c585e116284f0090349bf7e515ad13 : SUCCESS in 1h 05m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"54907d8a4f38e1060089065e23ec6048547a917b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-07 16:46:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2d3fd545119f9ce5ccf1595cb16af1e005f74244","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-13 16:16:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e49bc8454894f29f90e6691465bb8fb9b4e2dc6a","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-05-14 10:09:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e3b18cead9117744e8dd432dfa4d7201b7674395","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-05-14 10:10:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c8e64ec849dd3d6d3b2e7c2bb3ceb02928f5fe30","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-05-14 10:20:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"152cb8adb4264378382b8e12041bbae76286454f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 12:54:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f93670b51f04750ac4dd1fcfa3f8846 : SUCCESS in 9m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c27f9244f52494e8b1126e33338a1b1 : SUCCESS in 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d4429b54f844d728383f67fe7584b64 : FAILURE in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad5719a716f841ec9f6ecefabda53db5 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b30fed0ca2cd42eaabca7318c1eb7e5c : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1381680ad1fc42579241be84d6383dd5 : FAILURE in 12m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/650795c1a2d5469886cb8c421e9c1436 : SUCCESS in 10m 59s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9cf96907e72a4f95bb3b24ccb0efc2df : SUCCESS in 6m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e17ab82883fc4d3a9ff4b013e589e132 : SUCCESS in 8m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c1366c1a8c1b4ce99f2a739a67e66298 : SUCCESS in 8m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d0164339fb3c4ccfabdd45a47157c04c : SUCCESS in 44m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4f2f333a6a224258813e2de93b3a8a25 : FAILURE in 2h 31m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e1252776a60b483aab008d77b699a454 : SUCCESS in 52m 45s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c8e5f74f623c499f8fe4037ce68eec53 : SUCCESS in 51m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/07e0105c30aa4aec82b311ef3c8d9ca6 : SUCCESS in 57m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/59393e37f0cc4a658306ba98d8f1a3b5 : SUCCESS in 1h 23m 42s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/424ee928507d42418d26f6a915aecc43 : SUCCESS in 1h 40m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/df6d0cad8783427dbde3aa374893c135 : SUCCESS in 2h 20m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/45f32c7f88bf48f08dc04ce9bf5a5d96 : SUCCESS in 1h 49m 57s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/tests/functional/api/v2/test_member.py\n  Comments left for invalid file octavia/api/v2/controllers/health_monitor.py\n  Comments left for invalid file octavia/api/v2/controllers/load_balancer.py\n  Comments left for invalid file octavia/db/models.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_l7rule.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/rest_api_driver.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_load_balancer.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_quotas.py\n  Comments left for invalid file octavia/network/drivers/neutron/allowed_address_pairs.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_listener.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/loadbalancer.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_pool.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/database_tasks.py\n  Comments left for invalid file octavia/amphorae/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_l7policy.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_health_monitor.py\n  Comments left for invalid file octavia/api/v2/controllers/pool.py\n  Comments left for invalid file octavia/controller/worker/v2/controller_worker.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/database_tasks.py\n  Comments left for invalid file octavia/api/v2/controllers/member.py\n  Comments left for invalid file octavia/tests/functional/db/test_models.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_amphora.py","accounts_in_message":[],"_revision_number":3},{"id":"61fcaeac035525835ae09a38cbbd89fe6763ed5a","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-05-18 07:56:58.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"8dc4931a9894d03740af1774519d8c9564352bf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 10:14:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50aad2a5862741d69ef64bf0a04430d2 : SUCCESS in 10m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/894eba514531456ea307ffbf56d62cfa : SUCCESS in 10m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6eb1d66d2a364014bc012218d2c8417b : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9d9795d518d410aba09cc32bd9c6f58 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b24afd2ceaa4b319c2ac3ef6ab2f7d2 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45ba69bef39541ecb612f896f04e68d6 : SUCCESS in 12m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc4e972f94de4248ac4acb2ffa05dff0 : SUCCESS in 6m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0154cf60ad11417c8193c8334a454676 : SUCCESS in 6m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e5093e81380d4b05b4fec62d03088a2a : SUCCESS in 8m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1858459664b841698b15621c4f584c54 : SUCCESS in 6m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6216573121d44cec8fa5842856204718 : SUCCESS in 42m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c3dea532b43a4e90b22ffc7e9e1e3a70 : SUCCESS in 1h 46m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3b6ed662973d46108684399cb03bb5c3 : SUCCESS in 58m 39s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/58663f0d548740f2a66fdc718c7b8044 : SUCCESS in 56m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/387831865fdf4936907a381bd5a07089 : SUCCESS in 59m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/72db1c8dc6d445c2bb1490e9469c6487 : SUCCESS in 1h 35m 48s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5c67f5f97dbb4f7d9bd05576593970ba : SUCCESS in 1h 40m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/38e2e5282f5942a9a0b66764cb825fb2 : SUCCESS in 2h 11m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/786da647482840f5b23dc57abf7030fa : SUCCESS in 1h 30m 36s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/common/context.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_base.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_noop_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_driver_lib.py\n  Comments left for invalid file octavia/amphorae/backends/utils/keepalivedlvs_query.py","accounts_in_message":[],"_revision_number":4},{"id":"ed145f66de9e71a7d3894eb8c0b704eb783dc046","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-08-24 09:58:52.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8822e3e94866bb14cd6cd9c08ad5873423762822","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 12:17:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5dfd754ae8ca4ab988f5d5704160ec65 : SUCCESS in 10m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/411df9cf48e14c48b861f6ec97ea6a27 : SUCCESS in 9m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/053c0756be404b0d944e261d814f3732 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfb61558065c4cd4b16342e3b65def39 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38c6a703fb31408cb93d21e18064b444 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dc696ecf09a41bd83bfe787a10372f6 : SUCCESS in 11m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcedaaf54be34004aeaf617ff12ba9d9 : SUCCESS in 5m 59s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a8d807ccec644d1ca897351d8e411e4e : SUCCESS in 4m 50s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e3192cd5cbd448aa9761db1a95ddaf01 : SUCCESS in 8m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e8b60ba25ad5467786bf4fc1aad6cc0e : SUCCESS in 7m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/11d0e71782db46e78df24e07a7696aa6 : SUCCESS in 45m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/26dafb5144c74fe591ff55192423f8e9 : SUCCESS in 58m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4af893541b244dcebf65e441641b177d : SUCCESS in 39m 35s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b1cc5aca68e64416bc105e15dbd90132 : SUCCESS in 38m 41s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/00cbc1ec9ba54f4d89226445a2845bbd : SUCCESS in 47m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/175030e4a6b34450b296db4a854836c1 : SUCCESS in 2h 02m 51s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6b62cd1d96384857a89217b7f3437daf : SUCCESS in 1h 08m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6854fcb1c8a34b8aa7d4da6b9e033d34 : SUCCESS in 1h 14m 07s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/77b78cc6ed7f446990beac27e931ef09 : SUCCESS in 43m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7a0ae47b9ebc4e8a933472629297904d : SUCCESS in 53m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/e4f7b9ace43442d1a45466421c2ceb1d : SUCCESS in 1h 21m 11s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"441596972139885ec0d7ab82132cf3c9bdc71e13","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-25 19:13:19.000000000","message":"Patch Set 5: Code-Review-1\n\n(7 comments)\n\nJust some typos in the release note","accounts_in_message":[],"_revision_number":5},{"id":"6f00cf4ce326231b9f5353a2ccef146b4de4d237","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-08-27 09:09:08.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"ebe5741d0728d8ca8a1ed504a33abdfd80b99c2f","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-08-27 09:09:19.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"a362298a828d4f5eb228642f31f8857c0376cf4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 11:30:22.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2e20075b10d41a58bae58d3628d5711 : SUCCESS in 11m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c661509d82c747b6abf84fce2dd75e5f : SUCCESS in 11m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd8af6161c0d41b1bfef91c1c2e752a0 : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b549348856d466ab554fcb4fa9ae0b8 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80aa7bb98d07478d96c0498015e14ddc : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4589c897a884aee8d3d7b03a9e9723c : SUCCESS in 11m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e3b7aeaf54646108ab507207402f70f : SUCCESS in 6m 22s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3c1d1fbe6a424298988e158403f94e2f : SUCCESS in 6m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/827bcbd60ae44db9974e8080f40f9b22 : FAILURE in 9m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/de6fb4dd57c7439a95e7aaa336a6f85b : SUCCESS in 7m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/46a98c3053f04bb1b28e48af5fa196aa : SUCCESS in 41m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/65710ff602b44addbc6b71191dfe5082 : SUCCESS in 1h 30m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f4bac1c629ff4faea9e8d822e7d0e332 : SUCCESS in 1h 01m 29s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6068adc3aea64feaa7ba37648132d012 : SUCCESS in 43m 07s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4ba357975a68472b984cce4e22ea6f4b : SUCCESS in 44m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/552721cd8f724f2285657abc0356345c : SUCCESS in 1h 57m 31s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/38dc187f79e045d4b20e5cdd274f926a : SUCCESS in 1h 24m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/37d63477b0c54ef9b7f4b09ffe7d419d : SUCCESS in 1h 09m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/678f2cd6d1704b648df117c574e44475 : SUCCESS in 1h 07m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9d4c64d94ff34d8aad02d3110bf066e6 : SUCCESS in 1h 18m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/98907c73d43f428d9d56c06ccaa87ef3 : FAILURE in 11m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c17695a2d71b48eb90f7da68c271b903 : FAILURE in 13m 06s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f22948d339dd5e6730ba95b7453713053c547f43","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-02 07:58:42.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"b71304fe10b355922fb2ad9a652182353e43e228","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 09:52:35.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5da800370a2b412eb956509795aa45e2 : SUCCESS in 10m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/343fa63f1bce4eeba2b5601767fdf6ea : SUCCESS in 11m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68fe1a169e154eaabbb121f6595d7891 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ce4182a307f466fa7a717bed8479f2b : SUCCESS in 8m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2d89f1842004e588043d1676449a958 : SUCCESS in 9m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eab68c35a3f64d459e596930bc9cd4fa : SUCCESS in 13m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4685de92dd764632bf2927590d12a100 : SUCCESS in 7m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/50614d0049d1414a98e8e32a64244574 : SUCCESS in 6m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/befe67cb709e43fca6589c42237468f0 : SUCCESS in 9m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d07888e4beef4821bb660cd809b31998 : SUCCESS in 8m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d9cad7327fd54b34a2094096dd0ca4d8 : SUCCESS in 41m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/49170e3b8e97451fa5d12823f4975772 : SUCCESS in 1h 11m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4cc2d2c6696442a9ab16e7146dfa6110 : SUCCESS in 45m 42s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/07ebfba658d24bbbb295a6363c8d68af : SUCCESS in 38m 24s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5d8341845ec14d33838c63bf4636caca : SUCCESS in 40m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1eb6707ce77f4557a6c7bd85a821b18d : SUCCESS in 1h 30m 05s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/53aeb5c094a94d7989991d7ac2cb128f : SUCCESS in 1h 27m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d37bf2a7ba9e4ed9bc60a24363a09147 : SUCCESS in 55m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/53219ea4ae4c4988bbb3f710f1c428ee : SUCCESS in 59m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d37d329b5581495bb0ffd60cd47a4340 : SUCCESS in 56m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d096366a6ad5483a95f383c6c049e04c : SUCCESS in 1h 30m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/f16f34266e0f4a409407187f81e63c03 : FAILURE in 40m 26s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"eefc91a9dd803f1f2075f6800760593361fe84c7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-02 19:25:37.000000000","message":"Patch Set 7: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"d150cc45900b9cb2ea89ffd3183a2c29e4382529","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-09-23 13:05:05.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"276851431c5c55f2a7fade6f4b8ad0cd06f543a2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-23 23:47:24.000000000","message":"Patch Set 7: Code-Review-1\n\nMarking -1 to reflect the open comments.","accounts_in_message":[],"_revision_number":7},{"id":"22ab6b4bdc3e18416a0715dfe789b769944b8b40","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-23 23:51:54.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"c1264135f923c5f3a359f1a6b382c6a98a063a51","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-25 09:15:28.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"a4adc5aa4beb2e5b1496ba44885fa1baed3d1d2f","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-25 09:17:06.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"c94bf8254cc50357f47dcbe51aa71514ab9fb402","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 12:02:34.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dfe653f9aca84507b23903ddbe118ff2 : SUCCESS in 11m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/826dd250584b4f518c613c6580a37a8a : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dec30f478c9540a78eb5f06c25dfc33b : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e92a0335f9854e01890db1730bd25cf8 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b403daf54fb483b8ff61974f4aca3b0 : SUCCESS in 7m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89088b9f8275464d87acd48fe82a6d64 : SUCCESS in 13m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26d1e0892d0e48b497f6cfd0bf6c16f6 : SUCCESS in 5m 10s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/554e77f59a294ee18ca6440303d79671 : SUCCESS in 6m 22s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/cb64db66ee0f411689354eccfda6285e : SUCCESS in 8m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a6ddaabb97514d4ab816a79cce73fb24 : SUCCESS in 5m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/35c0afb7436d41d08ceb156530a0c6b0 : SUCCESS in 11m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3f431bc70d214af6a62a54823e769540 : SUCCESS in 1h 38m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/deba8a29907241f5a94d7d333d924d87 : SUCCESS in 1h 47m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/321caeb81cf74d028f8039669397a7fe : SUCCESS in 41m 07s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/af91c1c067b74615a549805a44961d0c : SUCCESS in 39m 14s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/57587809c66a49b6bd590805850ad00d : SUCCESS in 40m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0c6b0b4a1d214557b5e1bdbb75abb2d2 : SUCCESS in 1h 01m 31s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9a03410d90194676af6ad1b846439b20 : SUCCESS in 2h 01m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/04bee4171dc3487abf45293ed2dcd229 : TIMED_OUT in 2h 34m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e22b1173f9714df8810aa129957b6a60 : SUCCESS in 1h 43m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/357751157e744782be7b8004ffd95bf0 : SUCCESS in 2h 13m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/19658fbbbff546ae99494dade0f94e91 : TIMED_OUT in 2h 33m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3599219870734ef581686e28bd674c95 : FAILURE in 2h 30m 05s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"2f498c4614af2a38807a743ce596a58b6a6a6b0c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-25 21:16:23.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"50493b5f480f4cc089fcd580e212f4b41affc55c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-01-21 09:26:54.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"bedd7ca26d98920edf2f210e0ee8dca58a90cab6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 12:09:17.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1c59927c0794833be77dc9f852ba167 : SUCCESS in 9m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a9465e60ccb549499f4920579c78422f : SUCCESS in 8m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48b1c16b9aa340878ce0d15b390b8a88 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48df47de5e9b4471ae610dd3dcfa5724 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/190c463ac9aa42b1a11e7b7b1f0f8566 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5dd37b1f927d4ddf9175df2770443bc0 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a31acd059c44ccd9568808e547d61bf : SUCCESS in 11m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/edf8d3750b6f4a21af2e0318ba9c0eb9 : SUCCESS in 5m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/897e052f4a67422fa7546114b7f55582 : SUCCESS in 5m 43s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a1ac9cfe492e436e898a5463acc41ba3 : SUCCESS in 7m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a1e552dce0bf4f6e9d67cb51b6513d62 : SUCCESS in 4m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6422273785fa41a08e59918d3a5dafa0 : SUCCESS in 7m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/069f70ae77424f3885816c156d5b006b : SUCCESS in 1h 12m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/96036e9252b74588a8edb6e8665042ef : SUCCESS in 2h 03m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/328502fb25c54b0787a4b5c28b33b1fd : SUCCESS in 38m 36s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f4e5b297a599413d8559984514535ebe : SUCCESS in 46m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9f7434fe84de4189b30dcecc264439d2 : SUCCESS in 48m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/85fb2d534a05420492bd35777f8f1c1f : SUCCESS in 1h 06m 32s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/810cb947be73466b892f1052b9f1f093 : SUCCESS in 2h 04m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c129aaed61ea433b944915af7efe10f8 : FAILURE in 2h 30m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/159e4e2406414f168ba98e5e85924ed5 : SUCCESS in 1h 30m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/10c105de39cf4606885913496dfc1404 : FAILURE in 20m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0dbd49c0f8fb4e309d8b33d93a5ec187 : SUCCESS in 1h 57m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3ffbbc57eaf9458db95016ef4bf5c3c4 : FAILURE in 27m 38s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"e5fbaff19a0390151660acc4562f92b9587b5ccd","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-02-11 13:31:17.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"6c07854a4fd8f7b016c4829ace2a44405d97fe43","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-02-11 14:28:33.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"feb45f9afeec4b50842a5dc1836b82da974a76dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-03-01 09:42:00.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"ffbef897295cd8eb86a66cb2cd3ffa6b82874064","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-03-01 09:42:45.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"182a47fb93ecb63c0840576b7567b912ef160be4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-03-01 09:44:45.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"ee43ffed11f57f185747b8e590a5c45caed2ea2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 12:08:38.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d416a1ce2fba41c69fa86d7b8741faf8 : SUCCESS in 9m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba3d49ba688c41e8a6a7f5e2223ee6c6 : SUCCESS in 9m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b54b8f0d81c4418c85b51b1a716b1222 : SUCCESS in 8m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e633d4570fdc4ff7bef949b46d439b21 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33eac165a6064ea48967065fb1ed1229 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d547085a6f024ab89f94e7f4bc97454c : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7afc6c4de37b402ba25609b2eacbf618 : SUCCESS in 11m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04ef3ad1dd9d4524853eefe9a648d502 : SUCCESS in 6m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/307cb8f692924d1d837ef1d5cad3ce9b : SUCCESS in 7m 03s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7ded80828d2749b3ba772729031be6fb : SUCCESS in 9m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1db781b4982e473692d4c8bc38e1d73d : SUCCESS in 5m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ab08ab933df44319b98bdca662f454d7 : SUCCESS in 7m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f5ee591774ac421f869f1f63a79c0051 : SUCCESS in 1h 15m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e7564b54413b41eabb521c02dfb77721 : SUCCESS in 2h 12m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a5e20550a5b0451bb3a56f6376e1d6ce : SUCCESS in 44m 53s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a2f0cab6e4ac46308708cc5d2c3a0506 : SUCCESS in 36m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2f2296af533044cbb578efcd35c2578d : SUCCESS in 37m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/09006811e4124440a31855bb027f7665 : SUCCESS in 1h 00m 56s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dfc4409531b44bfc953031c5ae3c1ce8 : SUCCESS in 1h 47m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3e552b155eb74722972ab196962f487f : FAILURE in 1h 57m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/99117a9b7a924efeadbe49299941bbb1 : SUCCESS in 1h 41m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1a3acea4c9884b539d5645553d456306 : SUCCESS in 2h 16m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b1911661c01f41e48a9a89b585427b63 : SUCCESS in 2h 18m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/9b072da282cf4753bc56610609f73f9c : FAILURE in 44m 12s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"e9ee87785b716473707eeec43b9444d45e0e1769","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-01 17:01:05.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"e7db404702ae6db28858074c9dced5bdd1a3ae90","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-05 12:36:58.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"48cc219732a44882da4f5a568c53541a84c6dfa3","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-03-05 12:39:08.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"73a4fd81f92c6731fccb0a1ceaa61055100413c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 12:39:48.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"0a9b4d415a43b82917247045130668bfdaabb390","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 14:35:37.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2fcafce2373146da8d97b9c5347a0510 : SUCCESS in 10m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/468bb0c0bf0b417d91ce031d4289b282 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8874a91414214ab1809a0fb72a823673 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd379fc838344c8580ffd92f71457fa3 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a75535c34e294edda96ec7de461db2d4 : SUCCESS in 13m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05d02ca0929148f2ab1eb392e4e5c1f1 : SUCCESS in 5m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/178ef0d99e3f430394f85f954f890711 : SUCCESS in 6m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8f9f568d588f4cd0886dd069a3362584 : SUCCESS in 7m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dbf11de8cdc545e5bd18ef057c87bb78 : SUCCESS in 1h 21m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/10b151df19194013be9de8d74af35cd1 : SUCCESS in 1h 52m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/18d1cbb4ffdc42598a7f8d5e6142a93d : SUCCESS in 44m 44s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a546092d439844e785d3bd4676752710 : SUCCESS in 34m 42s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/31cf3caa9bb24f8da8b89751656587b5 : SUCCESS in 44m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8e4956e53347478ab2447a946aa35d02 : SUCCESS in 1h 01m 02s","accounts_in_message":[],"_revision_number":12},{"id":"d483d10f20b570277f2ba1fb754fa09970b50260","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 14:35:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"cc463a5313e5cd0390712cc26d2a238b4ce1511f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 14:39:26.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/143fa2aef43044f3a251ca6dd381f8e7 : SUCCESS in 1m 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bee1a714bf7643deb335f639f0fb9230 : SUCCESS in 1m 56s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"08e1d6d060f04a23849082bc8f7d08114798a3c5","revisions":{"8eacdd2e6317021a662eb27df2d5d874d7d2d629":{"kind":"REWORK","_number":1,"created":"2020-05-07 10:21:38.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/84/726084/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/726084/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/726084/1"}}},"commit":{"parents":[{"commit":"837b7bf94092438e486d3e18c5b9a21bb1298f98","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/837b7bf94092438e486d3e18c5b9a21bb1298f98"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-05-07 10:15:06.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-05-07 10:15:06.000000000","tz":240},"subject":"Add retry for getting amphora VM","message":"Add retry for getting amphora VM\n\nAn inssue has been observed several times when amphora VM failed\nto get, although its creation was successful.\nAdd retry to avoid stop loadbalancer creation.\n\nChange-Id: Ic94d226ad6e25323b94bc85f3ad847937b3be218\nStory: 2007637\nTask: 39690\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8eacdd2e6317021a662eb27df2d5d874d7d2d629"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8eacdd2e6317021a662eb27df2d5d874d7d2d629"}]},"branch":"refs/heads/master"},"b487e613177bca2f888dad6db08327b5b9467fdb":{"kind":"REWORK","_number":2,"created":"2020-05-14 10:10:56.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/84/726084/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/726084/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/726084/2"}}},"commit":{"parents":[{"commit":"837b7bf94092438e486d3e18c5b9a21bb1298f98","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/837b7bf94092438e486d3e18c5b9a21bb1298f98"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-05-07 10:15:06.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-05-14 10:09:37.000000000","tz":240},"subject":"Add retry for getting amphora VM","message":"Add retry for getting amphora VM\n\nAn inssue has been observed several times when amphora VM failed\nto get, although its creation was successful.\nAdd retry to avoid stop loadbalancer creation.\n\nChange-Id: Ic94d226ad6e25323b94bc85f3ad847937b3be218\nStory: 2007637\nTask: 39690\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b487e613177bca2f888dad6db08327b5b9467fdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b487e613177bca2f888dad6db08327b5b9467fdb"}]},"branch":"refs/heads/master"},"32c32f526805db2771b2691c26e2b15d74980986":{"kind":"REWORK","_number":3,"created":"2020-05-14 10:20:32.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/84/726084/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/726084/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/726084/3"}}},"commit":{"parents":[{"commit":"837b7bf94092438e486d3e18c5b9a21bb1298f98","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/837b7bf94092438e486d3e18c5b9a21bb1298f98"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-05-07 10:15:06.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-05-14 10:20:15.000000000","tz":240},"subject":"Add retry for getting amphora VM","message":"Add retry for getting amphora VM\n\nAn inssue has been observed several times when amphora VM failed\nto get, although its creation was successful.\nAdd retry to avoid stop loadbalancer creation.\n\nChange-Id: Ic94d226ad6e25323b94bc85f3ad847937b3be218\nStory: 2007637\nTask: 39690\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/32c32f526805db2771b2691c26e2b15d74980986"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/32c32f526805db2771b2691c26e2b15d74980986"}]},"branch":"refs/heads/master"},"e81d0a13c8b5ffbe7ffbf052f9dec0b714ee1531":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-05-18 07:56:58.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/84/726084/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/726084/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/726084/4"}}},"commit":{"parents":[{"commit":"9a1d6d3585070da570bcf6313e237b486bdac81b","subject":"Fix E741 pep8 errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a1d6d3585070da570bcf6313e237b486bdac81b"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-05-07 10:15:06.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-05-18 07:56:58.000000000","tz":0},"subject":"Add retry for getting amphora VM","message":"Add retry for getting amphora VM\n\nAn inssue has been observed several times when amphora VM failed\nto get, although its creation was successful.\nAdd retry to avoid stop loadbalancer creation.\n\nChange-Id: Ic94d226ad6e25323b94bc85f3ad847937b3be218\nStory: 2007637\nTask: 39690\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e81d0a13c8b5ffbe7ffbf052f9dec0b714ee1531"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e81d0a13c8b5ffbe7ffbf052f9dec0b714ee1531"}]},"branch":"refs/heads/master"},"9958631cb14a92f27499e07584db7ded6bd105ab":{"kind":"REWORK","_number":5,"created":"2020-08-24 09:58:52.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/84/726084/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/726084/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/726084/5"}}},"commit":{"parents":[{"commit":"13faf288e4c1044facc579c8e40e46c524bbe3f8","subject":"Merge \"Fix memory consumption issues with default connection_limit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/13faf288e4c1044facc579c8e40e46c524bbe3f8"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-05-07 10:15:06.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-08-24 09:58:35.000000000","tz":240},"subject":"Add retry for getting amphora VM","message":"Add retry for getting amphora VM\n\nAn inssue has been observed several times when amphora VM failed\nto get, although its creation was successful.\nAdd retry to avoid stop loadbalancer creation.\n\nChange-Id: Ic94d226ad6e25323b94bc85f3ad847937b3be218\nStory: 2007637\nTask: 39690\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9958631cb14a92f27499e07584db7ded6bd105ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9958631cb14a92f27499e07584db7ded6bd105ab"}]},"branch":"refs/heads/master"},"2eda9610f357cbd55112f930edc4dfb9fe9d5086":{"kind":"REWORK","_number":6,"created":"2020-08-27 09:09:08.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/84/726084/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/726084/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/726084/6"}}},"commit":{"parents":[{"commit":"13faf288e4c1044facc579c8e40e46c524bbe3f8","subject":"Merge \"Fix memory consumption issues with default connection_limit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/13faf288e4c1044facc579c8e40e46c524bbe3f8"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-05-07 10:15:06.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-08-27 09:08:49.000000000","tz":0},"subject":"Add retry for getting amphora VM","message":"Add retry for getting amphora VM\n\nAn issue has been observed several times when amphora VM failed\nto get, although its creation was successful.\nAdd retry to avoid stop loadbalancer creation.\n\nChange-Id: Ic94d226ad6e25323b94bc85f3ad847937b3be218\nStory: 2007637\nTask: 39690\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2eda9610f357cbd55112f930edc4dfb9fe9d5086"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2eda9610f357cbd55112f930edc4dfb9fe9d5086"}]},"branch":"refs/heads/master"},"5046717643bb3866a79508164ff4e2829d225293":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-09-02 07:58:42.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/84/726084/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/726084/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/726084/7"}}},"commit":{"parents":[{"commit":"0a1e834123ea3182b768a54bb6e65f8824458efe","subject":"Merge \"Add some details on enable_anti_affinity option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a1e834123ea3182b768a54bb6e65f8824458efe"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-05-07 10:15:06.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-09-02 07:58:42.000000000","tz":0},"subject":"Add retry for getting amphora VM","message":"Add retry for getting amphora VM\n\nAn issue has been observed several times when amphora VM failed\nto get, although its creation was successful.\nAdd retry to avoid stop loadbalancer creation.\n\nChange-Id: Ic94d226ad6e25323b94bc85f3ad847937b3be218\nStory: 2007637\nTask: 39690\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5046717643bb3866a79508164ff4e2829d225293"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5046717643bb3866a79508164ff4e2829d225293"}]},"branch":"refs/heads/master"},"e62093751c6ea738a1a25e3fe682242d3de8aa3c":{"kind":"REWORK","_number":8,"created":"2020-09-25 09:17:06.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/84/726084/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/726084/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/726084/8"}}},"commit":{"parents":[{"commit":"0a1e834123ea3182b768a54bb6e65f8824458efe","subject":"Merge \"Add some details on enable_anti_affinity option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a1e834123ea3182b768a54bb6e65f8824458efe"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-05-07 10:15:06.000000000","tz":240},"committer":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2020-09-25 09:16:47.000000000","tz":240},"subject":"Add retry for getting amphora VM","message":"Add retry for getting amphora VM\n\nAn issue has been observed several times when amphora VM failed\nto get, although its creation was successful.\nAdd retry to avoid stop loadbalancer creation.\n\nChange-Id: Ic94d226ad6e25323b94bc85f3ad847937b3be218\nStory: 2007637\nTask: 39690\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e62093751c6ea738a1a25e3fe682242d3de8aa3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e62093751c6ea738a1a25e3fe682242d3de8aa3c"}]},"branch":"refs/heads/master"},"3b162cf97b6a1b93d4a6fba5672e12925fa61fbe":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-01-21 09:26:54.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/84/726084/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/726084/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/726084/9"}}},"commit":{"parents":[{"commit":"a4aa03d3bc98eb27cc353140cd998a623baa505f","subject":"Fix the unit test tree structure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a4aa03d3bc98eb27cc353140cd998a623baa505f"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-05-07 10:15:06.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2021-01-21 09:26:54.000000000","tz":0},"subject":"Add retry for getting amphora VM","message":"Add retry for getting amphora VM\n\nAn issue has been observed several times when amphora VM failed\nto get, although its creation was successful.\nAdd retry to avoid stop loadbalancer creation.\n\nChange-Id: Ic94d226ad6e25323b94bc85f3ad847937b3be218\nStory: 2007637\nTask: 39690\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3b162cf97b6a1b93d4a6fba5672e12925fa61fbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3b162cf97b6a1b93d4a6fba5672e12925fa61fbe"}]},"branch":"refs/heads/master","description":"Rebase"},"6df87e25bba980f5b6cfaed95fa7c45f97b232a0":{"kind":"REWORK","_number":10,"created":"2021-03-01 09:42:00.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/84/726084/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/726084/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/726084/10"}}},"commit":{"parents":[{"commit":"a4aa03d3bc98eb27cc353140cd998a623baa505f","subject":"Fix the unit test tree structure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a4aa03d3bc98eb27cc353140cd998a623baa505f"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-05-07 10:15:06.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2021-03-01 09:40:36.000000000","tz":0},"subject":"Add retry for getting amphora VM","message":"Add retry for getting amphora VM\n\nAn issue has been observed several times when amphora VM failed\nto get, although its creation was successful.\nAdd retry to avoid stop loadbalancer creation.\n\nChange-Id: Ic94d226ad6e25323b94bc85f3ad847937b3be218\nStory: 2007637\nTask: 39690\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6df87e25bba980f5b6cfaed95fa7c45f97b232a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6df87e25bba980f5b6cfaed95fa7c45f97b232a0"}]},"branch":"refs/heads/master"},"8549cac8d97cf3a8a2acd70cfeb02b9b2d7887ec":{"kind":"REWORK","_number":11,"created":"2021-03-01 09:42:45.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/84/726084/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/726084/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/726084/11"}}},"commit":{"parents":[{"commit":"a4aa03d3bc98eb27cc353140cd998a623baa505f","subject":"Fix the unit test tree structure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a4aa03d3bc98eb27cc353140cd998a623baa505f"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-05-07 10:15:06.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2021-03-01 09:42:34.000000000","tz":0},"subject":"Add retry for getting amphora VM","message":"Add retry for getting amphora VM\n\nAn issue has been observed several times when amphora VM failed\nto get, although its creation was successful.\nAdd retry to avoid stop loadbalancer creation.\n\nChange-Id: Ic94d226ad6e25323b94bc85f3ad847937b3be218\nStory: 2007637\nTask: 39690\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8549cac8d97cf3a8a2acd70cfeb02b9b2d7887ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8549cac8d97cf3a8a2acd70cfeb02b9b2d7887ec"}]},"branch":"refs/heads/master"},"08e1d6d060f04a23849082bc8f7d08114798a3c5":{"kind":"REWORK","_number":12,"created":"2021-03-01 09:44:45.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/84/726084/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/726084/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/726084/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/726084/12"}}},"commit":{"parents":[{"commit":"a4aa03d3bc98eb27cc353140cd998a623baa505f","subject":"Fix the unit test tree structure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a4aa03d3bc98eb27cc353140cd998a623baa505f"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-05-07 10:15:06.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2021-03-01 09:44:41.000000000","tz":0},"subject":"Add retry for getting amphora VM","message":"Add retry for getting amphora VM\n\nAn issue has been observed several times when amphora VM failed\nto get, although its creation was successful.\nAdd retry to avoid stop loadbalancer creation.\n\nChange-Id: Ic94d226ad6e25323b94bc85f3ad847937b3be218\nStory: 2007637\nTask: 39690\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/08e1d6d060f04a23849082bc8f7d08114798a3c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/08e1d6d060f04a23849082bc8f7d08114798a3c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
